)]}'
{"id":"openstack%2Foctavia~905593","triplet_id":"openstack%2Foctavia~master~Ib160f988cea0c6a27b3c1efeb2b2953f8929dc5c","project":"openstack/octavia","branch":"master","topic":"pyupgrade","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2024-04-15 07:45:58.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"Ib160f988cea0c6a27b3c1efeb2b2953f8929dc5c","subject":"pyupgrade changes for Python3.8+ (3)","status":"MERGED","created":"2024-01-15 12:34:37.000000000","updated":"2024-04-15 09:23:41.000000000","submitted":"2024-04-15 09:22:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":6,"has_review_started":true,"submission_id":"905593-pyupgrade","meta_rev_id":"cb1408907c4faea1e84363048c3fd37fff956dcf","_number":905593,"virtual_id_number":905593,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-15 09:22:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"all":[{"value":1,"date":"2024-04-10 08:36:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-04-10 08:36:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2024-04-10 08:36:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":2,"date":"2024-04-15 07:45:58.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":1,"date":"2024-04-10 08:36:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":1,"date":"2024-04-15 07:45:58.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-15 15:43:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-23 01:15:35.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-02-23 01:24:22.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2024-02-23 07:49:34.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2024-02-27 07:35:03.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"}],"messages":[{"id":"118168ab94d6a61c99e7acb641d244a2b278e43a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-01-15 12:34:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d622afa293e8b932ed7b73c6952602a5e47dc82d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-15 15:43:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c17d7ac07f143eab822ef03637078b3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b34ff548d58243d787f946cf0ccc2aaf : SUCCESS in 12m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/549bd223a4e54be8a78e5fe0ec5507ba : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55b90bf603d440aba772e100e14db758 : SUCCESS in 4m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/54799be04e6146ce8ae6df888af4bc46 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9298e7814d44ce0bd048610d748c55a : SUCCESS in 12m 39s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/8639fb513cdc47df9b98570a79e85c52 : SUCCESS in 4m 34s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/4fee90c3b7324bbc92c83ed5d68e0ef6 : SUCCESS in 7m 56s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/58a8bb23fbfe46e786308f80d3f867b8 : SUCCESS in 9m 49s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a767ea7168634f27bd25483d74e46f90 : SUCCESS in 4m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6b2de0a717ed4f4683a6b2b510fab662 : SUCCESS in 7m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cca83b9f1b994d2b8e6d2254798e0729 : SUCCESS in 7m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/db590be31e9a456ab4dc89ed602074cb : SUCCESS in 7m 30s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/dc1daf181fcc4bc9bdf8ac686fec9d14 : SUCCESS in 1h 18m 58s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/57169315f1bc4e7790ada16165a31ffd : SUCCESS in 2h 06m 15s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/6c2d19fecf104a6386184e97b43f0801 : SUCCESS in 1h 24m 42s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/29548ff9f7bc4f1cbaf0c250f5e12137 : SUCCESS in 1h 40m 05s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c120843a987a423db99ba02659e97449 : SUCCESS in 1h 08m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/24501d9b3f3244b3aa1ccbc877fff56d : SUCCESS in 33m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a8ae3efa97874e129d628fd05ea5af25 : SUCCESS in 52m 52s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/1ba42afbe7a1473da42c5e59df62e21f : SUCCESS in 1h 02m 03s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/74990404507f4a528ed20c93cdfbbbe8 : SUCCESS in 1h 06m 48s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/64d7a14d8e044d129cf4853e01d9e617 : SUCCESS in 32m 25s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/eb8b535c24e64224945a697154bd6f6a : TIMED_OUT in 2h 32m 47s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5ce9e0419d224cf887200a69c56e58b1 : TIMED_OUT in 3h 04m 45s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/07ceedb32244454798cd6cd794d19fe7 : SUCCESS in 2h 20m 55s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/8d2671c8a0be427e807644fec9ff01ac : TIMED_OUT in 2h 33m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"274fd950b0f576da5c6b23b60da83bf86897685e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-02-23 01:15:35.000000000","message":"Patch Set 1: Code-Review+2\n\n(7 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c175bdc1f7bfa413870d1840b9ff2cdaf06e4863","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2024-02-23 01:24:22.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cb321d5a9edb8b806fb0f62b6bf798c7a0df7e16","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2024-02-23 07:49:34.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a8b05da324a963de2c99a84bb04e6328cb81cf5b","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2024-02-27 07:35:03.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"935211aae16583cf19511bd74eb9ee5e2dc31cda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-10 08:36:01.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"095909da5252114ad73e0e8304ad4e90c6396148","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-10 11:40:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b73e058daba14db3b5374842cc422ae2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/128f1dcab0ad4da498c679b2143572d5 : SUCCESS in 11m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f0128ecd5d246b3bca730d656a1d7ab : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a9886c1db42443895a6b3423fbf7f9c : SUCCESS in 6m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/23c4f3b45dea4f7884ad0ed8b8d19bf2 : SUCCESS in 4m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/81a5cf8b285541faad653263a96e99dc : SUCCESS in 8m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d551501e20e249dc8cf615ab0743d603 : SUCCESS in 11m 22s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/f0e0337c4c694159a2c1d9e1dff1d1ed : SUCCESS in 4m 51s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/6c1adc3cc00f41ef94c3389574015c04 : SUCCESS in 8m 23s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/2ebb0c61cf2b418a84ad449f2da4ad43 : SUCCESS in 8m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6689e8dbc1a24bdbbd1eadd141244b54 : SUCCESS in 4m 29s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2194fabffeef4fa8958c96dc8cfe1367 : SUCCESS in 7m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e23288c7e9bb4d96ab61a6e07d20a35c : SUCCESS in 9m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5e9a9bd397c7447f8e39b8f90689dbe9 : SUCCESS in 7m 34s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2636b72026224975b94a55c06e9614fe : SUCCESS in 1h 38m 29s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/1cf77ea93eda44ee86499d9696c68d35 : SUCCESS in 1h 47m 40s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/247c7c3763dc45feaeb256f26118603d : SUCCESS in 1h 07m 44s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/15a0201e5b364dbc98b1a824329d77e0 : SUCCESS in 1h 47m 10s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/21d5cbdf1b3d4f90b16c5674d479fc21 : SUCCESS in 1h 10m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ed9be4e72b7e4d68a2a39768a2238870 : SUCCESS in 48m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/dd94c11b0c8f448284c266e28f908580 : SUCCESS in 48m 04s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/1e0f04ac58a14c8bb7b031aba2d29e4c : SUCCESS in 1h 16m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/70ee110a0245442bad90bfb8e600bda1 : SUCCESS in 45m 06s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e21a0edbc5fc49429e324f6cafab320e : SUCCESS in 2h 16m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9bf9822a9b244302a5e228c3f7e5d205 : SUCCESS in 2h 50m 36s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8dbf7e0f62d648d685f7793556f2ef8f : FAILURE in 35m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/7a75c7669b5a436c9b96331d69031d20 : TIMED_OUT in 2h 33m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0024293ce057a364bccdfd88466234b6054e43b4","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-15 07:45:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"938b11708029060299a20783b78ee33746c72d7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 07:46:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fe1d3c018310974097e7fc7142074d3ea3e4ddbf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 09:22:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b066d1b45c24d929f941b964edec847\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75bf846801b3493e838ce976763b6c30 : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bdd41f23783f4ea78ee47901e21b0463 : SUCCESS in 4m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/97cc570bb50545b2a732a041caa31df7 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e57507bcaf44031b05895c2ed1b76ea : SUCCESS in 11m 28s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/224d520a677a4d78915bc035c2c9b190 : SUCCESS in 3m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8b93c4b7689b42afa9f605e38ff6ce33 : SUCCESS in 6m 56s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/609b7dc2e03a4097ab422062e484cdff : SUCCESS in 1h 26m 26s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/fcd3cb86f9694d26a8d57fe27aba63df : SUCCESS in 1h 35m 26s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/6964c64bb8ae40a5889da17b19cfed41 : SUCCESS in 54m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/50ffaeffc91d4bb7b220bdb95ecd13fb : SUCCESS in 45m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/876eda3995f14209a7142615bf3d0d59 : SUCCESS in 51m 20s","accounts_in_message":[],"_revision_number":2},{"id":"b31b899dcaf9c8caf87a902e8fcc0ebfd742b3dc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 09:22:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"cb1408907c4faea1e84363048c3fd37fff956dcf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 09:23:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3051b6a6a8ff4d1f9698dbeff6ba2665\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90587ae6f7014ac0ae3067f28d9d7c24 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3d3360953ba31b6bcafe79f3f2f88ad722fdc121","revisions":{"7a32ced80386eeb8981154781d3bcf94b06176ed":{"kind":"REWORK","_number":1,"created":"2024-01-15 12:34:37.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/93/905593/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/905593/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/905593/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/905593/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/905593/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/905593/1"}}},"commit":{"parents":[{"commit":"4940997d60b704a8b3b4bf5eab9f48b55975a98b","subject":"pyupgrade changes for Python3.8+ (2)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4940997d60b704a8b3b4bf5eab9f48b55975a98b"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-01-15 11:54:19.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-01-15 11:54:19.000000000","tz":60},"subject":"pyupgrade changes for Python3.8+ (3)","message":"pyupgrade changes for Python3.8+ (3)\n\nResult of running\n\n$ pyupgrade --py38-plus $(git ls-files | grep \".py$\")\n\nThis was inspired by Nova [1]\n\nFixed PEP8 errors introduced by pyupgrade by running:\n\n$ autopep8 --select\u003dE127,E128,E501 --max-line-length 79 -r \\\n  --in-place octavia\n\nand manual updates.\n\n[1]: https://review.opendev.org/c/openstack/nova/+/896986\n\nChange-Id: Ib160f988cea0c6a27b3c1efeb2b2953f8929dc5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7a32ced80386eeb8981154781d3bcf94b06176ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7a32ced80386eeb8981154781d3bcf94b06176ed"}]},"branch":"refs/heads/master"},"3d3360953ba31b6bcafe79f3f2f88ad722fdc121":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-04-10 08:36:01.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/93/905593/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/905593/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/905593/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/905593/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/905593/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/905593/2"}}},"commit":{"parents":[{"commit":"9b1b8a3b6f121b5aec5ba7a8418078631adec310","subject":"pyupgrade changes for Python3.8+ (2)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9b1b8a3b6f121b5aec5ba7a8418078631adec310"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-01-15 11:54:19.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-04-10 08:23:50.000000000","tz":120},"subject":"pyupgrade changes for Python3.8+ (3)","message":"pyupgrade changes for Python3.8+ (3)\n\nResult of running\n\n$ pyupgrade --py38-plus $(git ls-files | grep \".py$\")\n\nThis was inspired by Nova [1]\n\nFixed PEP8 errors introduced by pyupgrade by running:\n\n$ autopep8 --select\u003dE127,E128,E501 --max-line-length 79 -r \\\n  --in-place octavia\n\nand manual updates.\n\n[1]: https://review.opendev.org/c/openstack/nova/+/896986\n\nChange-Id: Ib160f988cea0c6a27b3c1efeb2b2953f8929dc5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3d3360953ba31b6bcafe79f3f2f88ad722fdc121"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3d3360953ba31b6bcafe79f3f2f88ad722fdc121"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
