)]}'
{"id":"openstack%2Foctavia~905662","triplet_id":"openstack%2Foctavia~stable%2F2023.2~I5a5d746657da3501fb332361fee6ea25186285f3","project":"openstack/octavia","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-01-18 17:49:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5a5d746657da3501fb332361fee6ea25186285f3","subject":"Stable-only: Cap hacking to \u003c 6.1.0","status":"MERGED","created":"2024-01-16 09:27:16.000000000","updated":"2024-01-18 17:50:11.000000000","submitted":"2024-01-18 17:49:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"905662","meta_rev_id":"a4665ab9cd8a4d7450d0fa4f8dbe5dfb429aec62","_number":905662,"virtual_id_number":905662,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-18 17:49:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"all":[{"value":1,"date":"2024-01-17 16:11:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":1,"date":"2024-01-18 09:45:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-18 15:53:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2024-01-17 10:40:50.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-01-18 15:53:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-16 12:41:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-17 10:40:50.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2024-01-17 16:11:00.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2024-01-18 09:45:44.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"state":"REVIEWER"},{"updated":"2024-01-18 15:53:21.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"25e2447f4d072833a353f0cfea2f0ff50dd988b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-16 09:27:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ad952097bc00610ede27df2b05b1dac07f6720c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 12:41:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2384dc435e0d40d8b77f70c139219761\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/5350e8382e044d04a1dfc320cf33f991 : SUCCESS in 2m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db92dcdf445342c3a497bddcbe736512 : SUCCESS in 10m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c49853a557d54e6bb9688a6ef9061991 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8f3b2a3e6fc49cf963783071654ae65 : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8379d9f98cdf4a1998c9377ce97095d8 : SUCCESS in 5m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/303db689722e4ba7b849fd373bf28dac : SUCCESS in 6m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ede97920fb094093b45a63be77a54ba6 : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57cddef8bc934a899e5470bebc3558d2 : SUCCESS in 11m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a240d3c31c624e67a7e01c9e0e702cc1 : SUCCESS in 5m 33s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/ded8dcd2342342e7a2701c03b21bbe2f : SUCCESS in 5m 19s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/f4bd03fdd15e479e89f9c0433e6559ff : SUCCESS in 9m 22s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d9455dc4d8f6498d825a113c3536ed76 : SUCCESS in 4m 52s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/93217e1f3fce464a92b809b2d23ec0a6 : SUCCESS in 7m 19s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5825f7ef37e442759c5a185f5dba9f60 : SUCCESS in 9m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9efef2738b1d4fb5b1da8c075e01c279 : SUCCESS in 6m 56s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/03d69985052f4e7893c3dbff8cca4915 : SUCCESS in 1h 13m 13s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/3e0c340e8553443c9a2937cef5cebf87 : SUCCESS in 1h 21m 21s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/053d15f07c6547e4852a63f07faa446c : SUCCESS in 1h 03m 52s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/90ec89ad81db4334b7abc0216e8fde9e : SUCCESS in 1h 56m 52s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/53dd0bb390a64ee2b18c92b726764832 : SUCCESS in 1h 06m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0564a9e928f64925992047627e26669e : SUCCESS in 48m 58s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/e954e82cba28469f9b8dbd01a853b58b : SUCCESS in 1h 16m 27s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8d71fb324792412cb19210300a72f4e8 : SUCCESS in 46m 34s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/433c52f932a347fa87473bfb453d4315 : SUCCESS in 1h 30m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/90f3ac37c8cc4dca85fa8b3057f3feee : TIMED_OUT in 3h 04m 26s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ff54fea82e854c8c872463f67fac9c56 : TIMED_OUT in 2h 34m 20s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/6820b715d0674c22b89e4be1840bc29f : SUCCESS in 2h 35m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e897ac438043e2121a98f9b4d9a33866c58c2bc3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-16 17:16:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"042ca1eb17b229218ab29fbb63b78d9be4ab7c0f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-16 17:18:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"81851e8daa5e04c176b04fffcdfd8c63c64760b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-17 02:50:48.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"3f99b05b8f572086f38848b9322053e59bb8e3d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 05:49:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e53f5058403d42c5b0df7c57df023797\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/f7ce129a536441be9428b6cc3f78f307 : SUCCESS in 2m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee651cf203ed4ce88c8f12f09f0c46bf : SUCCESS in 12m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c24947eb42684558836fdad32c2190ff : SUCCESS in 7m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22d86a1a37b44a9296b8be5ca38d10e2 : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d839887440ee495abb8058aff2902998 : SUCCESS in 5m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d2f4790c2fbb40479ef9edcf82d0392e : SUCCESS in 6m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4dcbd142c09b4f418f71aa20013bbf27 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a43c19d44dd4a9887b9c686b54d4e21 : SUCCESS in 12m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/686ae706f6f8481880e1c80eefff184c : SUCCESS in 5m 50s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/4fe26b7aab7f4e5580213f3947ea32b3 : SUCCESS in 5m 10s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/03d32a3be7864dc9a5437dcf6448d9bc : SUCCESS in 9m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1ea21d6df0c447979e31dc3ab0ca50bd : SUCCESS in 5m 34s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9dfa6faceb3a48b7995cd1bdfba884d0 : SUCCESS in 7m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d9f0c8823ce74faeb94ddc6482b96f2c : SUCCESS in 8m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ec21ebabe0d1464788c27f0c69172d2b : SUCCESS in 8m 24s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5a5f484d72d14c6687243d9c79a0c6bf : SUCCESS in 1h 10m 21s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/335f30acedf74a6a816d6309061fd96f : SUCCESS in 1h 27m 34s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/0ae040c192914c93bd9c0d763d2fc01d : SUCCESS in 58m 19s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/d08dfe2b8a4b4d7dbdaf42672b98a569 : SUCCESS in 1h 24m 26s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/bf9f0abd817741a4a5d7d420974fddef : SUCCESS in 1h 00m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8bd25b1001bc4a11aa7d2eacc9c298f0 : SUCCESS in 34m 53s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/1b7b5e5b74c1415894f7e0c301cdc94c : SUCCESS in 1h 15m 45s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/422b13f4e1b04665a7fa9cdcd8fe722e : SUCCESS in 30m 53s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f05aa6e611fd48dfb11004de4046ecdb : SUCCESS in 1h 48m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/86def571836247e8850417e5b3cdb0bc : SUCCESS in 1h 51m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/fafa702ea3d746fd9008c7a1c22fca14 : FAILURE in 20m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/c9d2f1b283ec4f259accd28bed9e61e1 : SUCCESS in 2h 14m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"16f974e40f9f3ff4edcf63eadcd9d7b7850eb01d","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-01-17 10:40:50.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"447c0d0aaebb1731b133339ad4fb1382a8ff3f1f","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-01-17 16:11:00.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"0fb83e28ace46944341030c8cd9ed9df0c4dacf3","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-01-18 09:45:44.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"7be7d60a60e68a70653b74cc192aa092895dd2c2","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-01-18 15:53:21.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"25edb72430392fdfe1c06f81d42fa1ce0af0ed15","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-01-18 15:53:34.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"41c7a1f0821caaaf41c64d0ff0537fab5ae1de18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 15:54:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"87c20de464d960997d42129c9e5df9abaf1f701f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 17:49:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8322073176ba449dba1655fa496c70ac\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/7b24ba5c804a4c21b22afbe82bbb51b0 : SUCCESS in 2m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b7d9cd4cbd2425fb5a07b1a2ff53f87 : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b159f3fe8838497bb7a88d1e7a0ef486 : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3682d6917ea641d8bd00ecaf4fcc2fad : SUCCESS in 5m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/907ebe62c1304f49a187bce65e3dd4cc : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cb222594a674ce59dac7a52671b1e38 : SUCCESS in 11m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c4fbe50f90b74ab58fe61a6670899abd : SUCCESS in 5m 26s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/33b56a4746b74ff7b64f742bf75c101b : SUCCESS in 4m 47s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/91b96e8c5e7b44c297efb2649a7564be : SUCCESS in 9m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1efa6fef997f4c6890b7e97fad8934c5 : SUCCESS in 1h 10m 49s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/514e601c809d4241ade41720637f1cd9 : SUCCESS in 1h 53m 08s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/fe1bd47eda1742a7818236c9bc262767 : SUCCESS in 1h 19m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/db5df5088e8f4b2d987d1be421f8ebbb : SUCCESS in 56m 22s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6ed2820ee6b54b27a2abfc5ff38439c0 : SUCCESS in 59m 00s","accounts_in_message":[],"_revision_number":2},{"id":"80ea66c8c664445e3c1a37e75c8945ad2128af18","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 17:49:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a4665ab9cd8a4d7450d0fa4f8dbe5dfb429aec62","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 17:50:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0328359b2f0f44be9593af50aed6468e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a34eb8871be845308c3258f7db1f94eb : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a14fe4d64be2435d94d0fa536dcc965f : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"421c7fdce87f2874fcab271cbc7e3ecafec9bd6a","revisions":{"7d3e7f98c07eb636e4c79b6299502e4d32bf591b":{"kind":"REWORK","_number":1,"created":"2024-01-16 09:27:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/62/905662/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/905662/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/905662/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/905662/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/905662/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/905662/1"}}},"commit":{"parents":[{"commit":"88d7315a60314e44fcce88ad198ceb3c0c107fe6","subject":"Add Noop Certificate Manager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/88d7315a60314e44fcce88ad198ceb3c0c107fe6"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-16 09:26:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-16 09:26:43.000000000","tz":540},"subject":"Stable-only: Pin hacking to \u003c 6.1.0","message":"Stable-only: Pin hacking to \u003c 6.1.0\n\nHacking 6.1.0 pulls flake8 6.1.x which is not compatible[1] with\nflake8-import-order 0.12.0 pulled in this branch.\n\n[1] https://github.com/PyCQA/flake8-import-order/issues/189\n\nChange-Id: I5a5d746657da3501fb332361fee6ea25186285f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7d3e7f98c07eb636e4c79b6299502e4d32bf591b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7d3e7f98c07eb636e4c79b6299502e4d32bf591b"}]},"branch":"refs/heads/stable/2023.2"},"421c7fdce87f2874fcab271cbc7e3ecafec9bd6a":{"kind":"REWORK","_number":2,"created":"2024-01-17 02:50:48.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/62/905662/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/905662/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/905662/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/905662/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/905662/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/905662/2"}}},"commit":{"parents":[{"commit":"88d7315a60314e44fcce88ad198ceb3c0c107fe6","subject":"Add Noop Certificate Manager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/88d7315a60314e44fcce88ad198ceb3c0c107fe6"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-16 09:26:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-17 02:50:34.000000000","tz":540},"subject":"Stable-only: Cap hacking to \u003c 6.1.0","message":"Stable-only: Cap hacking to \u003c 6.1.0\n\nHacking 6.1.0 pulls flake8 6.1.x which is not compatible[1] with\nflake8-import-order 0.12.0 pulled in this branch.\n\n[1] https://github.com/PyCQA/flake8-import-order/issues/189\n\nChange-Id: I5a5d746657da3501fb332361fee6ea25186285f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/421c7fdce87f2874fcab271cbc7e3ecafec9bd6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/421c7fdce87f2874fcab271cbc7e3ecafec9bd6a"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
