)]}'
{"id":"openstack%2Foctavia~905720","triplet_id":"openstack%2Foctavia~stable%2Fxena~I5a5d746657da3501fb332361fee6ea25186285f3","project":"openstack/octavia","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-01-18 20:10:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5a5d746657da3501fb332361fee6ea25186285f3","subject":"Stable-only: Cap hacking to \u003c 6.1.0","status":"MERGED","created":"2024-01-16 17:17:15.000000000","updated":"2024-01-18 20:11:45.000000000","submitted":"2024-01-18 20:10:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"905720","cherry_pick_of_change":905662,"cherry_pick_of_patch_set":2,"meta_rev_id":"d5f409e9773cfb5038491e0d6607855c9ea14791","_number":905720,"virtual_id_number":905720,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-18 20:10:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2024-01-17 16:11:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2024-01-18 15:54:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-17 10:42:46.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2024-01-18 15:54:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-16 19:33:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-17 10:42:46.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2024-01-17 16:11:14.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2024-01-18 15:54:08.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"8dc8dc3a5f3f9ec78159b5025b9c5be63a190ab6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-16 17:17:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e8f655272c7f9aa90148b5db450a860f84c6cd00","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-16 17:17:24.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"2f8f139d749050d567734ddc3fd7fb9d2975e6a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 19:33:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dc97c191a1e14b888dffe3d1184a625c\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/816c2ad6739345b09742d882190528bb : SUCCESS in 2m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc6505b6309f49db9b4a58b25e1f1112 : FAILURE in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50b2c72e30f04d599300b8aa8138e106 : FAILURE in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/92dc6cd5569545d8b2e31a8fd4f396da : FAILURE in 2m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ea6990a806e4ec2b01a6df022fa058a : FAILURE in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb4832c7cf654d4ab57e69e5ab08c689 : FAILURE in 3m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3630644203642759c830142cfd541a7 : FAILURE in 4m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/62f3de840e2648ee958c5c5fd517b876 : SUCCESS in 5m 12s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/b24adc1584124b599ba36f7cc020de77 : FAILURE in 3m 02s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/a7d8303a8aad4282b128784a3e9e3112 : FAILURE in 3m 32s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/045bd98ece874e6e817211b3311c1a29 : SUCCESS in 5m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/78c549f10348439ea6d8ffc659a163d6 : FAILURE in 5m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a1765ff431bc4fc9a5d0272c0a35de9a : SUCCESS in 2h 07m 32s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/aaf952a519fc4736bbe2e838d171b667 : SUCCESS in 1h 23m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fe4fefac7dcb47f1888bd5bc70e3d0b5 : SUCCESS in 32m 29s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/5505a22201ff4655911553f2b6dfe7ca : FAILURE in 10m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6d976974d03e44cb8a3469ad516b2684 : SUCCESS in 29m 14s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/07050ebbae7a4fc5a3313a05f800f075 : SUCCESS in 1h 46m 19s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/48af4ca9ce564589b8a24d44232e3abc : FAILURE in 10m 26s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/dc50e5368c3b47ccb194dc972fc78fa9 : FAILURE in 5m 29s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b350dbb6ef004b96b8ffd5d0e6fc9e25 : SUCCESS in 1h 49m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/b5f78c2e40a048ac948fcb187c26b307 : SUCCESS in 1h 23m 04s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/3dd5ee0a551b42128ab7683423c6b090 : NODE_FAILURE Node request 300-0023319993 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5a90eb80df72ebd3eff20e9c1f4fa514db23dfb9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-17 02:55:18.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"340104195a207879eea59e82b7114e14150317ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 05:13:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c45db28a1ad4ffa81780a00aea44d93\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a8add42fc3914f58b3cbc1d4ef8fd055 : SUCCESS in 2m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb1ee76cbe804fd3aa7e1d49930b146e : SUCCESS in 9m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98a1de2b8d9f40b7a4231b61e5c6ca53 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/23e832d484b349b4b3b50b42b9cb2773 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6bb40b821ad54ebb9f5e6bee0a09ee7f : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/10e0be7988bd4571b0d9edab21ffe0e6 : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b90a86a99c546ba8a81b2d1c503a194 : SUCCESS in 11m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8b42ad36432441c8a93d3ff9270db2d : SUCCESS in 5m 52s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/e3aa872368624036b8ad501fdfc7bcfc : SUCCESS in 4m 51s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/4d1884059f3f4257a8c1f61d08052710 : SUCCESS in 9m 04s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/17c33e4601a244d5a29d9a2b650d42dc : SUCCESS in 5m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/151917eba3aa4d7d954194e864973892 : SUCCESS in 8m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2d5a8fe7a4314c128a3452567197ee08 : SUCCESS in 1h 31m 29s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9811572394a94678bfa5e61fcd402416 : SUCCESS in 1h 27m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/68962ec216ac4b29b03a51986871dbe6 : SUCCESS in 33m 30s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/9940469da0d24c02bd2817f150d07ad3 : FAILURE in 13m 59s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7c903346640842df8379dc0d628f2c23 : SUCCESS in 28m 54s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/913a0bf388aa4106a5d1b06d10ae4c49 : SUCCESS in 1h 32m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0551db910e454f269c658d10b1ba8066 : FAILURE in 12m 24s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/aa8204373e2f448182cd613f899f3eba : FAILURE in 6m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3898ff3fb3124ae1ac97490db21248c5 : SUCCESS in 1h 45m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/30abcd007d0c4eb7960a1f2c29c8d7ce : SUCCESS in 1h 59m 46s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/f4c1c44ad48644e09a5ed36e758d98bc : NODE_FAILURE Node request 300-0023325685 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"16ae88cb068e3d4fcb01f1a5b912fab9a7966c5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-17 09:27:21.000000000","message":"Uploaded patch set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"71f38ca8998981f994164dc84f2e6665d63a79c7","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-01-17 10:42:46.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"18b4a45c5831575b426f732f33a3684c1f1a2872","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 11:25:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd38f62758f84e97bbfefe9a2b94e307\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/5312c7b1a18d41009fa8818a2cf1bb94 : SUCCESS in 2m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c290f1b11b34e859bcf5e2e900ad432 : SUCCESS in 11m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a02e8611fbc74dcb808821a88554cccd : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8df416e2578347caae3c4a8b1d0bf1c8 : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df74e3dec0a84c1491ebcbb44aed328b : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ca9559b8536460aa094adcea1586efb : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca036dc0832a47878974c9abafec0832 : SUCCESS in 12m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be0ed08a27a9487a9343521737927ff3 : SUCCESS in 5m 32s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/aadb42db11c44c9f87f8aae28f683d1b : SUCCESS in 5m 26s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/8fc1536445fe4286ad425c5a81256bf8 : SUCCESS in 10m 50s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ef5725836e3745b780f94450e74018bc : SUCCESS in 4m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d25c465964af4621b264b1cdff4999cb : SUCCESS in 7m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8bca269cadf049fb8e3f141de017595e : SUCCESS in 1h 20m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/73b2c32c10cd49ab82d38f0c8ce92d02 : SUCCESS in 1h 43m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/509f0f63d6b04d76b14b16264dcf25f6 : SUCCESS in 32m 49s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/34966664929444deae9af5bf611f55e2 : FAILURE in 10m 51s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0903fc0bd90249b8ac6f1839f251200d : SUCCESS in 29m 46s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/dd8a84ce251f45fd8edd0eeec9d90d31 : SUCCESS in 1h 38m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c3c5324eacb7426690ce99d6d0e2a79d : FAILURE in 11m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0e5d7839f8fe45cfabf6651cbac2a035 : FAILURE in 8m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b4c53437600e49ee95ea8b84fdb2badd : SUCCESS in 1h 51m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/392277960dc64c958e8bed20ab43a8d5 : SUCCESS in 1h 55m 48s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/e9f62c6e7d0d487684adcdcef86e52c7 : NODE_FAILURE Node request 300-0023327989 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"255734065f7e5bdc2d80f6dad320b0aaded8dcd0","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-01-17 16:11:14.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"235e51008e727d864a41a1c75e53764d976b07bc","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-01-18 15:54:08.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c8ff3a05544672e03641fbf519c3dec6070b9c68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 15:54:53.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"97b549e80604d78fe2417eef2a2407af6fffeb84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 20:10:28.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da1f91296cf24135b0758770dd13c7e2\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/6ccbaa082a274e67a3626f1876008da4 : SUCCESS in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05985b1c15aa4504acec890f00cbc961 : SUCCESS in 7m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07195b97831c4228a3b08bc49c827ad4 : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe112427d90f4309924f3084ffd093fe : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8167e8fe0524795a63aa721ac713116 : SUCCESS in 12m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f83a0de0dbdd46ac9cb1d1ea05902dee : SUCCESS in 5m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b37c6023a5f24bdf9fce0706fb072918 : SUCCESS in 6m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ce0b4d893607462ba1bb3a6faf1797cf : SUCCESS in 8m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/384a80317ac043fd9b4655951942ce41 : SUCCESS in 1h 33m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7f5444fdb28e4810af1af5d12a95d48d : SUCCESS in 1h 38m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4fdec0cd9faa400e88160defe968b3ce : SUCCESS in 33m 35s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9ae26d2b9cf44442b524c9c9c21b5cc8 : SUCCESS in 38m 58s","accounts_in_message":[],"_revision_number":3},{"id":"04b451b3e2e87072ffb6a5d53e6cdfd6609d3ad9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 20:10:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"d5f409e9773cfb5038491e0d6607855c9ea14791","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 20:11:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fb343e240234e0d96406ae9d1acdb64\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81da4a7f5df24db98533b00ccba37d72 : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f948d9bd891460bba8362e4b570ed24 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"457aa7a229061753cee050a9e66259bef78448e3","revisions":{"7878cc78132ca42f72e271758fc8fd65a8ed44fe":{"kind":"REWORK","_number":1,"created":"2024-01-16 17:17:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/20/905720/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/20/905720/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/905720/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/905720/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/905720/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/20/905720/1"}}},"commit":{"parents":[{"commit":"6079b9c4cdb336ad485f95cf3fa82520e22105cb","subject":"Merge \"Fix incorrect removal of IP rules in the amphora\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6079b9c4cdb336ad485f95cf3fa82520e22105cb"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-16 09:26:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-16 17:17:03.000000000","tz":540},"subject":"Stable-only: Pin hacking to \u003c 6.1.0","message":"Stable-only: Pin hacking to \u003c 6.1.0\n\nHacking 6.1.0 pulls flake8 6.1.x which is not compatible[1] with\nflake8-import-order 0.12.0 pulled in this branch.\n\n[1] https://github.com/PyCQA/flake8-import-order/issues/189\n\nChange-Id: I5a5d746657da3501fb332361fee6ea25186285f3\n(cherry picked from commit 7d3e7f98c07eb636e4c79b6299502e4d32bf591b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7878cc78132ca42f72e271758fc8fd65a8ed44fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7878cc78132ca42f72e271758fc8fd65a8ed44fe"}]},"branch":"refs/heads/stable/xena"},"98c5f656467773593744e8c73ce00030608733f0":{"kind":"REWORK","_number":2,"created":"2024-01-17 02:55:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/20/905720/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/20/905720/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/905720/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/905720/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/905720/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/20/905720/2"}}},"commit":{"parents":[{"commit":"6079b9c4cdb336ad485f95cf3fa82520e22105cb","subject":"Merge \"Fix incorrect removal of IP rules in the amphora\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6079b9c4cdb336ad485f95cf3fa82520e22105cb"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-16 09:26:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-17 02:55:18.000000000","tz":0},"subject":"Stable-only: Cap hacking to \u003c 6.1.0","message":"Stable-only: Cap hacking to \u003c 6.1.0\n\nHacking 6.1.0 pulls flake8 6.1.x which is not compatible[1] with\nflake8-import-order 0.12.0 pulled in this branch.\n\n[1] https://github.com/PyCQA/flake8-import-order/issues/189\n\nChange-Id: I5a5d746657da3501fb332361fee6ea25186285f3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/98c5f656467773593744e8c73ce00030608733f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/98c5f656467773593744e8c73ce00030608733f0"}]},"branch":"refs/heads/stable/xena"},"457aa7a229061753cee050a9e66259bef78448e3":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-01-17 09:27:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/20/905720/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/20/905720/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/905720/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/905720/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/20/905720/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/20/905720/3"}}},"commit":{"parents":[{"commit":"6079b9c4cdb336ad485f95cf3fa82520e22105cb","subject":"Merge \"Fix incorrect removal of IP rules in the amphora\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6079b9c4cdb336ad485f95cf3fa82520e22105cb"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-16 09:26:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-17 09:27:11.000000000","tz":540},"subject":"Stable-only: Cap hacking to \u003c 6.1.0","message":"Stable-only: Cap hacking to \u003c 6.1.0\n\nHacking 6.1.0 pulls flake8 6.1.x which is not compatible[1] with\nflake8-import-order 0.12.0 pulled in this branch.\n\n[1] https://github.com/PyCQA/flake8-import-order/issues/189\n\nChange-Id: I5a5d746657da3501fb332361fee6ea25186285f3\n(cherry picked from commit 421c7fdce87f2874fcab271cbc7e3ecafec9bd6a)\n(cherry picked from commit 7d9f2914af0dfb59188caf99d00b4013f69c1c2f)\n(cherry picked from commit 85f94c25e629877f92562e16aadeffab30a3ee0d)\n(cherry picked from commit fd5185f8ed4691437ac7053dd9402134ae995d72)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/457aa7a229061753cee050a9e66259bef78448e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/457aa7a229061753cee050a9e66259bef78448e3"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
