)]}'
{"id":"openstack%2Foctavia~905842","triplet_id":"openstack%2Foctavia~stable%2Fzed~I5a5d746657da3501fb332361fee6ea25186285f3","project":"openstack/octavia","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-01-18 17:59:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5a5d746657da3501fb332361fee6ea25186285f3","subject":"Stable-only: Cap hacking to \u003c 6.1.0","status":"MERGED","created":"2024-01-17 09:26:24.000000000","updated":"2024-01-18 18:00:42.000000000","submitted":"2024-01-18 17:59:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"905842","meta_rev_id":"9a23d9975a5773ac52913fcb4e0e60b9c4f90593","_number":905842,"virtual_id_number":905842,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-18 17:59:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2024-01-17 16:11:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":2,"date":"2024-01-18 15:53:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-17 10:41:59.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":1,"date":"2024-01-18 15:53:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-17 10:41:59.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2024-01-17 12:04:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-17 16:11:07.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2024-01-18 15:53:49.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"f938ace4431966c646fd0c99ae0810377ab11d84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-17 09:26:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"92af592c318bf28c4aa9b636857e7b2b244eb394","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-01-17 10:41:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f83b22c102aeb9ed82443b40a60c54193b88946e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 12:04:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/783baf2d4d384eabbc33a4540d77b617\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/1eef7922e23a4ae889193bc81639c6ad : SUCCESS in 1m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79f0a3b3b2bf46fa93517658ffc96720 : SUCCESS in 11m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14f4c6a7347548d98ba71da71b051f7c : SUCCESS in 7m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c9c02ff2f404efab05836b51107cd8c : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b49e5dd336884c7c862b5507f51c3447 : SUCCESS in 4m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ae5575a607b64043a991c8e16962adca : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37eb7561a2f64619856a4a980afc5ce5 : SUCCESS in 12m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8329d78d18f4798b39fefe5ed375fab : SUCCESS in 5m 46s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/b2c61fd3bb1a4ba0b4e1cac0186e8157 : SUCCESS in 6m 50s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/6d6d8f5f25c0469baaf246cb103dbab5 : SUCCESS in 8m 32s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b84d94757f194f999a043899797391b9 : SUCCESS in 6m 10s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b79479f3f4854b729e254d5e670f8499 : SUCCESS in 8m 49s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2768cafe4349482698ba64383ff10507 : SUCCESS in 8m 38s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/26ce2ea900304c02b3eba65d895f36e0 : SUCCESS in 8m 03s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7d58b551a56c4346b75e7950560b2eaf : SUCCESS in 1h 17m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a8b514f03c694b2c99d1545abae3e205 : TIMED_OUT in 2h 33m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b343d0ed016444e59d85c6840a2ef342 : SUCCESS in 48m 18s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/4d0dd4508ff145dc82a7396327aa8ece : SUCCESS in 1h 12m 06s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/880c003b1c3d4fc79b793b85a7ef53da : SUCCESS in 41m 13s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/64edbfea74194a58aa07894326aa7d65 : SUCCESS in 2h 27m 19s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3c689a82926c4969ad88c39321f70a43 : SUCCESS in 2h 04m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ffb23d6cc23a4f8a82ef3fa2d98980c5 : FAILURE in 22m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/65f890ef1c4644d881b393eb3c3978f1 : SUCCESS in 2h 01m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/84d15ade301e493dbf2bdfc9db892837 : SUCCESS in 1h 40m 35s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/4d981287b3494328ad22027f6bb3b32f : TIMED_OUT in 2h 34m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"347b3b09ab10e06236d7536ff969713e811868f0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-17 14:21:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b3e8388622ff33fcae342ec7377c01edc8aff936","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-01-17 16:11:07.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5cce68b99ffe7654f7fbc282ac6388d1709b15b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 17:18:07.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c5d5341459264f66bdd34dcf8a4a3332\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/7f7146ba50a04cf99a3a053c44382881 : SUCCESS in 2m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/627aeb609cc849288aa0beb5bd4015a9 : SUCCESS in 10m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a09fd4c3c0604abdad3897e72cc0bb14 : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86af059778874c7fa217c16ae7f811a4 : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/38377583210046c4b1608169b2a1172d : SUCCESS in 5m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a46d6aa3bbf3475582dab25948d1d032 : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f0db7626a194aafb1e7b512fb49d827 : SUCCESS in 12m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87fea41f45b14831a4d2d6bfa43c5750 : SUCCESS in 6m 19s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/53bfd653db434c80a8804ea3eefe67aa : SUCCESS in 5m 23s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/50359b4d06904db5b582975871775cb0 : SUCCESS in 8m 44s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f61ef1d0b6724a008e968bdb5efa2570 : SUCCESS in 4m 53s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/339b6e37475c46ee9c05c439ba388a06 : SUCCESS in 8m 16s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d4930f1ef3644dd0b39d6fb91e519767 : SUCCESS in 7m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/351a103a1f50422d8ecdf315e3223ba0 : SUCCESS in 8m 58s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fe50dfc2a4214f75a5a1f807360dfbfc : SUCCESS in 1h 44m 52s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b9168fe7a3634a729b8f1b85f4194f0a : TIMED_OUT in 2h 34m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f9de74c862f24df68e2d9429dd571141 : SUCCESS in 1h 01m 12s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/354818ff0c7e467a8c15930dba49bf43 : SUCCESS in 1h 15m 51s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/407ba281918c4d01bf37ade8a850d1c0 : SUCCESS in 33m 02s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/01d09bb0fd4c4251b4b28d348ba95a76 : SUCCESS in 1h 43m 16s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/113cf5963303411c8725852158f6bd21 : SUCCESS in 2h 00m 51s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/77079cf078f34880924aabf38d7caf5e : FAILURE in 23m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/37fffb8246b64cc39fd6d1f1d0985b38 : SUCCESS in 2h 16m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/080d1042b98e45d58d962111d5d16735 : SUCCESS in 2h 22m 12s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/72bb6e5f7fa74188bcfcb40bc4bc013f : SUCCESS in 2h 22m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6079d932695e04134600794791c5564a3419bab8","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-18 02:30:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"079c4a03af26c7ba4c4d2db169e4611d0adf58d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 05:22:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccac1a01b7f746d3a6cc4f3ec85a02a7\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/6fce9dfbadf64e0aaea8ce0402cfd136 : SUCCESS in 2m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b950e6eda1c4c0db076f4a096f12b47 : SUCCESS in 11m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/732e386dc84a43e2a5b4aa0120262e43 : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e95205e4cde40b5ae98a2b52b184e4a : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9c364a1d4c244fa2a9723203c17f6328 : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/85570be697df47cdac78557c83e2bfa5 : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40dc76c7f5fb4d1f8cb7b803bffef637 : SUCCESS in 11m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c6a21c61aef401e8b7f9f9d4c9ab421 : SUCCESS in 6m 31s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/60276a2c5bc245298a018c92626b1cc7 : SUCCESS in 5m 25s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/21bc7c873eb54d7c9a378d2e39b9be8e : SUCCESS in 9m 29s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/57461701ec224791ae170508f126fe7d : SUCCESS in 5m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9db73e761b2245edb20d7d09a94599d6 : SUCCESS in 9m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6da4732de42649aa8ba38504721a8ee8 : SUCCESS in 6m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8fd21eac9b454241aa9e8abfd1c8cee3 : SUCCESS in 8m 06s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/267525deaa574ca8aecd74c5c84974c1 : SUCCESS in 1h 53m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/500e3d051d0e4aafbdd32e4d7d6c1cb5 : SUCCESS in 1h 49m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f7b36cb6f99048ba8208f73ccd32d3b9 : SUCCESS in 37m 26s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/33f37b2022024078a1d31c8ccb958122 : SUCCESS in 1h 06m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/42efb88007a345d8b451b262318f1b73 : SUCCESS in 31m 29s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3a981bbb058642b7911efe3b815550f7 : SUCCESS in 1h 53m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/16b166f5c1684a579e5bd2f0cc3fbadd : SUCCESS in 1h 55m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/bd84325100734e3ba418b37439e219bf : FAILURE in 24m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/cbbadad518a547abbaa24f5032604a37 : SUCCESS in 1h 46m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/012a751f7f5f4522949a78d5cfd66959 : SUCCESS in 1h 27m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/177a35ae09eb4252839e2d80e17fda4c : FAILURE in 2h 16m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e52893e7fdae8d47498bcbd84d64683a76f29c18","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-01-18 15:53:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e1fd119ab3e7296b5bce2dee1fe600ce19ad4c0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 15:54:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"772a0f3d9fda4d5ed9b4d4aeb057886fd45e8305","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 17:59:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d80e66b563164794aa38dfe998e5ad45\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/bffb318ec437401398611e6e1c3bdbec : SUCCESS in 2m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1877f94dc90b442aab44b65f2634dad2 : SUCCESS in 7m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30b5975a32ae41f2ae7c341694825289 : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/92ad3b0129db4ea88d950438f305eb22 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13c79176130b448ab25fccb8bfedd8c1 : SUCCESS in 12m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8b7c37a24ff45639ba89229e57f6baa : SUCCESS in 5m 11s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/852522ed55be43ecb3c8bc66ad8e2791 : SUCCESS in 6m 27s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/77ad23d0dad84dcdb0960dabe723979a : SUCCESS in 9m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7f8b1718040245e1a59e545e471f7c0d : SUCCESS in 1h 18m 59s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3031d128f4694a6bb74a090a55b4a33a : SUCCESS in 1h 38m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5d424f1dec2b49d2950679b08f0cb121 : SUCCESS in 36m 12s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/de03c6a0c78a458fa65e6ed4c2657a7c : SUCCESS in 38m 35s","accounts_in_message":[],"_revision_number":1},{"id":"cd13bd30bc7036ab009c9074b03a66fd4ec308f6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 17:59:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9a23d9975a5773ac52913fcb4e0e60b9c4f90593","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 18:00:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5dc6f79afac24352b0727c51d8fa7f2e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b242a3b4b2c841cb85dd1cb6d403f2ea : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8eec59e363e04adebdea950565fe8722 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"85f94c25e629877f92562e16aadeffab30a3ee0d","revisions":{"85f94c25e629877f92562e16aadeffab30a3ee0d":{"kind":"REWORK","_number":1,"created":"2024-01-17 09:26:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/42/905842/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/42/905842/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/905842/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/905842/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/905842/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/42/905842/1"}}},"commit":{"parents":[{"commit":"07c33fec4578e2529a2a5fc8fad4cde929597849","subject":"Add Noop Certificate Manager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/07c33fec4578e2529a2a5fc8fad4cde929597849"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-16 09:26:02.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-17 09:26:15.000000000","tz":540},"subject":"Stable-only: Cap hacking to \u003c 6.1.0","message":"Stable-only: Cap hacking to \u003c 6.1.0\n\nHacking 6.1.0 pulls flake8 6.1.x which is not compatible[1] with\nflake8-import-order 0.12.0 pulled in this branch.\n\n[1] https://github.com/PyCQA/flake8-import-order/issues/189\n\nChange-Id: I5a5d746657da3501fb332361fee6ea25186285f3\n(cherry picked from commit 421c7fdce87f2874fcab271cbc7e3ecafec9bd6a)\n(cherry picked from commit 7d9f2914af0dfb59188caf99d00b4013f69c1c2f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/85f94c25e629877f92562e16aadeffab30a3ee0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/85f94c25e629877f92562e16aadeffab30a3ee0d"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
