)]}'
{"id":"openstack%2Foctavia~906327","triplet_id":"openstack%2Foctavia~master~I790bb48291e30a6aa233e082f9e211f18409b3e4","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-04-17 18:39:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I790bb48291e30a6aa233e082f9e211f18409b3e4","subject":"devstack: Drop removed services","status":"MERGED","created":"2024-01-23 03:50:40.000000000","updated":"2024-04-17 18:41:38.000000000","submitted":"2024-04-17 18:39:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"906327","meta_rev_id":"9fd1fb3c549dbfaabec35fc494be7cfbfe081b3d","_number":906327,"virtual_id_number":906327,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-17 18:39:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2024-02-23 00:29:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-04-17 16:41:59.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-04-17 16:41:59.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-23 06:35:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-23 00:29:10.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-04-17 16:41:59.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"024e8f4f10d9fbf10ff7a868e3fd3050a9f98d40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-23 03:50:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6c53786f45d6508cbe85e5e2e1d61c1c14d65f2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 06:35:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7beb3b383d574833b658fa91d2039d02\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84d52718d00f41bfb5db9dc290bc600b : SUCCESS in 10m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b3941f5915542f4980517cf05a5260c : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/98f523128e5f407ca5ebf36b14bf2e26 : SUCCESS in 6m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/015f28d4d0cb4803b5fb93fbaf8fbc3f : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef7bcfeea26a41e6aca86c5a8dad94a0 : SUCCESS in 14m 20s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/e98bb46ba5b641ebbf5212e686cda9cb : SUCCESS in 5m 34s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/0c8d7002971c416fa3051c4300df0435 : SUCCESS in 7m 06s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/fbf0541fd5f442f2839916b0b13ea10e : SUCCESS in 9m 48s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d5d87593623d4919ba54b77d5f5e3145 : SUCCESS in 5m 58s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/73d3b2b6cfbb45359ba66c826a2397a7 : SUCCESS in 11m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/804e6fea0c414482b1231483296613fd : SUCCESS in 7m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/47158faeb21345439f44ecdd0eb5ae33 : SUCCESS in 9m 17s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5f7f3270579a424bb802ac68aca5629e : SUCCESS in 1h 00m 10s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/892bfef5d60b42b0acac097decf5f8a0 : FAILURE in 1h 35m 14s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/c5ea3d4e544541d4a91770146615f5fe : SUCCESS in 57m 37s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/eb5603f728f74a2689dceb44314a809b : FAILURE in 1h 39m 02s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3c080eef63a24a858dd88fb1438d99cb : SUCCESS in 56m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c1efb017078f480ea425c826ed527d9c : SUCCESS in 35m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/60aaf7f1dfac404b855043db737d38d0 : SUCCESS in 38m 45s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/e047e69949be45338adb22ad061bf30e : SUCCESS in 44m 42s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/e2c451786663415abbb8342bde47a90b : SUCCESS in 1h 12m 14s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/64c40d933ace4bca91004ef656f5da64 : SUCCESS in 37m 20s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e2a2c033488943bcb8212502e5ded069 : FAILURE in 2h 27m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2c49e8bad99f4933a77d7f6568b64ce8 : SUCCESS in 2h 21m 53s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0cd226b47c69442bbb6210428e2ee3dd : FAILURE in 22m 10s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/7468d83eb4644cd0a8b165aaf8b91a00 : TIMED_OUT in 2h 34m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"65bd8400959b0edf5a426463e195b5547cece74e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-23 13:00:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3eea3728018b04aa7df6eb63b48b4601ebcadaa7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 15:48:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/224cfa5aa24e4e7990f139db112fb2b8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa6e8c61ca304bb296c634f4580c1577 : SUCCESS in 13m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51e4104980a9482b858c4d1d92be4d68 : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff539be008504b988ce0e77cfb584a0a : SUCCESS in 4m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/16d58a685eb249d8a35fd049548cf432 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60b1c3542870431cb34f92edaa8af8b6 : SUCCESS in 11m 53s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/202b6d02988245798eab2d7e17b06fc3 : SUCCESS in 5m 13s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/6f8dd9f1563546689b5090fa4a9b153f : SUCCESS in 8m 40s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/7d7b00cf313f406d824902d554319763 : SUCCESS in 10m 23s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/dcb5b4cf050d41739f99f71630fd0ac3 : SUCCESS in 4m 45s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9f6ed0333bac4ac8a28208d93c860314 : SUCCESS in 6m 53s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9bb59d65e10b4e3686c13ce983b1bbc2 : SUCCESS in 7m 25s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/457c27113cc34df4b8a71a9924c8af10 : SUCCESS in 8m 44s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/853da501ff2f4272b868791cdaec9549 : SUCCESS in 1h 18m 34s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/86c163f9991e46e2b7d75dcc4e9252c5 : FAILURE in 1h 34m 27s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e6d2d69bd26a4485b13f93ae0ab74883 : SUCCESS in 59m 08s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/4df8dd022e314382a3db86f7a9801973 : FAILURE in 1h 40m 02s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/edf5d971eabc4dfe88172b50b9272373 : SUCCESS in 1h 23m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7cb77db1fc6e448582126ac92bec725b : SUCCESS in 56m 08s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0c01b3f6a2dd4c1c90f2a04d423e9e88 : SUCCESS in 39m 25s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/3b92efb3c37e4fd4a37d01b7c4992b65 : SUCCESS in 59m 07s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c849778081014321b9fb4e480e594423 : SUCCESS in 1h 10m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ada7e0eaf0f54a438662a474791580a3 : SUCCESS in 33m 53s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3645e90c369b47eda057ee9f7c8ce6e8 : FAILURE in 2h 16m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ff13e6cffbf04ffaa2cd184d95a43208 : SUCCESS in 2h 45m 40s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5bd0088baba54d3da674ed29edaf1b50 : FAILURE in 37m 43s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/aed49230ba56497f9701dd5248cf2029 : TIMED_OUT in 2h 35m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4aeb9c11aa1b05bc59bb7008f1a2ad4cf90ef386","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-24 16:48:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bcf8a71cb14c94e02ee8589def81642130910be9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 19:41:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3c5b763f13ab4c8494e4ad77f84342e3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b98b2b2852243599166f2cd742ea962 : SUCCESS in 12m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82bf4fa9d8ef4ef2a438141a2c3ee526 : SUCCESS in 7m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/61f82433d5044a20b7d67bb6cb3f01bb : SUCCESS in 4m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aabf90d3b17d4a33b4f0d3031975675e : SUCCESS in 6m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d470d0d149b4667a4708f91ea9b4478 : SUCCESS in 13m 58s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/03183a5bce174cae93a8d90e7b8e4ef8 : SUCCESS in 5m 35s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/e6b5b80b8b194679834abbe8bdf29186 : SUCCESS in 9m 27s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/bb9c78c3b91c40c3848e13ea35b45839 : SUCCESS in 11m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d4ab932b78fd4ecabdcd956cccbdde3f : SUCCESS in 4m 41s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3b1e278cdb67466c83b965754a2e9755 : SUCCESS in 10m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/20cc03afe15d4762a6f254ac3890d26d : SUCCESS in 7m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/82490bc38bb9466fb4bf69992f1325a9 : SUCCESS in 7m 47s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7411136e9f034f6b8ed568b3c6c7d544 : SUCCESS in 1h 25m 55s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/7b206f5f16ae4a55962d0ccea496d0c9 : SUCCESS in 2h 05m 32s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/850d161df5ce4efcabefe285c73934d4 : FAILURE in 1h 23m 29s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/80bfde3a3e26461e8e82b2b96aa6afa2 : SUCCESS in 2h 00m 02s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f80f2951ed2145d59c65de84fe9402cd : FAILURE in 1h 58m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8670fcf738154589ae923e855c096fc9 : SUCCESS in 49m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/43a51ff2212f4c82868c368ed8e04fbc : SUCCESS in 49m 54s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/675a1ed5a78c449bab9a810bae7b8b6d : SUCCESS in 54m 18s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/260da90896e24547ae68bb684666460a : FAILURE in 56m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0517c702e13142d58003625a79daa238 : SUCCESS in 36m 13s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/757fb047036d4ee6954f8cb869421285 : TIMED_OUT in 2h 32m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2273c6b8990e47b49e24f2b6a4744dac : SUCCESS in 2h 51m 36s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/94554dc424364a5a85f3f47bd8a2b12d : FAILURE in 31m 46s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/33be7f3ae2324566aaedcfca1d9936c2 : TIMED_OUT in 2h 34m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e355f1f9eac8aad848e87e83acc6061e07788b4f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-28 13:02:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"97b9be835828d419e275b792bb64af883933ab19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-28 15:36:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/10f23a97960144b0b0022e802ccdc84a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6094d797b6c4eb285e6e1a0c9cea1a8 : SUCCESS in 10m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d80774c03484743898a2540b40a98b2 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c87df7c2fb4f4458b5888628b4daf25e : SUCCESS in 5m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5ba49e89d76647c08825a24444172532 : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a59c90a744de41419f5814f207316968 : SUCCESS in 11m 11s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/523ec176c6ce4bfbba3e14b0c961a010 : SUCCESS in 5m 14s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/636cc828fe564ffbb55c068964994c72 : SUCCESS in 8m 36s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/0f4a69203c994b088ba27cb1fc33871e : SUCCESS in 7m 49s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f7c05861a1f5409095d9b0ba0208541f : SUCCESS in 4m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0e3ca0404e81480ea188765881ce3899 : SUCCESS in 7m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/de71ebb78dfd45eea9ec9c88e1df16ef : SUCCESS in 7m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6dd354bf39ad46a8a64a37f873973b97 : SUCCESS in 11m 42s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b443f08cb9da480c9f6a8bfb6c30c78e : SUCCESS in 1h 15m 42s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/9306d868a7c04374a39e7eebbf9e0be7 : SUCCESS in 1h 33m 03s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/2042586cdf574ae9b499d8cc71205be0 : FAILURE in 1h 28m 55s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/40897aef84fd48a5ab05fc664d7b7915 : SUCCESS in 1h 58m 24s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f81f7c4817e946c38d0da93de1ceb93c : SUCCESS in 1h 03m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9ad12476f0a34f3fa95fba7b0b631188 : SUCCESS in 59m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/48276afcb7814a378eb1dd04650f1399 : SUCCESS in 55m 46s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/32c86b8cf49f4dab892743357b49fea6 : SUCCESS in 37m 35s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/262e077b14634bee9985b643e8e930b2 : SUCCESS in 1h 01m 23s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/54251926a2f54d79a331253db0ea2a31 : SUCCESS in 50m 24s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e127c1a57f724bc1a8c8f990b366463c : SUCCESS in 2h 18m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d24e2aef678d415b930d3e961d18625d : FAILURE in 1h 25m 54s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d3bffb3eee1b4856b83ce2c2773e96e4 : FAILURE in 33m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/486bd7a4a5ad4139b4603fb20247f11a : SUCCESS in 2h 06m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"72d06568333861f491aa888fe82000e983452c77","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-02-23 00:29:10.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fe5fa4faabd56d9c4e7eebec29bb434df2f06b24","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-29 17:00:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7b041b0464a0043c2f0ec2b4c58325f556ae9c2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 20:21:46.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/80eb534684b0403b8f9519c083707c4a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f431055e408d4c369769bd21f5356d0e : SUCCESS in 13m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a57eea622d3499a9eefd124cd192851 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f3f64ce66284339a20b54b9bd665a8c : SUCCESS in 4m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b3afe910ad144622951a41d7b4e8b5dd : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/473b2396e80c4294bdaa1e7e727dd14c : SUCCESS in 11m 31s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/a3bcc4c287bc4648b0ba676e618723d1 : SUCCESS in 4m 30s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/63368f8d6cca46a28691983ed4fcd687 : SUCCESS in 6m 34s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/8b7caabddf7f4a7992cd745bc8ceca62 : SUCCESS in 10m 23s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a8950e2ca9ba43aead86143a64cd9826 : SUCCESS in 4m 29s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/dc1391faa3c3470c874a7e48abe6f745 : SUCCESS in 6m 26s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f944278742ca41a8823d62e0243ce7fe : SUCCESS in 7m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/01d0d2d30a524ba7ae08d5e6f3bd8238 : SUCCESS in 7m 15s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/87d34a4768864ae8a2f56b0deecfa743 : SUCCESS in 1h 19m 24s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/95b1372e537c45e58272952e6a1891bc : SUCCESS in 2h 12m 57s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/d61ab85789f64e3c8e00d81c835528dc : FAILURE in 31m 32s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/618a5cc54113449fa261de48187331e5 : SUCCESS in 2h 15m 29s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c4b7c2be7981449b84a2f245b6f8fc33 : SUCCESS in 1h 22m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/92d1cd3b769a45e7965e6b233ca7951c : SUCCESS in 56m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e13793f644a8498d843c4eedfce3d1fb : SUCCESS in 59m 56s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/69b6501921bc4dcfacf489d1296e9492 : SUCCESS in 49m 52s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/473e0a290aa7452e9866ac6b6a9fe43b : SUCCESS in 1h 09m 51s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eb5f0ce771644111a63b0ed05dfbba8c : SUCCESS in 44m 35s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/39e7d7d0074c480da7a7bbe2b0d1cc51 : TIMED_OUT in 2h 34m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ff96685690c14ad486400c3e60185f76 : SUCCESS in 3h 04m 11s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e330d5f2cff642ff98e449b6e2a4fa34 : FAILURE in 31m 13s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/8ef8d357e4554c3b9b7c4e78a4cb84eb : TIMED_OUT in 2h 33m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3d7bee43989dbf6b1a34963a06b10d0da9f662cc","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-04 07:41:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0f9055db26887aa3a9ba42496b296b249fa3c608","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-04 10:50:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d960447f8a9b42468bc6b773d54300a7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7151b58992d44c8899acac7bf358cfe2 : SUCCESS in 10m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ff21be98b7e4f2494fed531cb667732 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/32947e7cefd04887836e17eed9271b07 : SUCCESS in 5m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3f29872caf3941a4b18ce7a88a87bc5a : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf28700868894f2bb7c61c2acf12bcf6 : SUCCESS in 12m 15s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/a65ac0e68e2b46edb9568f0c92f27829 : SUCCESS in 4m 49s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/e17e5f46dda34f329f8dc92347a415c7 : SUCCESS in 8m 15s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/d9171fd5c8da4a7daf1a2f6faab93911 : SUCCESS in 10m 45s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/107a6cf555fc45feab5a4168c7ef9193 : SUCCESS in 5m 17s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/387037dcbd3d4204936167ecacf8376b : SUCCESS in 5m 40s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/50ec5d755fe74ebca661226e1826ef6a : SUCCESS in 6m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/dfd3b9c5b9734865bd34999d51f4a0a6 : SUCCESS in 7m 06s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e4e19ba2d40241f4ba18e18635ea3121 : SUCCESS in 1h 39m 07s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/2306690e35264df4878e4b196b2da28e : SUCCESS in 1h 36m 31s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/8251bd72e7db46cb8977931223f40810 : SUCCESS in 1h 03m 23s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e16386eecde746bdb7d4028f00e3da6d : SUCCESS in 1h 41m 57s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/d732300e9f4b42d0af4e765f72f864d8 : SUCCESS in 1h 05m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a3977e5d50df4fe691c0b43ae8d64ae0 : SUCCESS in 48m 37s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9414f1b9eebe4e78b8f825de499190bc : SUCCESS in 53m 00s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/bd30715931a247648254c116a78ed5a7 : SUCCESS in 59m 51s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/39c9ee53ee494ec5afc38200a61b0933 : SUCCESS in 1h 10m 51s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1abe42945f014e369c9f6c6a64f773cf : SUCCESS in 41m 08s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/377535648f284320a90e85424def2353 : TIMED_OUT in 2h 32m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/724f0710d32f4e26b340efb183738d00 : TIMED_OUT in 3h 04m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2a57e637a68a4bd2aa7728d0eedf895b : FAILURE in 21m 04s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/34dc45e86de549ba9885275d4f2cf89c : SUCCESS in 2h 30m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3cde80b23eab921f5d8b3f91c4c04666704562f9","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-17 16:41:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8f7750dea1add5e816a1e4758791d4f983896cd5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 16:42:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4323589631892a5611b3aca28bec1b94afbdcd83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 18:39:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8de778ff7100447fab48be03885e216d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9526c3af16d04d868eae8bd72ef1c31c : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4f5ae2bdc6244113b2e018bd759ec710 : SUCCESS in 4m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8700c0035a6d42d591da78309d6bb5b5 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc1b750550f745e6a4727957fda87ad5 : SUCCESS in 10m 29s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f07a09b197fa4003bb90fff5456caa40 : SUCCESS in 4m 10s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/eea315b753394b7f8bd027a818c09b0e : SUCCESS in 7m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2a5f3857b5d74d2fa0379409f6d25baf : SUCCESS in 1h 49m 32s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/ad8378cc4e3f45a3bc4479dd31f3febd : SUCCESS in 1h 35m 53s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/babebf332a0741048e16a33f89d1fdca : SUCCESS in 1h 02m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6d68a8ab968b463189f4f41c96ac6d0d : SUCCESS in 33m 52s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b31543cefc9d4d19a153fe2804260cba : SUCCESS in 42m 57s","accounts_in_message":[],"_revision_number":1},{"id":"4c0a24294bd42c6a0e4d1df93346710466b69b29","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 18:39:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9fd1fb3c549dbfaabec35fc494be7cfbfe081b3d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 18:41:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90b2073aebae45128de1782082179a1f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32c5394ad9b94e3192508e230fb97e47 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1740e335afc797866f4c8760730cde42cb38efdd","revisions":{"1740e335afc797866f4c8760730cde42cb38efdd":{"kind":"REWORK","_number":1,"created":"2024-01-23 03:50:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/27/906327/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/906327/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/906327/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/906327/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/906327/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/906327/1"}}},"commit":{"parents":[{"commit":"5750e4512d622450f6ecab1c7001ebd5637e0d53","subject":"Merge \"Provide Amphora stats for Octavia no-op drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5750e4512d622450f6ecab1c7001ebd5637e0d53"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-23 03:49:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-23 03:50:22.000000000","tz":540},"subject":"devstack: Drop removed services","message":"devstack: Drop removed services\n\nglance-registry and nova-objectstore were removed some time ago.\n\nChange-Id: I790bb48291e30a6aa233e082f9e211f18409b3e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1740e335afc797866f4c8760730cde42cb38efdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1740e335afc797866f4c8760730cde42cb38efdd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
