)]}'
{"id":"openstack%2Foctavia~906495","triplet_id":"openstack%2Foctavia~stable%2Fzed~I43c1e1e48c86e2706fb8286ca11264434228e0b5","project":"openstack/octavia","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2024-04-25 23:54:52.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-04-24 16:51:11.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2024-01-29 14:55:00.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I43c1e1e48c86e2706fb8286ca11264434228e0b5","subject":"Don\u0027t update the management port when calculating delta [v1]","status":"MERGED","created":"2024-01-25 22:46:58.000000000","updated":"2024-04-25 23:56:01.000000000","submitted":"2024-04-25 23:54:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"906495","cherry_pick_of_change":906769,"cherry_pick_of_patch_set":1,"meta_rev_id":"cb6c79828b0fb573ae8665742eb9641f749ac07e","_number":906495,"virtual_id_number":906495,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-25 23:54:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2024-04-24 16:51:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-29 14:55:00.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2024-04-24 16:51:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"CC":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-25 22:49:40.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2024-01-25 22:49:40.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-01-26 01:19:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-25 19:00:05.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"}],"messages":[{"id":"8651fbf15b12c1c1a7c4f902741e47bb665c1922","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-01-25 22:46:58.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2023.1.","accounts_in_message":[],"_revision_number":1},{"id":"fff492d4c2a4bf8a2dbeb60de430778c4613750f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-01-25 22:52:16.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7dbafe68503bfd7b308e981ae9ed718fd7a8b445","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-26 01:19:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6141d5b822324340974d3a1e07fa29d0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/37299baa00954a7c867a7f3ba30b489c : SUCCESS in 9m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e730c73827404b33901dcebe6ce34344 : SUCCESS in 6m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6fadcc8009114bfb9af578577dbb7167 : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f07ba0832ea54f0488c1feba429aa46b : SUCCESS in 5m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b6c6a7a32b3a4eddb034e0059c05e2cc : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c872733eb9ff4c338748adcd2fcd9e4d : SUCCESS in 11m 58s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/0cf9c25837d244ed8998ebc5e856bb74 : SUCCESS in 4m 07s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/fd511416d34343d0bca173735994ee55 : SUCCESS in 7m 21s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c363346868e140ba94adab1efbb38d24 : SUCCESS in 3m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3f5a481ce9194e969fcf14976cb62a67 : SUCCESS in 7m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fe72a545b0e54909a60d7f1e14c98be2 : SUCCESS in 8m 56s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7de2df98b7374fd79347e49c7f9e712a : SUCCESS in 5m 24s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3bc940d74eb6429d9cae7150968a8115 : SUCCESS in 1h 26m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/58433162cc2d491280767ecb994ef288 : SUCCESS in 1h 47m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b56617cc6f814825bce545b887a35c71 : SUCCESS in 37m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a7ff8d2773a64f1899ca9b9a351e03bc : SUCCESS in 39m 32s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/de2b901e6e4f416cac9b0d906a1396a7 : SUCCESS in 1h 01m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2626c8b8a3d84f85bcab5e01d5d4e430 : SUCCESS in 31m 19s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3a8b90995ac74a20b032673609d4c394 : SUCCESS in 1h 37m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8827a2334d454360a4543382e1c74f2c : SUCCESS in 2h 19m 36s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f99a2a54b19f4766ad5d013771264592 : FAILURE in 24m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/69468411a1054a5d8fe9802756f24c81 : SUCCESS in 1h 35m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/cb6094af896b4a95b09fbe485acb8b2f : SUCCESS in 1h 59m 15s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/ac75e73b62034d02803cd9a8dcf844cc : SUCCESS in 2h 12m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"aaafffe311904583d6296884a0d3e1d87f538961","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-01-29 14:55:00.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f836acbc407d99ecbdab44c2045f02fd0f8371ee","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-01-30 11:06:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6f8fa6b9d754b2798efe5a56a54662a5f30a66bf","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-02-07 16:10:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7032da8f08c3843872b404596deb905b8f8a4e5a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-04-24 16:51:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9a087e9a2891f758bfb38ac884745e11258e1387","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 16:51:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5d641fd9a2ba44f1084296d37f50146d7e7bd5ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 19:26:57.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cc96b65e33814158bd7b5b48210f6713\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e846858b3ad5413fbb5a9d09df547b8a : SUCCESS in 6m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c52d9ec36e3f417d92a55df4c807099c : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aaa6a3eea55c450ba23d8a1ec10412df : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91e1d711fc484bacbffd10ad184e2338 : SUCCESS in 12m 06s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/846e644228b048149d0a05fb504fdcf9 : SUCCESS in 5m 14s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8457b5741fc14b9a8b2798d3ab854db7 : SUCCESS in 6m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4eb48069c3b642db8991ad28d0d1d6f4 : SUCCESS in 1h 57m 19s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9963d9996bfe406cafb7a74cb4343fd8 : TIMED_OUT in 2h 33m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/caf9e58b38304c32b9697b8c604cf760 : SUCCESS in 55m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/415ef96632df4890a96386c3978d3beb : SUCCESS in 59m 33s","accounts_in_message":[],"_revision_number":2},{"id":"e50d8c1b5a5b41bc3b37e24b8b83b41e9e1ef18e","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-04-25 19:00:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"19eb734f251a3c0402a16e6621d18f8683dfb4ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 21:30:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99ba65ff0547498f940a156f629f191e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e43966f382e940e1af2154766cbb4d21 : SUCCESS in 10m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e2c426fef724170b8a8b9d370a9d158 : SUCCESS in 6m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/662ad1776b374e5394aec81749d3c874 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b138602508a4d7c8eacfd7498e64d03 : SUCCESS in 4m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/297f3778b5424ca28eac242ed7ac475d : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/783bc3b37bdf426598d5f5888cb0388c : SUCCESS in 12m 09s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/b06eb3e201774616b6accc43803be53e : SUCCESS in 5m 36s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/8527dfccaba64fdeab9fc6e054602ed2 : SUCCESS in 8m 34s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/fe8e405f9fef4e7ebf5b8416fdacf9eb : SUCCESS in 5m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/22bc2c138e3a4fa594a111140181ce72 : SUCCESS in 6m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/dee212c06c67434a839dff5593b7bed7 : SUCCESS in 7m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f423dc8a81214eb8a169878d68cb1956 : SUCCESS in 7m 27s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4574517fad544674adef200a713cb2f8 : SUCCESS in 1h 22m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a5f2bf49c24646c4b2d8a8ca4fbdbe4e : SUCCESS in 2h 19m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ee3634131dec46b5ab332e2bc68b1e69 : SUCCESS in 51m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fc8ad26193dc4685b7a968f25664125f : SUCCESS in 58m 40s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/63dd1cd400dd4b6fac361562152436a5 : SUCCESS in 58m 50s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/69e8f883df2a495598d4880427ef7fd1 : SUCCESS in 39m 39s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e3920b2dda10417d923094e7ecf8cba6 : SUCCESS in 1h 57m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/19ca7969050447129a8b23e2c3df2113 : SUCCESS in 2h 12m 57s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6519bbc8068348468e5f5945f7125c5c : FAILURE in 25m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c5a72ba39d72462e983ade74e1d8c04e : SUCCESS in 1h 50m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/16bbe0490b844e4586ae3676a137b3a8 : SUCCESS in 2h 02m 12s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/f751a18d79d94f89a29054c26a7823b0 : SUCCESS in 2h 00m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e99e50aef086b4f13597daa6186179ed844eedb9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 21:31:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"65426c947bd7ecfd4b4f59f0296a1b564ec81c9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 23:54:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b85a3f04a1d74a95a46700fc2dd8596d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cace06ae8b284ea29052a586010d7448 : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea68c36ec53442718afd6a727c92173d : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e77b94162c89403f8160beab69f1cd18 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7960f334988f44bb946ffb382e1a9ca4 : SUCCESS in 9m 30s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/95f2d686c70d4c078aa79654c7c80588 : SUCCESS in 5m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/430f2f6cb34e416390e986cd52b21b86 : SUCCESS in 7m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a675edf6885b4c008b73bb17aa1ce07e : SUCCESS in 2h 15m 12s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c701a89346c0470dbd9a36c7c6128374 : SUCCESS in 1h 58m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b9731ed8799043408789536b67f24fad : SUCCESS in 49m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6cea555471dd4a979b75d2e23fccc37e : SUCCESS in 52m 54s","accounts_in_message":[],"_revision_number":2},{"id":"8388be3c3683197919ac60158f7d7b8f9d976320","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 23:54:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"cb6c79828b0fb573ae8665742eb9641f749ac07e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 23:56:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f975a68316e4529b1e2c37bd7cd7f03\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87fa194859874c0c9ab219334dffcfc0 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5b8f49ffaa0beda86f49ea7014996641314e7b32","revisions":{"bd0a300c4a38cc4e345d08d3402d3c52818840ee":{"kind":"REWORK","_number":1,"created":"2024-01-25 22:46:58.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/95/906495/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/95/906495/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/95/906495/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/95/906495/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/95/906495/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/95/906495/1"}}},"commit":{"parents":[{"commit":"000b577f3e9c9ff7cb893e9f6e635753017a78c6","subject":"Merge \"allowed_cidr validation for additional_vips\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/000b577f3e9c9ff7cb893e9f6e635753017a78c6"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-01-25 22:35:59.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-01-25 22:46:58.000000000","tz":0},"subject":"Don\u0027t update the management port when calculating delta [v1]","message":"Don\u0027t update the management port when calculating delta [v1]\n\nCalculateDelta should be used to get the list of subnets/networks to\nplug (or unplug) into the amphora for the VIP and the members.\nIt should never reconfigure the management port.\n\nStory 2010623\nTask 47564\n\nThis was originally fixed for *v2*, but not for v1 as this patch does.\nMaster v1 already removed from the code.\n\n[1] https://review.opendev.org/c/openstack/octavia/+/876041\n\nChange-Id: I43c1e1e48c86e2706fb8286ca11264434228e0b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bd0a300c4a38cc4e345d08d3402d3c52818840ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bd0a300c4a38cc4e345d08d3402d3c52818840ee"}]},"branch":"refs/heads/stable/zed"},"5b8f49ffaa0beda86f49ea7014996641314e7b32":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-01-25 22:52:16.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/95/906495/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/95/906495/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/95/906495/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/95/906495/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/95/906495/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/95/906495/2"}}},"commit":{"parents":[{"commit":"000b577f3e9c9ff7cb893e9f6e635753017a78c6","subject":"Merge \"allowed_cidr validation for additional_vips\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/000b577f3e9c9ff7cb893e9f6e635753017a78c6"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-01-25 22:35:59.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-01-25 22:52:16.000000000","tz":0},"subject":"Don\u0027t update the management port when calculating delta [v1]","message":"Don\u0027t update the management port when calculating delta [v1]\n\nCalculateDelta should be used to get the list of subnets/networks to\nplug (or unplug) into the amphora for the VIP and the members.\nIt should never reconfigure the management port.\n\nStory 2010623\nTask 47564\n\nThis was originally fixed for *v2*, but not for v1 as this patch does.\nMaster v1 already removed from the code.\n\n[1] https://review.opendev.org/c/openstack/octavia/+/876041\n\nChange-Id: I43c1e1e48c86e2706fb8286ca11264434228e0b5\n(cherry picked from commit b3cd58c305fc105192bcfff4e85b722c404820ac)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5b8f49ffaa0beda86f49ea7014996641314e7b32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5b8f49ffaa0beda86f49ea7014996641314e7b32"}]},"branch":"refs/heads/stable/zed","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
