)]}'
{"id":"openstack%2Foctavia~906496","triplet_id":"openstack%2Foctavia~stable%2Fyoga~I43c1e1e48c86e2706fb8286ca11264434228e0b5","project":"openstack/octavia","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2024-02-01 19:02:11.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-02-01 16:35:59.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2024-01-29 14:54:51.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I43c1e1e48c86e2706fb8286ca11264434228e0b5","subject":"Don\u0027t update the management port when calculating delta [v1]","status":"MERGED","created":"2024-01-25 22:47:38.000000000","updated":"2024-02-01 19:03:59.000000000","submitted":"2024-02-01 19:02:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"906496","cherry_pick_of_change":906495,"cherry_pick_of_patch_set":1,"meta_rev_id":"0f71b8dee8be2b881509c1e9fec4705db3ce18fb","_number":906496,"virtual_id_number":906496,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-01 19:02:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2024-02-01 16:35:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-29 14:54:51.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2024-02-01 16:35:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-25 22:49:14.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2024-01-25 22:49:14.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-01-26 00:45:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2833bcb6d809194af67fb4aaeef72d0e3dac5515","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-01-25 22:47:38.000000000","message":"Patch Set 1: Cherry Picked from branch stable/zed.","accounts_in_message":[],"_revision_number":1},{"id":"977338fc8304c6d9d1fa74ce1dafa6a13965d91c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-01-25 22:53:06.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"0c26472216cbadd8e1ee21589c11f6b1c4f1c2ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-26 00:45:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e88b4a7e42b64b4cb34892008d5a95da\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d568f367eb0346c08ebc5b9599621225 : SUCCESS in 11m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/082f224a4d864d8889414c52cf3d6e66 : SUCCESS in 7m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a87eade85ba4adc878ffe4d3857ab80 : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/acd6310385934b748f335824146b30cc : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b56b59187994c17ad5fd023b1affb70 : SUCCESS in 14m 46s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/58992b9c9e96431c9af9ab169ba58e78 : SUCCESS in 7m 21s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/d36b5b4d4bb14c768ad949d38b7d18ec : SUCCESS in 6m 02s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7f782d291ff248828f56a8c77c8ab48a : SUCCESS in 4m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7299e8c869b54e11b50ef90c3c4b264b : SUCCESS in 9m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/58beef46d1b241ebb9af80b3b564317f : SUCCESS in 1h 25m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8271f1b0c007406e919a8a74a8d26131 : SUCCESS in 1h 38m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a500780ae8624c80a8ad14b95a9a8886 : SUCCESS in 35m 09s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3b8b312deb784b10b2b0e6a3a40d56f8 : SUCCESS in 37m 16s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/fde99d124ee0427e9540e9f5dc3eff45 : FAILURE in 54m 43s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1807efa702664163b84432bf4db29562 : SUCCESS in 31m 47s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/269f29ff084e4c20a7d667bde9c0d4b1 : SUCCESS in 1h 30m 36s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/05e994416e124b6b83b94e46ab9d05fd : FAILURE in 9m 00s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a87f0dc1b14e4f0fbf6200b8a58339be : FAILURE in 7m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/efba82144c3644d6b2d802fc4d4eef9f : SUCCESS in 1h 41m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/a82743bef5a74ecd9fe6555746396021 : SUCCESS in 1h 50m 11s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/02ab0d6937c24e84aff2fa7812e7d890 : FAILURE in 43m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"04a943aad138c0a78e8516f81476abcd6cc9b7dd","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-01-29 14:54:51.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a2d9bcdc152f678a7859921a658aa122381f8163","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-01-30 11:05:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b67ee765cb8e6afb4bc1d75a85515df13bbf43c5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-02-01 16:35:59.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"13a853f87b4717c735c9ef89a18d0cb10ea4aea6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 16:36:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"65a5d3d87203d208d167d75a19bc247b59cccd69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 19:02:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d95b027d393e49958146e7f53846a377\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e2e01ae1b5f4e9fb397c26ffde72217 : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fbccad83a18844d59d4ebd694cd0b5f4 : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/664d69ab2aeb480e97a4080166d09098 : SUCCESS in 7m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/886d55e17ee44251ac5cd18b7eb23a84 : SUCCESS in 13m 58s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6e627342f3274d28881d6f38ddecdd9a : SUCCESS in 4m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/62d617d426f64028b54493c981c31de4 : SUCCESS in 11m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/67df4dbb40a340a08cd72741afa97c6b : SUCCESS in 1h 31m 25s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7f7ba45a1be548ff9149cfdf4030e236 : SUCCESS in 2h 23m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ef7f4ea42b0d4be59fc7575d85d38c97 : SUCCESS in 55m 35s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a6628541a1904d4ea0849211ebb0c1cc : SUCCESS in 1h 02m 27s","accounts_in_message":[],"_revision_number":2},{"id":"2f02961036a921e8dbf1caa5084e27680fc894cd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 19:02:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"0f71b8dee8be2b881509c1e9fec4705db3ce18fb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 19:03:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68a9f0bf1bec44d6ab14fbf2fab4f0ef\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d37b35275c784c9ea74c6980967df082 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1ceac5af85940ad485450aa52aecad6eb8e8462d","revisions":{"e4ceaed980e52c2801e16b11531d016301fd2454":{"kind":"REWORK","_number":1,"created":"2024-01-25 22:47:38.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/96/906496/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/906496/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/906496/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/906496/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/906496/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/906496/1"}}},"commit":{"parents":[{"commit":"8972dc3de8475a17853d918cc0deab285df78433","subject":"Merge \"Fix upgrade check not working\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8972dc3de8475a17853d918cc0deab285df78433"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-01-25 22:35:59.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-01-25 22:47:38.000000000","tz":0},"subject":"Don\u0027t update the management port when calculating delta [v1]","message":"Don\u0027t update the management port when calculating delta [v1]\n\nCalculateDelta should be used to get the list of subnets/networks to\nplug (or unplug) into the amphora for the VIP and the members.\nIt should never reconfigure the management port.\n\nStory 2010623\nTask 47564\n\nThis was originally fixed for *v2*, but not for v1 as this patch does.\nMaster v1 already removed from the code.\n\n[1] https://review.opendev.org/c/openstack/octavia/+/876041\n\nChange-Id: I43c1e1e48c86e2706fb8286ca11264434228e0b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e4ceaed980e52c2801e16b11531d016301fd2454"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e4ceaed980e52c2801e16b11531d016301fd2454"}]},"branch":"refs/heads/stable/yoga"},"1ceac5af85940ad485450aa52aecad6eb8e8462d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-01-25 22:53:06.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/96/906496/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/96/906496/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/906496/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/906496/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/96/906496/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/96/906496/2"}}},"commit":{"parents":[{"commit":"8972dc3de8475a17853d918cc0deab285df78433","subject":"Merge \"Fix upgrade check not working\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8972dc3de8475a17853d918cc0deab285df78433"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-01-25 22:35:59.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-01-25 22:53:06.000000000","tz":0},"subject":"Don\u0027t update the management port when calculating delta [v1]","message":"Don\u0027t update the management port when calculating delta [v1]\n\nCalculateDelta should be used to get the list of subnets/networks to\nplug (or unplug) into the amphora for the VIP and the members.\nIt should never reconfigure the management port.\n\nStory 2010623\nTask 47564\n\nThis was originally fixed for *v2*, but not for v1 as this patch does.\nMaster v1 already removed from the code.\n\n[1] https://review.opendev.org/c/openstack/octavia/+/876041\n\nChange-Id: I43c1e1e48c86e2706fb8286ca11264434228e0b5\n(cherry picked from commit 5b8f49ffaa0beda86f49ea7014996641314e7b32)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1ceac5af85940ad485450aa52aecad6eb8e8462d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1ceac5af85940ad485450aa52aecad6eb8e8462d"}]},"branch":"refs/heads/stable/yoga","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
