)]}'
{"id":"openstack%2Foctavia~906769","triplet_id":"openstack%2Foctavia~stable%2F2023.1~I43c1e1e48c86e2706fb8286ca11264434228e0b5","project":"openstack/octavia","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2024-05-02 18:25:03.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-05-02 15:55:13.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2024-01-29 14:54:34.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I43c1e1e48c86e2706fb8286ca11264434228e0b5","subject":"Don\u0027t update the management port when calculating delta [v1]","status":"MERGED","created":"2024-01-25 22:46:15.000000000","updated":"2024-05-02 18:26:21.000000000","submitted":"2024-05-02 18:25:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"906769","meta_rev_id":"185fae86a416d0c5baf78407e226f98dedde9884","_number":906769,"virtual_id_number":906769,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-02 18:25:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2024-05-02 15:55:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-29 14:54:34.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2024-05-02 15:55:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-25 22:50:01.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2024-01-25 22:50:01.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-01-26 01:02:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c4c6617a67c77c912e44785036a15e641ba89f11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-01-25 22:46:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c66f76fcc8b2e646018479c709a0d07d302e7b2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-26 01:02:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b79ac7b429da4d9b8cbd562041413e57\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c619bd25b56e4c1694beca545283bedb : SUCCESS in 8m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcd3ccec926c4c669b0fb320c9f1bb2f : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47c19ec91d914ea5be57770331f93210 : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/279b8091a3914343a47fd2a63e1add55 : SUCCESS in 8m 45s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/3f69f56b8349494e972c7db41423986d : POST_FAILURE in 5m 18s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/8082084116b64f42a144c990dc1b1562 : SUCCESS in 7m 21s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d8834ea5d79f4333a717c598802679eb : SUCCESS in 4m 44s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ed6143730a9e45f58794d12fe05231ac : SUCCESS in 7m 16s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/91a0bcf25c5a4f018016835ccf771f2b : SUCCESS in 8m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cc52f0a08611447ab68fbc3201ab60c6 : SUCCESS in 5m 29s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/67c867b6be01419e9713a682cd959b8e : SUCCESS in 1h 14m 32s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/4fdb2b2c11964e8c8a292e6008b25601 : SUCCESS in 1h 32m 08s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/f3b391cb32044527bfdaa6de4b6f1a3e : SUCCESS in 1h 16m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cd9786ba01224ad3b5d638f04c41629b : SUCCESS in 51m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c331927e608148a1b2faedbbfccd1497 : SUCCESS in 48m 58s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/8a134d3364894b05943e640f74e5191d : SUCCESS in 49m 30s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/8fd90b35db77487abf7c8e00af8a6468 : SUCCESS in 57m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f11f8054d30b4ed8a42cb1fc56edbc19 : SUCCESS in 31m 43s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/820f34376b7c47618e5593c88a3e72f4 : SUCCESS in 1h 43m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ab82fbc2f16043e284e1ad2f30b94a62 : SUCCESS in 2h 08m 21s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2b2d469218874e04a4be262f77b1ef67 : FAILURE in 22m 59s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fba5d5d9ec6341679adc32fde54768e4 : SUCCESS in 2h 02m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/a2186b3f6bfa4d279135cc53175af434 : SUCCESS in 1h 43m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/a9b2c90323e64a2ba57e28d1341153c7 : SUCCESS in 2h 04m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"28657d6515d47da825b6b530b8c7cc785772db64","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-01-26 08:34:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"097bbdb16ff917ab54de7bdf3451eccafec1f93e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-26 11:39:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80fc4404d416463a8d5efb6a91ee4b88\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7634fdde7293476ea88eb4acc869fac3 : SUCCESS in 10m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/869ab7f5a8c04cc99d0286106e14cd95 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1656ee7070a644e581797f8d213c5802 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1152e7e6b9145478bdc825feaf82aee : SUCCESS in 12m 39s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/44899fabd68a48dd8e18d39cc8d4779e : SUCCESS in 5m 03s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/500c56c97789436593e8bd5d09449935 : SUCCESS in 8m 44s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a002b9ad6a8c4f6da21ab83105bae260 : SUCCESS in 4m 49s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/37a3a0864b1044e7a794c75adffb8946 : SUCCESS in 7m 39s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/17d30c4e743a43879ea82e6bba959ecc : SUCCESS in 8m 33s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/73fb131e615742ac9a019af3ba018413 : SUCCESS in 6m 58s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/67b5d084e2f5409fb72e8a8559852179 : SUCCESS in 1h 28m 11s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/068221802296410e8fe46f30e1275187 : SUCCESS in 1h 39m 48s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/6816dad716144e1f860e62a568e100ba : SUCCESS in 1h 22m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5579ceb4b5584bb2ad8799312130c960 : SUCCESS in 47m 29s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b953c11e3f0e4954a569b93c37434cf5 : SUCCESS in 57m 34s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/72b053fbe81546ed801a30d57da35e1e : SUCCESS in 59m 07s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/50515a8082b04c6eb1e6239b2936c171 : SUCCESS in 1h 20m 13s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4cb9154e9f61400a84b7e0e0227bb686 : SUCCESS in 47m 04s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ae37a35697564b5dac4e616d745ab995 : SUCCESS in 2h 19m 15s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8264bf569e6a4c928c97e69dc49a7beb : SUCCESS in 2h 38m 44s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8edd9afec01e418688ac023ab47fc0e4 : FAILURE in 36m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/df39e00673f9410c930068302436e83e : SUCCESS in 2h 26m 12s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/fb943cf643c54e489090c8e76f9878d5 : SUCCESS in 2h 25m 59s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/675b271560b4420badb8c275cbde8bba : SUCCESS in 2h 02m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7ffddf376a18b33aa58cb0e4c71effdfcd7fa52c","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-01-29 14:54:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3b6b4d457125ebe352408cacd1a6347ac56310ae","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-01-29 14:56:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a871e0500060ee0abd4413cdbf799854bcc8589d","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-01-30 11:06:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"977cd62315e347f9e034f3dc94bef35c24afed3b","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-02-07 16:10:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"baeb79987e2e551e2d62c67672663ccd871b0e80","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-05-02 15:55:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a3f7be25f43d3f3995175a4ed83a08f5fec4451e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 15:55:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bed0a6875c8ee9f5dc0ecceea84c6de09a1df079","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 18:25:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5288a1513c8b4853b09c02fbb2c78036\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cedbd5a393ea48cc90242c3766dc68b0 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3dbf7d35e02d4db5b18f1defc145b016 : SUCCESS in 7m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99c1299f99464889816a080fd1bba00a : SUCCESS in 11m 29s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2f5647bedff74b55be92a4bd3a32d4fb : SUCCESS in 6m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6f7b57a01a7e4aeaa23987bc56cc85e3 : SUCCESS in 7m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3b7c43f382f84305b5e71e2eba733fae : SUCCESS in 1h 45m 49s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/5836de35cf8c489a84a2285921aa5e38 : SUCCESS in 1h 25m 38s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/5308b9946e2e4132847ac06132e7814d : SUCCESS in 1h 02m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7439cb5544094f289dc6e14420177dda : SUCCESS in 43m 22s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/dc61c4fb5f9c450e8e084bb28862a520 : SUCCESS in 45m 52s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/4231d75328bc4ec688f4e271af7a4692 : SUCCESS in 43m 59s","accounts_in_message":[],"_revision_number":1},{"id":"a2627b0fba9ffa372efbf59f57a81689b879df87","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 18:25:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"185fae86a416d0c5baf78407e226f98dedde9884","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 18:26:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbb7cca091e343a68279ca6b6b19964d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0db826e9c2fb4a0cb304d273704f392c : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b3cd58c305fc105192bcfff4e85b722c404820ac","revisions":{"b3cd58c305fc105192bcfff4e85b722c404820ac":{"kind":"REWORK","_number":1,"created":"2024-01-25 22:46:15.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/69/906769/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/69/906769/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/906769/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/906769/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/69/906769/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/69/906769/1"}}},"commit":{"parents":[{"commit":"42e421014de3003ce079380888510dc2f7eb1417","subject":"Merge \"allowed_cidr validation for additional_vips\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/42e421014de3003ce079380888510dc2f7eb1417"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-01-25 22:35:59.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-01-25 22:46:12.000000000","tz":60},"subject":"Don\u0027t update the management port when calculating delta [v1]","message":"Don\u0027t update the management port when calculating delta [v1]\n\nCalculateDelta should be used to get the list of subnets/networks to\nplug (or unplug) into the amphora for the VIP and the members.\nIt should never reconfigure the management port.\n\nStory 2010623\nTask 47564\n\nThis was originally fixed for *v2*, but not for v1 as this patch does.\nMaster v1 already removed from the code.\n\n[1] https://review.opendev.org/c/openstack/octavia/+/876041\n\nChange-Id: I43c1e1e48c86e2706fb8286ca11264434228e0b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b3cd58c305fc105192bcfff4e85b722c404820ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b3cd58c305fc105192bcfff4e85b722c404820ac"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
