)]}'
{"id":"openstack%2Foctavia~907017","triplet_id":"openstack%2Foctavia~master~I31c812b112f587d8273893e565417e9949288d16","project":"openstack/octavia","branch":"master","topic":"drop-direct-exec","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-03-08 14:50:56.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I31c812b112f587d8273893e565417e9949288d16","subject":"Drop direct execution of octavia/cmd/*.py","status":"MERGED","created":"2024-01-28 17:00:34.000000000","updated":"2024-10-22 15:48:06.000000000","submitted":"2024-03-08 19:37:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"907017-drop-direct-exec","meta_rev_id":"761b29f9599b25bc520a72e752ede950a81b6d2c","_number":907017,"virtual_id_number":907017,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-08 19:37:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2024-02-23 00:26:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2024-03-07 08:42:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2024-03-07 08:42:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-28 19:59:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-23 00:26:11.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-03-07 08:42:00.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"10769a474d5393b73a67ceb0193e9b551082a7ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-28 17:00:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c1152f6b46fd33bb50817534cc34455dee64b28","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-28 17:02:23.000000000","message":"Topic set to drop-direct-exec","accounts_in_message":[],"_revision_number":1},{"id":"eaf45f245a5c62d4d5f80db60c8e0f7684096c68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-28 17:18:28.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"d7ba329d85d937d2f588f52af615c1e0c457fa4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-28 19:59:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5ff92cf8e52f46aa93344a8d4009cd08\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f84a42ab1d343d69e5f901ffcbc4936 : SUCCESS in 9m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ee8fb0a9c924c85ba82430286202e42 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1292ac0a1a55450bbed50fada919b5fb : SUCCESS in 4m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c0ad6823384a4825b1d9de8a4642afa6 : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebf7afe1dfee47ea96016c0900e0abcd : SUCCESS in 11m 12s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/8ce03e3eff5b4121966eab94695cefbf : SUCCESS in 4m 02s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/e676a1b201444cf2a4f851f6cb27f78e : SUCCESS in 8m 31s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/351557141de84f60a53f0de2cd0e4ac0 : SUCCESS in 9m 49s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/81c28c1a08c14f68a160b33d055e3a0a : SUCCESS in 4m 54s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/21271cc7103b42baa012fe5b922e5116 : SUCCESS in 7m 10s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/249889019efe4ed79b9e6961f9ee6ea6 : SUCCESS in 6m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/15755d5816174fdd92260790b87d7463 : SUCCESS in 8m 06s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7bc6dd14ffb942eba715a88a01ee0455 : SUCCESS in 1h 02m 58s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/de9846b56e824ee3b54bb3bb5b21de3c : SUCCESS in 2h 02m 02s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/6d741bb355ce4ebf841a67825e9857bc : FAILURE in 1h 08m 19s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/40d889836a62459f8222366157609053 : SUCCESS in 1h 35m 43s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3ec11409a76f4b24a5767e62813eb101 : FAILURE in 1h 18m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d36f0567a5314632b27c37894267f82e : SUCCESS in 51m 37s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d902df7ba47d4d92ac562f120c8933e7 : SUCCESS in 37m 05s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/fe3f86f23ee14d8da3cf11256e168777 : SUCCESS in 49m 15s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/37174d950baa47deb5b7ded01d645d7c : SUCCESS in 1h 05m 03s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f7425cd00f78467db34ae30f365dd940 : SUCCESS in 38m 51s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/985b2ba71b4d453585e4e9adbb7b6eb3 : TIMED_OUT in 2h 33m 36s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/33c034382cf64a81be191234540e7f9c : FAILURE in 1h 28m 42s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e687257a81ec409bbbe0f69a462fa5f3 : FAILURE in 35m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/cee3fcb38596431287b6fb78ebcbf9ad : SUCCESS in 2h 28m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7abd6b3b22bd06c9ea9726d2dd35f3dc83ab48e8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-02-23 00:26:11.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"09a973e1d25cc49fc4c04132b290cce315c55a5b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-29 17:00:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6584d4c7269581ae5c6947539d6a355ae9a4ecaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 20:26:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca6fb76bbf104ad68fea91495444e28a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65e70d5b0e1b43d393f7d0a740ad5172 : SUCCESS in 13m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1f0339d4f3747a092cdd0eea9dd4b7f : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c341e44bb064127b87b2eef385b1cd1 : SUCCESS in 4m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3dfdb423233844688c48feb58e942934 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bcd8f854402426db64ff6aa79021e9b : SUCCESS in 14m 52s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/375a234bc8d743078b98eee2ec340081 : SUCCESS in 4m 57s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/1475e8f210c2429196960ad1462086b7 : SUCCESS in 7m 39s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/b42f4e0e86c74c9395d98ea82d2a00df : SUCCESS in 9m 56s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/46b628d18ab0416795139a974806c78d : SUCCESS in 4m 52s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f069002514cc423d9fd3dcd8b56b7d7b : SUCCESS in 11m 53s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/48af04849ba44d54b99fe78017400897 : SUCCESS in 8m 17s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a318f68dc8d243b09b970a8280f56181 : SUCCESS in 8m 50s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/35832be2e23943f494250cf79e464b0a : SUCCESS in 1h 41m 35s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/dd10ffa62a3b461ea79ba25415588b52 : SUCCESS in 1h 57m 52s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/2f4a4447382046898eced405d4de09cd : SUCCESS in 1h 13m 15s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3c7f50fd3dd2453f8136160d6d21d480 : SUCCESS in 2h 34m 17s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3ae5913e5cde449d94ad78f4fe793b8c : SUCCESS in 1h 17m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bbafc7262d3649f78548068fbfc93578 : SUCCESS in 52m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5dac39e8dca04d78bb976d0699b60039 : SUCCESS in 56m 27s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/131aa1201d3b4d36baa247aa33937121 : SUCCESS in 51m 39s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/f0635e7be5c74edeb44ec12af877398f : SUCCESS in 1h 07m 45s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/17f4f521bc8e4d0e9df4531b5f3e950c : SUCCESS in 50m 17s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f8e98740ef7e422e8e1a470ec1d69fa8 : TIMED_OUT in 2h 34m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e2f4775906614107a88e8829d9525156 : TIMED_OUT in 3h 04m 27s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6b0450013e6246bf839c8586f63cfad7 : FAILURE in 31m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/888a0db57d17401c9a26df0e22364dfb : TIMED_OUT in 2h 34m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"412cd727bbabcd5c34764e0d0219c33fb1961c74","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-03-07 08:42:00.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a3f94d5b1f365356b4bd638be308764ad99d3283","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 08:42:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"93f04ec2d3a48a0cf73d36d87c69e366903875fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 14:28:10.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f23a7410daf84f8ab4794ca0b4016316\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1ddc0bbd4a7435fa5b9f0c0b7f44a4a : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95e40d47de7a413688f1b293f3194d60 : SUCCESS in 5m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/53809c0d2615407ea337689856283340 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59c9944f02524f94a58da989ed2a30b9 : SUCCESS in 11m 04s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1290a0128950407da38ae74eb68db313 : SUCCESS in 5m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4070cc9a5b6e4282936cd8d747b4447e : SUCCESS in 8m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/44b68e46085045258a210617ab93b5b5 : FAILURE in 1h 38m 40s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/5808f90400f347c28a6c0779e624632b : CANCELED in 1h 39m 55s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ca88a23752c84f8ead37d553fe92bd04 : SUCCESS in 1h 19m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f8d68ee74964443c9fba2aee0e47de70 : SUCCESS in 37m 35s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/60cae5e1d5e84ab6bd797c29ea0c53a2 : SUCCESS in 1h 02m 36s","accounts_in_message":[],"_revision_number":2},{"id":"127ae6a4852299a625488f20b4ba390550ae2863","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-08 04:01:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"761a61441fcbf391ffc47fe1393960361b8ff939","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 06:52:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dde94b54350c41ae898dea98e81db231\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e36befd249c48bd895e5152f506a560 : SUCCESS in 10m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4bfb176316648a89f6d0967e1fb9542 : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/02da5c68d6ea4f96ac04f5459d2ff81a : SUCCESS in 6m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1bb7a01e89bc4222bed5654cda1b179b : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b47da5366e04e559eb95f035c3f1832 : SUCCESS in 12m 41s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/29a0eb47fe454095a152eba5ee397914 : SUCCESS in 6m 24s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/7fd11f717a1d4d9d915a288c5ffa2236 : SUCCESS in 9m 41s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/becbbe8f13284c4590c0912dea783678 : SUCCESS in 10m 13s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e6074f9331604d4aa758f835f31da32a : SUCCESS in 4m 31s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/267cb7503eb94535a8ab2ce7e31ac16f : SUCCESS in 6m 54s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9b2e27bb681947f5a9f4f0b48255e3a9 : SUCCESS in 7m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/416fdfc0047841b89cc5e1c10acf974d : SUCCESS in 7m 39s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5efca3227c784c8dba9cb940291a85e6 : SUCCESS in 1h 30m 20s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/828b43a260ec47549f7bad10da16b0d0 : SUCCESS in 2h 07m 59s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/0dd841c8a2ce4b0f9b0a51329ad0cdf1 : SUCCESS in 54m 26s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/7d9696ba028849ad93250cf564192b9d : SUCCESS in 2h 14m 45s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ce0bb00e9cb649acbefa3d710e910fb1 : SUCCESS in 57m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3da399fb529a4e459a150ab75839ed29 : SUCCESS in 35m 26s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2770440e679a473c944ae380bd41aa67 : SUCCESS in 1h 00m 22s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/e14ab590c00c442daf0e2a4e29a312d7 : SUCCESS in 1h 02m 08s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c8031d6278084b64be8ed23747d6e42d : SUCCESS in 1h 10m 50s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/686b22e18e19440d9afe39fc5334a134 : SUCCESS in 45m 10s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c24aae4abf7949b4b26d8ba7ad94d41e : SUCCESS in 2h 31m 00s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f69bd07316db481a98dc804bff691f67 : SUCCESS in 2h 47m 54s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8f7c296102f040a0808149ec7e722df9 : FAILURE in 35m 29s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/188eb36d672149b8bd03a5d711601331 : TIMED_OUT in 2h 34m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"47d1633838ebe6ebaa910441eff88f6763727e5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 06:53:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"450ffbf730d56903cf7f6c9f3231d43b5fa44f11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 09:14:53.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d994ca20987b4d4cb0cdb165551200e7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd1081fe276f469b8bb8641989b464c4 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b4ac192f120d43819f73ad49cfd7713f : SUCCESS in 4m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f61f640bedd746bfae57796b9edffb53 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29a15274222d4f8681e4b461faad9bc0 : SUCCESS in 11m 47s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2f18e5fdd3e040dd8e57a4febd99d0b8 : SUCCESS in 4m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/19e266d7ee0b442588b2e1e09a0528a6 : SUCCESS in 7m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1e5becd00a2149958aa6d857b7415d80 : CANCELED in 53m 50s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/0fc7cfee6fd944afa31eb4a90e50f0f9 : CANCELED in 1h 02m 55s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/caa2314f78cb4eb49f2718f6792d700e : FAILURE in 1h 02m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e30a7871cf0749638f644563f7ff2253 : SUCCESS in 35m 09s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ee7c73d22462412189e51560d1ea0740 : SUCCESS in 46m 49s","accounts_in_message":[],"_revision_number":2},{"id":"1e228e5995996475f452f4d1824274fa9a70ad7a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-08 09:33:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0611dc026d1c6922cf4e3d980e6f9c0d4fe18063","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 12:19:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fece8d0a5c14c95bb5e89e79d78257f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/047cee12ffff40448ccf0b69fee510fe : SUCCESS in 11m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4578de8aa4045bdb21c419d9d28e0e1 : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d857d4ea9a9410f923d127eb25a3496 : SUCCESS in 4m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a63e25c07d1544b8bd3d94282d8c58ad : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff75e7384ade45ce81be9ef831f7853a : SUCCESS in 10m 38s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/f312e9663fcb472f8065346b41ab4f49 : SUCCESS in 5m 21s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/a54c32c7358f44d19b311726823fef99 : SUCCESS in 7m 41s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/6c6d1e524fc44734b96b116a561dc10a : SUCCESS in 10m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0b8d22cd8bb74a738416cca0b7251115 : SUCCESS in 4m 07s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/59e046f9d43f4d218b2773205d38ea20 : SUCCESS in 6m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0c3626b96d3448c089bfa008e94316e2 : SUCCESS in 7m 18s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b0b92d9560194a60827ee801eadfc835 : SUCCESS in 6m 45s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9e765f0e97ab468b99db3a53a1b4d1a3 : SUCCESS in 1h 26m 44s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/d4fdee65096c488db4d817b361be9776 : SUCCESS in 1h 38m 47s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/6318fc28df3b43db85b74ef324bc7e89 : SUCCESS in 54m 24s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/64be5a89e60b480cb1cc36862b4ccc10 : SUCCESS in 1h 42m 42s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/4861c34d102d4044b0768e60cbe43a60 : SUCCESS in 1h 03m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f78a6a5e348443d1b08335d9bf3f8f05 : SUCCESS in 36m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ea36831aefa147e19bba5085294ce036 : SUCCESS in 41m 51s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/880cc6461374496bb3b42686aef49256 : SUCCESS in 38m 11s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ba08de43893940faad01b3c64d470509 : FAILURE in 47m 55s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2b63f2a7341247a29aaa5521f685f60c : SUCCESS in 32m 09s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/952b0601b00b4ecfa30e7d74d518dfe7 : SUCCESS in 2h 27m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/68430fb4f81640fba0e9ff68c672a5f5 : SUCCESS in 2h 18m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/92d8e4fe2c18499cb6bdc55dc667bcb0 : FAILURE in 22m 48s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/e90ea0c5aa4c4a5d916b90d3554fa48b : TIMED_OUT in 2h 33m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4db0bc913f3c78b1ce74ad412402b6632581c449","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 12:20:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"46419d8a9aed7d64cd26e3164adf6e100e821a69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 14:11:18.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/63ed6863c57c42b8afd52a8f97dbc3bb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a91f6e697904badbded38c9853c62cc : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c25ffb773c44a55a75099bf7093422b : SUCCESS in 4m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/df753f41fc49479194b49d14bb095a8c : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0379b3736a094615bbace6f64b914443 : SUCCESS in 9m 52s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/43f23e389f7844c89d5a22a8600d614d : SUCCESS in 4m 44s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8095af450cf3421db6875acd9c209455 : SUCCESS in 7m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/560efea0013a4a10be94670c35dbba1a : FAILURE in 1h 36m 36s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/ca9a9eabe2024817b11515210a990e0c : CANCELED in 1h 38m 47s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/648b0a50aa9a40e9a0dc9fba92122910 : CANCELED in 1h 22m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/91905e5f6ed143b29ede2b25240c8ee1 : SUCCESS in 48m 11s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3e59d64e107f4c59bc949a04b7a2dd4b : SUCCESS in 1h 06m 05s","accounts_in_message":[],"_revision_number":2},{"id":"6a04122fcb1bc3373c12353e78f148ce500f1b4d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-08 14:50:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4b69d6e8ff05840cb444cf0d091b699b0f9a234f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 17:34:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/780477c77eba4713aca7b491f990c2af\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b076b16f97bd4d408754f88ac93ea82b : SUCCESS in 11m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62a7f4094cb4465896de94f1987f81b8 : SUCCESS in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a1e6dda4e5d74c97bec87035767395ca : SUCCESS in 4m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/577ec904b83c4be2a2086e9ac18f5f0b : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8462e4a198334e3da469346399e9dc10 : SUCCESS in 11m 48s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/f6333f7d593f49af84986e04f5da75cd : SUCCESS in 5m 04s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/f535e91b9fe4464c8370dfdad59d1557 : SUCCESS in 7m 52s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/2a66b090349f43a1afc63e14db6654fb : SUCCESS in 9m 26s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/900d71241ee3464daa27499bd639c4d4 : SUCCESS in 4m 25s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cb1d6e2cd18542f1a56f3f84a983fc83 : SUCCESS in 8m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/17f71c41221f426e893b8b828f00a305 : SUCCESS in 8m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7a8090e245f44df2bd44ab0032d3049c : SUCCESS in 7m 06s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ea6240ac153c4afaa268aaaf7316db4c : SUCCESS in 1h 31m 36s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/3b05998c3ac0480b87a550183a370e62 : SUCCESS in 1h 42m 38s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/c40868fd3a2b46a3a792df59e40734c7 : SUCCESS in 1h 02m 37s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/341913fd07bd45e38f37d25255bc3747 : SUCCESS in 1h 46m 35s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/551a4ae400214cb094dda2a7ea95c84c : SUCCESS in 1h 04m 33s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/80f1ca3d5bf24725bbd06ed12c2599b7 : SUCCESS in 58m 52s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7f08aebbbc4f4f0abf40fd0921c59fa2 : SUCCESS in 50m 26s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/9cd2016dd07446a3a163bc6fcbff7156 : SUCCESS in 58m 06s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/b00343c9703e4efaa6eb32ef98e7109f : TIMED_OUT in 2h 33m 42s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4503fafa5d6b4af99ad8fa886d8927db : SUCCESS in 48m 30s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0f7360e6802340aeb9b2ee5f53bc04aa : SUCCESS in 2h 14m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9b91e8002de34657acc3a69d62c6642c : SUCCESS in 2h 22m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/843f3ac4140d4688a521f4052478185e : FAILURE in 26m 05s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/8c7993af3ae34381956c80c5465308e2 : TIMED_OUT in 2h 33m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"99731af915dc68770e8799d93d402c217dc40774","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 17:35:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e27d03c9fa9314f25aea1af1ae12f21eccdccbf8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 19:37:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6879e78247e44a8e8e4551dbb1c3fe4a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db04e664de6043f49d26faada8e98724 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4191b53b602b41489215e4dd7592212b : SUCCESS in 4m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9f8a4e1bb9984365a8ce9a9e2817b7d1 : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4d09992de204952b9fffffc78fa3c8f : SUCCESS in 11m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1831f037ed4e44139dbdc8ed70eb3c88 : SUCCESS in 4m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/627c32fd15e143178b773b2d20360eda : SUCCESS in 7m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a0d39a7d8f004c2ba9889db1b1f57836 : SUCCESS in 1h 19m 42s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/7f18b334eae34b19adc82d611085e4e6 : SUCCESS in 2h 00m 14s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/6f40f742c2b5499eaa7a020f6070ec7f : SUCCESS in 1h 05m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d4c4eff8dbc048f7beb27a3cb55989b2 : SUCCESS in 48m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7bb88b42ffcc4ce5a4b47f3d6649fb10 : SUCCESS in 53m 21s","accounts_in_message":[],"_revision_number":2},{"id":"2ed41edf8c5d109d8c00268570d795d60a563287","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 19:37:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"7621836ca7056cfa02c646a82a3eadaf72a52698","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 19:39:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/030a35a980554c3bbf703232b0352de5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12b90ba1f18143d09626e744dfc9db95 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":2},{"id":"761b29f9599b25bc520a72e752ede950a81b6d2c","tag":"autogenerated:gerrit:revert","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-22 15:48:06.000000000","message":"Created a revert of this change as I2e8ddaeb3f0cdc1bbfc5766fe13b45b5165526ca","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1e0f59f55833d404d7ebdb5a5f05763e876cec95","revisions":{"5ff95e785fdb2f6742cc613fade46bd40eb21904":{"kind":"REWORK","_number":1,"created":"2024-01-28 17:00:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/17/907017/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/907017/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/907017/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/907017/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/907017/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/907017/1"}}},"commit":{"parents":[{"commit":"b1c930d41f4390da2f33e91e613f70cbd3730a58","subject":"tests: correct use of has_calls -\u003e assert_has_calls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b1c930d41f4390da2f33e91e613f70cbd3730a58"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-28 17:00:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-28 17:00:10.000000000","tz":540},"subject":"Drop direct execution of octavia/cmd/*.py","message":"Drop direct execution of octavia/cmd/*.py\n\nWe generate scripts using setuptools, and we don\u0027t expect users may\nlaunch these processes directly from *.py files.\n\nChange-Id: I31c812b112f587d8273893e565417e9949288d16\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ff95e785fdb2f6742cc613fade46bd40eb21904"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ff95e785fdb2f6742cc613fade46bd40eb21904"}]},"branch":"refs/heads/master"},"1e0f59f55833d404d7ebdb5a5f05763e876cec95":{"kind":"REWORK","_number":2,"created":"2024-01-28 17:18:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/17/907017/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/17/907017/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/907017/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/907017/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/17/907017/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/17/907017/2"}}},"commit":{"parents":[{"commit":"b1c930d41f4390da2f33e91e613f70cbd3730a58","subject":"tests: correct use of has_calls -\u003e assert_has_calls","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b1c930d41f4390da2f33e91e613f70cbd3730a58"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-28 17:00:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-28 17:18:28.000000000","tz":0},"subject":"Drop direct execution of octavia/cmd/*.py","message":"Drop direct execution of octavia/cmd/*.py\n\nWe generate scripts using setuptools, and we don\u0027t expect users may\nlaunch these processes directly from *.py files.\n\nChange-Id: I31c812b112f587d8273893e565417e9949288d16\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1e0f59f55833d404d7ebdb5a5f05763e876cec95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1e0f59f55833d404d7ebdb5a5f05763e876cec95"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
