)]}'
{"id":"openstack%2Foctavia~907676","triplet_id":"openstack%2Foctavia~master~I6b281d2520db0048329b12b33108273ba2f96534","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-03-08 08:31:54.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I6b281d2520db0048329b12b33108273ba2f96534","subject":"redis: Support multiple sentinel servers","status":"MERGED","created":"2024-02-04 12:48:28.000000000","updated":"2024-03-08 14:12:57.000000000","submitted":"2024-03-08 14:11:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"907676","meta_rev_id":"21c22e1f8a145dbf73d543db8a077c02b5813715","_number":907676,"virtual_id_number":907676,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-08 14:11:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-02-29 16:57:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2024-03-07 08:41:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2024-03-07 08:41:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-04 13:24:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-22 23:37:04.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-03-07 08:41:18.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"1fde1e030376512de40e233a4e826dab043f433c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-04 12:48:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a3978bfdf2d57f9d05dcd97054e891e58dd6225","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-04 13:24:48.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/91cab63241b542e6a71a1a6c4f9955fe\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92eb12e3b9c8498a8084486631a922a9 : FAILURE in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d3519d1b8fe411bb94e35958154bf48 : FAILURE in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/53c57f7b9d4f4a13a21ebaf56135a783 : FAILURE in 3m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d27e02c04211494abb508c9a3d86a720 : FAILURE in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60cd7b52fc71406691cdcb6040fdde01 : FAILURE in 6m 26s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/d074f94ba46e47fb9847cdd8b609d70b : FAILURE in 3m 40s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/2b3c805769fe402cba1a7de6f8a7b627 : SUCCESS in 7m 57s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/368ad7acc0b04f3289b0cff255f1668a : SUCCESS in 7m 43s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7ab48f9d7bd94bfd80d16ebb31789ea5 : FAILURE in 4m 38s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1fba10a1f7ec4e3583eb2269b2b56eb3 : SUCCESS in 5m 37s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/95265c7efd6e4a0b96746d2bf4154f24 : SUCCESS in 8m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ab8438b57ef3476084e93e1afcf88750 : SUCCESS in 9m 01s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/11d057f2a1124475a226be489a29e5b4 : FAILURE in 13m 56s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/16ad3a8947fe4e74a81831f9765635f6 : FAILURE in 14m 42s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/7ed031645552478d835941c11281629c : FAILURE in 14m 26s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/952fc8b15b4d4f6e8ad5b533e90da0b2 : FAILURE in 16m 05s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/48b7eaa4e77843a696c3a60735aced25 : FAILURE in 16m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/933e9afee8d54c0aaba0ad44240accee : FAILURE in 19m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/50f18b631e504db482e34c83ee88bb55 : FAILURE in 34m 50s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/d174469e40554b9485aaf337747adb25 : FAILURE in 34m 12s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/defc6ac829474b45ac726fa85c6df748 : FAILURE in 23m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f019d22b901946ccb40b4263d8f19102 : FAILURE in 15m 25s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ca32cd70102b4daa8916b711d38d381c : FAILURE in 18m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cd5c412228984f99945151d8efbdbd7d : FAILURE in 19m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f169afa366794016803a0cc497d1ae09 : FAILURE in 14m 20s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/dfa8aa5ccd9045b1a905fd200c3a292d : FAILURE in 22m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ac5caa22cf7c7be1aff015984b7b6f2f36bb9d46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-04 14:26:39.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"67e6049db0cdf4b6a2587df8fddfed2686e3701d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-04 15:23:32.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"6ff11eb0c4b8fb431bdf17f5beb60831bad4d01f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-04 15:35:27.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6f1ee1a622e747f07aef5737e51cb4932bff8da3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-04 18:11:58.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dfa4ed2dde304e63930528d304ab4684\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/483444d5e5e644e080d7781387cb2057 : SUCCESS in 9m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20e9caeb08f949dfadc3f90749f2b605 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8eb0f42c62e34e03a4a0d6f367b8e684 : SUCCESS in 4m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ab76d6086b2942a8b67f77b3fa762e99 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d517316c93643079dc1d0404c9cf42d : SUCCESS in 9m 50s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/d0615d172bf84522947655b2f1b832fa : SUCCESS in 5m 13s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/24cdd5e1149643fcaf3607d2e9215f11 : SUCCESS in 7m 52s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/185d336ab54f4ed88053f7f5fd89d727 : SUCCESS in 9m 27s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5d5e12cafeb6453386cd326cbdb081e0 : SUCCESS in 4m 51s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9c93fb7e779d47c3b2e3e2ea44055189 : SUCCESS in 6m 35s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f4acb88f35b34156b04e0d32ee5e5eed : SUCCESS in 7m 17s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b6a10d5c5f62430aaa5dc8c197149152 : SUCCESS in 7m 33s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e2e5297d9bc04d75a1ec3c41456b41d2 : SUCCESS in 57m 21s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/a452f620e19e429b853c62cc75f356a9 : SUCCESS in 2h 08m 37s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/075e4cf6654743528fed86b4244f205a : FAILURE in 1h 16m 21s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c48b54c5edbc410fa51244b9c294c342 : SUCCESS in 2h 08m 02s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/051053b09e5040b1a918e5c35769c226 : FAILURE in 1h 13m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/135db205515442c1854bde77011d97a1 : SUCCESS in 52m 29s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/66181271e790413184dc4d54379773de : SUCCESS in 52m 04s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/5153587ed7664772a2e62d48817171e0 : SUCCESS in 57m 15s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/a4fbcbd83faf4ad1aeab1e0a342b1be4 : SUCCESS in 1h 01m 23s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/825bb50d608645b1b0d1130a61fc1de5 : SUCCESS in 33m 37s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/562152cc74e34e9f908c701d802278b1 : TIMED_OUT in 2h 33m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4259641fa69d4612863f000c86463b68 : SUCCESS in 2h 28m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ad85d67ffb4d47a2a04dc90230f75ca8 : FAILURE in 21m 40s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/05bd987d3feb49dabe5b307515ee21b1 : TIMED_OUT in 2h 35m 25s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"015aa6ab58f4ee38582c96a96957e7c3b53bc273","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-02-22 23:37:04.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ec33115ddb353399ec8831cdd841db9e9ec0a62f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-29 16:57:57.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"5f7f78db014ff73e82dab56041d26683f8c63659","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 20:17:58.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/27c3a2d295f34935b9f80d491262ee47\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c41027fb5eb64a95b90e63a607dffe05 : SUCCESS in 12m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ca36ed5bd354df6bb20e8137ffc8b8f : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1917bd98153647539bfce4067d06cc5a : SUCCESS in 4m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f3328036814c4e8db67c777fd44a5b74 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae8d4761954e490ca42135c48a57852d : SUCCESS in 12m 42s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/d7035d7af95d43789175dbc6d24977be : SUCCESS in 5m 24s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/590d72d63a044355a5ed2734dcc07531 : SUCCESS in 9m 20s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/5c05e049d25a4b639194baddb816ca8b : SUCCESS in 8m 58s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d12be4984fdc49fc805dc9175f491b78 : SUCCESS in 5m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/32e9e40cedbb4fd190e7ca523c582cbc : SUCCESS in 7m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1b735ec7fdda4357aa0a6bd31b5b2deb : SUCCESS in 8m 11s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/49521de46a234dc68c0817c7afe2d28c : SUCCESS in 7m 37s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/209107158f0b46778a4dd8053beadaf2 : SUCCESS in 1h 30m 45s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/6559a9f1dfdd48f1a2d992ab2df4ecd6 : SUCCESS in 1h 37m 04s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/8aca8597827642158b03c0eb88f83397 : FAILURE in 1h 31m 47s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/4e706521023c44f9890ffd2868a47aa6 : SUCCESS in 2h 01m 03s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b5e905190fca4df9a5ff798adbb16731 : SUCCESS in 1h 11m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/10f71ba0f9f5421c83ff22521c128754 : SUCCESS in 48m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5d147ed963224a0591f12140d8eb9282 : SUCCESS in 41m 03s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/1ed196c21944495c8afd2e8410fe62c0 : SUCCESS in 59m 56s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/1c5d528bd1854be5888962c5d573de36 : SUCCESS in 1h 12m 25s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/418c900612c249959ec91bcfea1c3099 : SUCCESS in 46m 11s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5d7fa28332ef4ff18bd8c1ed57321ed6 : SUCCESS in 2h 19m 29s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/86f5b6079ab0444ba1d32ec00254ad03 : TIMED_OUT in 3h 04m 49s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4abcd3ededbf4e408be4a91faf5ed1ca : FAILURE in 32m 00s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/9fbfca5c52404dad92558769eb2cac66 : TIMED_OUT in 2h 35m 04s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1d71dda8ccbe990264602c7ea1e4435063708954","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-01 00:51:47.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"330e0f29c626ad9a455524fa9c67492fc3adbf33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-01 03:55:48.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0349cba820a4c2b8371478ed38b087e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2c63a57abe54efaa987d14a2b89bd27 : SUCCESS in 10m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9125a2ce09d4c939a39475ba0e56306 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/beb98126f90d4e6a92968ebffc3cb0ef : SUCCESS in 4m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a4a18c1b7ccc4195b99551c019d88981 : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/262bd0e3406246c1a0dfed2ebf9941d5 : SUCCESS in 9m 33s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/ea74200cc3d34e239c3e4a42cd742a85 : SUCCESS in 4m 48s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/5d24009449714620b3ec4e9becb2e93f : SUCCESS in 7m 38s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/66d7c8d233da4f48b9f4b2e6a4adc8a0 : SUCCESS in 9m 05s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0217cf90cfe54372973cda731b993c63 : SUCCESS in 4m 44s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c194e18ab1f8489ea07a48b43187d91d : SUCCESS in 6m 15s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e6db20392d7147f4a59b2232d6b31092 : SUCCESS in 7m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ecc596bb82a941f1830b68db8d3e4da3 : SUCCESS in 6m 07s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e9a1a8cc65074e48816518de38fdc8e5 : SUCCESS in 1h 45m 14s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/7bfefcb9d10c414497f2f3b26bf1ab74 : SUCCESS in 1h 39m 51s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/b3384e074aea4cdeb6c5d90d49a63e3f : SUCCESS in 1h 12m 51s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/8dc86a1c0dd444d39c7e2165b46ac23f : SUCCESS in 2h 00m 38s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/06ae1c73cc4149f48b92566512673644 : SUCCESS in 1h 12m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/186bf62ab64149feb5df7592fe050c6a : SUCCESS in 42m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fcc3ccbac33c4b9ba2c89c3631cb684f : SUCCESS in 40m 19s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/a112984df32541acbe25552c2877a978 : SUCCESS in 40m 46s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/63ec5b4f834e47bdad169d0225467484 : SUCCESS in 1h 05m 04s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1414448c812d4f3b97103374f3bdcde8 : SUCCESS in 33m 47s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f782e9d629ee4f7dae93c2bef5145e86 : SUCCESS in 2h 19m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bb67a139a2c240c6828a60e2a6c4ca21 : TIMED_OUT in 3h 03m 15s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/92019cc273054e76ab5c696d3009490f : FAILURE in 27m 12s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/44f7aae5e4124cec81f0635de33c6857 : TIMED_OUT in 2h 34m 02s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"90e3bf9c8ddc3941f0a2a04334623886173f898f","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-03-07 08:41:18.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"7497cde05c2dd21ba313ccb36e2b9851ca5c86ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 08:42:01.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"b07f64b1ac487e7344ac8eab70f1e2a30839316e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 12:50:45.000000000","message":"Patch Set 5: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/907669 is needed.","accounts_in_message":[],"_revision_number":5},{"id":"281776fb9ab66763b386604541edec30309f244f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-07 13:30:28.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a62e597bf7451249cdae399ab9435df360fba77f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 16:14:14.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f233ce095a94393897ae06bcb88900f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a40a35a43008451bab883ec9f5ad0a49 : SUCCESS in 11m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/373b132b74734bc29fd43b0e616bf7f5 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/00b388c9e286497f86d61d0047d8df5c : SUCCESS in 5m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ea5ad0b5cb8a47dfa77c120cd6cb0aea : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41d88fd341b1454ab027582ceebe637c : SUCCESS in 12m 20s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/336de5e48b364b2394abddaa99411e57 : SUCCESS in 5m 30s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/c348bb5e15fa4a008343a772356be797 : SUCCESS in 7m 26s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/1a5ca87d4c984d508ede6228d7d99a40 : SUCCESS in 10m 05s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a58e6b66fc2646b2afde48122a88cb57 : SUCCESS in 3m 58s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5fd5c61eba024962be537fad3fda275d : SUCCESS in 7m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/063cd27c66a24da4b0a262ed691229e0 : SUCCESS in 6m 28s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6cc61f84f9da469395f4ecb0f3343f95 : SUCCESS in 7m 34s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f3460afb257f4e38a65bcccc186e179e : SUCCESS in 1h 31m 22s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/d0b46f14876a475a935be5f098b90f8b : SUCCESS in 1h 36m 42s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/0b118fe55a3b4df78ff51c584411e3be : SUCCESS in 57m 05s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/bc51358eaf3842b09869ab714d000559 : SUCCESS in 1h 44m 37s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e4655063ec3f4d80a0f77ccd4f76bdad : SUCCESS in 57m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4d4998fc6ebd4344be667dfc5cc6f7e4 : SUCCESS in 36m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/aaf137ec034f44f38b5b212292e2a0f2 : SUCCESS in 40m 39s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/8adbf82e9b504a0ebf3f4b7ab9f9b2db : SUCCESS in 40m 38s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/12d6ea8674124070b0af511801eb9998 : SUCCESS in 1h 16m 33s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6f6f7409e00a4a08bbc6e653bc95efc4 : SUCCESS in 34m 14s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/094217396e224dc1a282d8fc9458f522 : TIMED_OUT in 2h 33m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9d40e17d0d6d4ad799ed88859ded4bdf : SUCCESS in 2h 42m 12s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/614191c3dd19405a8293db04f3400a96 : FAILURE in 24m 35s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/cabcc91d1b8a4d7ba78cc4bb864d7a90 : SUCCESS in 2h 33m 19s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f9a5d10c6f73b0752cd6058c127edc6d7e53fb24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 16:23:17.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"c57058685282582d3aac9599c41d0db918493ede","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 16:59:20.000000000","message":"Patch Set 5: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/907669 is needed.","accounts_in_message":[],"_revision_number":5},{"id":"9a26317839ba528a0a44923d6538caed731301d2","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-08 08:31:54.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"adca84282d0e1d520da20e160a1ca073c8f732cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 11:36:21.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a56b6d9aae84050b410d8bc2b09d939\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae06075cee9442cb97c53198741aa04e : SUCCESS in 12m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f73979e9eb1b4719a4992a9fc70ede63 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/786259be055a4940823f187cfc60debe : SUCCESS in 4m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/56aa3c5bafcd474c9115c33d7e181073 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88cc4b30b4c844f38da7d191385b8939 : SUCCESS in 10m 55s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/6f902982c244423e9ce3c6d58e4f56b5 : SUCCESS in 4m 40s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/ac3828a889ce44b287e68c1120468021 : SUCCESS in 8m 31s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/b066e5031024482d8d23f427aef6de30 : SUCCESS in 9m 28s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/bc87304f87b7487e9b945ae54e5f4c11 : SUCCESS in 4m 29s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e0ef441201654b6d97bc68c7f8b057bd : SUCCESS in 6m 49s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5f8e8230204f493c9fa206226b9fd312 : SUCCESS in 6m 49s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2debca5bd1544ba6abd9db097e30e38e : SUCCESS in 8m 07s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7c91e6ce13ba428a9798e861c401f4ea : SUCCESS in 1h 38m 10s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/c35657824f7942d0a3ffba00fb242350 : SUCCESS in 2h 12m 26s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/f6788a803cf3423e9c452e68fa5e3296 : SUCCESS in 1h 22m 05s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/8bb75f4d347e407392d8d528541a0f4a : SUCCESS in 2h 16m 54s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/2d57904d67be416ea6bf44d4e97873d8 : SUCCESS in 1h 44m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/47d7f9e213ac4bafb1de1579280676bb : SUCCESS in 53m 26s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0747a0a8945b4395ad85ebaa29e7eca7 : SUCCESS in 1h 02m 02s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/d8c82a36bfbd4307807e9b9cedf06663 : SUCCESS in 57m 25s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/a9ad6220f5ef42f3afbe2cb5cd5c2c1e : SUCCESS in 1h 16m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/027d549fc2b247a6b80eaac6b3b26a90 : SUCCESS in 33m 22s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9c3ac2a27ecb411e92e4df66c091c48b : SUCCESS in 2h 16m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e5a27f5ab98a4182b461946349b0e309 : TIMED_OUT in 3h 03m 03s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e63711eb06ef491682a11e0cf3a9b362 : FAILURE in 23m 43s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/2988d4182681400c946129c380ddf338 : TIMED_OUT in 2h 33m 20s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"dbb713a3105d194bd17fae35697d69853effe7c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 11:36:57.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"fa2975b0c50a4b3329e99aef49b4b41d69566947","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 14:11:13.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1331bbdc532d4d75959b16954f485c09\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/048d957689af4c8ca45ae22a58ba6f95 : SUCCESS in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0039e8f073d249d4939f10e6bb0a1e0b : SUCCESS in 4m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dc5afddd9e2a476e99db9e46e35857f3 : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0481815c64884cebaa9f26bf707f352f : SUCCESS in 13m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/bbe1ea76738b44d6b8346bb94377bd86 : SUCCESS in 4m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/373352a5a6f24c55a2825cb5b1c8fc59 : SUCCESS in 10m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bbf8c36933b2429ca83af24fa6c79d9d : SUCCESS in 1h 23m 22s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/523dbcfccba748b1a40b13bf471909b0 : SUCCESS in 2h 28m 11s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/04a9fecde2e546288cf0f74b6dd1bcc7 : SUCCESS in 1h 40m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c87a2959537a434b872272ed39456cb8 : SUCCESS in 1h 00m 24s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5ff01b03f0264d9cb2e9d83572ee6b7c : SUCCESS in 59m 04s","accounts_in_message":[],"_revision_number":5},{"id":"aa0318c11775b8a1948a523f0cecc789c99e0a6a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 14:11:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"21c22e1f8a145dbf73d543db8a077c02b5813715","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 14:12:57.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad0d83d7ef4148f5a5fd4ca18db0f391\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ea741bf68c54dde84e73c0635603d13 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"bd3ef61a0c0042fb4198ea7746e567ae3ca8a17a","revisions":{"fc647b641723a0a3f687d8340b0f3639ae6ada68":{"kind":"REWORK","_number":1,"created":"2024-02-04 12:48:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/76/907676/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/76/907676/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/907676/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/907676/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/907676/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/76/907676/1"}}},"commit":{"parents":[{"commit":"d4cd3c18f9f51acea279082228ef65d092c31464","subject":"redis: Add username","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d4cd3c18f9f51acea279082228ef65d092c31464"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-04 12:46:45.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-04 12:46:45.000000000","tz":540},"subject":"redis: Support multiple sentinel servers","message":"redis: Support multiple sentinel servers\n\nRedis Sentinel client implementation support using multiple sentinel\nservers for redundancy, but only a single server from the servers list\nwas passed down to it.\n\nThis uses the new taskflow interface to add fallback servers, and\nregister the remaining servers in the list as fallbacks.\n\nDepends-on: https://review.opendev.org/c/openstack/taskflow/+/907674\nChange-Id: I6b281d2520db0048329b12b33108273ba2f96534\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fc647b641723a0a3f687d8340b0f3639ae6ada68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fc647b641723a0a3f687d8340b0f3639ae6ada68"}]},"branch":"refs/heads/master"},"bd25068a58353d5653232ec6f5304f4646381382":{"kind":"REWORK","_number":2,"created":"2024-02-04 14:26:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/76/907676/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/76/907676/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/907676/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/907676/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/907676/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/76/907676/2"}}},"commit":{"parents":[{"commit":"d4cd3c18f9f51acea279082228ef65d092c31464","subject":"redis: Add username","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d4cd3c18f9f51acea279082228ef65d092c31464"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-04 12:46:45.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-04 14:26:39.000000000","tz":0},"subject":"redis: Support multiple sentinel servers","message":"redis: Support multiple sentinel servers\n\nRedis Sentinel client implementation support using multiple sentinel\nservers for redundancy, but only a single server from the servers list\nwas passed down to it.\n\nThis uses the new taskflow interface to add fallback servers, and\nregister the remaining servers in the list as fallbacks.\n\nDepends-on: https://review.opendev.org/c/openstack/taskflow/+/907674\nChange-Id: I6b281d2520db0048329b12b33108273ba2f96534\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bd25068a58353d5653232ec6f5304f4646381382"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bd25068a58353d5653232ec6f5304f4646381382"}]},"branch":"refs/heads/master"},"5881fbba1d51f217ed518ac271bf5ead5a9d5954":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-02-04 15:23:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/76/907676/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/76/907676/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/907676/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/907676/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/907676/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/76/907676/3"}}},"commit":{"parents":[{"commit":"92ccbfd56b4393b579677b2c92ff6d57218f4696","subject":"redis: Add username","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/92ccbfd56b4393b579677b2c92ff6d57218f4696"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-04 12:46:45.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-04 15:23:32.000000000","tz":0},"subject":"redis: Support multiple sentinel servers","message":"redis: Support multiple sentinel servers\n\nRedis Sentinel client implementation support using multiple sentinel\nservers for redundancy, but only a single server from the servers list\nwas passed down to it.\n\nThis uses the new taskflow interface to add fallback servers, and\nregister the remaining servers in the list as fallbacks.\n\nDepends-on: https://review.opendev.org/c/openstack/taskflow/+/907674\nChange-Id: I6b281d2520db0048329b12b33108273ba2f96534\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5881fbba1d51f217ed518ac271bf5ead5a9d5954"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5881fbba1d51f217ed518ac271bf5ead5a9d5954"}]},"branch":"refs/heads/master","description":"Rebase"},"4166cad48bbfab7bb0584ef5e745e62696d6ce0e":{"kind":"REWORK","_number":4,"created":"2024-02-04 15:35:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/76/907676/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/76/907676/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/907676/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/907676/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/907676/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/76/907676/4"}}},"commit":{"parents":[{"commit":"119993be6ae4430abca91b741bad5f7a5beffbc2","subject":"redis: Add username","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/119993be6ae4430abca91b741bad5f7a5beffbc2"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-04 12:46:45.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-04 15:35:12.000000000","tz":540},"subject":"redis: Support multiple sentinel servers","message":"redis: Support multiple sentinel servers\n\nRedis Sentinel client implementation support using multiple sentinel\nservers for redundancy, but only a single server from the servers list\nwas passed down to it.\n\nThis uses the new taskflow interface to add fallback servers, and\nregister the remaining servers in the list as fallbacks.\n\nDepends-on: https://review.opendev.org/c/openstack/taskflow/+/907674\nChange-Id: I6b281d2520db0048329b12b33108273ba2f96534\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4166cad48bbfab7bb0584ef5e745e62696d6ce0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4166cad48bbfab7bb0584ef5e745e62696d6ce0e"}]},"branch":"refs/heads/master"},"bd3ef61a0c0042fb4198ea7746e567ae3ca8a17a":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-02-29 16:57:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/76/907676/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/76/907676/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/907676/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/907676/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/907676/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/76/907676/5"}}},"commit":{"parents":[{"commit":"16f6b2e8f65e12ba3bc51529d4a8a828ea2b8b16","subject":"redis: Add username","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/16f6b2e8f65e12ba3bc51529d4a8a828ea2b8b16"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-04 12:46:45.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-29 16:57:57.000000000","tz":0},"subject":"redis: Support multiple sentinel servers","message":"redis: Support multiple sentinel servers\n\nRedis Sentinel client implementation support using multiple sentinel\nservers for redundancy, but only a single server from the servers list\nwas passed down to it.\n\nThis uses the new taskflow interface to add fallback servers, and\nregister the remaining servers in the list as fallbacks.\n\nDepends-on: https://review.opendev.org/c/openstack/taskflow/+/907674\nChange-Id: I6b281d2520db0048329b12b33108273ba2f96534\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bd3ef61a0c0042fb4198ea7746e567ae3ca8a17a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bd3ef61a0c0042fb4198ea7746e567ae3ca8a17a"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
