)]}'
{"id":"openstack%2Foctavia~907678","triplet_id":"openstack%2Foctavia~master~I36daf64fe5df6682cf608eba6b4c7b7fe0652611","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-07-25 13:56:59.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I36daf64fe5df6682cf608eba6b4c7b7fe0652611","subject":"Drop entry point for non-existing container network driver","status":"ABANDONED","created":"2024-02-04 14:43:34.000000000","updated":"2024-07-25 13:56:59.000000000","total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5b817dd029a23da5bbebcaac6703238e277a2293","_number":907678,"virtual_id_number":907678,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-03-04 10:24:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"disliked":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":-1,"date":"2024-03-07 08:39:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2024-02-22 23:22:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-04 17:49:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-22 23:22:36.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-03-07 08:38:44.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2024-03-07 08:39:06.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"c28715476a7488be8e1b81241c16d0f69117e9fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-04 14:43:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"86fe6e534d3703decec0af16d7e9b585e317da6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-04 17:49:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7fb6c44cb32c4deda406eed5b5718479\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1bf14fc96669467fa366610e71b80ca7 : SUCCESS in 2m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/419ca4aa5b81469aae4d0cc61efc3403 : SUCCESS in 11m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/369889edf2c246cfbe2a3fd53877c686 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26c780eff95f4580b6060537db54b2a4 : SUCCESS in 4m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7187568fe29e470797734a64bb2e5697 : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a83e686c2314867aea0375c586ac3b3 : SUCCESS in 10m 48s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/8aad2c698bf0427c94a662b48005a08a : SUCCESS in 5m 08s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/0ff0d3f63f404412882cecb5234d1d9c : SUCCESS in 6m 37s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/514a311993a2462499c9e2ef285246c9 : SUCCESS in 14m 12s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a1cabd21fb8d436287466bcd0162c425 : SUCCESS in 4m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6045da3d63a34750b0d0e0ccb9a7b570 : SUCCESS in 6m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/77ad6aa9ffb44691b234add803acc266 : SUCCESS in 6m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7e8f909d18eb4946baf0d343530c13cd : SUCCESS in 7m 24s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/12e7baa7cfce433d92152d45b60a0fc0 : SUCCESS in 1h 14m 39s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/96c5d9566d6540398e2d71f900234b33 : SUCCESS in 2h 05m 28s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e76746cdcf11420c93d525b815e52ebf : FAILURE in 1h 24m 03s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/5ef145ce2e75494a8693a59fb4bd07af : FAILURE in 1h 02m 20s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/075cc3ad2e08409bacc12618038e990c : SUCCESS in 57m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/44cd5b6c1c8344f28c1363eff86be869 : FAILURE in 48m 34s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/b83c4aa359c747a29905e60d8285e2d9 : SUCCESS in 1h 06m 14s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/472618be5ec14a00aa5ddb750715056d : FAILURE in 1h 02m 38s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/cee3fcda76014be697e747d83adcabeb : SUCCESS in 2h 04m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/17187d7e4d68477abbc12f721de943f1 : SUCCESS in 3h 02m 18s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7135f5aaac3e4a40a3fbf0bca26659a6 : FAILURE in 36m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/34bd69e786b345a98e37a96821976965 : TIMED_OUT in 2h 34m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"17fd16d7e3697b378a7f61c3b58a55f2bb862e73","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-05 02:19:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8337705fbe73c63e336a70269aa68797875deb6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 05:44:40.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5456c1fcd88b4dc684e749227023b599\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1abcaee40a7045a0be506e37f55dadb2 : SUCCESS in 2m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/231dd544cb9448d599ca9eb91f71df7c : SUCCESS in 11m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99bf399fd418435e9a74fc57074fe5b2 : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c975640ea544f1186c8722caf391a77 : SUCCESS in 5m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2a897e1828684b6bbf9cce4087ab82f0 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/482aae92c43b4c0f8ef2f9fa5a8646ab : SUCCESS in 12m 13s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/2fae2418204d41f4a61454848bfa0db4 : SUCCESS in 5m 14s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/a387ef9ebf42421d8108aca0e03a2209 : SUCCESS in 8m 36s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/e5faeec2c0824025b8d92e232f4ff3a0 : SUCCESS in 10m 52s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3414788a26e94ca9bba0bfd9994622d2 : SUCCESS in 5m 10s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5ece017519b94b2aa342fc934c9d828b : SUCCESS in 9m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8fd967cea1454f91b8d4141080140fec : SUCCESS in 7m 49s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e3c4d953652147f4881c5dc4b901f129 : SUCCESS in 7m 41s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0545948c83b24cb7bdbb6ac95b0de663 : SUCCESS in 1h 22m 05s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/4afb3c8ce3e94ce6b5b5f385e595b4b2 : SUCCESS in 1h 36m 52s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/5c667ef7ada04ba49dfa1e67089078c2 : FAILURE in 1h 17m 51s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c7fe8367d75c473d950ae81ff3b19685 : SUCCESS in 2h 15m 20s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/6585652299bd48a484476b37aff77892 : SUCCESS in 56m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5615f03e50a5461eaee46508ce02214f : SUCCESS in 34m 15s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/08d479432a5c44aebe1d342a765ede19 : SUCCESS in 1h 03m 18s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/540e36d5dda14d1da8421ddf0ec4ec40 : SUCCESS in 50m 03s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/075f6ee1b2614c94925f943806229ab2 : TIMED_OUT in 2h 32m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/45b7ea864683439ebbc697081ffb671f : SUCCESS in 2h 49m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a7c04a6db3d040eba97ebf69a7a5368f : FAILURE in 33m 18s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/0d8b230006f14cc1b81d38e2a64aea74 : TIMED_OUT in 2h 34m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a50a13130e1b0faa3bd0f914ed115e6d28775bcc","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-02-22 23:22:36.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ba70d28d40c98b795f2f6475656f89ea57700acb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 02:23:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/89268dfe8978488d9bb354266b9c3c93\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6dadc20729a54676840d135774762dd0 : SUCCESS in 2m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27c318f9c0704f48a88e42b7758cce13 : SUCCESS in 9m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a183443e00c340ccbf89cb4c2a4dbcb8 : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e0343374e1c45e4a7cf09cf7795fe92 : SUCCESS in 4m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3ba59754467a4fd6aec0d7d97b56580a : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ee2676ec92840adae864af04789118a : SUCCESS in 10m 58s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/15c2dc196a0d4fc496e2cdf76948882b : SUCCESS in 5m 43s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/ab27b6691cba41f49d137ef135575b95 : SUCCESS in 7m 57s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/855debc62d6a4dab9231e116cabe4bf0 : SUCCESS in 11m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ade57bd8928c453f8d884edca8a0f9f0 : SUCCESS in 5m 18s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/558339673584478abf331448b58eaf50 : SUCCESS in 6m 22s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c0525c46917e4c0e84f21ea064c3e66d : SUCCESS in 5m 25s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/be6120273a144c6dbb3d7e949c739c77 : SUCCESS in 8m 15s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c228c8f662c34e168e265e265afdb7c5 : SUCCESS in 1h 25m 32s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/7915b5b16d57424bb4040ae52ddb5380 : SUCCESS in 1h 51m 01s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/d893d349af384fa9bd5f057e79a02179 : FAILURE in 1h 24m 31s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ab1043bf5541403c8c5425458addf264 : SUCCESS in 1h 55m 45s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/45e1b27acacd42ff9b8c37ccaa3ebe4b : FAILURE in 1h 27m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c4fc3ff6190a45debf2eaaac1c52e3bf : SUCCESS in 40m 59s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/66a02527eb2c48d5b570ff4a62b1ae21 : SUCCESS in 1h 11m 41s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/33878863703140a085b564038326b452 : SUCCESS in 38m 54s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/905fe3d41c4144aab1bd2ab9b77340d5 : TIMED_OUT in 2h 33m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/90e728c92d034ab4902d538e12fd6fd3 : SUCCESS in 2h 57m 57s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/580e3deaa9fa4b2aa344d4a3170e0868 : FAILURE in 35m 12s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/703f8dfd33364c1c95363b2b867f4924 : SUCCESS in 2h 33m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"af36031037e210b1b06812747591185794fb2608","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-29 16:59:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ada4aa361c8f77c9eaf2001888e02a4f8955b96e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 20:18:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/20b541a1cb7b40a89ae3e7af323c4933\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e545328a81354b5cbc0b168567f9b69f : SUCCESS in 2m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8df2e720b4ff4336976a6f47ef3f6119 : SUCCESS in 11m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45a72e5892aa4740a12077fb0f473a0d : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0044d32880c041379d6f70fba11c7cd3 : SUCCESS in 4m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4f68f2ac558e40aba383c01839e4150b : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/155d7d6dbf714e3c99994ca85815a8d2 : SUCCESS in 10m 43s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/8ca0183412b54f10be5fb4360145e4ce : SUCCESS in 5m 20s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/cf4ce609ab6a41389f18c68d7b02c4e7 : SUCCESS in 8m 18s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/6bcb60ebb7784e02b5f826dd2f416b0f : SUCCESS in 10m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/746378fa2b3c413b9e0f2f51a4948be6 : SUCCESS in 4m 59s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7522ede9696b4fd9b4d575c5b9de4073 : SUCCESS in 7m 51s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/26415fa8f08249d2960bb41550160324 : SUCCESS in 8m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d730de4670f54a1eb1a53a56f1d452e1 : SUCCESS in 9m 56s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/da89563cdd7e4dcfad828644747662a8 : SUCCESS in 1h 43m 59s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/6d7d03f2d0f645128b3dad688fd2a7f2 : SUCCESS in 1h 35m 18s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/bb5d10dc0cec44278550dbf519593da4 : SUCCESS in 1h 26m 17s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/cd3af1b603834aa4afd3b56ef3d13fc0 : SUCCESS in 2h 07m 38s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/d5cf265ff46e4e7a87170b42fb219d4e : FAILURE in 1h 37m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/868f326172834ac9a9b15ace47260704 : SUCCESS in 47m 54s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/bfef4beddc1a40a7a6e97c9f52f9a404 : SUCCESS in 1h 02m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/60ae62c88b2b4d8ba1c172c298f49d8e : SUCCESS in 45m 41s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d8bff6b62c8f4058aad614fe829f0893 : TIMED_OUT in 2h 33m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cbe84da4eb5b4c54b6e33987865beb70 : TIMED_OUT in 3h 02m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/745cab8c42384a7397b16f31bb04ff9c : FAILURE in 39m 25s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/8558707cbda04f49b72381539911c9af : TIMED_OUT in 2h 34m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ccf4d5e02b285a1d3f1533cefefe1bc9f9d78e17","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-03-04 07:41:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"20331bc0452fd98e6c5c56835f78b014f640e3b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-04 10:24:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af13bf1f5373423685cdafffc0dd5a84\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2309e988f9624bf48bd6f80375a947a9 : SUCCESS in 1m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cff4e42c4683475dbf14e900af44e7b2 : SUCCESS in 11m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8319e96d881b42bab2a98fd04a2a6e71 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f94b19eaa05495483823b6869d89ced : SUCCESS in 4m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bd9e17f3f7804cb4a5a5622bfd23ee99 : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de2a55e4770340169d2d09dc8514a221 : SUCCESS in 9m 30s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/499ef948b40c4de496118b1c817fba35 : SUCCESS in 6m 26s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/e8bcbe5e7ed6417a83531efdc50cf0b9 : SUCCESS in 7m 36s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/756f2bb4190440bb84b89605ee5e23d4 : SUCCESS in 8m 39s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/41b278eec7ca40fa928e4c9209b5197d : SUCCESS in 4m 20s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a823796ae02640bab5379bed7bb01f1b : SUCCESS in 8m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b97eb02d492a463abb910a634b7f16f1 : SUCCESS in 7m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7c98a31db7124e1b83e662665590fdb3 : SUCCESS in 7m 28s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/51ab07fcc2934fb19a5e186b58e883fe : SUCCESS in 1h 52m 08s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/30ecbea9d7b545c084617cb5b5b04aee : SUCCESS in 2h 11m 33s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/54d707e712914bc9b01a3845b845ca87 : SUCCESS in 1h 24m 57s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b36ff7bd09ac4df3bef7ac4f13ef55ae : SUCCESS in 2h 03m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/5e67dab77c08486abf32d06333ac237a : SUCCESS in 1h 35m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/abc1ca2c9f344e97a43ea370a04090da : SUCCESS in 42m 06s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/802df56d9e5f46079d70fe412a1fb561 : SUCCESS in 1h 09m 25s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/58384f807f984cc48b6d2f82be151a7a : SUCCESS in 33m 23s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/429b72e06f674ff8affa360df08b2e80 : TIMED_OUT in 2h 32m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/aebb00b433d749bf94720470e563d3c6 : SUCCESS in 2h 27m 00s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5553fad91d644c97af3c83540fd722ec : FAILURE in 23m 23s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/f35d294ad6534cabb051143ae7768bb1 : TIMED_OUT in 2h 34m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"55bc9d40bf39b85583a0786958ee33a26ff9cf38","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-03-07 08:38:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a81057981eab609ad48b297b89a4266ddb60627e","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-03-07 08:39:06.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"5b817dd029a23da5bbebcaac6703238e277a2293","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-25 13:56:59.000000000","message":"Abandoned\n\ndone by 75c1bdd104f3ac02cf05defc0655597b9d4e5e0b","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0350168e462f6943051541bd05d180ad8b02a1c0","revisions":{"0350168e462f6943051541bd05d180ad8b02a1c0":{"kind":"REWORK","_number":1,"created":"2024-02-04 14:43:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/78/907678/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/78/907678/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/907678/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/907678/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/907678/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/78/907678/1"}}},"commit":{"parents":[{"commit":"bbb4713464eee61add6ce32427621e36fa4086e9","subject":"Merge \"Declare we use threads and processes for coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bbb4713464eee61add6ce32427621e36fa4086e9"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-04 14:42:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-04 14:43:20.000000000","tz":540},"subject":"Drop entry point for non-existing container network driver","message":"Drop entry point for non-existing container network driver\n\nThis entry point was added by [1], but the actual implementation was\nnot included in the commit. The module path does not exist even now,\nso we assume that it was added by mistake.\n\n[1] fc427679bd9033dd5a97c785fc61b9d6ed4c7335\n\nChange-Id: I36daf64fe5df6682cf608eba6b4c7b7fe0652611\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0350168e462f6943051541bd05d180ad8b02a1c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0350168e462f6943051541bd05d180ad8b02a1c0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
