)]}'
{"id":"openstack%2Foctavia~907924","triplet_id":"openstack%2Foctavia~master~Ic263c3e316f7bdab832b9c7d22e9258924f8f9c7","project":"openstack/octavia","branch":"master","topic":"amp-cpu-pinning","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-02-07 08:55:17.000000000","reason":"Change was abandoned"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-02-07 08:55:17.000000000","reason":"Change was abandoned"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2024-02-07 08:09:40.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"Ic263c3e316f7bdab832b9c7d22e9258924f8f9c7","subject":"DNM Revert \"Optimize HAProxy SSL cache size\"","status":"ABANDONED","created":"2024-02-06 11:07:05.000000000","updated":"2024-02-07 08:55:17.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"revert_of":853606,"meta_rev_id":"1cb2eb15b19ffc4c10709d234ea0ac7426e282f2","_number":907924,"virtual_id_number":907924,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-02-06 19:34:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-06 11:07:05.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-02-06 11:07:05.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2024-02-06 11:07:05.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8b8be4d713cab44aab0c979ab68699ff165b1d6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-02-06 11:07:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"36e61c93cc952ac46ffcb4ca370a1d3ab64c501a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 13:47:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2f5652810fd4a058c1218c8dd7d708f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee7cf8a7078b4ff38f8e4cfea0324b02 : SUCCESS in 8m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86cc3e5b808d4ae68b6bc5e6ed7999ba : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/45599b19a7d0434fb2c5274785193ef4 : SUCCESS in 4m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/03780fc353714377bfa472d9d816a8ed : SUCCESS in 6m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/658da3674d3440df80082b0de4c46681 : SUCCESS in 10m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/735aa89e82934c46983fa3036cab6d4f : SUCCESS in 4m 32s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/af8e06dc8d8340679106e6fa4c8f55a3 : SUCCESS in 5m 10s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/0f24ad1600dc4ab5a4d0df8746331c5d : SUCCESS in 8m 47s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/1bab3b1762594d4aa2d674513a36c908 : SUCCESS in 9m 39s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ccf8c73c74eb442d8c1445ebcf8c4666 : SUCCESS in 4m 31s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ddb4afdc39524f9ca26ef5b02389fa66 : SUCCESS in 7m 04s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2771f5c28ea24d168e070cbbe2ddb55d : SUCCESS in 7m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d2d3fcd7ab4c42e4b550c2c0073f3a14 : SUCCESS in 7m 18s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d4e29b8dd25f4b569efbab9519cc9c8c : SUCCESS in 57m 07s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/f0707497050a408592311076e1b1da23 : SUCCESS in 1h 36m 36s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/b0fc99c9eb54484a85a25336f14f2afe : SUCCESS in 58m 29s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/48faf7181d7b4629815442fe5bfad702 : SUCCESS in 1h 40m 36s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/607f5eb991b04a63bfeb4ef7d0bfa72a : SUCCESS in 1h 03m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3287459bee5644cba7dd80ece9bd6ca6 : SUCCESS in 47m 01s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5e5f0cca846249098a61fcf451e92a41 : SUCCESS in 46m 43s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/f19edb98d5a84c1c92924908b92a305c : SUCCESS in 48m 03s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/a167ee40c4db4cbb95a8d8d02d98b657 : SUCCESS in 1h 01m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/774c48c5a7c24b769dac97717525f059 : SUCCESS in 31m 44s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/86f853838fb847008c16d310e4f8e77b : TIMED_OUT in 2h 32m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4ff2e0efe8d74dbab3f0e4ae27a2ec79 : SUCCESS in 2h 37m 54s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5a15e137ab1f4133bcac0e6f9f71c658 : FAILURE in 29m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/189766fd64a54d3b9201cd4418284516 : SUCCESS in 2h 23m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6078f3435757d382ec18820fefbb2c761742ac09","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-02-06 16:23:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"625e4d5e8c3f3688cc763d9cfc9300c432c994cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 19:34:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7c0d0cc9192a45d794bfbe49619544f1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b1b17f760f948a596c4c8353d778178 : SUCCESS in 11m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef4c4d748d9a455ba403321dfc0c4b38 : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/98dc5f6ff414402f8524185490d7fc3a : SUCCESS in 5m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9379058d56a1483b8d0fb1784f22440e : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfc887dffe204659a53d5b70cda8778a : SUCCESS in 11m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dada4a434a884af8a91d63b1695530bf : SUCCESS in 5m 01s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/1c83fe9e11d44a44b978fdb976cc4175 : SUCCESS in 6m 34s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/69ba60569c3c45aa99f92c0ba6b4548d : SUCCESS in 10m 33s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/0d81cf2707e44e9c9e569a954534f0ce : SUCCESS in 11m 32s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0241653a4e2d46cfac5e9359aabf7915 : SUCCESS in 4m 54s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d3bfcc7decf949ecad2f80d0bea990e3 : SUCCESS in 9m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7960b1aa7f804395a97819391301ed8b : SUCCESS in 9m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3413c80312904f8e81f97d72d5920761 : SUCCESS in 9m 55s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/237a2995a48e4a1ebdeb7bfe965b8de0 : SUCCESS in 1h 19m 54s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/b0a444b0d3a445b18f40d0a7978c435e : SUCCESS in 2h 05m 48s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/450dc3fe501346fba0da245b4e4c54b0 : SUCCESS in 57m 23s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c65ad0d6c0ec4e0ca108bc70782b56db : SUCCESS in 2h 05m 30s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/2fb9d09a8b354ffaa574d53ae0fc6713 : FAILURE in 1h 26m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/35843dc5166346659b0904a39a0518a0 : SUCCESS in 56m 06s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9fd9539a9bbe4181a2a3c4d2a3edc04d : SUCCESS in 39m 48s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/7f9ed6ffa0e6473c946598fef57f032d : SUCCESS in 58m 28s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/f72da0851f304b90a2612b48e6e26b9e : FAILURE in 1h 01m 17s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/876a35e8f8f144abb2bfabbbc3aa7498 : SUCCESS in 31m 37s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/25979c578a9e4583ba0607ba0cea63d4 : SUCCESS in 2h 04m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8ba9738618c54b08b23ddd43d1f76e12 : TIMED_OUT in 3h 03m 51s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/16a84c0ea5fc4a70bb9a03de1ee923ca : FAILURE in 20m 48s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/06bac78b8fcd4aac9d59ea45b0848504 : FAILURE in 44m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2842dca2d1965b0142bd0e026f48d538aca2942e","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-02-07 08:09:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1cb2eb15b19ffc4c10709d234ea0ac7426e282f2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-02-07 08:55:17.000000000","message":"Abandoned\n\napparently it has no impact on job stability","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9e56d8b86cdc6d7489308453579bd12fc285646d","revisions":{"9e56d8b86cdc6d7489308453579bd12fc285646d":{"kind":"REWORK","_number":1,"created":"2024-02-06 11:07:05.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/24/907924/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/24/907924/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/24/907924/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/24/907924/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/24/907924/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/24/907924/1"}}},"commit":{"parents":[{"commit":"454cff587ed10b5e504da93b074b77cb85055b13","subject":"Optimize HAProxy SSL cache size","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/454cff587ed10b5e504da93b074b77cb85055b13"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-02-06 11:07:05.000000000","tz":0},"committer":{"name":"Gerrit Code Review","email":"review@openstack.org","date":"2024-02-06 11:07:05.000000000","tz":0},"subject":"DNM Revert \"Optimize HAProxy SSL cache size\"","message":"DNM Revert \"Optimize HAProxy SSL cache size\"\n\nThis reverts commit 454cff587ed10b5e504da93b074b77cb85055b13.\n\nReason for revert: Just to test whether it has an impact on the traffic-ops test stability\n\nChange-Id: Ic263c3e316f7bdab832b9c7d22e9258924f8f9c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9e56d8b86cdc6d7489308453579bd12fc285646d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9e56d8b86cdc6d7489308453579bd12fc285646d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
