)]}'
{"id":"openstack%2Foctavia~909849","triplet_id":"openstack%2Foctavia~master~Ibadf6e529d53fb5a45b73af57243cee5a3f70d9b","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-04-24 21:40:55.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-04-24 15:16:49.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}},"hashtags":[],"change_id":"Ibadf6e529d53fb5a45b73af57243cee5a3f70d9b","subject":"Check haproxy status on reload","status":"MERGED","created":"2024-02-22 09:27:00.000000000","updated":"2024-04-24 21:42:04.000000000","submitted":"2024-04-24 21:40:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"909849","meta_rev_id":"5ce2d2d955b3e1155ace8ac6cc24e24906decb60","_number":909849,"virtual_id_number":909849,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-24 21:40:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-04-24 15:16:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2024-04-24 19:33:46.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2024-04-24 19:33:46.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-22 11:21:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-14 04:53:59.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-04-24 15:32:24.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"CC"},{"updated":"2024-04-24 19:33:46.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"4c4bc36cd2559381ed7c2af896d221b5343769ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-02-22 09:27:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cafda916cc5f9f5502f53649c8714244caa82d83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-02-22 09:39:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"57f609f858df2a0a7d78970a6a8b8061666160f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 11:21:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e7c4ceb00b7f403c914e06944aa17265\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1247387b73f4999a6440603f9c34114 : SUCCESS in 13m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67330610210b48f3853ed12710c2884e : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6cc66191d52848e09c4d3a03ac280814 : SUCCESS in 4m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/68262df7aa6442a8b678b6f9765c6d62 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a0eeb0cbb0d48659d9e549df85ae31d : SUCCESS in 11m 59s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/b0369adef8af44e6b6b01e6913ee0a4b : SUCCESS in 5m 36s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/ee0f3620c44c421895614bcf42be21fc : SUCCESS in 7m 38s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/16e3ef066e27482eb99391d4edbe88e9 : SUCCESS in 13m 34s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/5f938ceade0242b48ef90545ba17c29e : SUCCESS in 1h 31m 05s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/d4a75b5952774560bd1951c237f14621 : FAILURE in 1h 05m 02s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/67a5bc60183f40fba7b1d00091a7556f : SUCCESS in 1h 36m 01s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/be554226dc6f44f8bd27f234eaddde26 : FAILURE in 1h 39m 23s","accounts_in_message":[],"_revision_number":2},{"id":"29536597609b1eec62dcc004db8b6d6bc95f49a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-02-22 11:57:06.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"33f597d542e7db9fbbf1c8008e98cd5390fd431c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 13:37:53.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1c8b86ab020740278aba9dd1ea9f72e4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8faa58b205e1410ebc9c8805996de781 : SUCCESS in 14m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/200828f231034823b882adaaa5fe9966 : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a293224b7344f2cbf593febcae01b7c : SUCCESS in 5m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4943cce4e8af4f9297074ac0093c5100 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b7a9699bf4e4d57a2b77bd75c04e3c0 : SUCCESS in 10m 31s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/7cc1f18c0e964c4daabd0489c850fc3b : SUCCESS in 5m 41s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/048ca692a77c4b1aa884b3b6e7f21058 : SUCCESS in 8m 32s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/ccb7cef5447049e2a41fd12e673c6eca : SUCCESS in 10m 34s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/32fb05a2356348ae97485f5493e0596f : SUCCESS in 1h 33m 33s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/2091dc6589884c76b8a06b3bb1a4729c : FAILURE in 1h 38m 12s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ff35b9da684745e99d8a00d3ab88a870 : SUCCESS in 1h 37m 36s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a086d76651c94e8fb970bcb9657aa282 : FAILURE in 1h 22m 09s","accounts_in_message":[],"_revision_number":3},{"id":"ebf0590e227bc61388d3c570b284af8aef3eb59d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-02-22 14:27:11.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"1671ffdfde3702eeefc1ded674aab9f48a67f8f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 17:02:59.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1769565a405c47fcada9d4d801ee7cca\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d18030727ff4809b9769f006ee508e1 : SUCCESS in 13m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4e0580c33d8494b8afe47e31dc19041 : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/582f98bd58b34c8a96c4003d66be22eb : SUCCESS in 4m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3bf3ff99d8774bf195c3e9579be879cf : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc39941cbc3e454baa89afbeba4a3b96 : SUCCESS in 14m 14s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/724d80a8bf99405baba23c00f0d8e498 : SUCCESS in 4m 43s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/a12e6e80367149e8b85bde201226a902 : SUCCESS in 7m 53s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/c16fa60da4294bf6969befedf2af2563 : SUCCESS in 8m 40s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/64f7af14db414ab4ad4af97424146034 : SUCCESS in 2h 11m 41s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/fcf61cc45ea54c39ae483754b1804db7 : SUCCESS in 1h 45m 35s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/064313817c3541df9ab8b2644d9c2bde : SUCCESS in 2h 31m 17s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/693a6367fe9845e99c1619cde0e7493f : SUCCESS in 1h 37m 29s","accounts_in_message":[],"_revision_number":4},{"id":"14d5fae692b8353653e16407955e9c92c5459146","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-02-25 10:38:36.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b479aa97f6830e24d3679a22104d043bdf7825a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-25 12:25:46.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5feb323cf4c543889d8174a4905d9f0a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b09d6858ea64d18a896ff9850a74dcf : SUCCESS in 7m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c826e58ae0dc43fcbb5b93bba66ab2a0 : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a3f4724a0cb843819e9557e5deafb53d : SUCCESS in 3m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8228f9b4da1042159d46f77d10205c71 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b869f2fee9a4fc0aefe9749ec23d8f4 : SUCCESS in 12m 49s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/c5618927396e4a8e9da548b125437118 : SUCCESS in 5m 35s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/8e476225821f487782cc10a6501ba552 : SUCCESS in 5m 30s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/6f3706f546154768a7ba6133a4399dd1 : SUCCESS in 10m 05s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/93b5947a6bfd4492a0a89720217f5d55 : SUCCESS in 1h 41m 21s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/8df6eaa270334f108106557af14a16ca : SUCCESS in 1h 04m 03s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f9955bfbee1e4ee9a9e76722a7510bf8 : SUCCESS in 1h 45m 54s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/7b70bf133d9b42bb84ee3d8b7c406407 : SUCCESS in 1h 05m 48s","accounts_in_message":[],"_revision_number":4},{"id":"46d3c74a1f2960168567576c74c606fcc9477065","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-02-26 08:38:09.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"179fafe9661625ff9dca0d6227444269878a4f21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 11:31:18.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d86107a1b1d74fc199502b0643e36480\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c7212c516a2549b6a0b781e54d708996 : SUCCESS in 10m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb771b2365414001ba1e90ffcde3426b : FAILURE in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de1aac014600462588194ef80e25b080 : SUCCESS in 4m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/712e9cc900cd475597b686f5a2580c4f : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcf12b43210f4cfe8a86a62ad29ae707 : SUCCESS in 10m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/536ae51a374145c0a488e38c8d3b8ef9 : SUCCESS in 5m 46s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/10d75e79ab0e40c38605293ff4538a11 : SUCCESS in 4m 31s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/ee2fc7f685f44aaf86b7e7f5d1c8fff5 : SUCCESS in 7m 57s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/4ac63d90be3c4eaf9800c0544dcaa575 : SUCCESS in 10m 01s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/09aa4228846741fba80557e7da623deb : SUCCESS in 4m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2c3a723e067242dbbda75cb9028ed5aa : SUCCESS in 6m 04s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/563c97f837a64c16a7850495b28d4fc5 : SUCCESS in 7m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2a7e4abd67684d51a4e05500d4fd88a9 : SUCCESS in 6m 47s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/44927f3ac7ad4422a839e2f8ce6e510a : SUCCESS in 1h 11m 22s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/192d8f09a2ad4d3d842d7500c9b5594a : SUCCESS in 2h 08m 09s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/91954b3ccc0149c4be48d810cab0dee0 : SUCCESS in 1h 46m 51s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/31e010b10789490dae549fdeee56cfa8 : SUCCESS in 2h 15m 03s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/160613a6bcfe4489b9e1cc48f43fd558 : SUCCESS in 1h 08m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/430e8e3273be4d8bae835d9d1b15e441 : SUCCESS in 34m 55s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/305374279d4f4220910ecaec072b7d03 : SUCCESS in 39m 11s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/71e8e729d3e34fceb75691063a35f8d4 : SUCCESS in 39m 04s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ea5ea6187b214f5e9b22dbdf98129ebb : SUCCESS in 1h 07m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0f9081933d5542c790f48a03da540cdb : SUCCESS in 35m 25s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/74b7a6020fea4259a3ca670af250ead6 : SUCCESS in 2h 04m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6be406dfb2434adc85a8c00bba622fcf : SUCCESS in 2h 51m 51s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a4c655003e9f4003891b0d192455e7f1 : FAILURE in 23m 36s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/7fa75df5de624f2889a79ddc2f4c4801 : TIMED_OUT in 2h 33m 14s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/osutils.py\n  Comments left for invalid file octavia/common/tls_utils/cert_parser.py\n  Comments left for invalid file octavia/network/drivers/noop_driver/driver.py","accounts_in_message":[],"_revision_number":5},{"id":"a4993ff403f128b4d089fccc51b8c858c50c6726","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-03-07 10:25:24.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4a661fe2caf98caf564acd5043f345268315ebc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 13:18:51.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f21494eab6be4cc497773c2018d2536b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af97063b0f674ec6b6111c352a94708e : SUCCESS in 10m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad5fc40942ae4673ad2f3b80efce197c : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c8f4101cdfe4437da77d2587dc114556 : SUCCESS in 4m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fd94c86537ae4e0a8313598285276c63 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0629da4388b48e99e5e8aa6f03a34f4 : SUCCESS in 10m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/89d3ed0133d546fc969df8bb5fd864c5 : SUCCESS in 5m 24s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/3235d988cd7049eeb7b43fce3dcb5fd5 : SUCCESS in 5m 42s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/b5083e04a445488294d9fe2823131f62 : SUCCESS in 7m 18s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/c30bf078cca547e3a8d54ebca6e5d82c : SUCCESS in 9m 52s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e58e0850ca6243d08e5b9fd98c49c18c : SUCCESS in 4m 45s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/aedc0e4130a2406aa305d86e681ce9eb : SUCCESS in 8m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/def33c22418a4c40a781d1fd7d388d0c : SUCCESS in 8m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fefc04b168c04c60821a19a343643e79 : SUCCESS in 8m 05s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/472292bf062d498582223749ae39b753 : SUCCESS in 1h 38m 48s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/736bff0058fc44469f7c2275214293b8 : SUCCESS in 1h 38m 15s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/6edf53c217b24558bccde505b556183d : SUCCESS in 1h 24m 11s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/5b5807a7f8014e768d8abc908766660d : SUCCESS in 1h 48m 04s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/236144bfc2df40aabf4739fd6a8f7d21 : SUCCESS in 1h 07m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4667836410544789a7f904beb14a4c30 : SUCCESS in 44m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/668ce2ae8e2a4073abe3866774452b49 : SUCCESS in 45m 58s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/5347162b27814bc6b41225a7d9bf4ec7 : SUCCESS in 42m 10s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/d690fa1b1c3e4af88cd2acb5d250dc72 : SUCCESS in 1h 06m 21s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/564c1feb913645ca8e334f5f9b439bf1 : SUCCESS in 40m 05s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e4fcc606ee414fb2b5829782e8c748f7 : SUCCESS in 2h 26m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/dc283515c0a44f37a88d9390615d389c : SUCCESS in 2h 51m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/38bdef2390ac4e00bd21fac146b9c5e8 : FAILURE in 24m 36s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/20d05c4187154baa971f66326a5d2916 : TIMED_OUT in 2h 33m 32s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"aa041cf5f7697be753c7600f7ebe8c4b86d45aea","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-03-14 04:53:59.000000000","message":"Patch Set 5: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":5},{"id":"c3d36ea19b2668f3b724cf68a8ebfb5dd2d5a817","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-04-24 09:24:38.000000000","message":"Patch Set 5:\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"3fe0b97f715837dda1f19d5dc1c98ba255a10a4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-04-24 09:32:27.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"be5764d83ee6947257d3ec776dd76d55483099d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 12:17:34.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf4f24d530d04eec98f70fc619e668a8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/299687844b1a417da090fdc46956a8f9 : SUCCESS in 10m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba876b61ff3044d7ba010f44362d618b : SUCCESS in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/85a84919bf904801bef029a393a287d4 : SUCCESS in 6m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4ba0854c675f46409313a83ce29fcc8a : SUCCESS in 7m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bcd98ced1bec401780cd4780b4a6b7da : SUCCESS in 8m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/643c4a87ca21446890840a972371795c : SUCCESS in 10m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c2324dccf9c4291aa473264532737c6 : SUCCESS in 5m 52s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/a19a804f8171406ebc9b935b72c59fe5 : SUCCESS in 5m 40s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/8bbeb5f529854142a37ddafbf34b694d : SUCCESS in 8m 14s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/93f20f2242214062b5047d4fc414068e : SUCCESS in 11m 32s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/98e659b0be0d43d68d103108ec753824 : SUCCESS in 4m 54s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c13d5bfec7c042b2ade617d525ab136a : SUCCESS in 7m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fc027cf1a24c43698da1811fcf3c1f57 : SUCCESS in 8m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3e0a885cf79440569f7ff61d7f5ef291 : SUCCESS in 7m 15s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/43fc2465723b41b2a466f766b35fd7ec : SUCCESS in 1h 50m 26s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/1033956a8b3042f18f18c866d733cf79 : SUCCESS in 1h 40m 15s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/45eb4874f3e946888664a592abb6bc75 : SUCCESS in 58m 45s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b50d94238f43468b84711d4b1b34fb45 : SUCCESS in 2h 22m 01s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/23ce042bd26d4a159cafe7bb6941ec86 : SUCCESS in 1h 27m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2ad36880392f4e5387c6d8da7ffe384d : SUCCESS in 57m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4b75c41972fd461f8c88369cba5dbe36 : SUCCESS in 1h 00m 46s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/4934dfec9ab2444986023afd150835d8 : SUCCESS in 1h 10m 48s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/89d1c6a1d649404684a83a92d317f54a : SUCCESS in 34m 06s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/89644294974e400f99185e2d713f8729 : TIMED_OUT in 2h 34m 15s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3b21e13f7e814d5791b2e116c05baa4c : SUCCESS in 2h 43m 44s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3f602419f8c64ca683a6a25e8bc01b49 : FAILURE in 23m 54s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/84929269cfe84b648c1c3d7f5171081f : SUCCESS in 2h 31m 38s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"53929aa2d97ea5fe0be44b9932c34706cc5f9902","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-04-24 15:16:49.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f7abef1220b502d7a4fad9d18f0a2e0e6638b056","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-24 15:32:24.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"2384aa0d070c648baf1d0dc349ef9d5f77a3d99e","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-24 19:33:46.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"72b1badb10f34da95a79f5b773ca97c053ad9cb0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 19:34:21.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"4281b994148748f5f7a36f3b7d5c4ace1510eb70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 21:40:55.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21ef8a7d19ec41508cfb7c4b635bea19\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3cced5028184cafaf10a9f836a08c96 : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/90032c0400544f2db0b1d3bdbb8bac7d : SUCCESS in 5m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2865d953be454f3fa59ccf5b075717ef : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f31abeae798e43678baa81242d4fa6f4 : SUCCESS in 11m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d3d3867b4b34b619e23b04ad39bd141 : SUCCESS in 5m 19s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/cf99aeb93bd3477c9cd10bc08d7dc05b : SUCCESS in 4m 33s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/282e0d32bac948eb802c63adde47ba76 : SUCCESS in 7m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/48a8fd99dcf9477a9eeaeacec1782064 : SUCCESS in 1h 27m 22s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/9621d910e875484588df62ca302fe758 : SUCCESS in 2h 04m 59s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/51c9140d0915455aac9eecefc18a1f3f : SUCCESS in 1h 16m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7d4d3ff6d77e4bbc9e8aa3c5cec3693c : SUCCESS in 43m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5aac229bdca742efa27676c8345d6b5f : SUCCESS in 48m 09s","accounts_in_message":[],"_revision_number":6},{"id":"24126621072e9eb3cb6a86223b58aa2d4ec5fbdb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 21:40:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"5ce2d2d955b3e1155ace8ac6cc24e24906decb60","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 21:42:04.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/101bbbf8c8bc4878970cc55a008617ea\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/890d8f84e99e4c969d9dc8a8fe57c62d : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/141cf3f4d8654b009f3d22b6f4caecc1 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"f5349aca00d04222d838e6f55edc8c2131faf3c0","revisions":{"f2c1ef91f25b1e03ec50314052e0057624b00d0a":{"kind":"REWORK","_number":1,"created":"2024-02-22 09:27:00.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/49/909849/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/49/909849/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/909849/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/909849/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/909849/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/49/909849/1"}}},"commit":{"parents":[{"commit":"bbb4713464eee61add6ce32427621e36fa4086e9","subject":"Merge \"Declare we use threads and processes for coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bbb4713464eee61add6ce32427621e36fa4086e9"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-02-22 09:16:45.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-02-22 09:18:50.000000000","tz":-300},"subject":"WIP Check haproxy status on reload","message":"WIP Check haproxy status on reload\n\nChange-Id: Ibadf6e529d53fb5a45b73af57243cee5a3f70d9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f2c1ef91f25b1e03ec50314052e0057624b00d0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f2c1ef91f25b1e03ec50314052e0057624b00d0a"}]},"branch":"refs/heads/master"},"32804a6723047e40368e8a45628a944ba69fdf7b":{"kind":"REWORK","_number":2,"created":"2024-02-22 09:39:00.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/49/909849/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/49/909849/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/909849/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/909849/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/909849/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/49/909849/2"}}},"commit":{"parents":[{"commit":"bbb4713464eee61add6ce32427621e36fa4086e9","subject":"Merge \"Declare we use threads and processes for coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bbb4713464eee61add6ce32427621e36fa4086e9"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-02-22 09:16:45.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-02-22 09:30:51.000000000","tz":-300},"subject":"WIP Check haproxy status on reload","message":"WIP Check haproxy status on reload\n\nChange-Id: Ibadf6e529d53fb5a45b73af57243cee5a3f70d9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/32804a6723047e40368e8a45628a944ba69fdf7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/32804a6723047e40368e8a45628a944ba69fdf7b"}]},"branch":"refs/heads/master"},"e30019768bcdff73be79e4af746f60f8a8786f48":{"kind":"REWORK","_number":3,"created":"2024-02-22 11:57:06.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/49/909849/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/49/909849/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/909849/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/909849/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/909849/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/49/909849/3"}}},"commit":{"parents":[{"commit":"bbb4713464eee61add6ce32427621e36fa4086e9","subject":"Merge \"Declare we use threads and processes for coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bbb4713464eee61add6ce32427621e36fa4086e9"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-02-22 09:16:45.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-02-22 11:48:52.000000000","tz":-300},"subject":"WIP Check haproxy status on reload","message":"WIP Check haproxy status on reload\n\nChange-Id: Ibadf6e529d53fb5a45b73af57243cee5a3f70d9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e30019768bcdff73be79e4af746f60f8a8786f48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e30019768bcdff73be79e4af746f60f8a8786f48"}]},"branch":"refs/heads/master"},"1422a80de2c615ac073171044faaabbef43f86ff":{"kind":"REWORK","_number":4,"created":"2024-02-22 14:27:11.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/49/909849/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/49/909849/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/909849/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/909849/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/909849/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/49/909849/4"}}},"commit":{"parents":[{"commit":"bbb4713464eee61add6ce32427621e36fa4086e9","subject":"Merge \"Declare we use threads and processes for coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bbb4713464eee61add6ce32427621e36fa4086e9"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-02-22 09:16:45.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-02-22 14:18:55.000000000","tz":-300},"subject":"WIP Check haproxy status on reload","message":"WIP Check haproxy status on reload\n\nChange-Id: Ibadf6e529d53fb5a45b73af57243cee5a3f70d9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1422a80de2c615ac073171044faaabbef43f86ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1422a80de2c615ac073171044faaabbef43f86ff"}]},"branch":"refs/heads/master"},"4ae245a67abf35cb791eee8b1ebb9c917b4164d0":{"kind":"REWORK","_number":5,"created":"2024-02-26 08:38:09.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/49/909849/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/49/909849/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/909849/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/909849/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/909849/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/49/909849/5"}}},"commit":{"parents":[{"commit":"bbb4713464eee61add6ce32427621e36fa4086e9","subject":"Merge \"Declare we use threads and processes for coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bbb4713464eee61add6ce32427621e36fa4086e9"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-02-22 09:16:45.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-02-26 08:29:32.000000000","tz":-300},"subject":"Check haproxy status on reload","message":"Check haproxy status on reload\n\nWhen reloading haproxy, check if a \"\u003cservice\u003e is not active, cannot\nreload.\" error is triggered by systemd, it means that haproxy crashed\nduring the reload. When this error is detected, verify if haproxy has\nreloaded correctly (check the socket and its uptime).\n\nRelated-Bug: #2054666\n\nChange-Id: Ibadf6e529d53fb5a45b73af57243cee5a3f70d9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4ae245a67abf35cb791eee8b1ebb9c917b4164d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4ae245a67abf35cb791eee8b1ebb9c917b4164d0"}]},"branch":"refs/heads/master"},"f5349aca00d04222d838e6f55edc8c2131faf3c0":{"kind":"REWORK","_number":6,"created":"2024-04-24 09:32:27.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/49/909849/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/49/909849/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/909849/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/909849/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/49/909849/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/49/909849/6"}}},"commit":{"parents":[{"commit":"93dd9916c3d9407fd6e8580a166c6a56250389ad","subject":"Merge \"devstack: Drop removed services\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/93dd9916c3d9407fd6e8580a166c6a56250389ad"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-02-22 09:16:45.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-04-24 09:31:00.000000000","tz":-240},"subject":"Check haproxy status on reload","message":"Check haproxy status on reload\n\nWhen reloading haproxy, check if a \"\u003cservice\u003e is not active, cannot\nreload.\" error is triggered by systemd, it means that haproxy crashed\nduring the reload. When this error is detected, verify if haproxy has\nreloaded correctly (check the socket and its uptime).\n\nRelated-Bug: #2054666\n\nChange-Id: Ibadf6e529d53fb5a45b73af57243cee5a3f70d9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f5349aca00d04222d838e6f55edc8c2131faf3c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f5349aca00d04222d838e6f55edc8c2131faf3c0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
