)]}'
{"id":"openstack%2Foctavia~910133","triplet_id":"openstack%2Foctavia~master~Ia5396895b27e4b28a7d9d8d85a85a8449c21d493","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-02-27 03:27:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia5396895b27e4b28a7d9d8d85a85a8449c21d493","subject":"Fix pylint error","status":"MERGED","created":"2024-02-26 04:02:38.000000000","updated":"2024-02-27 03:29:11.000000000","submitted":"2024-02-27 03:27:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"910133","meta_rev_id":"7b83fc1ea6ffc41d9b9dfdfe53cf6feb47efb200","_number":910133,"virtual_id_number":910133,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-27 03:27:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"all":[{"value":1,"date":"2024-02-26 07:59:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-02-26 08:05:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2024-02-26 19:12:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2024-02-26 08:09:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-02-26 08:05:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2024-02-26 19:12:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-26 07:21:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-26 07:59:11.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"},{"updated":"2024-02-26 08:05:04.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2024-02-26 08:09:48.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2024-02-26 19:12:15.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"f65742f73c1ee8d0d2aa699ffed301250cb3673d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-26 04:02:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e7bb99f7d6ced4fdc86e13b98ddd685d7742026d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-26 04:16:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f107295cb81a1b56913893d0b1c476cd36b56c23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 07:21:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/334b4c7e3e7c48f198f19bb23e405bc2\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/20592f2675fb454185c87ac96eab96e8 : SUCCESS in 4m 42s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2aa14401c3354b17aeee1c69df56551e : SUCCESS in 1m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/72b48ce16c164691bb9c6de964d1e7bb : SUCCESS in 9m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0a85586a101466097518fdd19223659 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3524383211694ed29911a339575b82a4 : SUCCESS in 4m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6b0c83d5c8ad4d86be97e302d25536e3 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efd4d4da43734b9b9cd9b6f382e55628 : SUCCESS in 10m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ee8de273e9f4765811b92e6ae8735eb : SUCCESS in 5m 40s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/4cccab264f0947d49e68f65fc3b7f54a : SUCCESS in 5m 45s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/498f59ce43994c4992a4da9b3482e8e4 : SUCCESS in 8m 51s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/215c2cc3e3844e91acf0350332414783 : SUCCESS in 12m 43s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/dc42f4c3c7d24272811ecced5376ef69 : SUCCESS in 4m 28s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d701b99e7b6d4fbabf82fba5053fffab : SUCCESS in 6m 16s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9c526a25bd5343048f313bdb37a41c48 : SUCCESS in 8m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0d29440e33f14e0ea34cfa8388d373c8 : SUCCESS in 8m 10s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/eda2f5ab35c544f9af3963eaa09afc5a : SUCCESS in 1h 34m 45s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/3955470ce4804038ae0167fed4eddb37 : SUCCESS in 2h 06m 13s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e397eee965d14b60a4e28585ac614c11 : SUCCESS in 1h 00m 49s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/80828212bff14d19956e7778027ae800 : SUCCESS in 1h 51m 56s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/63ac413e90c94f699ef575614aaa725a : FAILURE in 1h 20m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0eb311bf1c024a8c8c7ce63a96c646ff : SUCCESS in 40m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/972134d3f6254bd199d3960128fb1d93 : SUCCESS in 45m 22s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/0db17b0e69a5465a8562cd84a6bf9190 : SUCCESS in 46m 59s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/22dcafc96c5e4adfafca3581fffcd36f : SUCCESS in 1h 06m 51s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4cdc17b84ff74e84a1709a568092c788 : SUCCESS in 39m 22s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d56da891ebec44558bcd3b4d833395b1 : SUCCESS in 1h 59m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9cafb3371a5847d2bc416e6b69fa6c72 : SUCCESS in 3h 02m 23s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c857f71109104907913db9bc29873102 : FAILURE in 28m 05s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/ddde8ee486174d80a7a5ae953aa15628 : TIMED_OUT in 2h 34m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"88bfa16344bdef478a5e6a90a2284ed5af91276c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-26 07:51:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b2621073387c8b8f50dabbbc47e3d2dbe9983a7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-26 07:58:49.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"a2f23f8507569a1be7469484a167bc1dd5c69b4b","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2024-02-26 07:59:11.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"2327a7decfa583315b6a9eb2c92c620a4ede1491","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-02-26 08:05:04.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2ca80d2ad800b2b11c4a53e85952b3cf1be24ca0","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2024-02-26 08:09:48.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"7eef70d2e8865b6cc99dc31e35f29aacba21d3f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 10:52:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60aef4cfa7264182ad5da565e20d1d87\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cbce8c5a11c349808ecaf63689147708 : SUCCESS in 11m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc60a896ee5947c7b7fe98ddbb72fcb2 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75f59c54c78545e489614bac684085e6 : SUCCESS in 4m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4b149abe0cb24d0091c569bcb3320356 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fba2660105c24571854678bb452c730a : SUCCESS in 11m 16s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/98e222e7b92b4ad5b394a8e5d02e8b74 : SUCCESS in 3m 56s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/a04990f0346a4e1c9042decbeb527eb9 : SUCCESS in 6m 52s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/c491af162a2145518eec1c2e7b35d488 : SUCCESS in 8m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1f29a702144e43168150afa2045db881 : SUCCESS in 4m 19s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2c3a437cecc144308025b923a1fe1b9a : SUCCESS in 6m 26s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/df8c6333530a4c7597626a8f2468f40d : SUCCESS in 8m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/460d97ebecbc46439a302ee4f60e0a12 : SUCCESS in 6m 31s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/566dacb9792d471a8a3d080104eb54bb : SUCCESS in 1h 14m 43s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/804a42b098ae4eb9b75b47cfb0541bb8 : SUCCESS in 1h 35m 09s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/872346378be144fea21908bea6090f9a : SUCCESS in 54m 20s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/1d0fb36ac86143faa8c11bda2ed40721 : SUCCESS in 1h 38m 21s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b4e7102acbc149848eee36ec16a08c05 : SUCCESS in 1h 05m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9d23c8d13a3c43da99a2b0e498aad542 : SUCCESS in 35m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/be8e4b064c9649c39ae8f170d3334307 : SUCCESS in 39m 00s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/71bd7533c0bd4c28b7577f30a086a306 : SUCCESS in 37m 43s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/b7daa8431d854bc781d0986f173d2fa0 : SUCCESS in 1h 04m 55s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a40779bcbf3e4ffa8ed0db531ce33e84 : SUCCESS in 43m 16s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d35ecf707fef45bb88ce4db4709a28eb : TIMED_OUT in 2h 32m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/172ea3e9b21945dc91ac3ee439e99ac4 : SUCCESS in 2h 50m 23s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/714665440862434c99d3b9febbc12819 : FAILURE in 29m 32s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/f47a384b64e44e9c867c820a75ed10d4 : TIMED_OUT in 2h 34m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7e6645da1cbca2de98a9367c2d66ef317f3ca6a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 10:52:42.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"354b17fa4cad9af9e8f4e7d251c5c7344204ffe2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 12:13:49.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/90a010562ae24a9997bf904104b6289f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/108c608b12f244838b55c8ea7be86154 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3268dcf42ae34cbd838dd20e33e80ef5 : SUCCESS in 5m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4943fe5d61204f7db9dfb70dd7140d3d : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3aad6b4551d4c7b9a35cac7a254ff33 : SUCCESS in 12m 20s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7317e1c9fe0645319ce0c8b613f4e16f : SUCCESS in 5m 33s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a1d28bf7ca86405fb4d2701073830391 : SUCCESS in 10m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/842efc2395c54db59f713350e3def200 : SUCCESS in 1h 10m 47s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/d0d492bfaf6d468ca43894b0b04c166d : CANCELED\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/4f5835289454478f9017a7ccb67b12f9 : FAILURE in 1h 19m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9ad8cd42eada47feae49191ab0474a16 : SUCCESS in 52m 15s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8460c7d47c554818a497fbfeb4a2d3e8 : SUCCESS in 55m 15s","accounts_in_message":[],"_revision_number":3},{"id":"20bfa995721b987dacfce883d4872a52434de5fe","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-26 13:28:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8864f83dca0b846ecfae9432b4837272df110187","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 16:57:52.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7e23c8b2f453472ca0e01f091808d63c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dac0dc1e18d94aaab96e7d326489ee4a : SUCCESS in 10m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18625a5bb05444a7bdcdb39b080ea0ca : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a781f58d72404ff4bbd2786fece77bcd : SUCCESS in 5m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e2e2dca5d9b944368da91e2137883ae8 : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/026e9cde80754ce18504f489c5eff3ba : SUCCESS in 9m 10s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/951ef3a7ad964e58837504751a158f0e : SUCCESS in 4m 44s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/a88028ee2bb44d52bb5c81957f566a15 : SUCCESS in 6m 33s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/a6ecbcd66a784b7d8c94c114c2b20978 : SUCCESS in 11m 57s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/23ef368368344a3c84ca5fee7f67673d : SUCCESS in 4m 51s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/abc26f67b44c461aaa45a364eca39b4c : SUCCESS in 7m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a353957f447443c3aa25715090f23bff : SUCCESS in 7m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/822ea39271754e2bb39254abbe126959 : SUCCESS in 8m 24s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9939a0a196704d64af1cad06e8b33118 : SUCCESS in 1h 15m 11s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/b5b3df3cd306485b842dcb5ae6e011f0 : SUCCESS in 1h 33m 19s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/6d8bcc873fa34bce837f80adaab5b0ef : SUCCESS in 57m 24s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/081e194e957645998523565bf17ae311 : SUCCESS in 1h 35m 05s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ab42bc5fa78245258cabd725ad14039f : FAILURE in 1h 52m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/686f8b693f474d179e88b3c96f75b0fc : SUCCESS in 35m 36s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/28930c7e08f44c9cb129c175709cf5cf : SUCCESS in 59m 05s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/44ad764fe1194f31be6ea2a784148888 : SUCCESS in 37m 56s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/825cd325cfdc48389ab18ca202f8a474 : SUCCESS in 1h 07m 23s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/23ddff228dfc432a90311550167cca3f : SUCCESS in 34m 01s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e1fe4d9563da4e6dbba9c871e6e6e7d2 : SUCCESS in 2h 10m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6dabdb551ba4430b90f7b777a4114848 : TIMED_OUT in 3h 04m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0392c665b4bc44509466d91275e9d47c : FAILURE in 24m 25s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/1a8f989eb0c8427b804efe9e92a2c22e : SUCCESS in 2h 28m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c0e5416ea02c035922d54e60efabc94e4ec1e68d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-02-26 19:12:15.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a618c36f49c08f91d3d13c071d56f3bd9f3513b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 21:40:33.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c80617d63d1f4ec58e75f639a4383f2c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c3dc072bd71347faa49c27f14b4f42e9 : SUCCESS in 8m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb8ea3c41ee84894bbf96b2a6b20a371 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/863597902bfe4c999ced1b41781bbad8 : SUCCESS in 4m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/96297d1a752f46a5be5a42df8311c190 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7578c0e8ee244f794ea1e1db55b0337 : SUCCESS in 10m 53s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/591a82dc766f49ef86a00bbcae05b504 : SUCCESS in 4m 44s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/ab0ee545124a4edd855f030d0fd869e4 : SUCCESS in 8m 14s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/91273f8844214a2e89c6f93871c3b8b0 : SUCCESS in 11m 44s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/dd89fe9522924532af6f10f4ac68eaf1 : SUCCESS in 4m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6532739ef42740a8badca32908c8598c : SUCCESS in 7m 01s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/10ab7d8599364d71a597c44dc2661721 : SUCCESS in 7m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/81b078be082d43c69d996a99ddc8241a : SUCCESS in 6m 17s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/abb223f1ee004b0e916eda550438dc27 : FAILURE in 1h 58m 29s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/10b2622f63814dfaa3a8f33558141121 : SUCCESS in 1h 41m 34s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/3afebecb0755448097c110054b4cc748 : SUCCESS in 1h 10m 24s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ca5f11fb9b99487fa5cd18afc15a4394 : SUCCESS in 1h 43m 18s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/38fc1a43a20c4e21a3f551b8177d604f : SUCCESS in 1h 03m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/174809ea8e66485689b220bdb8298b2e : SUCCESS in 38m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/131ba062b71049c3a209f81eb2a1dfbf : SUCCESS in 43m 36s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/c09fc57483ee49efb890818a8ead935c : SUCCESS in 42m 40s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/6c075981bbde40beb36a6005898d0a67 : SUCCESS in 57m 02s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9661747d081842d7877fe0f657a55140 : SUCCESS in 36m 24s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c10a162fb71f49eeb648a1d394d97b64 : SUCCESS in 2h 05m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2252d4320abe4aaba426c88de27449a9 : SUCCESS in 2h 26m 48s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/28d6cd70d73847a5a2874613634aa33e : FAILURE in 30m 37s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/74ccef9a6c70482da83ec100d1597ded : SUCCESS in 2h 19m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"eebf3f39781fb942b95336eb9561e29801aff134","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-02-26 22:30:14.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e267d23fcad2943c174a9b293da6ca9b45f6be69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 01:10:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27445fc4aa1740bdb6cbbc30fa2a8ee3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c79e3c136ab4a2b9d27829cef4e10e0 : SUCCESS in 10m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1011e8b2cc3e41c794d78429fd3b07cc : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a27261fbf6a84742b743eedaa2388595 : SUCCESS in 6m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5bde1c25183b40788fd4ea1b1c401a95 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00e3b1fae7f44959aba403faeff775ef : SUCCESS in 11m 20s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/1ec293045d03432089edf92e7ad01f30 : SUCCESS in 6m 42s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/d21876a17d7849488316555ddc54632d : SUCCESS in 7m 46s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/12a31b0d887745a58e541ede028e6739 : SUCCESS in 9m 12s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f80b16482eee41428696744fcc4fab8a : SUCCESS in 5m 33s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/369f92f5be8e4615ad46e9ad1af6fc32 : SUCCESS in 6m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/90a501ba94ef4140bd9cd52d6d73d21f : SUCCESS in 8m 44s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c23f95dde56c4456b9c7e74f2950d018 : SUCCESS in 7m 55s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4882fee46e24492796b4f1de215e3b53 : SUCCESS in 1h 15m 59s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/b4a76043f5f24bb6921281d049c28430 : SUCCESS in 2h 06m 33s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/28a885685bb141dc8bf8abb861a8efa2 : SUCCESS in 1h 30m 34s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/42faa49f2a5a4c8998cae844346caf73 : SUCCESS in 2h 04m 59s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f971ae110f784480bac452a4315c666a : SUCCESS in 1h 36m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/43b7659bd39d4595af9890eca7eb1df0 : SUCCESS in 51m 15s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/571b06d3b770422389607317436b3d46 : SUCCESS in 1h 01m 44s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/a8766e4c45b9425a874c0ea497badedc : SUCCESS in 55m 58s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/72875e9d36a543ceaa1b7e0897145bbc : SUCCESS in 1h 11m 55s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/925c664f4eab476f9ba48481877f0732 : SUCCESS in 42m 57s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/69a0da56dbee4a85916da41121f0354e : TIMED_OUT in 2h 36m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/86ffe8bfeb124459aacfca0f0100397f : SUCCESS in 2h 23m 30s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/24ee3d4e020d4c69b876ec45d4bb70d6 : FAILURE in 30m 22s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/b9ff7e92ca9643b28e56f9ee05e4cac8 : TIMED_OUT in 2h 35m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0adbf9a3c03da3200398be4bcb797da8a722e5cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 01:11:27.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c817c60cdbc6e550baf09faf093c4880791a8c1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 03:27:53.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/139545f0f1954bc8b99f9d5c50ea2d0f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89d06decf9cd4e528b1dd48326c031a6 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/567608341d3b4dcc8da91a6852df59e1 : SUCCESS in 4m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a9cb8e48d25e4668ac41f9a0d4837fef : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dd9f128141248c7a802aa236e09df57 : SUCCESS in 11m 18s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e4dca0be86b345f9962242a3cdbe8d37 : SUCCESS in 5m 30s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/516e39d0e7a14cc8a5b9d301cc8a0875 : SUCCESS in 6m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c357404150354213aa7c8039bafb2fb7 : SUCCESS in 1h 44m 26s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/3e2bc52052d6489baf383e21a71a1c45 : SUCCESS in 2h 15m 08s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/18cd9c7ad685446e8dfed841d1fce92d : SUCCESS in 1h 15m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c00e88f3b91a4262a4ecd6e15b8fddcc : SUCCESS in 57m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a534bd7ae022407fa18d486be833aae8 : SUCCESS in 1h 08m 26s","accounts_in_message":[],"_revision_number":3},{"id":"88397be1b4a59c9cff5ece8e887fbee21da653f5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 03:27:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"7b83fc1ea6ffc41d9b9dfdfe53cf6feb47efb200","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 03:29:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f023b7fefb8642d0962af91afd271c0c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/427a9783448e458fa41a2f03d4eefd81 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6d352dd3a665ce8bafe5a7d97e4da0a33d4dd2ff","revisions":{"bdbe0ddb808fdfc8d4627eb0d6aefd5b8fe82cc2":{"kind":"REWORK","_number":1,"created":"2024-02-26 04:02:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/33/910133/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/33/910133/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/910133/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/910133/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/910133/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/33/910133/1"}}},"commit":{"parents":[{"commit":"bbb4713464eee61add6ce32427621e36fa4086e9","subject":"Merge \"Declare we use threads and processes for coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bbb4713464eee61add6ce32427621e36fa4086e9"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-26 04:01:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-26 04:01:38.000000000","tz":540},"subject":"Cap pylint","message":"Cap pylint\n\npylint 3.1.0 introduced the new check (use-yield-from) and this detects\na few failures in current code. This caps pylint so that we can also\napply the same method for stable branches. The failures will be fixed\nin master later.\n\nChange-Id: Ia5396895b27e4b28a7d9d8d85a85a8449c21d493\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bdbe0ddb808fdfc8d4627eb0d6aefd5b8fe82cc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bdbe0ddb808fdfc8d4627eb0d6aefd5b8fe82cc2"}]},"branch":"refs/heads/master"},"719132c523bdf6e5fde08e0f12f513e2c5d893d0":{"kind":"REWORK","_number":2,"created":"2024-02-26 04:16:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/33/910133/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/33/910133/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/910133/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/910133/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/910133/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/33/910133/2"}}},"commit":{"parents":[{"commit":"bbb4713464eee61add6ce32427621e36fa4086e9","subject":"Merge \"Declare we use threads and processes for coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bbb4713464eee61add6ce32427621e36fa4086e9"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-26 04:01:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-26 04:16:14.000000000","tz":540},"subject":"Fix pylint error","message":"Fix pylint error\n\npylint 3.1.0 introduced the new check (use-yield-from) and this detects\na few failures in current code. This fixes these failures.\n\nChange-Id: Ia5396895b27e4b28a7d9d8d85a85a8449c21d493\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/719132c523bdf6e5fde08e0f12f513e2c5d893d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/719132c523bdf6e5fde08e0f12f513e2c5d893d0"}]},"branch":"refs/heads/master"},"6d352dd3a665ce8bafe5a7d97e4da0a33d4dd2ff":{"kind":"REWORK","_number":3,"created":"2024-02-26 07:58:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/33/910133/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/33/910133/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/910133/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/910133/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/910133/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/33/910133/3"}}},"commit":{"parents":[{"commit":"bbb4713464eee61add6ce32427621e36fa4086e9","subject":"Merge \"Declare we use threads and processes for coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bbb4713464eee61add6ce32427621e36fa4086e9"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-26 04:01:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-02-26 07:58:39.000000000","tz":540},"subject":"Fix pylint error","message":"Fix pylint error\n\npylint 3.1.0 introduced the new check (use-yield-from) and this detects\na few failures in current code. This fixes these failures.\n\nChange-Id: Ia5396895b27e4b28a7d9d8d85a85a8449c21d493\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6d352dd3a665ce8bafe5a7d97e4da0a33d4dd2ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6d352dd3a665ce8bafe5a7d97e4da0a33d4dd2ff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
