)]}'
{"id":"openstack%2Foctavia~910372","triplet_id":"openstack%2Foctavia~master~I5fe49aa5bab2c9093edf22fc6b1b90d5b243b04a","project":"openstack/octavia","branch":"master","attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-02-27 20:21:34.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I5fe49aa5bab2c9093edf22fc6b1b90d5b243b04a","subject":"Remove unused exception logger class","status":"NEW","created":"2024-02-27 17:12:57.000000000","updated":"2024-09-02 16:11:55.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3c4e6d05a5e77ca494ae29d6b674ca66ab9f826a","_number":910372,"virtual_id_number":910372,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-08-23 01:51:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"all":[{"value":1,"date":"2024-09-02 16:11:55.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-27 20:21:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-02 16:11:55.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"}],"messages":[{"id":"c09fe613f5452269b5cd80b12d2150ddb4c4a81c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-02-27 17:12:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e3189bbf84d8118d1413cda39d0ffecd4531dabb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 20:21:34.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3fd92c339c7945c5850f4e54beaf22ef\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/29bf3d27bb3343569d7b83b21074f358 : SUCCESS in 10m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fcb14a484a24053b5228c0a999fe491 : FAILURE in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/93c85b390a954584904120de65001cce : SUCCESS in 4m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/814db351372e480984a6a3966dbce949 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a971705706f49278158c674884a0973 : SUCCESS in 11m 56s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/3e2c6d2209294c939f46564122b5241c : SUCCESS in 5m 20s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/d7e19e1011164e2aa73b325690340e85 : SUCCESS in 9m 10s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/d37e942d12a0453583ae33864d9d1bde : SUCCESS in 11m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2061cc81dcb3460898e86ee842ba6f70 : SUCCESS in 5m 52s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b5af02c53e324af6b14fa6f1435143c5 : SUCCESS in 8m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7a1fcd51d8444c929052f127f7df717f : SUCCESS in 7m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2fa8208476564a30ad0cff2adb3334fd : SUCCESS in 7m 04s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/77453720584041c39f4d5dba4432983f : SUCCESS in 1h 38m 10s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/f485f8fc1c2342a09a636135a43bc199 : SUCCESS in 1h 36m 39s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ca8b9f54b02642a8879174ee91be14dd : SUCCESS in 59m 56s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/8a41c67f71cf468eb14afda8ac9f543d : SUCCESS in 1h 41m 05s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c67076f190604d19864fc7e6f9ad649b : SUCCESS in 1h 01m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/74d9e9221ea944259762ee7533222877 : SUCCESS in 34m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/75c8c2c023f143a595aa2da992670bca : SUCCESS in 53m 43s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/a3019c477a434ef2894809f6691f4990 : SUCCESS in 53m 54s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/e3adfd9d019b4edfaf6307e4f01b7280 : SUCCESS in 1h 04m 03s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/22d064b09b174e099fea7857fd266d64 : SUCCESS in 47m 39s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9036d0d92ace463c833213041aae4be4 : TIMED_OUT in 2h 33m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/15089da081bb4552abb3fb40ad5446cb : TIMED_OUT in 3h 03m 33s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b50fc0cefc3d4afbbe5595b6f3f7a904 : FAILURE in 30m 32s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/a9f9fb33cab647fcbb1e5c434f19c50c : TIMED_OUT in 2h 34m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c3f9f2b5dfeabfb5234403348f52892969744d9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-08-13 15:48:54.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a7393beaa282d3ed1ae541dacad6b170535b8b15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-13 18:36:44.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f5acb0e22c524945b3cfb1fe52d74bb7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e805a698dae4e509c8dfd13998d3fd6 : SUCCESS in 10m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d277c021bdc4ff28869cb04a796837e : FAILURE in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/43795639fdda403aafd35e1052a43cb1 : SUCCESS in 3m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/420c621121bc485198c15591d7fc643b : SUCCESS in 5m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9bd87d4f7d0e4045b1fd17ceee248615 : SUCCESS in 7m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1df8f3ddd4548d4a34a1acfd0553148 : SUCCESS in 10m 50s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/6b954cbc7e8b4f43a54b66703cffa20a : SUCCESS in 4m 04s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/4d1dfce52def44b9933aa295efa045a4 : SUCCESS in 7m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f3f9a9c1f8ee465d941f0d095d5cb654 : SUCCESS in 5m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/02162858da274724ae75ba512faa0d4c : SUCCESS in 9m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/74c12205ead344019fd00dfd888ced34 : SUCCESS in 9m 55s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f932e3403ebf4f25bacdd898a3d84d9f : SUCCESS in 8m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fcde1d42cdc346c4b487995bb5fb44a9 : SUCCESS in 1h 24m 02s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/045c172594584f5d9aa49eaedea13200 : SUCCESS in 1h 42m 15s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/217ef526b8c54238894548c33e725f0e : SUCCESS in 54m 33s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/46f2490e8cc84fe5bdcf3e4d7a9376ce : SUCCESS in 1h 46m 42s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b431e9e6926c45b7a8038ea7be7f70c6 : SUCCESS in 1h 10m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/745d22c0626343f5a8d5f4dd4d524827 : SUCCESS in 45m 05s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/af5a8cf359d0415f9adb68ed3a27dbe0 : SUCCESS in 43m 15s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/b8aaa530bce745e789d1764cad2860e5 : SUCCESS in 1h 01m 25s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d32d4bcfc5c0469a9c5f20670e85980a : SUCCESS in 42m 40s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/25247372f4344aed87c155fde582abe4 : SUCCESS in 2h 30m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c0ede04764984f18bf7e02de4826c5c2 : SUCCESS in 2h 45m 33s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/baece9f37afb4bd0be87f6ff365a0d3f : TIMED_OUT in 2h 33m 43s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/81a1e18adead48eaa41f86d10715a695 : TIMED_OUT in 2h 33m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ae8ded98d76bee23df3e98a845a2010413b7e19a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-08-14 15:19:19.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"ddf5ec5efc576e52f112affc298735efcbe51abd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-14 18:03:09.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0329efa4befa427abd68870b5684a8f2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f65ba707d99e4f50b680d09742c96fa3 : SUCCESS in 14m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07719fbf1592409f915d1923e7c16409 : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f92dda02dade47e8a4d4c43374d3781f : SUCCESS in 4m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9b7842153f054479a53c082503c922f8 : SUCCESS in 4m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a1b38f9c8e974c5da03f3187eec55d1f : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d90668b5340e423ab27da82fecca56c4 : SUCCESS in 12m 00s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/814f6dc0618243d0bd1a8a75cf1ea18d : SUCCESS in 4m 20s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/c9ca254b567640cb84cc2615d41867b2 : FAILURE in 8m 11s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/28395314a826457db7efbbb84eec2de3 : SUCCESS in 4m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/843c10370bec4e79a6805395e3a5d851 : SUCCESS in 8m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b47cd8a6374641d9af13f967a766b5a2 : SUCCESS in 9m 08s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/639d6c9947c348508836d89944d1fe30 : SUCCESS in 7m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9a71e6a6282b46229b3ee45b3d9e607f : SUCCESS in 1h 27m 17s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/b4bf5ae1bbbd41d99a4a05ca45de1be3 : SUCCESS in 2h 29m 56s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/5b77514d0a0b4f2f82669f5d59dcb6d3 : SUCCESS in 1h 31m 41s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/eea5bcd5c6984157924f7bf91e038543 : SUCCESS in 2h 08m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/6821927bc78d430cbde0afdb618e30d1 : SUCCESS in 1h 40m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a2e85a36741a45119f9df391ee82626b : SUCCESS in 42m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1befaa67435d4526b2d888616c708bc5 : SUCCESS in 1h 00m 45s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/b549d2224e0342cbbdfc35dd9acf1580 : SUCCESS in 1h 00m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c2a2b4b7d46b4d3eaa425f3f15a27f5e : SUCCESS in 33m 49s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5637889653624c0e827c7dce1d7ee47e : SUCCESS in 2h 15m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/676e31911f0d4135a1df5ef759b14e66 : SUCCESS in 2h 42m 32s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1658da4e43ea4daf9bf47ed0db389193 : TIMED_OUT in 2h 33m 54s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/e28216e044fc4c8a8346f8cf9776847d : TIMED_OUT in 2h 34m 15s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7544b19724d7ba6f4c4b46d2a20e498b484b5a33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-08-22 22:46:51.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"8ae98e56d045801ee91e70d938d8d96f255ce96e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 01:51:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd37e83df12e4a7e870ec11e79d2cc47\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9d00d25cc4b4dd5ab41945518d3014e : SUCCESS in 11m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbc3429de2cd4fba80c70f137180d6a9 : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9393d59eac9e45e4aaf16dc9111b96ee : SUCCESS in 6m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/21b7ba9fec6b4512b5f77aa1784d80f6 : SUCCESS in 6m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2b9a0c077cf84a16b0eb9471ded8e62c : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/307876c30cbe4949a1c55546a7a57a1a : SUCCESS in 10m 19s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/51ccd98fd19e4dbcb48a111c7b9a4a29 : SUCCESS in 6m 19s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/28581a9bd35f41acb64cf3b15bbb5b02 : SUCCESS in 7m 35s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3d2bcb0b311347428ef36d10b090a618 : SUCCESS in 4m 05s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/de169cf12207490fba6e75920b7ca9a7 : SUCCESS in 7m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/11a7ccf4fbc84bac8bbbe5ff0d5cae50 : SUCCESS in 6m 32s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f08e24ff81634cf4aea75381f833abde : SUCCESS in 5m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/42fa85648314486ca129cbf8ba7b4bc1 : SUCCESS in 59m 08s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/20f85ff1291644f4b33a3f789eeea661 : SUCCESS in 1h 44m 45s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/b568ca5fecbe4b4a9077547260a02683 : SUCCESS in 55m 27s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/4f366678bdc0451f894bd59fc24b4033 : SUCCESS in 1h 52m 02s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/896165b7f376454c8c70b8baf5e2f873 : SUCCESS in 1h 16m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/367008742700462187df1da1e19f292a : SUCCESS in 51m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6db5239e58a04a028e2870cf27b226ac : SUCCESS in 1h 02m 27s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/e9793c0d1ef540c0bc7e6ddb68a91f78 : SUCCESS in 1h 07m 19s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/06a62fbf0e6749caa1fc6a668e213e1f : SUCCESS in 33m 16s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0ba0c0eb80234715a7ffe613b93aa6f5 : SUCCESS in 2h 14m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/33e5e25d2425460890bcd7ed6bc99487 : TIMED_OUT in 3h 03m 00s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e8f68be5bdb84d53a3f3a59f2ff373a6 : SUCCESS in 2h 26m 17s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/ac7d8f751adc43b5a3df2d3ebfa3c5f1 : TIMED_OUT in 2h 33m 58s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3c4e6d05a5e77ca494ae29d6b674ca66ab9f826a","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2024-09-02 16:11:55.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"50960b6bdeac2a811dd0c3ec9ddb5ea96a38468e","revisions":{"0e22575cab940665f4c8efcce46eb01c99ac1dd0":{"kind":"REWORK","_number":1,"created":"2024-02-27 17:12:57.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/72/910372/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/72/910372/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/910372/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/910372/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/910372/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/72/910372/1"}}},"commit":{"parents":[{"commit":"3f4eddc5cc5c431a4d7884a60562d3ec20b8be9a","subject":"Merge \"Add h2 section to Octavia cookbook\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f4eddc5cc5c431a4d7884a60562d3ec20b8be9a"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-02-27 17:11:56.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-02-27 17:12:56.000000000","tz":0},"subject":"Remove unused exception logger class","message":"Remove unused exception logger class\n\nThere is an unused and untested exception_logger class in common/utils.py.\nThis patch removes this unused and untested code.\n\nChange-Id: I5fe49aa5bab2c9093edf22fc6b1b90d5b243b04a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0e22575cab940665f4c8efcce46eb01c99ac1dd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0e22575cab940665f4c8efcce46eb01c99ac1dd0"}]},"branch":"refs/heads/master"},"bc9715b7acefe6c49c2e4c7c96f49d8f495dc2f0":{"kind":"REWORK","_number":2,"created":"2024-08-13 15:48:54.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/72/910372/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/72/910372/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/910372/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/910372/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/910372/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/72/910372/2"}}},"commit":{"parents":[{"commit":"1b09ba8f378a61bc1d913be26fd7e0a28e950c27","subject":"Add spec for custom SGs on VIP ports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1b09ba8f378a61bc1d913be26fd7e0a28e950c27"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-02-27 17:11:56.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-08-13 15:48:53.000000000","tz":0},"subject":"Remove unused exception logger class","message":"Remove unused exception logger class\n\nThere is an unused and untested exception_logger class in common/utils.py.\nThis patch removes this unused and untested code.\n\nChange-Id: I5fe49aa5bab2c9093edf22fc6b1b90d5b243b04a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bc9715b7acefe6c49c2e4c7c96f49d8f495dc2f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bc9715b7acefe6c49c2e4c7c96f49d8f495dc2f0"}]},"branch":"refs/heads/master"},"c2f8af9d70de6bbaec0b5f73e22760cae244925d":{"kind":"REWORK","_number":3,"created":"2024-08-14 15:19:19.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/72/910372/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/72/910372/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/910372/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/910372/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/910372/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/72/910372/3"}}},"commit":{"parents":[{"commit":"1b09ba8f378a61bc1d913be26fd7e0a28e950c27","subject":"Add spec for custom SGs on VIP ports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1b09ba8f378a61bc1d913be26fd7e0a28e950c27"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-02-27 17:11:56.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-08-14 15:19:19.000000000","tz":0},"subject":"Remove unused exception logger class","message":"Remove unused exception logger class\n\nThere is an unused and untested exception_logger class in common/utils.py.\nThis patch removes this unused and untested code.\n\nChange-Id: I5fe49aa5bab2c9093edf22fc6b1b90d5b243b04a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c2f8af9d70de6bbaec0b5f73e22760cae244925d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c2f8af9d70de6bbaec0b5f73e22760cae244925d"}]},"branch":"refs/heads/master"},"50960b6bdeac2a811dd0c3ec9ddb5ea96a38468e":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-08-22 22:46:51.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/72/910372/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/72/910372/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/910372/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/910372/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/910372/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/72/910372/4"}}},"commit":{"parents":[{"commit":"b71e752c0fff2939e4347a22e55c7cbb3e9c7f9a","subject":"Merge \"Fix test_update_load_balancer_status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b71e752c0fff2939e4347a22e55c7cbb3e9c7f9a"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-02-27 17:11:56.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-08-22 22:46:51.000000000","tz":0},"subject":"Remove unused exception logger class","message":"Remove unused exception logger class\n\nThere is an unused and untested exception_logger class in common/utils.py.\nThis patch removes this unused and untested code.\n\nChange-Id: I5fe49aa5bab2c9093edf22fc6b1b90d5b243b04a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/50960b6bdeac2a811dd0c3ec9ddb5ea96a38468e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/50960b6bdeac2a811dd0c3ec9ddb5ea96a38468e"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
