)]}'
{"id":"openstack%2Foctavia~910464","triplet_id":"openstack%2Foctavia~master~I9bb45a1aca6b7b18644752a3dccc3ebfb7c106ef","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"6577":{"account":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"last_update":"2024-04-13 12:45:53.000000000","reason":"Change was submitted"},"10273":{"account":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"last_update":"2024-04-13 12:45:53.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-04-13 12:45:53.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-04-11 07:47:53.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2024-04-12 16:02:15.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"I9bb45a1aca6b7b18644752a3dccc3ebfb7c106ef","subject":"Fix negative or 0 limit parameter in pagination","status":"MERGED","created":"2024-02-28 04:38:31.000000000","updated":"2024-04-13 12:47:01.000000000","submitted":"2024-04-13 12:45:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"910464","meta_rev_id":"9e6c52556119fe87210cb95161aba4f04c881128","_number":910464,"virtual_id_number":910464,"owner":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-13 12:45:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-04-12 19:23:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2024-04-13 10:58:02.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2024-04-13 10:58:02.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-04-12 19:23:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"reviewers":{"REVIEWER":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-28 04:40:14.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-02-28 04:40:14.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2024-02-28 07:29:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-02 06:35:42.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2024-04-02 06:35:42.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"fb68d33c0ddfa418cbe7b0c5ed480f14ea120ef8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-02-28 04:38:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7c91d57c5227f6a5d25503a1c99f077eb1e53e6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 07:29:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/33cd435241974ddab9fb705230c1ae8a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c2acd8597b740e6855ea645be9930d0 : SUCCESS in 12m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4aa1dc983c4b4c04b94be7267d419a0d : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89af6c70120b4759b6c39c47e3f4fd09 : SUCCESS in 5m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e6d964d0f0fd4090bbb794159a9ae3d5 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c88c65ff4014414a675bafec131c2b8 : SUCCESS in 11m 15s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/69efef9113ad42cf8900b42d67ef6e98 : SUCCESS in 5m 04s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/5e20b47a49d940a597e57bf0c281546d : SUCCESS in 6m 25s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/8d1bfeef229049caada59e0066bc524b : SUCCESS in 11m 50s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0f1ca03932a54c4f968b5720538df219 : SUCCESS in 4m 43s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b60930bb481a452b89969bc728a5e4cd : SUCCESS in 8m 05s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b2f0fe3b753b4334a2171c14ef848404 : SUCCESS in 7m 33s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8c42440ac2424028b5ef026c00956907 : SUCCESS in 6m 53s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1c56433322934faf984f2ce713b5ef90 : SUCCESS in 1h 32m 41s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/7efdbcac5d15405da81549b200e8f234 : SUCCESS in 2h 27m 28s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/41e6377b52ef45f28e84e6c3c227a9aa : SUCCESS in 1h 20m 20s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/2e1be2e82265419ea5789fce003d6f76 : SUCCESS in 2h 19m 50s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c7317e332a7742bd8a3ec738f9ff7f4e : FAILURE in 1h 37m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a397c391fe3f45ad8fdd8d9284b86d1b : SUCCESS in 55m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5db14f2fac2a4b35b2d0f583d26190c2 : SUCCESS in 1h 02m 39s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/d434b711f7264a8f97a1b8e84fb820bc : SUCCESS in 1h 05m 48s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/47b2d4947b884c138cbba39fc06b3806 : SUCCESS in 1h 00m 39s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e8ad88656ac44585a0753962e28f82f2 : SUCCESS in 44m 20s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/408b1d7ef9184bf692a1884418d48a6d : SUCCESS in 2h 14m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ac946bb192c44536bcc70f2c8cd1f5d4 : SUCCESS in 2h 48m 46s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9185738547ac4ae1b18986ea61147dee : FAILURE in 21m 53s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/2c42fb14645849239930dfe1d834ae4a : SUCCESS in 2h 29m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1e2c6e92e7611b1441c0ca8e0cb479ec59876a76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-03-14 17:28:23.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"88e9e5365d02d8a6f122b7e7449e3be1308ec39f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 19:56:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5e78fe7e515490d81b53c2c6be56e53\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9748131140f845ceb605204c60e717ca : SUCCESS in 11m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb0e7761b5e9434baa7763b32e1d0bfa : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1290b23b68ee436abd9f781d9d3ad964 : SUCCESS in 4m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/df303b400c244c12bb9bb9748d7123bd : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6c6f26b5e424a2fbcb09b07693ca104 : SUCCESS in 11m 34s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/d5cdfa31bca549bbb755eb8233c125ff : SUCCESS in 4m 58s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/c7e3ae616fe94890bd816ddc02b6621f : SUCCESS in 9m 06s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/b493fe9fe2e84cf8a284342c035e8016 : SUCCESS in 10m 02s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a5cc747f17f74f118465232b9c8da7c2 : SUCCESS in 4m 33s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/eb14169fee604966953335a23cac249d : SUCCESS in 8m 02s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/df3370c14f474086be20662153cae5c2 : SUCCESS in 6m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/644ff260fc064e83b0a682fe25dc8e5a : SUCCESS in 6m 57s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b219d52e77b04fde82d0821892103a9b : SUCCESS in 1h 35m 50s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/338895f67f554b908607a81d550ac63e : SUCCESS in 1h 55m 01s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/c014fae71cfb47bc8830215530ae79ca : SUCCESS in 1h 00m 41s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/18470d93b2d14424ac49da602aa870f1 : SUCCESS in 1h 45m 02s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/63fe33df16cf4cd484c2c2bf05cbfaac : SUCCESS in 1h 22m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4a4e6909a9434d76be83796c70aec161 : SUCCESS in 54m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1afc26de18e44c5c97a6d9f09757e8ba : SUCCESS in 1h 04m 11s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/de146d581e9b4021a50eed623a50f7c6 : SUCCESS in 1h 02m 05s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/2cebb2b8d9c44efba8ac5db808929eb7 : SUCCESS in 1h 02m 36s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cf03be031dfd473a85141120a27f813c : SUCCESS in 33m 01s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5362e4eec18f480bb691c89b9e1d96ff : SUCCESS in 2h 25m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/89f384c52e204365b1765274bd015275 : SUCCESS in 2h 25m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/43d5a8b95f284f69875c0b5910022b61 : FAILURE in 31m 37s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/66b98c3414cb4105b2cad6fb8a755812 : SUCCESS in 2h 22m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"201077024a7489d1ea4fec48101de99cbc36a705","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-04-04 05:56:27.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"20fcae516c358bebcc812a03d1847199c7e67cac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-04 08:43:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ad47f7c79814c9d903b3e0fc77f0c55\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/644b0a85c8e9434e92361a3ecb90f2a2 : SUCCESS in 8m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5af208273eac4d5199e56e1f042e4f2a : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8fe3baa2fee94c5283bb5bf2a8138c12 : SUCCESS in 5m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d30cdfa6b79749e0b3e4807e69764acf : SUCCESS in 5m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f0f546a5726f44aa89e2630c56ebe85f : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fe8ed30e900485a878fb2d2b9467827 : SUCCESS in 8m 51s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/db633c566c1141e6a7ae607f0d83963f : SUCCESS in 4m 56s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/d54f15333426460d8dd8766e25c06ea9 : SUCCESS in 9m 06s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/0ec47f9aff0f48cb8d713d9f800f8c72 : SUCCESS in 8m 41s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/bce8261c6a3048599550b7ad83b5eb38 : SUCCESS in 4m 18s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b1c0eaae21774d69b55720f48e1458a4 : SUCCESS in 10m 02s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/12def670011f4da8a7480f9c807eab9d : SUCCESS in 6m 45s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3070883c023c4119bda5f5e78a80a9f4 : SUCCESS in 8m 13s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/41a869d95b7a44bbb12cf8cd20657985 : SUCCESS in 1h 35m 29s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/63a36cb1874841d896b395e34643a52e : SUCCESS in 2h 09m 12s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/330fced90cb240b8b88bbe5c950428ce : SUCCESS in 1h 34m 35s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c438b56579704749a7bf51fd17ca3d7d : SUCCESS in 2h 06m 00s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b11954e93ae04948ac6ce2cb8f6ece6e : SUCCESS in 1h 11m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cf4f5973eac94966bc4cc597d470a2c6 : SUCCESS in 47m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3c9fb21c9fc246cc8f41d4d55d56e79a : SUCCESS in 51m 41s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/7b807128c6fc47b9a56c501d0267ceeb : SUCCESS in 1h 10m 18s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ad9be7c846d741ba917b769a2346074b : SUCCESS in 43m 02s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/185ebfbe77304fc0bab43e9057350e2e : SUCCESS in 2h 30m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9c13042556eb43aeb7949d08f17b3cc9 : SUCCESS in 2h 45m 48s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/77bdfd10ed4f40cea1ab7ca6b8ebcd2e : FAILURE in 22m 29s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/d15403b5c5ed4a378c7314bf0bf689a7 : TIMED_OUT in 2h 33m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1181b13cb4632f902b87a7e48a5757e38097ec39","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-04-10 12:29:26.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"47c1b4a55022cbb4645a2ea619e85478b22255a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-04-11 07:38:44.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"710c1c76f8364f746d4779f07ee2cef55bf82622","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-04-11 07:47:53.000000000","message":"Patch Set 4: Backport-Candidate+2 Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"0c872143211f3998839bf3aa7fc54f518ea43a71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-11 10:49:48.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ddab3c9aa7f84a5ea0cc70d13d2f7b45\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be7c9dccb0ca4ae8bb1baee702075079 : SUCCESS in 11m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff859f9253b745f287a05953ad6f2549 : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ed13e10eeb03465480d0fb5760b1e322 : SUCCESS in 4m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/669b17fa2077446eb6f43e6e8b65a4e7 : SUCCESS in 4m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/105537c3c1274dfcb4ef4941f0d29008 : SUCCESS in 9m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6eb5c1e75dcd44348c8e5e9879b0e28b : SUCCESS in 12m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8183c44002b24b159d61d70867bd4747 : SUCCESS in 5m 48s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/7cff599e344349c7bee3704c901fe5e3 : SUCCESS in 5m 09s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/495f74f2f950404cba96e93bde998d78 : SUCCESS in 10m 29s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/7cd1fb0309dc4c1f9d6169c3a834b066 : SUCCESS in 10m 04s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/10c38cb954e1430bbd29a123e6d73d19 : SUCCESS in 4m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2d91cfd878eb459e817ca89c2617dc0d : SUCCESS in 8m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3c100880080447a689db24d600aaf102 : SUCCESS in 8m 50s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2f18b7e256154db5a1f7b8bdd28e59e0 : SUCCESS in 7m 00s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6f3f263b7e54488b819426b4a59512a1 : SUCCESS in 1h 14m 36s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/2cdf55d643e84522947153517263420c : SUCCESS in 2h 08m 19s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e7e074634b474943a4b307afb4cfa038 : SUCCESS in 1h 19m 34s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/5751fb8b479c4302902cd1bd9c8f733d : SUCCESS in 2h 18m 54s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/64c39c0a6d744635b90947628430fe2e : SUCCESS in 1h 34m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/352d36dec4404df5bd65e2151e540ec8 : SUCCESS in 54m 49s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/af97c2c73b7c42628877f06949fc6ba4 : SUCCESS in 58m 53s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/e1be84d007964ac582153cd35f871659 : SUCCESS in 59m 48s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b159dda25c724f68835b353697aa11e4 : SUCCESS in 46m 51s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5293036c39ba4bc6863d84c816a97279 : TIMED_OUT in 2h 33m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/86a8135f83aa434bb2bf41818d5f9960 : TIMED_OUT in 3h 04m 10s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/556bfd4b564a4882935dd4187e036dcc : FAILURE in 37m 26s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/7b27654a1fe04a0fbcb4acccca1fac06 : TIMED_OUT in 2h 34m 27s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e255745e8629ae3d6dae925472f764ee29a326ff","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-12 16:02:15.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e6f4429ad6c8950024d928dc5702952971aebd0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-04-12 19:23:28.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nCopied Votes:\n* Backport-Candidate+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE**\")\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"787507664f3c4e38143b2f38533c60a3842786ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-12 22:15:22.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9c262d83c6946278c53b12fee9a4dea\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77eefc3af445407eb48c759a78d63435 : SUCCESS in 8m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5a6eac754a74376a97d355150c0ca2d : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/16fff784c4264058ad899f217289e42f : SUCCESS in 4m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/97848d93877948c99e0f76297e794473 : SUCCESS in 4m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fdcc674457ea4509bd0e9af192e42cd3 : SUCCESS in 7m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/915ce5f7f36a4a168705167a44f073df : SUCCESS in 12m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cbe0c0c6c6054246824842fe77248e7d : SUCCESS in 6m 34s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/c97b723615e8469cb15fc6baf3cc28ac : SUCCESS in 5m 49s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/8822498343cf412aa60dd47d2410207d : SUCCESS in 8m 27s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/b0a7bfaa222745a286e61a6c111e4e87 : SUCCESS in 9m 00s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3d7fac3931284c42929ec795334ef9b7 : SUCCESS in 4m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ca240f597e0f497b82c9eaa57ae3e101 : SUCCESS in 7m 26s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/eb728f95364b4043b9b8134d20ff4c8b : SUCCESS in 8m 07s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/60c22feb0c5345fe83e3d2cdfed246a7 : SUCCESS in 7m 07s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/addb63255c7d4a16b1fd33f3d14a3b10 : SUCCESS in 1h 08m 29s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/00787315ca7e42a7b36b338d9a6eaba8 : SUCCESS in 2h 08m 11s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/f5539df463684400b2473fdece499382 : SUCCESS in 1h 11m 10s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/aa2527c83d6245109278f91179f1ae49 : SUCCESS in 2h 10m 23s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b917b5cf357f4b0d8b7cb96c3addb42a : SUCCESS in 1h 16m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ac38fb06c6f14ff09d3fc1b023255fab : SUCCESS in 42m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/14ec134a54df4721adca0b9cc94c9034 : SUCCESS in 42m 52s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ecfcce9d8cfc40e7ad2472246efee56b : SUCCESS in 1h 10m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/602ce8fd2c02469fb27dcfc5afa81690 : SUCCESS in 37m 00s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/643a0ba44d414af0a24e2f2b59f91885 : SUCCESS in 2h 11m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/dbd83441ad264150a3db6daad2864964 : SUCCESS in 2h 43m 26s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/fa27bd73f918489082e533227248b54d : FAILURE in 38m 05s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/8b00fe29d899476d9244a1bec6748784 : TIMED_OUT in 2h 34m 32s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"7ba4d7a8eeee74b7cc4e66f34db2c9dafdde0c4e","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-13 10:58:02.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"210511965034b35a725a21e36dc065a20149ce3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-13 10:58:38.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"8a1458e91535ce931cf05f6dfed692e1c03e9b55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-13 12:45:53.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28509c3e67134841b23f06746551ea76\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f61ecf365cb4747a93c9578d09cfa11 : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/250f25c8995644cfb1d7a0b1f7be7e90 : SUCCESS in 4m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e1f36544f8434c59ad0a3cfe09f9dc76 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41291167fbfc4631aa836f7b9596a4f7 : SUCCESS in 8m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/12193b34a8154812be2251e2ae5a72e3 : SUCCESS in 4m 24s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1713ebdd373340d983321d4081fbdec6 : SUCCESS in 4m 08s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e26b784f47f54abe9bad05d48b75aa8a : SUCCESS in 8m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e2052b37c13243a9a7cea7afc0f5099a : SUCCESS in 1h 33m 16s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/6ee32a14d1e1447e81c482ac64dede75 : SUCCESS in 1h 44m 56s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/01fbc4d315e34ff6965ebbbeb185f3a3 : SUCCESS in 1h 03m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1b9991b2e612425bad272579748b912d : SUCCESS in 43m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f853854ddbfe4f12bc400ea950b7703d : SUCCESS in 46m 22s","accounts_in_message":[],"_revision_number":5},{"id":"9bfa75c181d8e02f302528477cde3880bd026b6e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-13 12:45:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"9e6c52556119fe87210cb95161aba4f04c881128","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-13 12:47:01.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc273f221009408fa62123f6cf945f8f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43774170dcb7424b8030c14937b86b3a : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/96ce5b7e1f7f43dbadc212a0dcf7d146 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"85cfb6c2ae974d110712b5ea673cdafcfecb9e69","revisions":{"91371befd3f5b2bfdb8f02638eb8c171eb00c6dd":{"kind":"REWORK","_number":1,"created":"2024-02-28 04:38:31.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/64/910464/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/64/910464/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/910464/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/910464/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/910464/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/64/910464/1"}}},"commit":{"parents":[{"commit":"3f4eddc5cc5c431a4d7884a60562d3ec20b8be9a","subject":"Merge \"Add h2 section to Octavia cookbook\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f4eddc5cc5c431a4d7884a60562d3ec20b8be9a"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2024-02-28 04:38:14.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2024-02-28 04:38:14.000000000","tz":240},"subject":"Fix negative limit parameter in pagination","message":"Fix negative limit parameter in pagination\n\nChange-Id: I9bb45a1aca6b7b18644752a3dccc3ebfb7c106ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/91371befd3f5b2bfdb8f02638eb8c171eb00c6dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/91371befd3f5b2bfdb8f02638eb8c171eb00c6dd"}]},"branch":"refs/heads/master"},"e41ecde6a63a1b6a6afc6d31ec51bfa388fc8c96":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-03-14 17:28:23.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/64/910464/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/64/910464/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/910464/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/910464/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/910464/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/64/910464/2"}}},"commit":{"parents":[{"commit":"9f1a6e47d7eb77492bfcc68adb1392f163ce6d26","subject":"Merge \"dib: Remove Ubuntu Forcal support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9f1a6e47d7eb77492bfcc68adb1392f163ce6d26"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2024-02-28 04:38:14.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2024-03-14 17:28:23.000000000","tz":0},"subject":"Fix negative limit parameter in pagination","message":"Fix negative limit parameter in pagination\n\nChange-Id: I9bb45a1aca6b7b18644752a3dccc3ebfb7c106ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e41ecde6a63a1b6a6afc6d31ec51bfa388fc8c96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e41ecde6a63a1b6a6afc6d31ec51bfa388fc8c96"}]},"branch":"refs/heads/master","description":"Rebase"},"4822b2c5c5e7a14d1ea7fcfdbde0af81970b643e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-04-04 05:56:27.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/64/910464/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/64/910464/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/910464/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/910464/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/910464/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/64/910464/3"}}},"commit":{"parents":[{"commit":"fa7cc0178783ad4ab9f3ddce893042d5952d0bd8","subject":"Merge \"Fix fully-populated API with allowed_cidrs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fa7cc0178783ad4ab9f3ddce893042d5952d0bd8"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2024-02-28 04:38:14.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2024-04-04 05:56:27.000000000","tz":0},"subject":"Fix negative limit parameter in pagination","message":"Fix negative limit parameter in pagination\n\nChange-Id: I9bb45a1aca6b7b18644752a3dccc3ebfb7c106ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4822b2c5c5e7a14d1ea7fcfdbde0af81970b643e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4822b2c5c5e7a14d1ea7fcfdbde0af81970b643e"}]},"branch":"refs/heads/master","description":"Rebase"},"6e0ee484d98f796c3b36edfd6384a82076847be0":{"kind":"REWORK","_number":4,"created":"2024-04-11 07:38:44.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/64/910464/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/64/910464/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/910464/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/910464/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/910464/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/64/910464/4"}}},"commit":{"parents":[{"commit":"fa7cc0178783ad4ab9f3ddce893042d5952d0bd8","subject":"Merge \"Fix fully-populated API with allowed_cidrs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fa7cc0178783ad4ab9f3ddce893042d5952d0bd8"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2024-04-11 07:38:05.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2024-04-11 07:38:05.000000000","tz":240},"subject":"Fix negative or 0 limit parameter in pagination","message":"Fix negative or 0 limit parameter in pagination\n\nOctavia replace \"limit\" with None when it is less 1. (for example 0, -1)\nHowever the further code failed to compare None and int values.\nThis patch fixes it by validation, that limit is None.\n\nCo-Authored-By: Roman Goncharov \u003cgadzhet007@gmail.com\u003e\n\nCloses-Bug: #2060917\nChange-Id: I9bb45a1aca6b7b18644752a3dccc3ebfb7c106ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e0ee484d98f796c3b36edfd6384a82076847be0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6e0ee484d98f796c3b36edfd6384a82076847be0"}]},"branch":"refs/heads/master"},"85cfb6c2ae974d110712b5ea673cdafcfecb9e69":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-04-12 19:23:28.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/64/910464/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/64/910464/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/910464/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/910464/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/910464/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/64/910464/5"}}},"commit":{"parents":[{"commit":"f91ff6c5a616261f57a15f612e00ba2b879b827e","subject":"Merge \"pyupgrade changes for Python3.8+ (2)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f91ff6c5a616261f57a15f612e00ba2b879b827e"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2024-04-11 07:38:05.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2024-04-12 19:23:28.000000000","tz":0},"subject":"Fix negative or 0 limit parameter in pagination","message":"Fix negative or 0 limit parameter in pagination\n\nOctavia replace \"limit\" with None when it is less 1. (for example 0, -1)\nHowever the further code failed to compare None and int values.\nThis patch fixes it by validation, that limit is None.\n\nCo-Authored-By: Roman Goncharov \u003cgadzhet007@gmail.com\u003e\n\nCloses-Bug: #2060917\nChange-Id: I9bb45a1aca6b7b18644752a3dccc3ebfb7c106ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/85cfb6c2ae974d110712b5ea673cdafcfecb9e69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/85cfb6c2ae974d110712b5ea673cdafcfecb9e69"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
