)]}'
{"id":"openstack%2Foctavia~912062","triplet_id":"openstack%2Foctavia~stable%2F2023.2~I80a266e500958415a70d462ddfe57e9e03e6ef13","project":"openstack/octavia","branch":"stable/2023.2","topic":"honor-neutron-interface","attention_set":{},"removed_from_attention_set":{"12404":{"account":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"last_update":"2024-05-03 14:13:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I80a266e500958415a70d462ddfe57e9e03e6ef13","subject":"fix: specify endpoint info. for neutron client","status":"MERGED","created":"2024-03-08 06:57:42.000000000","updated":"2024-05-03 14:14:53.000000000","submitted":"2024-05-03 14:13:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"912062-honor-neutron-interface","cherry_pick_of_change":905794,"cherry_pick_of_patch_set":11,"meta_rev_id":"6fbf0fc50dde17d6bc083838691671ac03e2728d","_number":912062,"virtual_id_number":912062,"owner":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-03 14:13:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-04-30 11:39:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2024-05-02 16:23:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2024-05-02 16:23:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-08 06:57:42.000000000","updated_by":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"CC"},{"updated":"2024-03-08 09:20:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-30 11:39:01.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2024-05-02 16:23:41.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"8e11cce44e6bd8e4f35ea185adae936de0cf923b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2024-03-08 06:57:42.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9609a35caf4b984070bd1963fc4e4b2fa8fc1f50","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2024-03-08 06:58:17.000000000","message":"Topic honor-neutron-interface-stable/2023.2 removed","accounts_in_message":[],"_revision_number":1},{"id":"f20229c633d91b9f2d95a99687fe814a79ed5f14","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2024-03-08 06:58:22.000000000","message":"Topic set to honor-neutron-interface","accounts_in_message":[],"_revision_number":1},{"id":"6059ef221fd8d097e32a34a93267afddf1185cf7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 09:20:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f13d850000aa4daab5822084576b0613\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/345bf28020164bcca167ea47c00efc5a : SUCCESS in 10m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cda25d4221094faf93810a09aad8bf0a : FAILURE in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2dbc2022da0d45bf92e160ee0afe2cc0 : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e65d19701b9444e38aa8ac6fcedb11c4 : SUCCESS in 4m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ac29ea94529b4b9ab2121ad4f78d4a2c : SUCCESS in 5m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6332c2bf8d8a4180a7fbb2283a38e601 : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a494a2ff7ab4fe6a444c0f316e432ac : SUCCESS in 10m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a312f77c81ae4ddb89e5f4924c668a9f : SUCCESS in 5m 16s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/1b52ef0c71844a4d94f5ea4c35df215d : SUCCESS in 5m 15s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/27e2ee73f8ec4532847b612166425e3b : SUCCESS in 8m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/86f8e5b9a4344fb39a6442ebc6d1e28f : SUCCESS in 4m 33s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7af97a1f264147fbbdcb0217ba19de97 : SUCCESS in 7m 53s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c5dbd3525bb94adca6c1002b172c5b0b : SUCCESS in 6m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/36926e55f26146fca3175479aa0ab671 : SUCCESS in 6m 50s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5afb62ae31ec41129ced95959d4ca235 : SUCCESS in 1h 59m 15s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/d7dc2a4a0c3446a2b56e75726981a6e3 : SUCCESS in 1h 41m 05s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/80737da857494ba187d4d7b257b15688 : SUCCESS in 1h 03m 28s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c5f5648d36064b31bb068d22c27d3cc9 : SUCCESS in 1h 29m 39s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e39be93806534bbca1a0ac719f1a061d : SUCCESS in 1h 01m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/05d34f05b18644ccae9b3443556284c3 : SUCCESS in 34m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b39c7894d355437db5d17684207b484e : SUCCESS in 47m 00s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/8d123c48534f40edb67092ff01d848dc : SUCCESS in 1h 07m 40s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b8a23641907c4720ac713fb951d2fcbd : SUCCESS in 38m 27s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/281f99e9fa514811a8f169d4e286b6b2 : SUCCESS in 1h 54m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f40b049a490849358065cda62409fee3 : SUCCESS in 2h 20m 43s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b505389bde564796ac10a53bf99584c4 : FAILURE in 20m 06s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/578632a7b9f449c19ba4ec3342dc1c52 : SUCCESS in 1h 55m 55s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/osutils.py\n  Comments left for invalid file octavia/common/tls_utils/cert_parser.py\n  Comments left for invalid file octavia/network/drivers/noop_driver/driver.py","accounts_in_message":[],"_revision_number":1},{"id":"26220b34cab311c9d01fd8bae1c9371917144349","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2024-03-18 07:48:38.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"cc3a06cd92c1ad1bf1bb875cf6e61b9107586e71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-18 11:05:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d9656814ecd946e99896bab632a8af70\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2295703183534253ab20fab8b75ba718 : SUCCESS in 10m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/110b85bf740d4d7d976a8496ad09b1f2 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99d1a2821a0d42ac8b3217a77a65d89c : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0908d453c5cc455aaf2d9f1d08a4b058 : SUCCESS in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/74014f3ae4464f388905e8e44ac55635 : SUCCESS in 5m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c799c32073c346c6b456f4888187f508 : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/624ee054c7904ef7b039086d3688f9c0 : SUCCESS in 10m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b2505242fe64ac49e9cf1656328b4a9 : SUCCESS in 6m 01s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/abde5e2b837e4954b26bc69c35d4189c : SUCCESS in 4m 19s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/8fee86ce13b049c2a24dfa6bcaef2d94 : SUCCESS in 7m 12s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/099bf80a59914b5bbc8b407b11d3fdca : SUCCESS in 5m 38s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c9801872204d4d1a88511a6a91ec2ffa : SUCCESS in 6m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4bd0c9b7e304411a943499e3966fc40f : SUCCESS in 8m 02s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a717f4288a704614b83d38f96d0e3605 : SUCCESS in 6m 58s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3fe791a85b9a4c4f8517e3f246246383 : SUCCESS in 1h 33m 38s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/a993586a5af541d68ba48ad1f205ec87 : FAILURE in 47m 12s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/a0cc055863664dd09625b8b91e794593 : SUCCESS in 1h 01m 44s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a69940b5c0df4e14a35f5606ede48bb4 : SUCCESS in 1h 57m 13s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/7c1dee4a70b84f2595dd139b97159d50 : SUCCESS in 1h 04m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/67f82997440c438aab17b12c2e9317ed : SUCCESS in 39m 42s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4ffbbf984b1c49feb4a5dfa90c8cc867 : SUCCESS in 43m 07s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/e4f00ddd809b4ab8a032d4cba6368f4c : SUCCESS in 1h 09m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9e5e8f7a389d40108f93a2446167aff6 : SUCCESS in 41m 01s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d9c22493e3de45d2a3054690bd97600c : SUCCESS in 2h 12m 28s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/dd1ffd02bcc24f58bb48ada582c7ac40 : SUCCESS in 2h 14m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/458e6bc9329b4ec3bbe5345f39bf3fa3 : FAILURE in 23m 37s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/ae7c0daf44af441f8b2c65071af45f95 : TIMED_OUT in 2h 34m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b027b89938efdc36aee55cfde819520715ccf8f9","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2024-03-18 17:19:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"88ade7f24c21bdc76d9862e09496247b898c5684","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-18 19:58:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/113431727489475fa3c217838460e4fd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71bd2b87487041aa92a5dcf14b540dbe : SUCCESS in 12m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d42d52b411e44b783b276c2d7ab33b4 : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4094df6415964a79a4f141706e76aab2 : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/197dd0829b5d4216b5636c89ec457803 : SUCCESS in 5m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4dba7fd9359c4014bc6e0130a843d5cb : SUCCESS in 5m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/749cb02f1d4c4e6c802f198eb3ef686f : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98c15727cf32445db89d5e13456e1182 : SUCCESS in 12m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9641b47fef2447289a58d0ef854b40d6 : SUCCESS in 5m 20s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/0761e392a3d24fe087f64132dd75a53f : SUCCESS in 5m 09s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/c71d2a2022fc47db997344d5751d886b : SUCCESS in 7m 57s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/33eca56033234c5a9cc28db805a1b19a : SUCCESS in 4m 36s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/879ad9512afe4fa5afe3969c78a8b235 : SUCCESS in 7m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2a5036d8d17341568779d1851a648b14 : SUCCESS in 7m 07s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/20f54e9d878f4e0bb441b7fbb51c96fd : SUCCESS in 7m 05s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/554b0b44b6e44f5db7f6495c6d693084 : SUCCESS in 1h 25m 47s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/98b515baad544972a9378a64684c8c8b : SUCCESS in 1h 17m 09s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/6025d506bf7348bd9020290a5e430075 : SUCCESS in 57m 22s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/038550cead87424d9fc924547ac6b466 : SUCCESS in 2h 00m 56s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/d534d9509c1b45e99f7121a4546c6c07 : SUCCESS in 1h 05m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e36ed781be3346b2af2057adfa83914b : SUCCESS in 52m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ee9e20cafe58480ea18082c00b9c2d79 : SUCCESS in 52m 38s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/4600e992df68498d873278533c3ae62b : FAILURE in 36m 36s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3aa421c921414e058d35877ac17e0a87 : SUCCESS in 44m 04s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e6345cb38fb44127ad18ee7158e6c584 : SUCCESS in 2h 03m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/966f4d0b28c9424fa95c9d205832a8f6 : SUCCESS in 2h 36m 48s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9cf8d57d5add47309698ae3edcbb7146 : FAILURE in 37m 25s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/42604ceae59b49c79c8057f83129c1b2 : FAILURE in 27m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"905a378b0d2d9226403ced8c0c86a462de82b7f2","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-04-30 11:39:01.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"61bb39ed1192aa5425b859e27f35325b09741129","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-05-02 16:23:41.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6a41e4065004805c18804ffb88c66a52fd6122fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 12:36:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f82818eaab452b17e24d7a01bbd1d3451ff7b6aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 14:13:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7089cafb3cb04cfaa107b7e581a9c760\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6022fe1b1d834cb3961369b0e11e8cee : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3823427ad3c748c2b2b2e1776e5bb6a1 : SUCCESS in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f26e9075bd9e40739096ae9a25bbc369 : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/09f4e65d93074089b53b2355179982b4 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25656f21116d4ef59411b5429adb9992 : SUCCESS in 11m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c521a5410c50433489f5d43074dc2277 : SUCCESS in 4m 39s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9ed95913fad44ddea13bb960bdfe27da : SUCCESS in 4m 20s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8ad2cfba967d4f7d90becbb081065984 : SUCCESS in 8m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/10eff7ddd64e42fbaa552d824b740c64 : SUCCESS in 1h 15m 13s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/1e408340114345e784533429699b44ee : SUCCESS in 1h 18m 47s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/cdb94f03a118422f972795806fcd5205 : SUCCESS in 1h 35m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/675cef6208e94a228283f1168513af81 : SUCCESS in 33m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f2a99f30f3cd4510a773f03adfc40bf4 : SUCCESS in 48m 49s","accounts_in_message":[],"_revision_number":2},{"id":"25bb3fbfc7caf275d5e73acba2034d0587e8ee9e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 14:13:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"6fbf0fc50dde17d6bc083838691671ac03e2728d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 14:14:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fdf9e6c3edf40159db12f004b20df59\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/091251550125429b92f441f418fbc45e : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fddf3a56dc3b49249c81b5237612d6dd : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c664c865b8dfbb5065bbb34c8adb6ed1df5ce28d","revisions":{"57f1f2cd5ce0168d3521d557c62c970980e5aeb9":{"kind":"REWORK","_number":1,"created":"2024-03-08 06:57:42.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/62/912062/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/912062/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/912062/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/912062/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/912062/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/912062/1"}}},"commit":{"parents":[{"commit":"3f31a50f264c0e58691b1be7e90d324c13588b63","subject":"Merge \"Remove publish-openstack-octavia-amphora-image jobs\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f31a50f264c0e58691b1be7e90d324c13588b63"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2024-01-16 22:13:19.000000000","tz":-300},"committer":{"name":"Rico Lin","email":"ricolin@ricolky.com","date":"2024-03-08 06:57:42.000000000","tz":0},"subject":"fix: specify endpoint info. for neutron client","message":"fix: specify endpoint info. for neutron client\n\nCloses bug: #2049551\n\nChange-Id: I80a266e500958415a70d462ddfe57e9e03e6ef13\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/57f1f2cd5ce0168d3521d557c62c970980e5aeb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/57f1f2cd5ce0168d3521d557c62c970980e5aeb9"}]},"branch":"refs/heads/stable/2023.2"},"c664c865b8dfbb5065bbb34c8adb6ed1df5ce28d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-03-18 07:48:38.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/62/912062/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/912062/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/912062/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/912062/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/912062/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/912062/2"}}},"commit":{"parents":[{"commit":"acf4a306480aaa76f9386b14b1df46b765761ada","subject":"fix pep8 use-yield-from rule","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/acf4a306480aaa76f9386b14b1df46b765761ada"}]}],"author":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2024-01-16 22:13:19.000000000","tz":-300},"committer":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2024-03-18 07:47:33.000000000","tz":480},"subject":"fix: specify endpoint info. for neutron client","message":"fix: specify endpoint info. for neutron client\n\nCloses bug: #2049551\n\nChange-Id: I80a266e500958415a70d462ddfe57e9e03e6ef13\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c664c865b8dfbb5065bbb34c8adb6ed1df5ce28d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c664c865b8dfbb5065bbb34c8adb6ed1df5ce28d"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
