)]}'
{"id":"openstack%2Foctavia~913259","triplet_id":"openstack%2Foctavia~stable%2F2024.1~Ie206b0bc62039298942a7028e3f5a6d3a04eb4a7","project":"openstack/octavia","branch":"stable/2024.1","topic":"create-2024.1","hashtags":[],"change_id":"Ie206b0bc62039298942a7028e3f5a6d3a04eb4a7","subject":"Update .gitreview for stable/2024.1","status":"MERGED","created":"2024-03-14 19:54:06.000000000","updated":"2024-03-16 08:16:28.000000000","submitted":"2024-03-16 08:15:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"913259-create-2024.1","meta_rev_id":"ae1783fac969806f1987eb56abf8b5adff0bd7c9","_number":913259,"virtual_id_number":913259,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-16 08:15:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2024-03-16 07:51:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2024-03-16 07:51:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-14 20:12:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-16 07:51:57.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"3bfc147095afd4dbdb19ad052ee6954a9f996385","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2024-03-14 19:54:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b9852ae23deaf0d78958f0388254c0d5d40e8557","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-14 20:12:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a0386ff5aba4e64a95ebc241f1d536f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f76cc154cc644170b490d6c6cb5f9213 : SUCCESS in 11m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dcd8e4c2c4c48f78ab50c2a2b6a4d38 : SUCCESS in 6m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/69b09ef642574827be23dd8a396744e8 : SUCCESS in 4m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/613d53deeed14430b8dd00d7236d85f7 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2dcd94e3b7749f19eaf0262601745f6 : SUCCESS in 10m 56s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/91e63ca525ed44499ce0c8aea61d2f01 : SUCCESS in 5m 02s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/9813f67881904f1797506228527b617a : SUCCESS in 8m 06s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/cd95ca6511f24441a7ab51c2d1015b7e : SUCCESS in 4m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d76a8acbc77e4681a22f27bac4897d63 : SUCCESS in 7m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a40f4df392674e7e9a44677c2938cf59 : SUCCESS in 7m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ebc317e953474eee9fa03b04a791b7fe : SUCCESS in 8m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f851713079f802ef1bafe4b138f6872290370e1c","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-03-16 07:51:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9cc7b752b70c9a6452e8b897c2608daaf9f2b225","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-16 07:52:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f3b770a13323db83d05844065d660734d76f7be6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-16 08:15:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c672daa8ed5d4171afe3d3795e275fc9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05c727d18ce34052b496bc3cd990b1b3 : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/567cc26e311f4dacb75555f65f3559bf : SUCCESS in 4m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/690c6d28c40c42abbdaa791eff7f96e5 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c822aab9010443cb9002cb8193f8aa5c : SUCCESS in 13m 00s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e9f08d133e3d4b30a4b847abb03fdc00 : SUCCESS in 4m 09s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/305cdb26a93d418ca00a14b4a991973f : SUCCESS in 7m 54s","accounts_in_message":[],"_revision_number":1},{"id":"6a246962a59cd8878978e1a610c3755d44841d2a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-16 08:15:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ae1783fac969806f1987eb56abf8b5adff0bd7c9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-16 08:16:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5ae20b2440541f9b4e188a20f725817\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b14e6ecadc04a35922dd3091c451028 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"894023ec95a413b5a615a8f6b8b80c69059fc7ce","revisions":{"894023ec95a413b5a615a8f6b8b80c69059fc7ce":{"kind":"REWORK","_number":1,"created":"2024-03-14 19:54:06.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/59/913259/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/59/913259/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/913259/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/913259/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/59/913259/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/59/913259/1"}}},"commit":{"parents":[{"commit":"9f1a6e47d7eb77492bfcc68adb1392f163ce6d26","subject":"Merge \"dib: Remove Ubuntu Forcal support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9f1a6e47d7eb77492bfcc68adb1392f163ce6d26"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-03-14 19:54:05.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-03-14 19:54:05.000000000","tz":0},"subject":"Update .gitreview for stable/2024.1","message":"Update .gitreview for stable/2024.1\n\nChange-Id: Ie206b0bc62039298942a7028e3f5a6d3a04eb4a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/894023ec95a413b5a615a8f6b8b80c69059fc7ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/894023ec95a413b5a615a8f6b8b80c69059fc7ce"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
