)]}'
{"id":"openstack%2Foctavia~914962","triplet_id":"openstack%2Foctavia~master~I5c78b5c1e8d6abce7df34c5e3e0304a887daf5f9","project":"openstack/octavia","branch":"master","hashtags":[],"change_id":"I5c78b5c1e8d6abce7df34c5e3e0304a887daf5f9","subject":"Add workaround for taskflow REVERT_ALL bug with A/S LBs [1]","status":"NEW","created":"2024-04-03 12:49:32.000000000","updated":"2024-05-02 15:16:59.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"cc0af6ef9835ec244cb3f2427ed9c9554aff120a","_number":914962,"virtual_id_number":914962,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-04-03 17:12:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":-1,"date":"2024-05-02 15:16:59.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-03 17:12:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2c46d710feb5f7acc5ca8076cf86145dd90909d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-04-03 12:49:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"162bc73fcc1e7e685e4a03abe2467d3f3f2b78ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-03 17:12:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb0fbcb4f2bd4ff3a7c901958bae197b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6714914c1db4f99b6c708890d8538cc : SUCCESS in 11m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c7356484e0240d8b2f0720e1ca8c27e : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c3e97be304a04fa99790e6c5115ea72e : SUCCESS in 5m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f8aa07d7dacd4f4990211bc9f8057195 : SUCCESS in 4m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/555730d8359f47a099ade88741e0eb66 : SUCCESS in 7m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adff3d269768432782daea1dd839d388 : SUCCESS in 12m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/922761275fd94cefaf348c0ae4ed07c8 : SUCCESS in 5m 06s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/a86c778488eb47e889248f9636df1cbe : SUCCESS in 4m 54s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/739782f36cbf44ec829da240423586d5 : SUCCESS in 7m 38s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/7ebc1454321e4df7b7b7a362991dd695 : SUCCESS in 9m 44s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/39b49e9edbbf425ca2a336d65fbb0f00 : SUCCESS in 4m 44s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a9a2f55769d0403b8596af782860f1e7 : SUCCESS in 9m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bffbacbd429d4a5f8f9a56d21c9e73cb : SUCCESS in 8m 50s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fc21506a52ea40e8b54b2d8a4eac1bfd : SUCCESS in 7m 15s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b49437362e524d83b7e340779230e033 : SUCCESS in 1h 52m 00s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/0b03227b43124099b76ad64686d4173d : SUCCESS in 1h 41m 21s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/d096ea51d26447dc9ae59abf244db1ed : SUCCESS in 55m 24s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/146388cf4329407d8e4407743601590f : SUCCESS in 1h 47m 21s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/06d61775e47d4129a4f8bc16c0c4d63c : SUCCESS in 1h 11m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cf45d197b92945caa867600972379cf4 : SUCCESS in 40m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2e387b29111743a8a9b8df3e19e1d4e5 : SUCCESS in 41m 24s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/fdff8a265b004825a9fbb73e87e46672 : SUCCESS in 1h 05m 24s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1808eaf0724349edae0c0d4eba45cab5 : SUCCESS in 37m 12s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/30cd74f46b864cd2bffd4544bc2773e3 : SUCCESS in 1h 56m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/dc2228245ebd4214a348e70e4c15bdec : SUCCESS in 2h 27m 05s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4de07b3f3a7d4bcd9b18bc0c4040577e : FAILURE in 24m 14s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/15a2a815dcde4440b501c9339f10ca15 : TIMED_OUT in 2h 33m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cc0af6ef9835ec244cb3f2427ed9c9554aff120a","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-05-02 15:16:59.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"669434d4750e56bc4e63ff1318afd74179b9b313","revisions":{"669434d4750e56bc4e63ff1318afd74179b9b313":{"kind":"REWORK","_number":1,"created":"2024-04-03 12:49:32.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/62/914962/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/914962/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/914962/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/914962/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/914962/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/914962/1"}}},"commit":{"parents":[{"commit":"fa7cc0178783ad4ab9f3ddce893042d5952d0bd8","subject":"Merge \"Fix fully-populated API with allowed_cidrs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fa7cc0178783ad4ab9f3ddce893042d5952d0bd8"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-04-03 11:46:23.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-04-03 12:31:07.000000000","tz":-240},"subject":"Add workaround for taskflow REVERT_ALL bug with A/S LBs [1]","message":"Add workaround for taskflow REVERT_ALL bug with A/S LBs [1]\n\nThe REVERT_ALL feature of taskflow doesn\u0027t work as expected when using\nretries in unordered flows [0]. It means that the revert flow might be\nincorrectly executed, leaving a LB in PENDING_* status (because the\ntask that sets the LB to ERROR is not executed).\nThe retries are not really mandatory in Octavia, some tasks may run for\na few minutes without any issues (we fixed some issues regarding the\nduration of the tasks with the redis backend a few releases ago).\nSo it\u0027s perfectly fine to remove the retries.\n\nThis commit removes retries for AmphoraComputeConnectivityWait\n\n[0] https://bugs.launchpad.net/taskflow/+bug/2043808\n\nCloses-Bug: 2043360\n\nChange-Id: I5c78b5c1e8d6abce7df34c5e3e0304a887daf5f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/669434d4750e56bc4e63ff1318afd74179b9b313"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/669434d4750e56bc4e63ff1318afd74179b9b313"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
