)]}'
{"id":"openstack%2Foctavia~915360","triplet_id":"openstack%2Foctavia~stable%2F2023.2~Ia853d43dc273019c76da09104f31aa7e1b154fec","project":"openstack/octavia","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"6577":{"account":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"last_update":"2024-04-16 09:41:57.000000000","reason":"Change was submitted"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2024-04-16 07:45:43.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"Ia853d43dc273019c76da09104f31aa7e1b154fec","subject":"Handle empty delay on update healthmonitor","status":"MERGED","created":"2024-04-11 19:05:33.000000000","updated":"2024-04-16 09:43:03.000000000","submitted":"2024-04-16 09:41:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"915360","cherry_pick_of_change":914811,"cherry_pick_of_patch_set":3,"meta_rev_id":"a68be0e10069719c59596bf1973bfed26a73a382","_number":915360,"virtual_id_number":915360,"owner":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-16 09:41:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2024-04-16 07:01:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-04-16 07:45:43.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-04-16 07:45:43.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-11 21:46:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-16 07:01:02.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2024-04-16 07:41:12.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"623dccc288b11205ce97f538955853e387f17e7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-04-11 19:05:33.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"3991ad5c2ccdb5c50438d35eb2c71635f7d8c7ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-11 21:46:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/054a8fc7cab543a99dfa7d78b941b5ca\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a93b58adb68f42daa0c0c8d4245933a0 : SUCCESS in 10m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22855d850f5f49c19efa7dac9ee7ef20 : FAILURE in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea0bd5e8699b4436a95036662dc5c423 : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21851f9650e04ad2977d5daa2058591f : SUCCESS in 5m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fe0e5ec031f14b8fbbce928a91018622 : SUCCESS in 3m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/56213bd112c949fa857739117ebef623 : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efca09fad0644b9f9a7746b9b5ef3c04 : SUCCESS in 11m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd0a6555014548f5bae3d4dacddbf90d : SUCCESS in 5m 19s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/dd6c80c772644eb38fac828080221b7b : SUCCESS in 5m 05s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/8fc1a19119724b988bb278a3659919ee : SUCCESS in 7m 31s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/576033f8f5df4c2d99de6c570f51a756 : SUCCESS in 4m 18s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fe8f04095540420a8ffcaa416ad6ca54 : SUCCESS in 6m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/99dc9ea99be04368943601a0e2b60aaf : SUCCESS in 7m 46s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/95cd5b2b7df34979a5b1f66ce37a2098 : SUCCESS in 7m 21s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4d8fff99ad614b42a0b80f447f4cbfd2 : SUCCESS in 1h 47m 41s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/a633d01fd38348fc917f77157051ce08 : SUCCESS in 1h 19m 06s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/6da033bee8554652a33de58dadbb00df : SUCCESS in 1h 00m 20s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b304c5fd7f5f4f2a89cddfc8577f74a8 : SUCCESS in 1h 23m 30s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c4936c84f635498c9f323a70e4226e02 : SUCCESS in 1h 14m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5730f21421c046a7abf88dabb85898e3 : SUCCESS in 52m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f77883cec45e452d993ef4272cdfc12a : SUCCESS in 40m 00s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/1aa1557a258c41f6b38640d8dcec653f : SUCCESS in 1h 11m 10s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8fbab8e9093b4c829b0ba260a11cc224 : SUCCESS in 42m 34s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/48ee4d2ff68d4df18355365c0bcdda1b : SUCCESS in 2h 27m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/97f9cec327e246039365dffc9ef8ddb1 : SUCCESS in 2h 14m 39s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4a882edf24d844468e6723302d1da98e : FAILURE in 20m 46s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/f12a81970f06416fac58b93ec575802a : SUCCESS in 2h 11m 29s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/osutils.py\n  Comments left for invalid file octavia/common/tls_utils/cert_parser.py\n  Comments left for invalid file octavia/network/drivers/noop_driver/driver.py","accounts_in_message":[],"_revision_number":1},{"id":"eec5348a4d516a8efcdeb56aefd3d91eaf8c21d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-04-16 03:22:15.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c984de3a2e32d1d6d52c066e9e5b05df03242dca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 05:53:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cf14fddc66a4f588b0801a171bb8687\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8537df3609b74738a148a4f81d0f1952 : SUCCESS in 9m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f6b6fd1518d4ca5b510941025930fc8 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c50debebb2f463aad2914ba7f8551d7 : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e493a2b9a1374e7f904ba035faeca27b : SUCCESS in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6ea27b3f7d2d4057af383f7abc2f69ca : SUCCESS in 5m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/da21a26f76164625968396e398045b66 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aaaea07a96de4cbfa66c8bcb77dd70cd : SUCCESS in 11m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/35500bb84da54f07898e980d8d4e97c7 : SUCCESS in 5m 19s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/dbd00d22bdc149b99766398fac7849ce : SUCCESS in 7m 04s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/00afbb4b3c3f4627be4b3470355a60e1 : SUCCESS in 8m 05s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ac9667f4508d4205bbbeba972786ff63 : SUCCESS in 5m 52s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8b32b30f0ad6428dad8ad099b348b8e1 : SUCCESS in 7m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/987672253cda42e99f7f9e1e6d666044 : SUCCESS in 7m 44s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6946cc73c4ac48e1b9bd93f715fafe7c : SUCCESS in 7m 27s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f03797a1e03e42098a3b0a3742a8732d : SUCCESS in 1h 07m 38s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/ef811d0a31244eae9c38b0e22bb55517 : SUCCESS in 1h 16m 46s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/9e5a9024f0ec433ea95d6b86df9f806c : SUCCESS in 56m 02s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/dd8004d780654e6d85cc59069e5b63b0 : SUCCESS in 1h 20m 25s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/48961f5a0ffd472492795491c35a1322 : SUCCESS in 1h 34m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d11fb0b20cfe4acf8b094fdd547badef : SUCCESS in 36m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f81f00e818e14ae49eb44aa631bee6f2 : SUCCESS in 41m 14s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ffea32edff1e4a9babfa7f8fb6182590 : SUCCESS in 58m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/004165c9d3a64231a7a018dd5fce47a5 : SUCCESS in 48m 15s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1482919d1c1e4969bdac838cb6200f67 : SUCCESS in 1h 56m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c37a1907ed71479e814cfd330675a3e5 : SUCCESS in 2h 07m 56s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6e88cde8ed1648c4b23d511b2e23f00b : FAILURE in 22m 34s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/67fa1eea024e4a318a888a460b3c395b : SUCCESS in 2h 29m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a157025ae7336986cad2d2f90b3fb38088cad635","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-04-16 07:01:02.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f84fbebf705387bddda61d3b15e8b1a687a61c9b","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-16 07:45:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"506f9fcff2f2a39a304720b4efbadbc2b0bf9252","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 07:46:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c1bc6b61d0f34f27b4d75f6aae3217981dbc4226","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 09:41:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08d65b7ee4554cf6b514be985a12e4a6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38469760319c4222aa14ec5e189c89be : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30dbb8009b684216aa61334650b6e5f5 : SUCCESS in 6m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ba8e891c21da472c8d7d045727d0ce27 : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/12bc04e15b3c4863b7d73ddb0bd84358 : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df20696540ed4c4798845b118782f5d5 : SUCCESS in 13m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0608019b6f8448c0b7b7947a46942d2c : SUCCESS in 5m 36s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ccbd1aa06ad344a1b722f7232ae9a308 : SUCCESS in 4m 38s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/acf1c9867b2e48938277f94ba21042cf : SUCCESS in 6m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4b554f129441454e86d0468418679608 : SUCCESS in 1h 21m 52s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/a0a209c76ee1426dbcd5abff8567124a : SUCCESS in 1h 18m 39s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/6e3e12e5202a4c2f93749f8485f17522 : SUCCESS in 1h 14m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/629e63c736fb460d81cb86d4c517c5e4 : SUCCESS in 33m 27s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fb940b576f964097918cdd82b521a99c : SUCCESS in 1h 02m 39s","accounts_in_message":[],"_revision_number":2},{"id":"0df350f3fc358b8dec2acf51260deac66b9ae3d5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 09:41:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a68be0e10069719c59596bf1973bfed26a73a382","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 09:43:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee50fbd089fd4ac3b9e7c5881fda3109\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9398c3db8dd4c168f59e5e672b28555 : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb51ccfa14c34c6e9565280406d56f88 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b509b07951437d4f33b814cc639a5aa9748397b2","revisions":{"931ca6168c19da4e10142cbf98ef85321f3ae935":{"kind":"REWORK","_number":1,"created":"2024-04-11 19:05:33.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/60/915360/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/60/915360/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/915360/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/915360/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/915360/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/60/915360/1"}}},"commit":{"parents":[{"commit":"3f31a50f264c0e58691b1be7e90d324c13588b63","subject":"Merge \"Remove publish-openstack-octavia-amphora-image jobs\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f31a50f264c0e58691b1be7e90d324c13588b63"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2024-04-01 13:37:45.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2024-04-11 19:05:33.000000000","tz":0},"subject":"Handle empty delay on update healthmonitor","message":"Handle empty delay on update healthmonitor\n\nCheck that delay field is not UnsetType before further validation\n\nCloses-Bug: #2059894\nChange-Id: Ia853d43dc273019c76da09104f31aa7e1b154fec\n(cherry picked from commit 824b51a1dad80292b7a8ad5d61bf3ce706b1fb29)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/931ca6168c19da4e10142cbf98ef85321f3ae935"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/931ca6168c19da4e10142cbf98ef85321f3ae935"}]},"branch":"refs/heads/stable/2023.2"},"b509b07951437d4f33b814cc639a5aa9748397b2":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-04-16 03:22:15.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/60/915360/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/60/915360/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/915360/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/915360/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/60/915360/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/60/915360/2"}}},"commit":{"parents":[{"commit":"56cd4d3a969752f05f623078e19a1d0060daafe3","subject":"Pin pylint (\u003c\u003d3.0.4)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/56cd4d3a969752f05f623078e19a1d0060daafe3"}]}],"author":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2024-04-01 13:37:45.000000000","tz":240},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2024-04-16 03:22:15.000000000","tz":0},"subject":"Handle empty delay on update healthmonitor","message":"Handle empty delay on update healthmonitor\n\nCheck that delay field is not UnsetType before further validation\n\nCloses-Bug: #2059894\nChange-Id: Ia853d43dc273019c76da09104f31aa7e1b154fec\n(cherry picked from commit 824b51a1dad80292b7a8ad5d61bf3ce706b1fb29)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b509b07951437d4f33b814cc639a5aa9748397b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b509b07951437d4f33b814cc639a5aa9748397b2"}]},"branch":"refs/heads/stable/2023.2","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
