)]}'
{"id":"openstack%2Foctavia~915423","triplet_id":"openstack%2Foctavia~stable%2F2023.1~Ic330e5e57d674e8c50fb56ef04d670f11632e546","project":"openstack/octavia","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-04-10 21:07:44.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-04-10 21:07:44.000000000","reason":"Change was submitted"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2024-04-10 16:39:33.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"Ic330e5e57d674e8c50fb56ef04d670f11632e546","subject":"Pin pylint (\u003c\u003d3.0.4)","status":"MERGED","created":"2024-04-10 13:24:10.000000000","updated":"2024-04-10 21:08:51.000000000","submitted":"2024-04-10 21:07:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"915423","meta_rev_id":"e84538f1c353716e3b5c263fabb8b19271ce1203","_number":915423,"virtual_id_number":915423,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-10 21:07:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-04-10 16:31:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2024-04-10 16:39:33.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2024-04-10 16:39:33.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-10 14:47:03.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2024-04-10 16:00:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-10 16:28:10.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2024-04-10 16:31:06.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"b549f80b3416c356243a639d632b8509c3b58791","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-04-10 13:24:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d3e5db24ec9ee9bdd8183178abf4b99a27909955","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-10 14:47:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1464d1cd90512543444ee3e0f767944ba92862ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-10 16:00:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de6a0f4425fc4186950fddc8d2178e7a\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ae6f61d029a24d86b50e55746d17d797 : SUCCESS in 2m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cdea36ed96654bd296972109b8709c3e : SUCCESS in 10m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4060d40c40db46b7b282c26942c8933a : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f2b7b6f300b46c4bfb739a106e2cd70 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e11ca1c35b942f99076ad9dc213eb68 : SUCCESS in 12m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d97a325269c24fb4a82020b795455249 : SUCCESS in 5m 24s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/d001ed2c19ff428a967fe480ce7ba365 : SUCCESS in 4m 55s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/dbe38cb23b514f9c8f61ce8155d85fe2 : SUCCESS in 8m 10s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/cd40da338aca4e7d9ad45daae47481a4 : SUCCESS in 4m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5b71c87de5e9401696ca0729c1d69060 : SUCCESS in 8m 11s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4904618b75fa4a8eb55be59983dcb417 : SUCCESS in 8m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b1caf0a64efc4b5e806a8cecc9aec5df : SUCCESS in 7m 03s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/de4dc00003ee4b9db5c14621726df1f0 : SUCCESS in 1h 45m 34s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/1c1943dcbbb94bf6a44c374b4a0345f2 : SUCCESS in 1h 22m 44s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/64e50156efa6452ba74ff48d679792d7 : SUCCESS in 53m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/55f699e86ef849e997480adcd7951f76 : SUCCESS in 54m 10s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/6770e5570b734189b223dafd0430d594 : SUCCESS in 1h 09m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d2af6984f7d5440abf05c0aa0bbbcc21 : SUCCESS in 32m 59s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3a5eceeb65f3471b9336847801f5cbea : TIMED_OUT in 2h 34m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5c91ff0c27a445a78c1007bbf3f5992b : SUCCESS in 2h 17m 28s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ca022bcc466b4fdd9a399530deee5d38 : FAILURE in 36m 03s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a616494628764f6d876cfdcbdaf9c557 : SUCCESS in 1h 52m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/8013e3ae31f744ccbb494fc0d8525092 : TIMED_OUT in 2h 33m 56s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/6f8c66175efc47f0a5aa0221ccbe84f7 : SUCCESS in 2h 28m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5eea66911b585bca972f8bcb77543c1e1d373b06","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-04-10 16:28:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a216ec36d2bbcfbfe76ce90825022dfdf4aa16bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-04-10 16:30:51.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"56a1141505f124e0458d6e51440733e2304406cd","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-04-10 16:31:06.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"128616451bd453f597f1e8f3c2964641152a75e1","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-10 16:39:33.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b564bcf4ebfd3d17e52b1ae26a863efd20279293","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-10 19:12:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/490fa191e3314b2fa553ef0f07d14a9b\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3637bee5ad7546918d9bc0b524cc441f : SUCCESS in 1m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6f52dd8b324d43b98a01a02a07bc766f : SUCCESS in 9m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/daa4a08ff08141f8a54648c5656391d8 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ffe5cb126714098b853ce534d1fbf6c : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68f388b8f34146e9bb1767ac34589232 : SUCCESS in 11m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd6fe8e85c714ae7886c1205b291d0c5 : SUCCESS in 5m 51s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/e80eaf170fb7481c8759fe02ec8b97c5 : SUCCESS in 6m 42s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/6e446944c1774fa0b56e0dc39c60ac66 : SUCCESS in 8m 38s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/43f36ff9ab5e42738c34a4f0bc373424 : SUCCESS in 4m 52s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/309c63a6977847c1a2407c0227460f41 : SUCCESS in 6m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a08d33679fd44fe2999e802e7edc502f : SUCCESS in 6m 09s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d9437f1fa5b14015bff124306a0d6b7f : SUCCESS in 7m 27s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a8a925d66faa41f89df95f3c65cff614 : SUCCESS in 2h 25m 50s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/e87501a7cbf44c40b7ecf59a221c9082 : SUCCESS in 1h 54m 50s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/96b7006101ca4b71babe17326506efc4 : SUCCESS in 1h 14m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/458af81613f74303bca1fe590588f1f5 : SUCCESS in 38m 54s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ccd7f8e290294ce8ae12a5717cabb5d9 : SUCCESS in 1h 12m 18s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2c9d60de83534cdb81909cad85e69a04 : SUCCESS in 29m 21s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4ea642b01b9f43e6a4cabef405bc2bfd : SUCCESS in 1h 47m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/748e3826050544ff9c6f067ce5055630 : SUCCESS in 2h 16m 00s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d64ec152b33f4f39b29ad07e269794b0 : FAILURE in 29m 06s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/966e58ba7dfe407f81aab63bb0d0da90 : TIMED_OUT in 2h 33m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/16959f0b0b304c99974f4fd04e8f1e93 : SUCCESS in 2h 03m 38s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/fc64c870d8ad4c17a68cb10c2a66708e : SUCCESS in 2h 28m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3c967e0e26b9f0b11bf2ab68dec8bca0a186f05e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-10 19:13:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8ae08c65c828dba16fc1c78e53e6121781cc6876","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-10 21:07:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86cf3a232b6248d3a7e7dfe4163ebaca\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/f41f87dbff0443a7adbd00e2017d89e3 : SUCCESS in 1m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96ba73034a264cbb9dc6bd1fe70a730f : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b6529478b4f4792949ee2c3a93467d2 : SUCCESS in 7m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e147aa1a78e4992b2809cdc758cd084 : SUCCESS in 10m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea3075909b474d80b354adea77c7fb35 : SUCCESS in 6m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d904afc2dda541a9bad8e23aaded1efc : SUCCESS in 4m 20s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/00378788ffc34b9299e58a77085b3e25 : SUCCESS in 7m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2d1599dbaa194824bf665ec1a408fb8e : SUCCESS in 1h 17m 12s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/cbcdebcdc3c84d24a911ffd295c29052 : SUCCESS in 1h 51m 17s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/92ef04ac7fcd4c7d8d79bb0c6c6f5a32 : SUCCESS in 1h 19m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/71ad55cf482441df93d74eef51b7a68c : SUCCESS in 50m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e2d348db62014b97b1b1b3c523ea7051 : SUCCESS in 58m 38s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/0c8fd2b927744b4dab509652dd79144c : SUCCESS in 37m 04s","accounts_in_message":[],"_revision_number":2},{"id":"d175a94dae57aae9cb2186d668e8b0884b716628","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-10 21:07:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e84538f1c353716e3b5c263fabb8b19271ce1203","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-10 21:08:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9dea5dd009e0409c955e5a5f4857aa4b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15fabc134a8544749efdab238e78c5e0 : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64ec54e1eb624bdcafbed32724a7e9a4 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d6f32b19fb56493f5a3b68b4b6f27e376b4d3979","revisions":{"de400267fa71167cd0d4ec9992911778f4a3cfaf":{"kind":"REWORK","_number":1,"created":"2024-04-10 13:24:10.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/23/915423/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/23/915423/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/915423/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/915423/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/915423/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/23/915423/1"}}},"commit":{"parents":[{"commit":"42e421014de3003ce079380888510dc2f7eb1417","subject":"Merge \"allowed_cidr validation for additional_vips\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/42e421014de3003ce079380888510dc2f7eb1417"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-04-10 13:23:15.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-04-10 13:23:28.000000000","tz":120},"subject":"Pin pylint (\u003c\u003d3.0.4)","message":"Pin pylint (\u003c\u003d3.0.4)\n\nnew pylint release 3.1.0 adds new checkers and fails on stable branches,\npin pylint.\n\nChange-Id: Ic330e5e57d674e8c50fb56ef04d670f11632e546\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/de400267fa71167cd0d4ec9992911778f4a3cfaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/de400267fa71167cd0d4ec9992911778f4a3cfaf"}]},"branch":"refs/heads/stable/2023.1"},"d6f32b19fb56493f5a3b68b4b6f27e376b4d3979":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-04-10 16:30:51.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/23/915423/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/23/915423/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/915423/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/915423/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/23/915423/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/23/915423/2"}}},"commit":{"parents":[{"commit":"42e421014de3003ce079380888510dc2f7eb1417","subject":"Merge \"allowed_cidr validation for additional_vips\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/42e421014de3003ce079380888510dc2f7eb1417"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-04-10 13:23:15.000000000","tz":120},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-04-10 16:30:51.000000000","tz":0},"subject":"Pin pylint (\u003c\u003d3.0.4)","message":"Pin pylint (\u003c\u003d3.0.4)\n\nnew pylint release 3.1.0 adds new checkers and fails on stable branches,\npin pylint.\n\nR1737 - Use \u0027yield from\u0027 directly instead of yielding each element one by one (use-yield-from)\n\nChange-Id: Ic330e5e57d674e8c50fb56ef04d670f11632e546\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d6f32b19fb56493f5a3b68b4b6f27e376b4d3979"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d6f32b19fb56493f5a3b68b4b6f27e376b4d3979"}]},"branch":"refs/heads/stable/2023.1","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
