)]}'
{"id":"openstack%2Foctavia~915772","triplet_id":"openstack%2Foctavia~master~Ic18044b8065d2350c180ad26f5307a77aa99037b","project":"openstack/octavia","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-04-16 10:18:17.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2024-04-16 10:18:17.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-04-16 10:18:17.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-04-15 09:19:13.000000000","reason":"removed on reply"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-04-15 09:05:29.000000000","reason":"Only change owner, uploader, reviewers, and cc can be in the attention set"}},"hashtags":[],"change_id":"Ic18044b8065d2350c180ad26f5307a77aa99037b","subject":"SQLAlchemy 2.0: Drop use of removed autoload","status":"MERGED","created":"2024-04-14 17:16:16.000000000","updated":"2024-04-16 10:19:51.000000000","submitted":"2024-04-16 10:18:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"915772-sqlalchemy-20","meta_rev_id":"7a1ba969fca3f5c92acfb2c227bb71407d8f4077","_number":915772,"virtual_id_number":915772,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-16 10:18:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-04-15 09:19:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2024-04-15 09:27:06.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2024-04-16 08:16:43.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-14 20:31:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-15 05:51:11.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2024-04-15 06:18:12.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2024-04-15 08:30:34.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"},{"updated":"2024-04-15 09:27:06.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"510213aea9ba8bbcbf38a325e46a0d0d7e59bef5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-14 17:16:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3d937d8219534a55a96c75e1bb9eda5bf1647585","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-14 17:18:27.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"b15c8c962a81c58c2da6ba508bdc3dc88092b1d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-14 20:31:08.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/07339ca9530a406497ea6c808dd0924e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/530c280f7b1a430e9ec5ab6c0a117f11 : SUCCESS in 15m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33b047e035634c72b74f52ad82a9eefc : FAILURE in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b817efc462a349bd88c91205a287b8b6 : SUCCESS in 5m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ade84815086449b681ac8b2338420292 : SUCCESS in 4m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5de3f591592f44178078f1db7ff7b4a2 : SUCCESS in 8m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a768cd98e363406aa8078aaea1b02247 : SUCCESS in 9m 36s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/0a937ce8518e4e7f92bd7f41c75aecda : SUCCESS in 4m 55s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/cb25ac09853d4163a8f25c093a04600f : SUCCESS in 7m 34s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/b0d61735007c48f7990703131d2e1d0b : SUCCESS in 9m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/eaf54a205336420bbeefdcf88e90f8ce : SUCCESS in 4m 10s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8dcefca6121e44c1819afb437dd97dca : SUCCESS in 5m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/79681650179a476fb97747af1106b794 : SUCCESS in 7m 50s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/20652380eaec45cbb5ff3ad3b5c7f43e : SUCCESS in 6m 28s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b1890910674147529459ae09a2fc057e : SUCCESS in 1h 00m 53s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/faa9cd036e454d54b6657fee3682033e : SUCCESS in 1h 53m 29s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/17b6326a8d7d4448b0fb7b7e5d4887ca : SUCCESS in 1h 09m 56s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/2fb71aed742a49e2ad92a3e574f7c483 : SUCCESS in 1h 38m 44s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c09e89d68f2441baab1be75c7c0bdbaf : SUCCESS in 1h 24m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1d1bd0a86b3e483eb846640b310acee9 : SUCCESS in 43m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0c89107f6ece441ca83ce3691d17a440 : SUCCESS in 38m 22s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/657f563f3548445894bb29b74466696d : SUCCESS in 1h 06m 44s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fa3d5fc417464ddd87aa2990e420d63b : SUCCESS in 32m 29s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/976b642400564d5192809c2a09c6b0c1 : TIMED_OUT in 2h 33m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/dc990f2540ac40d984676aa7179598a5 : TIMED_OUT in 3h 04m 06s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f6ed8b5b0c1b440082c075cf25a333a5 : FAILURE in 36m 48s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/838d6b3b23864e528233a11a341d3817 : SUCCESS in 2h 28m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a33f21f74df3d5b93c32eed468b0eb24ab85b959","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-04-14 22:04:34.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"73c5ec26a8e872a18ecaf38783d757e3394de03e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 00:38:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2dbd2b133b14e9a90179c11a5d38ecc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/846414d707ff411caccd1c3adfe2036d : SUCCESS in 9m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e27eb2d4e1d1413e93f6d08ab6a37cf8 : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a7d29b091e6747438671b7044cb2f4d3 : SUCCESS in 5m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/17089346bfd8495880e68f750873873f : SUCCESS in 5m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/06be4b4e18e94677821b633ee5c94575 : SUCCESS in 8m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/400f22d9819e492fb2731d86c39c7de3 : SUCCESS in 9m 42s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/8fb1153cae6c41f5b8646ab9694c1811 : SUCCESS in 4m 27s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/fb102e0f5caf4f2a8d241e621448c895 : SUCCESS in 6m 51s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/b4ddba3757284983989e14a440d916d3 : SUCCESS in 7m 57s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6ea1a52566ce48ae9fec9eba128e5e95 : SUCCESS in 4m 09s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2b2cb06cc7994e39a023e5f272c6bc7c : SUCCESS in 6m 55s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/15f00eca062e47ebba6dbb044bee2cbe : SUCCESS in 7m 02s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a40b3dc641644337ad8a650a780e9678 : SUCCESS in 6m 11s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f4143ba5c9324f439696bc079eab9f01 : SUCCESS in 1h 03m 47s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/d18bf7ef6855407282840f3378548ae8 : SUCCESS in 1h 38m 58s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/97526d6b71f641ec826501e163562552 : SUCCESS in 1h 04m 18s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/5f3d140046e7421d9a6adc99099a6aaa : SUCCESS in 1h 42m 43s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/494f455367ab49b49ea70abd5895bb92 : SUCCESS in 58m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3df1ec542f804dac930c546039272086 : SUCCESS in 51m 14s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a2d83f4a776f4992a89beb3cba33a726 : SUCCESS in 42m 08s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/8f01b181bd034578a62b55b8d63c7d7e : SUCCESS in 1h 08m 18s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6c4c3c4624e8454abe68b828d3fc0a76 : SUCCESS in 36m 07s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/602062fb83c64532a1cabb0afb663fff : SUCCESS in 2h 06m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f50e0328c73547cbb4d8a74d14efae19 : SUCCESS in 2h 25m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/bc9a7616e58147a894679b6f454d407a : FAILURE in 29m 50s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/870c96d5210b430db4ec340dec3f6711 : TIMED_OUT in 2h 33m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3ba8e9cc442ba8bc34a67c53930e7071233530c3","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-15 02:50:25.000000000","message":"Topic bug/2061303 removed","accounts_in_message":[],"_revision_number":3},{"id":"cd24dcf3e7188a45d0700c925e010b4054450d6f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-15 02:50:33.000000000","message":"Topic set to sqlalchemy-2","accounts_in_message":[],"_revision_number":3},{"id":"0850eec32a70f70aeacea681c7e15db9307576fd","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-15 02:50:38.000000000","message":"Topic sqlalchemy-2 removed","accounts_in_message":[],"_revision_number":3},{"id":"723126abb978f0143bb59a90e635e4332891ffb0","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-15 02:51:05.000000000","message":"Topic set to sqlalchemy-20","accounts_in_message":[],"_revision_number":3},{"id":"7ac6bfe3ad968cdecfa3ddd459683fe70fc74df8","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-04-15 05:51:11.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"b106837c1ab6195ed6e69d4fa0dc9e20363095d0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-15 05:58:02.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"b46b479e90e73d3f6d82b3e188794bc6cf77bf41","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-04-15 06:18:12.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"56763f40a43e2fffa48b5fc73a4a857d2c76a9cc","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-15 08:30:34.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"4ef64684202286c7b7d56e7d8130ab9a5e3f5064","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-15 08:41:41.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d88c81684d14a811f0f5c6f55c079fd9d241bf49","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-15 08:59:25.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7215121d6d6e9956abfb65702c48364ca4e5e644","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-04-15 09:01:11.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6a9223ad720afd14da227cd341309ff2e7b014bc","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-15 09:02:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f7550d7e0b1584c0a25cd9dd62298b2c40b24ccd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-15 09:05:04.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"8fe5cc01322d4a9b286c4fb89e6a3b405b5db07b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-15 09:05:29.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"b950e9748d43d04ebc0bc0b9f0e870c3b2b1f8b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-15 09:05:46.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"fe687d1a1b4e764f8e9dcbbce5023941e3cfe5de","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-15 09:06:57.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"282479ad8f943d06ccb477e54f4cfba482e1516b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-04-15 09:19:13.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"508c648021b204e88b7edec0c5804bb5850a6215","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-15 09:27:06.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"257c295da08624f55da7dbf6d0740fc0d8449edd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 12:18:14.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4a6da8900d6d4661bb097f69b6b15c85\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/06f707b3feef469ab858408e30f26ec0 : SUCCESS in 11m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45d5221ba0c040e488f0a50f4ab0eb9c : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f4a37bf3c3b2445eb2b233a2a7bf0c5f : SUCCESS in 4m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1f94363a243b4333909867ec6fb08053 : SUCCESS in 5m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3da8f721995f4c0b8af586668eea5a94 : SUCCESS in 8m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f721d623b7f41919380e64bcd274f91 : SUCCESS in 10m 14s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/68baabff6eec417481738f63a5077e43 : SUCCESS in 5m 07s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/b00e3ec5f59d4314b59208944182445c : SUCCESS in 9m 33s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/4eedceb2d7a8446ebfa8dcca5b32afe5 : SUCCESS in 9m 32s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/dc62a4fc77a5454488b24552aa3c1e62 : SUCCESS in 5m 33s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/32d71aee8b27452dabb18c1c24bb9381 : SUCCESS in 7m 54s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/de9ef131acd54899b1bb8b46c9c83efa : SUCCESS in 6m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2c026d5d1c3c4820bc83e25774548e4e : SUCCESS in 6m 49s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/24f63343f3bd4fd7a1779efd5a0aae01 : SUCCESS in 1h 17m 30s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/e99e1078fa3e4ce7a65f7c0fe43acedf : SUCCESS in 2h 22m 13s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/49390aeeaa864635bd7d41e8b7c095db : FAILURE in 1h 30m 09s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/0cb381ca9b0842bd845542534c30fdd9 : SUCCESS in 2h 23m 17s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/df6e3abed7a3419d824eb6bcaa6dac60 : SUCCESS in 1h 14m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ab573e80ef2c45da91fa557351ebccf7 : SUCCESS in 56m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6fd13005b10043f2ba0d7c4120d3f8a2 : SUCCESS in 41m 49s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/364176e6b8e2417087fc114162d58f24 : SUCCESS in 1h 15m 06s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2a6db9d41cd94329bfc1d990cf6898ee : SUCCESS in 37m 17s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0b615fb0aef74f79a99183447cda93ee : SUCCESS in 2h 03m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/08c61a78f494499392e00e1679690426 : TIMED_OUT in 3h 04m 51s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6fdb2df152e6423c81946f9fc4f1a80d : FAILURE in 25m 02s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/b77bc0022d844249a0ba505bbbf88b17 : TIMED_OUT in 2h 35m 29s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"c07829f06743193e335829e8d5d53c822e8f732b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-15 12:21:06.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"fc827b2c1eec87bb89ba468d983b60352f3d479d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 15:05:18.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40fd257dee9c457f91ee3cb44b6d0dde\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a9b1ec138900465a9b1526a2dc8c06e2 : SUCCESS in 9m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ae00cfa2a5d4c9582a52b084588ff4b : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2fd7dee34222415fae5c11a5e371e35d : SUCCESS in 4m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/99aed11002414d70b5c9e988eeb07fe5 : SUCCESS in 5m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/89cf6ac5098d425abc2aa5ecaa875fa3 : SUCCESS in 9m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64ab22685e1f43efa121b0670ba96d01 : SUCCESS in 12m 31s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/1aa197a0f5814df3922f8fc82679e627 : SUCCESS in 4m 53s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/8a442ad1fd4545cba707add67a745bef : SUCCESS in 8m 35s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/f99bea74bdf44b9f93d36d97f39d8e31 : SUCCESS in 9m 43s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/463def320eec4854901841eac9e01cf2 : SUCCESS in 4m 41s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/05454deee4d34f528ce6f6abc23f00b1 : SUCCESS in 8m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9980d22c593442ba9a93349472a5aff9 : SUCCESS in 7m 44s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/33af61a7805f47a3946f3a67482ffede : SUCCESS in 9m 22s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/eb7853c07a6740eb94e0b5814325f7ff : SUCCESS in 1h 13m 06s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/fe004c2f04bb4ece861f8a4ed6458c3b : SUCCESS in 1h 39m 10s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/1f59917b452e40cb8a72bf90d6f2046c : SUCCESS in 1h 02m 31s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3c146df90b4f4731aa778a3dabc4e26c : SUCCESS in 2h 01m 05s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/504214feddd04581bae2bc365a51cb6a : SUCCESS in 1h 23m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9f4621eb3ce6489aae340dd4ff25348a : SUCCESS in 39m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6d47d43cfc7b4463b74d35bea342eb1e : SUCCESS in 39m 37s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/bf73c67a8cb24c519a5ad89c2bcff89a : SUCCESS in 1h 11m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fb466b5f46f340f49cd73accb0bcce14 : SUCCESS in 43m 59s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f0a95597a57f4bfa99fa0f884713cc66 : SUCCESS in 2h 22m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/11570415e7e1483f84914c10fdcb1763 : SUCCESS in 2h 40m 51s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ac39d74f26d14cf495b933d405b8d213 : FAILURE in 24m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/d2882a8f939c47eeb85c3d65790aadc0 : TIMED_OUT in 2h 35m 56s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"9fed67d003914c97cac819e8e06cf9bba5cf43cb","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-16 08:16:43.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"794b96b000020d026ce239a0e703715581e0d4c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 08:17:30.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"5b04ca67d7e3394a32b77ce0ef8de21fc53e921a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 10:18:17.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a326a91a68274676a749039b6653fb6d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3f61f57858242f19d71aeb6282a5e71 : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/76e06b8bbecf41d4b74e9d52625e8b7e : SUCCESS in 4m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c81bca0d47ff40c2b3e3a3fd622f9fae : SUCCESS in 6m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1b068a4bcbe4489b01d34794448e93f : SUCCESS in 10m 26s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8c7edfe0934d4fa18458db214e9325b7 : SUCCESS in 4m 58s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4eb581ec5acc4144b4f201bddecade3a : SUCCESS in 7m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d65c111ef25b4258b8498178b581a288 : SUCCESS in 1h 54m 47s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/27b5c62c939b42c6bc3e4d9afb175902 : SUCCESS in 1h 39m 44s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/522b0d67932c438988f31595a5808b1e : SUCCESS in 57m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/abcaad3dfc744655b2ee7b269b003493 : SUCCESS in 35m 24s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5b4a24db697146dc83da639ea3cd3f11 : SUCCESS in 59m 31s","accounts_in_message":[],"_revision_number":6},{"id":"3ee62663b653ac8cadd0ba43e066616da5a4a2be","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 10:18:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"7a1ba969fca3f5c92acfb2c227bb71407d8f4077","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 10:19:51.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cdfbfff9d8d4be9b702617ce50f6572\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad618cab289f4311905d3c4d5bc2dccb : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"d0c29009e68a28f46b5ac962449746ca87011fa3","revisions":{"9de48f5bc5f76228c58fefec2bc3174aca8c27f7":{"kind":"REWORK","_number":1,"created":"2024-04-14 17:16:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/72/915772/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/72/915772/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/915772/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/915772/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/915772/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/72/915772/1"}}},"commit":{"parents":[{"commit":"fa7cc0178783ad4ab9f3ddce893042d5952d0bd8","subject":"Merge \"Fix fully-populated API with allowed_cidrs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fa7cc0178783ad4ab9f3ddce893042d5952d0bd8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-14 17:14:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-14 17:15:52.000000000","tz":540},"subject":"Fix compatibility with SQLAlchemy 2.0","message":"Fix compatibility with SQLAlchemy 2.0\n\nThe autoload arugment was removed[1] in SQLAlchemy and only\nthe autoload_with argument should be passed.\nThe fallback logic is added to keep the compatibility with SQLAlchemy\n\u003c 2.0, but it should be removed once the SQLAlchemy in upper\nconstraints is bumped to 2.0+.\n\nCloses-Bug: #2061303\nChange-Id: Ic18044b8065d2350c180ad26f5307a77aa99037b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9de48f5bc5f76228c58fefec2bc3174aca8c27f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9de48f5bc5f76228c58fefec2bc3174aca8c27f7"}]},"branch":"refs/heads/master"},"609b04b2001a5e78cebc7f40cced99eb3ea04326":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-04-14 17:18:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/72/915772/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/72/915772/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/915772/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/915772/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/915772/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/72/915772/2"}}},"commit":{"parents":[{"commit":"fa7cc0178783ad4ab9f3ddce893042d5952d0bd8","subject":"Merge \"Fix fully-populated API with allowed_cidrs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fa7cc0178783ad4ab9f3ddce893042d5952d0bd8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-14 17:14:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-14 17:18:27.000000000","tz":0},"subject":"Fix compatibility with SQLAlchemy 2.0","message":"Fix compatibility with SQLAlchemy 2.0\n\nThe autoload arugment was removed[1] in SQLAlchemy and only\nthe autoload_with argument should be passed.\nThe fallback is added to keep the compatibility with SQLAlchemy \u003c 2.0,\nbut it should be removed once the SQLAlchemy in upper constraints is\nbumped to 2.0+.\n\n[1] https://github.com/sqlalchemy/sqlalchemy/commit/c932123bacad9bf047d160b85e3f95d396c513ae\n\nCloses-Bug: #2061303\nChange-Id: Ic18044b8065d2350c180ad26f5307a77aa99037b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/609b04b2001a5e78cebc7f40cced99eb3ea04326"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/609b04b2001a5e78cebc7f40cced99eb3ea04326"}]},"branch":"refs/heads/master","description":"Edit commit message"},"c3aa6363d7f5eabcdb0601bbb85d69c0ff9e9d91":{"kind":"REWORK","_number":3,"created":"2024-04-14 22:04:34.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/72/915772/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/72/915772/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/915772/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/915772/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/915772/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/72/915772/3"}}},"commit":{"parents":[{"commit":"fa7cc0178783ad4ab9f3ddce893042d5952d0bd8","subject":"Merge \"Fix fully-populated API with allowed_cidrs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fa7cc0178783ad4ab9f3ddce893042d5952d0bd8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-14 17:14:32.000000000","tz":540},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-04-14 22:04:33.000000000","tz":0},"subject":"Fix compatibility with SQLAlchemy 2.0","message":"Fix compatibility with SQLAlchemy 2.0\n\nThe autoload arugment was removed[1] in SQLAlchemy and only\nthe autoload_with argument should be passed.\nThe fallback is added to keep the compatibility with SQLAlchemy \u003c 2.0,\nbut it should be removed once the SQLAlchemy in upper constraints is\nbumped to 2.0+.\n\n[1] https://github.com/sqlalchemy/sqlalchemy/commit/c932123bacad9bf047d160b85e3f95d396c513ae\n\nCloses-Bug: #2061303\nChange-Id: Ic18044b8065d2350c180ad26f5307a77aa99037b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c3aa6363d7f5eabcdb0601bbb85d69c0ff9e9d91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c3aa6363d7f5eabcdb0601bbb85d69c0ff9e9d91"}]},"branch":"refs/heads/master"},"573b1bebe7e1a49194ea74f9e79ebe39c56212a8":{"kind":"REWORK","_number":4,"created":"2024-04-15 09:05:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/72/915772/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/72/915772/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/915772/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/915772/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/915772/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/72/915772/4"}}},"commit":{"parents":[{"commit":"fa7cc0178783ad4ab9f3ddce893042d5952d0bd8","subject":"Merge \"Fix fully-populated API with allowed_cidrs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fa7cc0178783ad4ab9f3ddce893042d5952d0bd8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-14 17:14:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-15 09:04:18.000000000","tz":540},"subject":"Fix compatibility with SQLAlchemy 2.0","message":"Fix compatibility with SQLAlchemy 2.0\n\nThe autoload arugment was removed[1] in SQLAlchemy and only\nthe autoload_with argument should be passed.\n\nThe autoload argument is set according to the autoload_with argument\nautomatically even in SQLAlchemy 1.x so is not at all needed.\n\n[1] https://github.com/sqlalchemy/sqlalchemy/commit/c932123bacad9bf047d160b85e3f95d396c513ae\n[2] https://github.com/sqlalchemy/sqlalchemy/commit/ad8f921e969b6f735dc8e08d882c961dde78f2b1\n\nCloses-Bug: #2061303\nChange-Id: Ic18044b8065d2350c180ad26f5307a77aa99037b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/573b1bebe7e1a49194ea74f9e79ebe39c56212a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/573b1bebe7e1a49194ea74f9e79ebe39c56212a8"}]},"branch":"refs/heads/master"},"3bb55cb27136e27da9526e61b5593abbfbef75a4":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2024-04-15 09:05:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/72/915772/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/72/915772/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/915772/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/915772/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/915772/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/72/915772/5"}}},"commit":{"parents":[{"commit":"fa7cc0178783ad4ab9f3ddce893042d5952d0bd8","subject":"Merge \"Fix fully-populated API with allowed_cidrs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fa7cc0178783ad4ab9f3ddce893042d5952d0bd8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-14 17:14:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-15 09:05:10.000000000","tz":540},"subject":"SQLAlchemy 2.0: Drop use of removed autoload","message":"SQLAlchemy 2.0: Drop use of removed autoload\n\nThe autoload arugment was removed[1] in SQLAlchemy and only\nthe autoload_with argument should be passed.\n\nThe autoload argument is set according to the autoload_with argument\nautomatically even in SQLAlchemy 1.x so is not at all needed.\n\n[1] https://github.com/sqlalchemy/sqlalchemy/commit/c932123bacad9bf047d160b85e3f95d396c513ae\n[2] https://github.com/sqlalchemy/sqlalchemy/commit/ad8f921e969b6f735dc8e08d882c961dde78f2b1\n\nCloses-Bug: #2061303\nChange-Id: Ic18044b8065d2350c180ad26f5307a77aa99037b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3bb55cb27136e27da9526e61b5593abbfbef75a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3bb55cb27136e27da9526e61b5593abbfbef75a4"}]},"branch":"refs/heads/master"},"d0c29009e68a28f46b5ac962449746ca87011fa3":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2024-04-15 09:05:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/72/915772/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/72/915772/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/915772/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/915772/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/915772/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/72/915772/6"}}},"commit":{"parents":[{"commit":"fa7cc0178783ad4ab9f3ddce893042d5952d0bd8","subject":"Merge \"Fix fully-populated API with allowed_cidrs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fa7cc0178783ad4ab9f3ddce893042d5952d0bd8"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-14 17:14:32.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-15 09:05:33.000000000","tz":540},"subject":"SQLAlchemy 2.0: Drop use of removed autoload","message":"SQLAlchemy 2.0: Drop use of removed autoload\n\nThe autoload arugment was removed[1] in SQLAlchemy and only\nthe autoload_with argument should be passed.\n\nThe autoload argument is set according to the autoload_with argument\nautomatically even in SQLAlchemy 1.x[2] so is not at all needed.\n\n[1] https://github.com/sqlalchemy/sqlalchemy/commit/c932123bacad9bf047d160b85e3f95d396c513ae\n[2] https://github.com/sqlalchemy/sqlalchemy/commit/ad8f921e969b6f735dc8e08d882c961dde78f2b1\n\nCloses-Bug: #2061303\nChange-Id: Ic18044b8065d2350c180ad26f5307a77aa99037b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d0c29009e68a28f46b5ac962449746ca87011fa3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d0c29009e68a28f46b5ac962449746ca87011fa3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
