)]}'
{"id":"openstack%2Foctavia~915834","triplet_id":"openstack%2Foctavia~master~Ib557a9cf938fcf4257d2c2848fff78a62d82109a","project":"openstack/octavia","branch":"master","topic":"job_backend_etcd","attention_set":{},"removed_from_attention_set":{"27442":{"account":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"last_update":"2024-09-16 17:11:26.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-09-16 17:11:26.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-09-16 15:01:03.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2024-09-16 10:51:27.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ib557a9cf938fcf4257d2c2848fff78a62d82109a","subject":"Support for Jobboard etcd backend","status":"MERGED","created":"2024-04-15 08:31:03.000000000","updated":"2024-09-16 17:14:17.000000000","submitted":"2024-09-16 17:11:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"915834-job_backend_etcd","meta_rev_id":"55ad5829ed6312b9c3aa5ce7ee9fa0ba915cc066","_number":915834,"virtual_id_number":915834,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-16 17:11:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-09-16 15:01:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2024-09-16 10:51:27.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-09-16 15:01:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-15 09:19:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-14 10:51:06.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2024-08-15 22:06:10.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-09-02 16:15:15.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"}],"messages":[{"id":"1dc9f8ea17626e71fc1a022479413b019e2a2064","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-04-15 08:31:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0f280df960af65db31714aefb8b6215175073bab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-04-15 09:01:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f0c4a772954f050890c7dfadbe13e5fec31e8bae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 09:19:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/757e9104f9d647bf97d5e817fc86a3ac\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b1f45e63facb43d6aabdaff503ba5d84 : SUCCESS in 1m 58s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-etcd https://zuul.opendev.org/t/openstack/build/a6041bfd5cb8411d80da25811735d8bb : FAILURE in 16m 40s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-etcd https://zuul.opendev.org/t/openstack/build/11eda53155044a92ae55a8f803b9513f : FAILURE in 16m 05s","accounts_in_message":[],"_revision_number":2},{"id":"8e603f9d77bad0daafb7cdef34b2ab980627ba66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-04-15 09:20:42.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"ecb58540d66a52f10c4acf6ec7ebac0c8a2d536c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 11:50:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e063732957d4ef9b56653f07e0d4645\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/76f5b571987049129162a73c224a6da9 : SUCCESS in 2m 41s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-etcd https://zuul.opendev.org/t/openstack/build/e50950731af94ac19f4a42886775f1f7 : SUCCESS in 2h 27m 37s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-etcd https://zuul.opendev.org/t/openstack/build/0f4d4305a6664bf995e8de2dcb66cd90 : SUCCESS in 1h 26m 11s","accounts_in_message":[],"_revision_number":3},{"id":"6e99de21776238de3b83cf418a4cca3a6da0adc3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-04-15 11:58:05.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"7c12d8ba3e5bce592974ad5412e23e48155b413c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-04-15 13:34:45.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"9d187595f694c6c6b71314385f2b56b14befe6ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 14:42:33.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5dd46aba66ea47c4877e09d8572f3cff\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/055fbffa2f9840848810f9df45a70448 : SUCCESS in 2m 00s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-etcd https://zuul.opendev.org/t/openstack/build/1022c22bc9564dd1a57cd502c947a7b1 : FAILURE in 1h 05m 41s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-etcd https://zuul.opendev.org/t/openstack/build/a28010e6386e4d0dbee723d0ac830ce9 : FAILURE in 44m 43s","accounts_in_message":[],"_revision_number":5},{"id":"6b026806d14a9716dce5c92275fd5df15eb83e67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-04-15 14:47:45.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"378f0991bbfc430fdaf66b9f85bcc8bdfa1b7c84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 15:19:07.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8b71acc4e074509b96a7cde9bb57050\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/72c9a88dddb242f6bf5d1b81c429a6a3 : SUCCESS in 2m 03s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-etcd https://zuul.opendev.org/t/openstack/build/f05664f05fff4f7ebbf50db715cd78af : SUCCESS in 28m 41s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-etcd https://zuul.opendev.org/t/openstack/build/7d148c3dc3bd48fdb50ecdf8b953c0ef : SUCCESS in 29m 06s","accounts_in_message":[],"_revision_number":6},{"id":"8bd3a301ec57e9e1fe610edb2f810f0d9356b291","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-04-19 16:23:37.000000000","message":"Topic set to job_backend_etcd","accounts_in_message":[],"_revision_number":6},{"id":"163d32ce0489a565893e1f937baae5b3987f7528","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-04-19 16:23:52.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d95c5c0002ced49de3f2df4f5c243ad9243ba9a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 17:08:08.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b491b3b0fe341e98041fb491787db2e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d8fe74ca5cb548b8b496464af11a9931 : SUCCESS in 2m 20s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-etcd https://zuul.opendev.org/t/openstack/build/1f179ed59c114088be658846f096502f : SUCCESS in 41m 15s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-etcd https://zuul.opendev.org/t/openstack/build/eebf4e6aa1f7478090559ea9cdc40e62 : SUCCESS in 39m 53s","accounts_in_message":[],"_revision_number":6},{"id":"ba06b419babc4f2481b9c36f61b46a6190116ff3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-04-25 12:56:11.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"b5b3c6205a0740ede0c85323e30076eafe00acc9","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-04-25 14:35:09.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"48844dc62c1f901ef419063f4d9f1a5dd58eede8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 15:24:21.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02f47ce52d9044a19d1d4eb296fd8c91\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a4542943ff844df8bdf3b23bbe2fabad : SUCCESS in 2m 54s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/505a6bd7748646b399f7b00c90bacd4c : SUCCESS in 2h 25m 37s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3b2206a14b7c4a1cb4b7c9fcdac8bb30 : SUCCESS in 1h 42m 24s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-etcd https://zuul.opendev.org/t/openstack/build/ea00590d390f454393c446ea21b283b7 : SUCCESS in 49m 21s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-etcd https://zuul.opendev.org/t/openstack/build/f7b21836d1e649b09b0c137ed923b8ab : SUCCESS in 35m 07s","accounts_in_message":[],"_revision_number":7},{"id":"a3bd1eee3e803c3ba55f09a9e5d4e4d34c151505","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-04-25 15:56:10.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"0e328259e2d5fed917b4d5ab11e5bcb8a62cc1d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 17:18:14.000000000","message":"Patch Set 8: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/915492 is needed.","accounts_in_message":[],"_revision_number":8},{"id":"44a4dd06f3493777e944615b4a64554f7f637a01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-04-26 13:23:43.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"c9fe02cba8565dd12f5a9f05437e10b7d7e8c77b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 15:38:25.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60bef059a53c4112ba3ce11b1cfa8d32\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5d4c7277209c451d9dd91d17d0905a23 : SUCCESS in 2m 06s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/2184f5603e19481780df91b49bc0f654 : SUCCESS in 2h 12m 32s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/64a65d836afd4ba1820f5db5145c0051 : SUCCESS in 1h 28m 50s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-etcd https://zuul.opendev.org/t/openstack/build/123bfe49b49f4f05b7e15e50dbda7ba3 : SUCCESS in 41m 47s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-etcd https://zuul.opendev.org/t/openstack/build/bc2d117e966646c3a327450bb599026f : SUCCESS in 43m 13s","accounts_in_message":[],"_revision_number":9},{"id":"cfaa37fd4f434f17d3f8055ab9d543513a5be1fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-04-29 09:40:19.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"991813e401e66d83091de34430c100e96d5ff02b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 10:04:27.000000000","message":"Patch Set 10: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/915492 is needed.","accounts_in_message":[],"_revision_number":10},{"id":"c22e24e4bb4809c12b03ea868d94621dde548592","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-06-05 09:48:02.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"11938b06641308631ad4707d6abd8de1557040d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 12:24:44.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5bad069d4360441eaaa831d4d0a7bff2\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/df1c2160b7d946aab636930b279fe9e0 : SUCCESS in 2m 48s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/94377e42a3c444e2a544f6935bb51783 : TIMED_OUT in 2h 34m 33s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/6bc57087bde44f2fa4c6cc59d80a49c3 : SUCCESS in 1h 33m 54s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-etcd https://zuul.opendev.org/t/openstack/build/a48b036e3b7446d386197ca4f046b0c1 : SUCCESS in 49m 12s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-etcd https://zuul.opendev.org/t/openstack/build/f660877a03764a23a79711552857b055 : SUCCESS in 49m 53s","accounts_in_message":[],"_revision_number":10},{"id":"281c7a45f40f97793000323d8a02c3d5980acf0d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-06-05 12:54:38.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"eab6ca4e2ebb297c9c51904696744f8c03ea0f73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 14:57:54.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cfa37f689398453b8889c6572a7a4a15\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/78d568a3f208444cb02ad4a2e324a061 : SUCCESS in 1m 45s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/759baa1770974af8b7adc5fc859ef60e : SUCCESS in 2h 01m 53s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/d099a196d82b4803bb822af647875e08 : SUCCESS in 1h 09m 23s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-etcd https://zuul.opendev.org/t/openstack/build/6c57c551a84d4c8a9598a9856e5654ee : SUCCESS in 39m 25s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-etcd https://zuul.opendev.org/t/openstack/build/37a2c4b83470491996e899ac434182db : SUCCESS in 37m 51s","accounts_in_message":[],"_revision_number":10},{"id":"46d8737af0dbab27b08de36e66d6856c1e620dc6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-06-10 13:59:01.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"81408b2d9651f7fb81c83f918557694a227d01d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-10 16:18:48.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7f9c80eb9504ed98c31f740fa14c65c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ec4a9131d71a4f5b9282a3a2a139c44e : SUCCESS in 1m 56s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/93ad05304b884a05a21a6e79bdf19088 : SUCCESS in 2h 17m 51s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/5b76853adaba4c1c9b3178e5ba2901b4 : SUCCESS in 1h 13m 46s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-etcd https://zuul.opendev.org/t/openstack/build/6e6044494387406dac43d088ac62eef7 : SUCCESS in 42m 45s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-etcd https://zuul.opendev.org/t/openstack/build/839a18fded3d4069a71b218df350f541 : SUCCESS in 42m 00s","accounts_in_message":[],"_revision_number":11},{"id":"f0269298964f7cb2f1d1eff921188ffc31ff12e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-06-11 11:53:42.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"1ce8ae35101367a47f10fbd9ebc6b4b0de6651f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-11 12:08:51.000000000","message":"Patch Set 12: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/915492 is needed.","accounts_in_message":[],"_revision_number":12},{"id":"8d243c1d71e7fcc4e7f5e0af07ce35b5a0f6d626","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-07-05 07:06:04.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"b2cda8ac599b79ba58a5e0838f66b43de5042af8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 09:41:59.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1388f4fd9fad45bfb692846456f09a93\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e2ea59221ca041d9b1a7b76795ac3e96 : SUCCESS in 2m 09s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b97a3161eb7941a0a49ca1091ab53874 : SUCCESS in 2h 26m 18s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/7e3477832ca0456ab75493a326403e27 : SUCCESS in 1h 19m 35s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-etcd https://zuul.opendev.org/t/openstack/build/4141c48ba23447468acfabdb2c3e4389 : TIMED_OUT in 2h 33m 12s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-etcd https://zuul.opendev.org/t/openstack/build/64bdf59068344f069264bb71dbe6a063 : TIMED_OUT in 2h 33m 26s","accounts_in_message":[],"_revision_number":12},{"id":"fb88673da21a83a9f8021ef145bb8af84d3e50b2","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-07-08 06:22:51.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"c4565ce63e0d1872662f84fa960e4b78ddac5e37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-08 08:56:56.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bfaf7861c7a54133aa8be2851fb69732\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4c3b768f97ae40ecbbd5415332ddf6f1 : SUCCESS in 1m 54s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/4336daa2fc05410faa442c9d49d0d2e9 : SUCCESS in 1h 50m 20s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/bfa20fbc3c5d4523aea476679561d330 : SUCCESS in 57m 48s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-etcd https://zuul.opendev.org/t/openstack/build/121b9f0811894ee4b5337ebd91273319 : TIMED_OUT in 2h 32m 44s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-etcd https://zuul.opendev.org/t/openstack/build/ba13a60e2f394495bb183482cf0303bd : SUCCESS in 2h 16m 53s","accounts_in_message":[],"_revision_number":12},{"id":"93e9d615f6aeee059221abb004b1ebd20c79a207","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-07-09 07:32:02.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"eba93fabb78f6bb0f3c42dda160174e4aa15e239","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 09:22:20.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c10113813e04d6f8e91ec70365825fc\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/41c43878a45a400c9a28f4956d9583e4 : SUCCESS in 1m 49s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/536fbe944b214234a4d36dd2b960d808 : SUCCESS in 1h 46m 17s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ed380153f2f5424780c9aab76100dc6b : SUCCESS in 1h 01m 54s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-etcd https://zuul.opendev.org/t/openstack/build/141fd422edc6466d9fcbc9ae899bc1fc : SUCCESS in 1h 47m 45s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-etcd https://zuul.opendev.org/t/openstack/build/af1941e710bc4122a19b9a736e4a989a : SUCCESS in 1h 06m 06s","accounts_in_message":[],"_revision_number":13},{"id":"628b6a32212125f5daa5e925310c4d40bb7e6dde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-07-09 15:32:39.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"0ae04c93bf2f96fc90109b75e9b8debff297a811","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 18:10:44.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9cf290b1d7974c04845fd6089e53028f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ac231b49ab954c1fb70eb219a8ec9ceb : SUCCESS in 1m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/187b585ecca040b4b84e775d3ecd9101 : SUCCESS in 11m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38d34f8442224a7f9c7ac6391c40a196 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da90fadc035c461fa5aa610c4e559a68 : SUCCESS in 4m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/484634f6a1a745e68c22dc362baf66ba : SUCCESS in 5m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1fcbb9a5dfbc440f80d58d5ba708c42b : SUCCESS in 8m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/040713dca5d8408583d51319e16fe4ff : SUCCESS in 11m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd543f5090af4fac90e21f644167c2ff : SUCCESS in 5m 49s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/b41761d2d89c465a8f0ea8eb8912ea59 : SUCCESS in 5m 05s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/ce80f93fde144673ba5a8e6699d04d04 : SUCCESS in 9m 45s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f4864ec111bd4cb38869ab4c55a233bb : SUCCESS in 4m 18s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1f1845c047de4231a7ed75219a10d8cc : SUCCESS in 9m 39s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/481e3b1ab3504822819ef68eb1e0f717 : SUCCESS in 6m 44s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3df64585ec694548a2cb58cd3046eb03 : SUCCESS in 10m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1c6b25739f50492f94d516a4d11aea7b : SUCCESS in 1h 21m 33s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/977f22562699497191b549b377d46d91 : SUCCESS in 2h 01m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/74126573f02343d9b738976fdc784e98 : SUCCESS in 1h 22m 15s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e63e941783a2466fae133a5c8871cc6a : SUCCESS in 1h 54m 41s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/1bbddb5dc1da4fe29d640bd1306d76d0 : SUCCESS in 1h 11m 48s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-etcd https://zuul.opendev.org/t/openstack/build/0aaeab76228940dea817ae8c8d41d273 : TIMED_OUT in 2h 34m 05s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-etcd https://zuul.opendev.org/t/openstack/build/fec9bc25185e4c1fa60ee47ed4454d7e : SUCCESS in 2h 12m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d865aaad11fe44038cf1b9b7d79e34a7 : SUCCESS in 43m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/16b27b20525e41259cd744e8e6017ae9 : SUCCESS in 55m 21s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/0624835254d844e1857781d77205245c : SUCCESS in 1h 01m 23s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e014aea1b60a4ad99dbbdb3121e4c976 : SUCCESS in 41m 34s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/94fb781b65734ca3a1a75a9f476e57bf : TIMED_OUT in 2h 34m 28s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c7b7eac08ece402a9cd8ca89885c5db5 : SUCCESS in 2h 21m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0e7ec73a65c445e2841e57636d86e81c : FAILURE in 32m 20s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/f02201b0cb4a409d8704577658826db9 : TIMED_OUT in 2h 35m 07s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"0720fa953537b00b792686ee7bff82e57f6d4485","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-07-10 14:33:58.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"6e75d8a68eed65057da65bf5607bd5c0d8906957","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 17:43:00.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/780998ef84da4788acf30c06e2a3bf79\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6c2f6f581d5d47f59c9fd7fcfb0dd90b : SUCCESS in 1m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a6c2022845947ddbf1207f68fac1140 : SUCCESS in 9m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4adb5d449b2c4341928edd405d90d51c : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6004278d213146a19ea8b7b61b5da4bf : SUCCESS in 4m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2f95caf51cfb4c9988e3d30a87e95e92 : SUCCESS in 5m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/83f7668bdd91422a8f125372ab1748a3 : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/389c808a43e541bda8fe60d6b5a5173b : SUCCESS in 10m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a5736fc695a845a080c1470c2cd36344 : SUCCESS in 5m 16s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/10a3a83f82f84ae88cc9dd33abe50164 : SUCCESS in 5m 35s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/df09ab3a7f6841ce8d5b85e35ecf2eb2 : SUCCESS in 8m 28s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/11581bc9078d464ca811b5076c679f3d : SUCCESS in 4m 20s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b0035a17692541aa8942988086648818 : SUCCESS in 7m 08s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1e6e644da1c948588a600f6342be3f37 : SUCCESS in 7m 23s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d0b59fbdba5641c7957155710cc31df6 : SUCCESS in 7m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/406a5821460c46559b840ff69758ea00 : SUCCESS in 1h 30m 06s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/f9ceac86ddf748f591491636f87f52b8 : SUCCESS in 1h 42m 29s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/b3c2bc1502e141b6bea71a6e005e238b : SUCCESS in 1h 23m 45s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/6986f758eb0043ec9c0f5b46222405ef : SUCCESS in 1h 46m 27s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/4f3c2aee89b04072961e18140aae0a9d : SUCCESS in 1h 24m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9823efa73b334ea9ac2f48dea5c71e0f : SUCCESS in 32m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e62c0841e5a040b5b15f38b15b448ca4 : SUCCESS in 1h 04m 58s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/331674e0e7fd41468ad42f0cbeecd458 : SUCCESS in 1h 10m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6a2f6de4c3294e46bbf30b2998978121 : SUCCESS in 49m 22s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e9d33dd46688470fb362cbaa54104cf4 : TIMED_OUT in 2h 34m 37s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a3485abbbf074a85b9f71a3f43f87aae : TIMED_OUT in 3h 05m 42s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/09b118693c244c748d7df4daf65f1c52 : FAILURE in 26m 21s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/cc20868d5d2745219e6f2d86bd4eb178 : TIMED_OUT in 2h 34m 49s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"2255deb511fc0dcc43a481da0c53298d96c2ec77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-08-02 15:27:43.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"df427b9045920a5eca61ab65d51967b616afcba6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-02 17:17:32.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63a996b628c04299856fac61769829b5\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2dffc1bdfe0e4f698154de92d22bfc12 : SUCCESS in 1m 45s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-etcd https://zuul.opendev.org/t/openstack/build/22f6c33e1b174666a4c333eb04cf5b85 : SUCCESS in 1h 48m 30s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-etcd https://zuul.opendev.org/t/openstack/build/7936650ac9284fd7a6c8ca6b7a66f333 : SUCCESS in 57m 52s","accounts_in_message":[],"_revision_number":16},{"id":"48fd6fd2dcc7c05a81adbd06aa74f5dad76c8443","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-08-05 09:16:24.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"a31a60e1adaf3e87cdaf8d851a801b729ddd19e2","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-08-05 09:17:18.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"10df0629cdf9dfb972e5da9a010f179d6336b2a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-05 12:20:48.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae24b1934a8e425fad53e39340bc6eb6\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ce39755f92dc40228c142d797d01b803 : SUCCESS in 1m 36s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e02a52d742eb47678a51aa261ad56f4d : SUCCESS in 1m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b85bb7def5b462988c4fc35bb4ead06 : SUCCESS in 16m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11749e1b416b4506a0a6d4a7bdc042fc : SUCCESS in 6m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/76add272845d4dfebff5cd07e263f24c : SUCCESS in 4m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7023a3bad69b4ed4b140595b2e0b88d8 : SUCCESS in 4m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/97ad07a64e494b18a25d337d3ff250cf : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3dce8755694d4469b8ebcf1514acae12 : SUCCESS in 10m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da048da4c3334046b992005d1c24ac50 : SUCCESS in 4m 57s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/36ee2c853f9448d491e050f388e58db4 : SUCCESS in 4m 29s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/30053fbf669d4583bbd595b0e502b4fd : SUCCESS in 5m 25s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d63e8ed8b58145aca005f4b74cc2e4f1 : SUCCESS in 6m 25s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8785cf15848747d0af703a6165b9385a : SUCCESS in 6m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f7fd1df1dc5c47d188a4011e079fab27 : SUCCESS in 6m 45s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/6a9a78f70a0942769c6a97ee0b85835a : SUCCESS in 6m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a37965234e2548949b4a7fcb7a0209df : SUCCESS in 1h 50m 34s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/3c58e38ae292463bb239774376bb07b5 : SUCCESS in 2h 20m 27s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/c13d855e3b7d47e9a589ff5198df7c36 : SUCCESS in 1h 05m 00s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/12ea9b130f784213bc571c770f3b0a71 : SUCCESS in 2h 19m 32s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/cb67f6e668f64484909f40bcfdbb6a34 : SUCCESS in 1h 11m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/62cebf28c1854bb2b90c2ee62755497c : SUCCESS in 50m 40s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d3bbae7b366d4f939cfe95660333d7e8 : SUCCESS in 54m 39s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/eae37aa2b6474d418b16919dae81a439 : FAILURE in 14m 07s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0de92260d61f44108d9c0170dd3b5631 : SUCCESS in 39m 49s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/440093e81ac04a32b21433123810c6b0 : TIMED_OUT in 2h 34m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/48928d59d00d4beba5d045912499248d : SUCCESS in 3h 00m 42s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a2fae5c49b1d4a3d9bced937282e2b1a : TIMED_OUT in 2h 34m 06s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/8701ff1e202543158c53b784b220d7ab : FAILURE in 8m 42s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"bb357388b9bde1da2f0d5fbfcec26fe613c6beb2","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-08-14 10:51:06.000000000","message":"Patch Set 17: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"3298bb7138ac99f1d407d8e8ecad8df79fb52dad","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-08-14 11:13:23.000000000","message":"Patch Set 17: Code-Review+2","accounts_in_message":[],"_revision_number":17},{"id":"097178368d9228a494d19d1db24b5f53fa7c7bd7","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-08-15 22:06:10.000000000","message":"Patch Set 17: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":17},{"id":"f2a62cf6747afbc4a960b25cbf22a57d2b80a3a5","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-08-21 16:18:27.000000000","message":"Patch Set 17: Workflow-1","accounts_in_message":[],"_revision_number":17},{"id":"a316a56d650e34f53eccaf8dcae2a3097e16d759","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2024-09-02 16:15:15.000000000","message":"Patch Set 17: Code-Review+1","accounts_in_message":[],"_revision_number":17},{"id":"d2d3c3c207e3dfb7d94f1f15d474dbc16785bf1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-09-16 07:51:01.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":18},{"id":"38e3693c3b0925d4250e2272168fa5dfa61afec2","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-09-16 07:51:17.000000000","message":"Patch Set 17: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"a9bcf1530f59e101c1008763b827dee10b3fa196","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-09-16 08:48:58.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"b6110dfe586366399033ab1372219970a46736d1","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-09-16 10:51:27.000000000","message":"Patch Set 19: Code-Review+2","accounts_in_message":[],"_revision_number":19},{"id":"6a3ba95e43823a97f23bb529f82401447ea11659","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-16 11:26:16.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21d459190035472b94b84b6702d83ca2\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/17df11f5d182451ca8524c7e749813da : SUCCESS in 2m 13s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3ccaf9e624b440a4bf021a3c26aa6f9b : SUCCESS in 2m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f5ef749e660f410d9f850c06f737bbef : SUCCESS in 10m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bc9965fa910403799c3df9997f9a803 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce890103597f4c45b741dbc0eb52fd6b : SUCCESS in 4m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/74331d5738494c4397d91f9c44e7bd42 : SUCCESS in 4m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8c570ba6384c479ea907847124d2752f : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7b1578762da4070abb32c205274008d : SUCCESS in 11m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/baf789d2efe0435ba626160a2c791e0f : SUCCESS in 4m 55s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/f8fd4d8dd01b41d798d3882e2e0a10f5 : SUCCESS in 4m 47s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/0201083a6cef4ff0973eb6be427456bc : SUCCESS in 6m 49s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/95bf85df932d4f9285f0f4130068b783 : SUCCESS in 4m 00s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/818252453e4f4478bdf4c17edd64a5c0 : SUCCESS in 7m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d4bf584aa0894f3bacbd52e2314db8d3 : SUCCESS in 8m 16s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2d19b0878de3451eb5a7300d0089a03b : SUCCESS in 6m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4b6e6acb0902420ea74ba1625801f48a : SUCCESS in 1h 54m 33s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/32b813a9f76e4da595d9b4ab2e21a53d : SUCCESS in 1h 43m 43s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/45410795aa2e42fa859414f2d481cf1e : SUCCESS in 55m 07s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c57e7bb0e128409eb5ae32e93062e16d : SUCCESS in 1h 47m 06s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/9a645aa1cc2046f3ba1315ff88fc58c0 : SUCCESS in 1h 11m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/777ba8e18387444188d526881fd053bc : SUCCESS in 33m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/faee008de8e840a8b0e6225f7dc47f7a : SUCCESS in 38m 49s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/9f6f91fafc0e412094ad16e5df4e53fa : SUCCESS in 56m 19s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/66c2bd694c384fe695abfb51ef75d346 : SUCCESS in 29m 35s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5279f35e66f84a3ca619d9fd4dee5dec : SUCCESS in 2h 29m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/496522d5ae134596b7ae518f3569b093 : SUCCESS in 2h 31m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d570b30f1b794a3baa10f95f2df94f8b : SUCCESS in 2h 19m 19s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/612f4be41e534691ae5941ab4aefd374 : SUCCESS in 2h 27m 33s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"1c213df54ffa8a602f6a49104fbab96ed3a4efac","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-09-16 15:01:03.000000000","message":"Patch Set 19: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"bcaf123d9a287bba4bb97075818c9d8d83644fcd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-16 15:02:22.000000000","message":"Patch Set 19: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":19},{"id":"500e52b55395cd2c2864914eecf18d8fea7d3ec4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-16 17:11:26.000000000","message":"Patch Set 19: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1a0bead0ec94a6bab383719d1fff42c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6cf14351db834499aaeb8ba29dc8b5fe : SUCCESS in 1m 54s\n- requirements-check https://zuul.opendev.org/t/openstack/build/844abcf498094de59d29a55ccf12cac1 : SUCCESS in 1m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93698ffd73b9487f935edb3c9e94fd90 : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f1f19066fbc4640bed2862822a5aaa4 : SUCCESS in 4m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/77ff3fe1f06e47c496badd2f1a1c56c9 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef108725d9db4f2597a7761f850ee32d : SUCCESS in 11m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2fb52c37dd434f31a35b6c315383371f : SUCCESS in 6m 38s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/25267f554573408788cb93a634dd000d : SUCCESS in 4m 54s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c916582efa4e45ca85fbab956a5e5d9e : SUCCESS in 6m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a54b2b3f6380488e98b76dc6871b61b5 : SUCCESS in 1h 27m 37s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/abb415b75f8e4c19ab3fad2cff528dea : SUCCESS in 2h 07m 55s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/681beeccf626425eaf6bbb0af10c40fd : SUCCESS in 1h 18m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/01181b7ae85d40ea9e0e8a96b674c4fb : SUCCESS in 45m 50s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f0eaf64da85146e5bab9ff56d11ff5c3 : SUCCESS in 1h 04m 48s","accounts_in_message":[],"_revision_number":19},{"id":"0eeee07426fa5b91f4b38508a0a0be1c52788aa4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-16 17:11:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":19},{"id":"55ad5829ed6312b9c3aa5ce7ee9fa0ba915cc066","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-16 17:14:17.000000000","message":"Patch Set 19:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/936b1e0749ab47209c6b77df4c605421\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c99b214d60d24505b48afd48d4342fc4 : SUCCESS in 1m 03s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/887104db685e44da8774d9f96365c6d4 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":19}],"current_revision_number":19,"current_revision":"56cbf437b67265ce028a166019e5ad4f7d6a7417","revisions":{"b51dea1cc666e826b1b2f519bef09b15a3205af4":{"kind":"REWORK","_number":1,"created":"2024-04-15 08:31:03.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/34/915834/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/34/915834/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/34/915834/1"}}},"commit":{"parents":[{"commit":"85cfb6c2ae974d110712b5ea673cdafcfecb9e69","subject":"Fix negative or 0 limit parameter in pagination","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/85cfb6c2ae974d110712b5ea673cdafcfecb9e69"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-04-15 08:19:29.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-04-15 08:19:29.000000000","tz":-240},"subject":"WIP Initial support for Jobboard etcd backend","message":"WIP Initial support for Jobboard etcd backend\n\nDepends-On: https://review.opendev.org/c/openstack/taskflow/+/915492\nChange-Id: Ib557a9cf938fcf4257d2c2848fff78a62d82109a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b51dea1cc666e826b1b2f519bef09b15a3205af4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b51dea1cc666e826b1b2f519bef09b15a3205af4"}]},"branch":"refs/heads/master"},"a229af827eaa26f41bde9e0a3bcd1ce58dc007cd":{"kind":"REWORK","_number":2,"created":"2024-04-15 09:01:48.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/34/915834/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/34/915834/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/34/915834/2"}}},"commit":{"parents":[{"commit":"85cfb6c2ae974d110712b5ea673cdafcfecb9e69","subject":"Fix negative or 0 limit parameter in pagination","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/85cfb6c2ae974d110712b5ea673cdafcfecb9e69"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-04-15 08:19:29.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-04-15 08:51:16.000000000","tz":-240},"subject":"WIP Initial support for Jobboard etcd backend","message":"WIP Initial support for Jobboard etcd backend\n\nDepends-On: https://review.opendev.org/c/openstack/taskflow/+/915492\nChange-Id: Ib557a9cf938fcf4257d2c2848fff78a62d82109a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a229af827eaa26f41bde9e0a3bcd1ce58dc007cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a229af827eaa26f41bde9e0a3bcd1ce58dc007cd"}]},"branch":"refs/heads/master"},"8fa3afedf4eef57a65c2a7ace88132339aa426c1":{"kind":"REWORK","_number":3,"created":"2024-04-15 09:20:42.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/34/915834/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/34/915834/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/34/915834/3"}}},"commit":{"parents":[{"commit":"85cfb6c2ae974d110712b5ea673cdafcfecb9e69","subject":"Fix negative or 0 limit parameter in pagination","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/85cfb6c2ae974d110712b5ea673cdafcfecb9e69"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-04-15 08:19:29.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-04-15 09:10:07.000000000","tz":-240},"subject":"WIP Initial support for Jobboard etcd backend","message":"WIP Initial support for Jobboard etcd backend\n\nDepends-On: https://review.opendev.org/c/openstack/taskflow/+/915492\nChange-Id: Ib557a9cf938fcf4257d2c2848fff78a62d82109a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8fa3afedf4eef57a65c2a7ace88132339aa426c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8fa3afedf4eef57a65c2a7ace88132339aa426c1"}]},"branch":"refs/heads/master"},"a7f4613067f115f03959b9131a2bc66cd5233977":{"kind":"REWORK","_number":4,"created":"2024-04-15 11:58:05.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/34/915834/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/34/915834/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/34/915834/4"}}},"commit":{"parents":[{"commit":"85cfb6c2ae974d110712b5ea673cdafcfecb9e69","subject":"Fix negative or 0 limit parameter in pagination","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/85cfb6c2ae974d110712b5ea673cdafcfecb9e69"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-04-15 08:19:29.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-04-15 11:47:27.000000000","tz":-240},"subject":"WIP Initial support for Jobboard etcd backend","message":"WIP Initial support for Jobboard etcd backend\n\nDepends-On: https://review.opendev.org/c/openstack/taskflow/+/915492\nChange-Id: Ib557a9cf938fcf4257d2c2848fff78a62d82109a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a7f4613067f115f03959b9131a2bc66cd5233977"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a7f4613067f115f03959b9131a2bc66cd5233977"}]},"branch":"refs/heads/master"},"b8bba08231709ece7cdcf35b3af9755993658729":{"kind":"REWORK","_number":5,"created":"2024-04-15 13:34:45.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/34/915834/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/34/915834/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/34/915834/5"}}},"commit":{"parents":[{"commit":"85cfb6c2ae974d110712b5ea673cdafcfecb9e69","subject":"Fix negative or 0 limit parameter in pagination","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/85cfb6c2ae974d110712b5ea673cdafcfecb9e69"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-04-15 08:19:29.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-04-15 13:24:12.000000000","tz":-240},"subject":"WIP Initial support for Jobboard etcd backend","message":"WIP Initial support for Jobboard etcd backend\n\nDepends-On: https://review.opendev.org/c/openstack/taskflow/+/915492\nChange-Id: Ib557a9cf938fcf4257d2c2848fff78a62d82109a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b8bba08231709ece7cdcf35b3af9755993658729"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b8bba08231709ece7cdcf35b3af9755993658729"}]},"branch":"refs/heads/master"},"506e8aa0f997589dcce96b3602e54cd2c48c5d9c":{"kind":"REWORK","_number":6,"created":"2024-04-15 14:47:45.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/34/915834/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/34/915834/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/34/915834/6"}}},"commit":{"parents":[{"commit":"85cfb6c2ae974d110712b5ea673cdafcfecb9e69","subject":"Fix negative or 0 limit parameter in pagination","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/85cfb6c2ae974d110712b5ea673cdafcfecb9e69"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-04-15 08:19:29.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-04-15 14:37:03.000000000","tz":-240},"subject":"WIP Initial support for Jobboard etcd backend","message":"WIP Initial support for Jobboard etcd backend\n\nDepends-On: https://review.opendev.org/c/openstack/taskflow/+/915492\nChange-Id: Ib557a9cf938fcf4257d2c2848fff78a62d82109a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/506e8aa0f997589dcce96b3602e54cd2c48c5d9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/506e8aa0f997589dcce96b3602e54cd2c48c5d9c"}]},"branch":"refs/heads/master"},"7d69c39096866a1528caa76efe8c09a4090c6aca":{"kind":"REWORK","_number":7,"created":"2024-04-25 12:56:11.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/34/915834/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/34/915834/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/34/915834/7"}}},"commit":{"parents":[{"commit":"85cfb6c2ae974d110712b5ea673cdafcfecb9e69","subject":"Fix negative or 0 limit parameter in pagination","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/85cfb6c2ae974d110712b5ea673cdafcfecb9e69"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-04-15 08:19:29.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-04-25 12:55:36.000000000","tz":-240},"subject":"Support for Jobboard etcd backend","message":"Support for Jobboard etcd backend\n\nTODO: docs\n\nDepends-On: https://review.opendev.org/c/openstack/taskflow/+/915492\nChange-Id: Ib557a9cf938fcf4257d2c2848fff78a62d82109a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7d69c39096866a1528caa76efe8c09a4090c6aca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7d69c39096866a1528caa76efe8c09a4090c6aca"}]},"branch":"refs/heads/master"},"d53a44b6915c9dd1efc1e1913039f6a4d56e6481":{"kind":"REWORK","_number":8,"created":"2024-04-25 15:56:10.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/34/915834/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/34/915834/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/34/915834/8"}}},"commit":{"parents":[{"commit":"85cfb6c2ae974d110712b5ea673cdafcfecb9e69","subject":"Fix negative or 0 limit parameter in pagination","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/85cfb6c2ae974d110712b5ea673cdafcfecb9e69"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-04-15 08:19:29.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-04-25 15:55:36.000000000","tz":-240},"subject":"Support for Jobboard etcd backend","message":"Support for Jobboard etcd backend\n\nTODO: docs\n\nDepends-On: https://review.opendev.org/c/openstack/taskflow/+/915492\nChange-Id: Ib557a9cf938fcf4257d2c2848fff78a62d82109a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d53a44b6915c9dd1efc1e1913039f6a4d56e6481"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d53a44b6915c9dd1efc1e1913039f6a4d56e6481"}]},"branch":"refs/heads/master"},"01a503911fd4012498916e3ae6a06e10bfc5f838":{"kind":"REWORK","_number":9,"created":"2024-04-26 13:23:43.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/34/915834/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/34/915834/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/34/915834/9"}}},"commit":{"parents":[{"commit":"33c755d753c6a7502aede5901b3e7ea75fa0203c","subject":"Merge \"Updates for new Python runtimes in 2024.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/33c755d753c6a7502aede5901b3e7ea75fa0203c"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-04-15 08:19:29.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-04-26 13:22:10.000000000","tz":-240},"subject":"Support for Jobboard etcd backend","message":"Support for Jobboard etcd backend\n\nTODO: docs\n\nDepends-On: https://review.opendev.org/c/openstack/taskflow/+/915492\nChange-Id: Ib557a9cf938fcf4257d2c2848fff78a62d82109a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/01a503911fd4012498916e3ae6a06e10bfc5f838"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/01a503911fd4012498916e3ae6a06e10bfc5f838"}]},"branch":"refs/heads/master"},"a28d96961b1191f608e72dc0392c5b6ab8df181f":{"kind":"REWORK","_number":10,"created":"2024-04-29 09:40:19.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/34/915834/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/34/915834/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/34/915834/10"}}},"commit":{"parents":[{"commit":"33c755d753c6a7502aede5901b3e7ea75fa0203c","subject":"Merge \"Updates for new Python runtimes in 2024.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/33c755d753c6a7502aede5901b3e7ea75fa0203c"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-04-15 08:19:29.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-04-29 09:39:30.000000000","tz":-240},"subject":"Support for Jobboard etcd backend","message":"Support for Jobboard etcd backend\n\nTODO: docs\n\nDepends-On: https://review.opendev.org/c/openstack/taskflow/+/915492\nChange-Id: Ib557a9cf938fcf4257d2c2848fff78a62d82109a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a28d96961b1191f608e72dc0392c5b6ab8df181f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a28d96961b1191f608e72dc0392c5b6ab8df181f"}]},"branch":"refs/heads/master"},"aa4be8a0d743950c3fa540c68de748f2cac60196":{"kind":"REWORK","_number":11,"created":"2024-06-10 13:59:01.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/34/915834/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/34/915834/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/34/915834/11"}}},"commit":{"parents":[{"commit":"fcc61e9ca8e79674cf641eefcb65972e60a25e7d","subject":"Merge \"Replace use of deprecated cert.not_valid_after\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fcc61e9ca8e79674cf641eefcb65972e60a25e7d"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-04-15 08:19:29.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-06-10 13:58:37.000000000","tz":120},"subject":"Support for Jobboard etcd backend","message":"Support for Jobboard etcd backend\n\nTODO: docs\n\nDepends-On: https://review.opendev.org/c/openstack/taskflow/+/915492\nChange-Id: Ib557a9cf938fcf4257d2c2848fff78a62d82109a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aa4be8a0d743950c3fa540c68de748f2cac60196"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aa4be8a0d743950c3fa540c68de748f2cac60196"}]},"branch":"refs/heads/master"},"9529eacd5ed07bb5897dd4fb172bc0c4d6d0f511":{"kind":"REWORK","_number":12,"created":"2024-06-11 11:53:42.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/34/915834/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/34/915834/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/34/915834/12"}}},"commit":{"parents":[{"commit":"fcc61e9ca8e79674cf641eefcb65972e60a25e7d","subject":"Merge \"Replace use of deprecated cert.not_valid_after\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fcc61e9ca8e79674cf641eefcb65972e60a25e7d"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-04-15 08:19:29.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-06-11 11:53:04.000000000","tz":120},"subject":"WIP Support for Jobboard etcd backend","message":"WIP Support for Jobboard etcd backend\n\nTODO: docs\n\nDepends-On: https://review.opendev.org/c/openstack/taskflow/+/915492\nChange-Id: Ib557a9cf938fcf4257d2c2848fff78a62d82109a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9529eacd5ed07bb5897dd4fb172bc0c4d6d0f511"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9529eacd5ed07bb5897dd4fb172bc0c4d6d0f511"}]},"branch":"refs/heads/master"},"c3b9f99d79cc1bbf28e64d45fbc4c3b51e138267":{"kind":"REWORK","_number":13,"created":"2024-07-09 07:32:02.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/34/915834/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/34/915834/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/34/915834/13"}}},"commit":{"parents":[{"commit":"fcc61e9ca8e79674cf641eefcb65972e60a25e7d","subject":"Merge \"Replace use of deprecated cert.not_valid_after\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fcc61e9ca8e79674cf641eefcb65972e60a25e7d"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-04-15 08:19:29.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-07-09 07:30:18.000000000","tz":120},"subject":"WIP Support for Jobboard etcd backend","message":"WIP Support for Jobboard etcd backend\n\nIncludes a workaround for etcd3gw\n\nTODO:\n- docs\n- remove workaround\n\nDepends-On: https://review.opendev.org/c/openstack/taskflow/+/915492\nChange-Id: Ib557a9cf938fcf4257d2c2848fff78a62d82109a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c3b9f99d79cc1bbf28e64d45fbc4c3b51e138267"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c3b9f99d79cc1bbf28e64d45fbc4c3b51e138267"}]},"branch":"refs/heads/master"},"b17bdc2755c4034d99229ceed5b672a45b81007b":{"kind":"REWORK","_number":14,"created":"2024-07-09 15:32:39.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/34/915834/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/34/915834/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/34/915834/14"}}},"commit":{"parents":[{"commit":"7974e755e18bed376ea5b1bce2a927eeb34494b2","subject":"Merge \"jobboard: Support authentication and SSL for Redis Sentinel\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7974e755e18bed376ea5b1bce2a927eeb34494b2"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-04-15 08:19:29.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-07-09 15:28:08.000000000","tz":120},"subject":"Support for Jobboard etcd backend","message":"Support for Jobboard etcd backend\n\n- add a new etcd_taskflow_driver for the octavia worker\n- enable etcd in devstack\n\nDepends-On: https://review.opendev.org/c/openstack/taskflow/+/915492\nChange-Id: Ib557a9cf938fcf4257d2c2848fff78a62d82109a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b17bdc2755c4034d99229ceed5b672a45b81007b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b17bdc2755c4034d99229ceed5b672a45b81007b"}]},"branch":"refs/heads/master"},"23586af88497e49ef3bc8b24157d2ff0330bf055":{"kind":"REWORK","_number":15,"created":"2024-07-10 14:33:58.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/34/915834/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/34/915834/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/34/915834/15"}}},"commit":{"parents":[{"commit":"7974e755e18bed376ea5b1bce2a927eeb34494b2","subject":"Merge \"jobboard: Support authentication and SSL for Redis Sentinel\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7974e755e18bed376ea5b1bce2a927eeb34494b2"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-04-15 08:19:29.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-07-10 14:33:32.000000000","tz":120},"subject":"Support for Jobboard etcd backend","message":"Support for Jobboard etcd backend\n\n- add a new etcd_taskflow_driver for the octavia worker\n- enable etcd in devstack\n- add experimental jobs for jobboard/etcd\n\nDepends-On: https://review.opendev.org/c/openstack/taskflow/+/915492\nChange-Id: Ib557a9cf938fcf4257d2c2848fff78a62d82109a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/23586af88497e49ef3bc8b24157d2ff0330bf055"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/23586af88497e49ef3bc8b24157d2ff0330bf055"}]},"branch":"refs/heads/master"},"7a0de513b0084c29898c889ac8a25c0fa065331a":{"kind":"REWORK","_number":16,"created":"2024-08-02 15:27:43.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/34/915834/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/34/915834/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/34/915834/16"}}},"commit":{"parents":[{"commit":"c2c5a4ce820abb9251556ca0f31c7c2ac34c66bb","subject":"Merge \"Add specs to resize a load balancer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c2c5a4ce820abb9251556ca0f31c7c2ac34c66bb"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-04-15 08:19:29.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-08-02 15:27:35.000000000","tz":120},"subject":"[WIP] Support for Jobboard etcd backend","message":"[WIP] Support for Jobboard etcd backend\n\n- add a new etcd_taskflow_driver for the octavia worker\n- bump taskflow and etcd3gw\n- enable etcd in devstack\n- add experimental jobs for jobboard/etcd\n\nChange-Id: Ib557a9cf938fcf4257d2c2848fff78a62d82109a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7a0de513b0084c29898c889ac8a25c0fa065331a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7a0de513b0084c29898c889ac8a25c0fa065331a"}]},"branch":"refs/heads/master"},"425a46eaf5fefe7bb54b2e667225442d09378038":{"kind":"REWORK","_number":17,"created":"2024-08-05 09:16:24.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/34/915834/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/34/915834/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/34/915834/17"}}},"commit":{"parents":[{"commit":"c2c5a4ce820abb9251556ca0f31c7c2ac34c66bb","subject":"Merge \"Add specs to resize a load balancer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c2c5a4ce820abb9251556ca0f31c7c2ac34c66bb"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-04-15 08:19:29.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-08-05 09:14:58.000000000","tz":120},"subject":"Support for Jobboard etcd backend","message":"Support for Jobboard etcd backend\n\n- add a new etcd_taskflow_driver for the octavia worker\n- bump taskflow (Etcd support) and etcd3gw (important fixes for the\n  backend)\n- enable etcd in devstack if requested\n- add experimental jobs for jobboard/etcd\n\nChange-Id: Ib557a9cf938fcf4257d2c2848fff78a62d82109a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/425a46eaf5fefe7bb54b2e667225442d09378038"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/425a46eaf5fefe7bb54b2e667225442d09378038"}]},"branch":"refs/heads/master"},"02b0e7973b531b2c26b0dba62012855dc3c650f6":{"kind":"REWORK","_number":18,"created":"2024-09-16 07:51:01.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/34/915834/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/34/915834/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/34/915834/18"}}},"commit":{"parents":[{"commit":"dda68b1cf6387170c3c68aad7dec82e6d84596a8","subject":"Merge \"Fix SR-IOV when VIP interface is used for members\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dda68b1cf6387170c3c68aad7dec82e6d84596a8"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-04-15 08:19:29.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-09-16 07:50:51.000000000","tz":120},"subject":"Support for Jobboard etcd backend","message":"Support for Jobboard etcd backend\n\n- add a new etcd_taskflow_driver for the octavia worker\n- bump taskflow (Etcd support) and etcd3gw (important fixes for the\n  backend)\n- enable etcd in devstack if requested\n- add experimental jobs for jobboard/etcd\n\nChange-Id: Ib557a9cf938fcf4257d2c2848fff78a62d82109a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/02b0e7973b531b2c26b0dba62012855dc3c650f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/02b0e7973b531b2c26b0dba62012855dc3c650f6"}]},"branch":"refs/heads/master"},"56cbf437b67265ce028a166019e5ad4f7d6a7417":{"kind":"REWORK","_number":19,"created":"2024-09-16 08:48:58.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/34/915834/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/34/915834/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/915834/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/34/915834/19"}}},"commit":{"parents":[{"commit":"dda68b1cf6387170c3c68aad7dec82e6d84596a8","subject":"Merge \"Fix SR-IOV when VIP interface is used for members\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dda68b1cf6387170c3c68aad7dec82e6d84596a8"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-04-15 08:19:29.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-09-16 08:48:53.000000000","tz":120},"subject":"Support for Jobboard etcd backend","message":"Support for Jobboard etcd backend\n\n- add a new etcd_taskflow_driver for the octavia worker\n- bump taskflow (Etcd support) and etcd3gw (important fixes for the\n  backend)\n- enable etcd in devstack if requested\n- add experimental jobs for jobboard/etcd\n\nChange-Id: Ib557a9cf938fcf4257d2c2848fff78a62d82109a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/56cbf437b67265ce028a166019e5ad4f7d6a7417"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/56cbf437b67265ce028a166019e5ad4f7d6a7417"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
