)]}'
{"id":"openstack%2Foctavia~915970","triplet_id":"openstack%2Foctavia~master~I502df0b5e819fc9f1fb8c3ca258bb35136ebf35f","project":"openstack/octavia","branch":"master","topic":"2024-2-runtimes","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-04-25 17:23:56.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2024-04-25 19:03:55.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I502df0b5e819fc9f1fb8c3ca258bb35136ebf35f","subject":"Updates for new Python runtimes in 2024.2","status":"MERGED","created":"2024-04-16 07:32:35.000000000","updated":"2024-04-25 19:06:26.000000000","submitted":"2024-04-25 19:03:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"915970-2024-2-runtimes","meta_rev_id":"9898a30782cd0c4e111d416891b405c4711af852","_number":915970,"virtual_id_number":915970,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-25 19:03:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-04-24 21:13:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2024-04-25 17:23:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2024-04-25 17:23:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-16 07:33:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-16 08:42:37.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2024-04-24 21:13:26.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-04-25 17:23:56.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"f1ecd2238558f0767e5fb44081d9a44bc7fa1f5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-16 07:32:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6863f7ce0bf0d704c98a6fded222d5d0232f8297","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 07:33:07.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/octavia on branch master.  The\nproblem was:\n\n  Job openstack-tox-functional-py312 not defined\n\nThe problem appears in the the \"opendev.org/openstack/octavia\" project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - periodic-stable-jobs-neutron\n        - openstack-cover-jobs\n        - openstack-python3-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n        - octavia-tox-tips\n  ...\n\n  in \"openstack/octavia/zuul.d/projects.yaml@master\", line 3","accounts_in_message":[],"_revision_number":1},{"id":"ca876621b14811e1095945052a1e043305d13bee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-16 07:40:58.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"16424c2e9d241833555d6fe245602e343537d3be","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-04-16 08:42:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"00af11bbd57331c9b5309c5e425c5cb1067829d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-16 09:00:57.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3d8559b6022192ff31a269e77090cf2c21b1e2ba","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-16 09:01:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d6377ea51c7fd5422c1a2cdbc0a7e6c5dda4f224","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 11:58:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bdb76bd8a5c2468f93479054184c3677\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c9c559f5fc3f41fb818bc1fdc7238ec7 : SUCCESS in 3m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c503335afe94c31badb8c7ce7e7a379 : SUCCESS in 11m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77b4c24f59f4467689e67d5b610a2be3 : SUCCESS in 5m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/feaadffa5af140c9bd4b3a9684cf7479 : SUCCESS in 5m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/76e0e59bcc3d4915bc6ef7684c9b325c : SUCCESS in 6m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d775abd6c9b64fa09fe828aa378b4631 : SUCCESS in 7m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16b0bdb0d0364e968a05dea8b3e4503d : SUCCESS in 11m 53s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/105579984d634441b440a61857d35be3 : SUCCESS in 5m 07s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/888dadf8262f4f1c928d3060a2d08c9f : SUCCESS in 4m 33s\n- octavia-tox-functional-py311-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/74939501ad0a452184479352c7d3ae67 : SUCCESS in 8m 57s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a32847a5dbc44d12b1d73085ee58b503 : SUCCESS in 4m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ac3d16f30c4e41759ac1d51c88b7c76b : SUCCESS in 8m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/419abbe0667545e9b1bea0f01a1e10d3 : SUCCESS in 7m 04s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/6b0f0bf6b3854e95bc750f2ee170c4f1 : SUCCESS in 7m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4b78fc681dd94132ba7e0f9ce211ecbc : SUCCESS in 1h 43m 23s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/67578cb2055449058b56498434cc19e9 : SUCCESS in 1h 37m 46s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e9817ded6ff14e4d8785a3d4a379d12e : SUCCESS in 1h 17m 50s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/7b3f338297fa4ac5b84167adbb782261 : SUCCESS in 1h 41m 30s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3ec0b643149c405bb9269d7604a49ef1 : SUCCESS in 59m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4905ded58e4b4bb89ddcbff626053420 : SUCCESS in 37m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8179d2101b07483698608a55724f7bac : SUCCESS in 59m 40s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/a17ff6455a3640bfa07d9836d5970c83 : FAILURE in 35m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/15316acb84424a15a7fb839c41da4964 : SUCCESS in 33m 22s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/17d699e975a0497b86326f5b88b2f0f9 : SUCCESS in 2h 02m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/35b05720661e49be8fa8365d4ac7a911 : SUCCESS in 2h 54m 44s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c67fed19a4914c84be316e099bfdc51c : FAILURE in 22m 51s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/e4cc5b46baec4f4ea7f43286b7767156 : FAILURE in 31m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"dad448d7decc834193d9122210526dd583187b3a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-04-24 21:13:26.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"dbff0457a6c44fa0b7e4f27b24eac240351127d4","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-04-25 17:23:56.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9686421bff21885b771b5c06ef50702f476f471d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 17:24:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"dbc40f06197c7f3f5cfb7431e00bdb5533478d6e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 19:03:55.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee2d29217c204ec1a23662ecd2df4113\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d144af53809d45c29c4872faf4b506ef : SUCCESS in 1m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9fe673341be4f799de8e5965963ebca : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7d1a35a442cb4eab9b2508c205da36a0 : SUCCESS in 4m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/61a98e3da5cd49de923b24549a351410 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62db728dfcd0412b8f0b16c2ef603c48 : SUCCESS in 11m 26s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ca774cb7a20c4fed96dc8e359a98262e : SUCCESS in 4m 18s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/cfd78fb61e0e4661a5353a3a78878475 : SUCCESS in 3m 58s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8dabe163aade45ab8b86d5fb84c05bc4 : SUCCESS in 1h 11m 21s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/0ca85044e05842819bdd46c0f90586e2 : SUCCESS in 1h 38m 50s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/dd5e638d13ed4147a7c77640401f23c4 : SUCCESS in 55m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/93d85bc2ed0b451c862ca76f8a06f94a : SUCCESS in 35m 49s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/05d8cf213861433280d97b9d6fe2b0cf : SUCCESS in 40m 14s","accounts_in_message":[],"_revision_number":3},{"id":"9502c22ba95704dddf735be2a1ba11c5dcf85eb9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 19:03:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"9898a30782cd0c4e111d416891b405c4711af852","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 19:06:26.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f4f3a2fdf81413dbde286ac26ad094a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72f650d3660349069cce5755687d5393 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"176fc1c4e7238c91b1ab14b4e035923e60c2f504","revisions":{"e3e7f19cd8eb5799047e0fd9ede2b396dbca3043":{"kind":"REWORK","_number":1,"created":"2024-04-16 07:32:35.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/70/915970/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/70/915970/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/915970/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/915970/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/915970/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/70/915970/1"}}},"commit":{"parents":[{"commit":"603d437aefc5234af5790933185c135c50553080","subject":"Merge \"Add pyupgrade hook to pre-commit config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/603d437aefc5234af5790933185c135c50553080"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-04-16 07:31:34.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-04-16 07:31:34.000000000","tz":120},"subject":"Updates for new Python runtimes in 2024.2","message":"Updates for new Python runtimes in 2024.2\n\nSee https://governance.openstack.org/tc/reference/runtimes/2024.2.html\n\nChange-Id: I502df0b5e819fc9f1fb8c3ca258bb35136ebf35f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e3e7f19cd8eb5799047e0fd9ede2b396dbca3043"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e3e7f19cd8eb5799047e0fd9ede2b396dbca3043"}]},"branch":"refs/heads/master"},"e6928d0c5f4510e4406bee62e440ab5f75c29039":{"kind":"REWORK","_number":2,"created":"2024-04-16 07:40:58.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/70/915970/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/70/915970/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/915970/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/915970/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/915970/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/70/915970/2"}}},"commit":{"parents":[{"commit":"603d437aefc5234af5790933185c135c50553080","subject":"Merge \"Add pyupgrade hook to pre-commit config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/603d437aefc5234af5790933185c135c50553080"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-04-16 07:31:34.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-04-16 07:40:44.000000000","tz":120},"subject":"Updates for new Python runtimes in 2024.2","message":"Updates for new Python runtimes in 2024.2\n\nSee https://governance.openstack.org/tc/reference/runtimes/2024.2.html\n\nChange-Id: I502df0b5e819fc9f1fb8c3ca258bb35136ebf35f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e6928d0c5f4510e4406bee62e440ab5f75c29039"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e6928d0c5f4510e4406bee62e440ab5f75c29039"}]},"branch":"refs/heads/master"},"176fc1c4e7238c91b1ab14b4e035923e60c2f504":{"kind":"REWORK","_number":3,"created":"2024-04-16 09:00:57.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/70/915970/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/70/915970/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/915970/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/915970/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/70/915970/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/70/915970/3"}}},"commit":{"parents":[{"commit":"603d437aefc5234af5790933185c135c50553080","subject":"Merge \"Add pyupgrade hook to pre-commit config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/603d437aefc5234af5790933185c135c50553080"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-04-16 07:31:34.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-04-16 09:00:49.000000000","tz":120},"subject":"Updates for new Python runtimes in 2024.2","message":"Updates for new Python runtimes in 2024.2\n\nSee https://governance.openstack.org/tc/reference/runtimes/2024.2.html\n\nChange-Id: I502df0b5e819fc9f1fb8c3ca258bb35136ebf35f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/176fc1c4e7238c91b1ab14b4e035923e60c2f504"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/176fc1c4e7238c91b1ab14b4e035923e60c2f504"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
