)]}'
{"id":"openstack%2Foctavia~916099","triplet_id":"openstack%2Foctavia~stable%2F2024.1~Ia106d81c1b2588e5d938d2238c8a2f6660bf5ef1","project":"openstack/octavia","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-04-22 15:36:30.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"167":{"account":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"last_update":"2024-04-23 10:37:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia106d81c1b2588e5d938d2238c8a2f6660bf5ef1","subject":"Fix fully-populated API with allowed_cidrs","status":"MERGED","created":"2024-04-19 11:57:33.000000000","updated":"2024-04-23 10:39:14.000000000","submitted":"2024-04-23 10:37:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"916099","cherry_pick_of_change":912721,"cherry_pick_of_patch_set":1,"meta_rev_id":"ca9d98dcfa32bf8f4179fa6bce4c37a06f476092","_number":916099,"virtual_id_number":916099,"owner":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-23 10:37:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2024-04-23 08:26:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-04-22 07:46:01.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":1,"date":"2024-04-23 08:26:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-19 11:57:33.000000000","updated_by":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2024-04-19 12:18:57.000000000","updated_by":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-04-19 14:34:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-22 07:46:01.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2024-04-23 08:26:38.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"fc5b3f41b17425141c44e203b809d33497d45672","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2024-04-19 11:57:33.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ee2a8c6e315fb398d7eaf8c343794ad63eda7aa2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 14:34:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a4c694fcba74ea0b56ddddd893b8638\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/991471d3a3924a4cacb1c4a50074cc68 : SUCCESS in 10m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aea85c4f5e69468d9b73dd00d07a1da5 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c980296807e3490a91bf3cffa69b56f4 : SUCCESS in 5m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0b681131b7cb48668a6822508cc9a9b4 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a2ba9a381954f9689219971250f5da7 : SUCCESS in 10m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac00f084d3864dc7955b6e73b254cfd7 : SUCCESS in 5m 48s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/412786ffbe034a3fa8c15f84a62c2d48 : SUCCESS in 4m 49s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/443801a4287644cbbbb136269155c3ad : SUCCESS in 7m 18s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e56ca8ed889047eb8a751a41fe7b2d1d : SUCCESS in 5m 28s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1caa90e10acb42289664355d4e7293f0 : SUCCESS in 7m 19s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a30f1fd3476b42658b84536912e4607c : SUCCESS in 6m 28s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/476107d91cfa4ec3a98d906f069a0624 : SUCCESS in 9m 02s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d92a3d05431149f1b3f97aeab7ebd057 : SUCCESS in 1h 16m 30s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/3b28c24f51424377bfc9dc6d578a58eb : SUCCESS in 1h 38m 16s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/b62cd2f025134818a5d092310497c638 : SUCCESS in 55m 23s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b6c2cbd35de44fbebf0f7ace69c8e8c8 : SUCCESS in 1h 43m 17s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/babac5f14dea4b8ea6bd55818a17639d : SUCCESS in 1h 15m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/19dba6098b6e48b8a2fe6d73800a5aa2 : SUCCESS in 47m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bc178ddbb1624116a895d527f97b37c1 : SUCCESS in 53m 15s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/3fb8af4ceb50491baa5ee0b348a3f284 : SUCCESS in 54m 47s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ae45c8fc78484c67b683d813b49c984c : FAILURE in 13m 32s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7dd58b2bcec2414f9e6e629aa552b1bd : SUCCESS in 47m 37s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/754561a93d844361ac8edc54556c8de0 : TIMED_OUT in 2h 34m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fced69b222034e75b16c37f59f8c7c86 : SUCCESS in 2h 27m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0f111c5638924326aaf4246848c83914 : FAILURE in 35m 31s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/58f69a8c84b044ffa08911dec01722c3 : FAILURE in 9m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c13c55d5296d9fa387a96ae1de704eb04bd7cd5c","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-22 07:46:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2c3b89558f9aa3d19d967ce1580326b002e630d6","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-04-23 08:26:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9a91dbbdafa7cf77c5fcdf367bbe2b2cb330b843","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 08:27:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"843933e1107405aeaf5bb3c53e3de3bf1bd55a56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 10:37:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/282fb4b880064f879c7a7bac28601b9c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba8fa345126442de8b3720b31622e51b : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4045d7c2d2ed4475b120d0d948cb9c45 : SUCCESS in 5m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f5c7473e439244c69552358ddb86cf48 : SUCCESS in 7m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d65a06981694e509f8c2fa8447d6aa0 : SUCCESS in 11m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee42322812e84594adef88d3277ff931 : SUCCESS in 5m 18s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d5292287a17541d6a10a3366c23202a0 : SUCCESS in 4m 21s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0485fe3897f64ab7881ae54894a3a76e : SUCCESS in 7m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1af42c90302a4ba2ba0841f79d55627e : SUCCESS in 1h 18m 30s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/fbe14edf5b864df3b141043fdc0fbb74 : SUCCESS in 2h 09m 27s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/5614f38f50904bc4be92f0b065d87972 : SUCCESS in 1h 20m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/30d9a04e4e754471953906237a0fa807 : SUCCESS in 48m 00s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/11f830fc2474457bb47c20427f580430 : SUCCESS in 56m 40s","accounts_in_message":[],"_revision_number":1},{"id":"ce5e8024a475ebd9b3df3c5a68a407f5daf8fa72","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 10:37:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ca9d98dcfa32bf8f4179fa6bce4c37a06f476092","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 10:39:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50f1743a60f14a168d2b2d1c370fe7d1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6eedd8e6370f4c5f8afec934f418456f : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb95d6fe957d442fac506a9d49a034d0 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8d1998299f9972e6244d92c4937a19b6fa01c0a1","revisions":{"8d1998299f9972e6244d92c4937a19b6fa01c0a1":{"kind":"REWORK","_number":1,"created":"2024-04-19 11:57:33.000000000","uploader":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"ref":"refs/changes/99/916099/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/99/916099/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/916099/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/916099/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/916099/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/99/916099/1"}}},"commit":{"parents":[{"commit":"fb7b56048d62058c8f79653e1345a0ed4804ca74","subject":"Merge \"Update TOX_CONSTRAINTS_FILE for stable/2024.1\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fb7b56048d62058c8f79653e1345a0ed4804ca74"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-03-13 08:31:17.000000000","tz":-240},"committer":{"name":"Christian Berendt","email":"berendt@osism.tech","date":"2024-04-19 11:57:33.000000000","tz":0},"subject":"Fix fully-populated API with allowed_cidrs","message":"Fix fully-populated API with allowed_cidrs\n\nWhen creating a LB + a listener with an allowed_cidr with the\nfully-populated API, an issue happened when Octavia validated that the\nallowed_cidrs and the VIP ip address have the same IP version. The\nvip.ip_address value was not updated in the load balancer object,\nforcing the expiration of the DB object before entering _graph_create\nfixes this issue.\n\nNote: there\u0027s no change in the tests, the test function for this feature\nexists, looks correct, and passes successfully, the bug is only\nreproducible in octavia-api.\n\nCloses-Bug: 2057751\nChange-Id: Ia106d81c1b2588e5d938d2238c8a2f6660bf5ef1\n(cherry picked from commit 5d9b23c6021ad961fafc17866c6b8ea50f2dc824)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8d1998299f9972e6244d92c4937a19b6fa01c0a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8d1998299f9972e6244d92c4937a19b6fa01c0a1"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
