)]}'
{"id":"openstack%2Foctavia~916100","triplet_id":"openstack%2Foctavia~stable%2F2023.2~Ia106d81c1b2588e5d938d2238c8a2f6660bf5ef1","project":"openstack/octavia","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-04-22 15:36:37.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"167":{"account":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"last_update":"2024-04-23 10:37:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia106d81c1b2588e5d938d2238c8a2f6660bf5ef1","subject":"Fix fully-populated API with allowed_cidrs","status":"MERGED","created":"2024-04-19 11:58:06.000000000","updated":"2024-04-23 10:39:22.000000000","submitted":"2024-04-23 10:37:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"916100","cherry_pick_of_change":912721,"cherry_pick_of_patch_set":1,"meta_rev_id":"02d835d3e6458977b692eb261966d3755b0ef953","_number":916100,"virtual_id_number":916100,"owner":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-23 10:37:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2024-04-23 08:26:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-04-22 07:46:12.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":1,"date":"2024-04-23 08:26:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-19 11:58:06.000000000","updated_by":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2024-04-19 12:19:04.000000000","updated_by":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-04-19 14:43:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-22 07:46:12.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2024-04-23 08:26:48.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"dd3e14c64b27913f5f09b93a0f74922e95d2df3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2024-04-19 11:58:06.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"eddbf87cf0280b0724ed86e7047ecb618e55977f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 14:43:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01474522578e4d4493bbc909c5bd6498\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc740256bea34faf99b52a684435cc1e : SUCCESS in 10m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb6283287313443291e277cc6bfa92ee : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/93ae232a60df41a2ad3f32f70b0721d3 : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b2c407447044a459dffb6634d81fefa : SUCCESS in 4m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/42080ba346034e0cbe9f74d4066bc2aa : SUCCESS in 5m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e7bc7a25bc0f411abd866b2aa6199af7 : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99a8ba0dcb8144639fb3adac70a9a950 : SUCCESS in 10m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e20e7e963b34414b565d4d45e0f0e7f : SUCCESS in 4m 47s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/29ae90995afb44438f1ae68f5c731771 : SUCCESS in 5m 03s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/257dbbadf882461db3fafb9be511e1a2 : SUCCESS in 8m 43s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/df8dc2991a7e4acd8e7d62e0637df9a1 : SUCCESS in 4m 52s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/619ce8b9e701459bb2388e33e417b31a : SUCCESS in 6m 59s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f3eed5a8e5e94be7b17a18b17d6dbdd6 : SUCCESS in 7m 08s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/51b8d49beaec46bea330462c54bdc494 : SUCCESS in 7m 53s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/561780959cdc46428ed0ae9d9b4414dd : SUCCESS in 1h 32m 46s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/1425d5002561439ca20a2207197eae43 : SUCCESS in 1h 17m 51s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/d6b75159c3094badb868af840b18e3d5 : SUCCESS in 59m 02s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/19b51241199247458eb10a0a521a41b4 : SUCCESS in 1h 20m 20s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e2253518fedb4df391ae84c98c12744e : SUCCESS in 1h 31m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f272819c8b5044d59239d7897af07051 : SUCCESS in 56m 22s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1d1181fb3ac6401f85b00f0fe61bc3b6 : SUCCESS in 56m 53s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/51241baa06cb4497a986b2ca576fabc1 : SUCCESS in 1h 07m 50s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/28b6d19ec9ac4055a74e87c517c2ab2b : SUCCESS in 49m 20s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1fd26904b15c4eeda370a34bf29ebf0c : SUCCESS in 1h 48m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/15be9254c52347e9b0451a5389401162 : SUCCESS in 2h 30m 54s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0fb21c94bcf744c392d3730c0a6138d1 : FAILURE in 30m 06s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/ef9d8490a689439d8ed790277ec8d4ee : SUCCESS in 2h 20m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d3cf4872ac575b8eacf21dd209f9d5ecf3463b8a","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-22 07:46:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"71171538af2e89517b14582d266dc746e8ea1f8a","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-04-23 08:26:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d1a062c4fd3e4fd54294805387c11a7577ab094a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 08:27:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c01d6b2e312dc01162d60c8b757e230c90bf4872","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 10:37:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05cd639402884d2fac0ebfb90713db36\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a74a12b062494ec289a5d8a1fa6c959a : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/907a8a8e2cee4f64a3f285632379541f : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/217c991b4c2b4fcca81cdb44fd3a1d4e : SUCCESS in 5m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/058a15adb24247c49a5247d22df66bd2 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/111baa9e28c74baea0b28213d1a59339 : SUCCESS in 9m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/660bde0fb4cf4dd1906b1a6ac59487b3 : SUCCESS in 4m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/bcb07d2990494e63975d3b60f1c81bcf : SUCCESS in 4m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8d4c6ff33ffa4021b91fbf85faa6a7ff : SUCCESS in 5m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ff3e88692ea447b08718f629f139f401 : SUCCESS in 1h 26m 06s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/a63d5dd04a62496bbaee7706b7fb65ba : SUCCESS in 1h 18m 04s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/c83fe83fb36143a0abc8e759d43c5de0 : SUCCESS in 55m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b1b935e5d12f40d5990a5f61fd5a8fab : SUCCESS in 35m 12s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/52379c6dc49f4bbab234e65d37720979 : SUCCESS in 37m 54s","accounts_in_message":[],"_revision_number":1},{"id":"0c2e6c3411529f9c40d308c47d075f61042de5f0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 10:37:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"02d835d3e6458977b692eb261966d3755b0ef953","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 10:39:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/841fe878f9cc440096ae818a10bc7366\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44ed538104ce456d9e77841c154f6acc : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae29cc146bc74086b3cef23fa828ccd4 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a1d422fe7fbc8a7e5d4c8a54a7ffcd1ba0f15352","revisions":{"a1d422fe7fbc8a7e5d4c8a54a7ffcd1ba0f15352":{"kind":"REWORK","_number":1,"created":"2024-04-19 11:58:06.000000000","uploader":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"ref":"refs/changes/00/916100/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/00/916100/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/916100/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/916100/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/00/916100/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/00/916100/1"}}},"commit":{"parents":[{"commit":"ac5a0470f284c8ffa0cde8ca6cacde3b3099b050","subject":"Merge \"Handle empty delay on update healthmonitor\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ac5a0470f284c8ffa0cde8ca6cacde3b3099b050"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-03-13 08:31:17.000000000","tz":-240},"committer":{"name":"Christian Berendt","email":"berendt@osism.tech","date":"2024-04-19 11:58:06.000000000","tz":0},"subject":"Fix fully-populated API with allowed_cidrs","message":"Fix fully-populated API with allowed_cidrs\n\nWhen creating a LB + a listener with an allowed_cidr with the\nfully-populated API, an issue happened when Octavia validated that the\nallowed_cidrs and the VIP ip address have the same IP version. The\nvip.ip_address value was not updated in the load balancer object,\nforcing the expiration of the DB object before entering _graph_create\nfixes this issue.\n\nNote: there\u0027s no change in the tests, the test function for this feature\nexists, looks correct, and passes successfully, the bug is only\nreproducible in octavia-api.\n\nCloses-Bug: 2057751\nChange-Id: Ia106d81c1b2588e5d938d2238c8a2f6660bf5ef1\n(cherry picked from commit 5d9b23c6021ad961fafc17866c6b8ea50f2dc824)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a1d422fe7fbc8a7e5d4c8a54a7ffcd1ba0f15352"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a1d422fe7fbc8a7e5d4c8a54a7ffcd1ba0f15352"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
