)]}'
{"id":"openstack%2Foctavia~916124","triplet_id":"openstack%2Foctavia~master~Ic19bc7d2146433f0f7a502e384c689a74fb74bf6","project":"openstack/octavia","branch":"master","topic":"remove-vagrant","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2026-02-17 12:12:47.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2026-02-17 09:51:27.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2026-02-17 12:12:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic19bc7d2146433f0f7a502e384c689a74fb74bf6","subject":"Delete Vagrant related files for devstack","status":"MERGED","created":"2024-04-17 16:33:28.000000000","updated":"2026-02-17 12:13:55.000000000","submitted":"2026-02-17 12:12:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"916124-remove-vagrant","meta_rev_id":"8773a0dab18dae1088c38474b377b650593d70fb","_number":916124,"virtual_id_number":916124,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-17 12:12:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"all":[{"value":1,"date":"2024-09-02 16:12:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-08-22 22:43:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2026-02-17 09:51:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2026-01-23 21:45:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2026-02-17 09:51:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-17 20:03:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-22 22:43:11.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-09-02 16:12:53.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"},{"updated":"2026-01-23 21:45:53.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-01-23 21:45:53.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"24f9ae1cc513408d9bdf27bc0e7f8974be5c0483","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-17 16:33:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f6aeeff9712030533c595d7e7b1e5b311fec0984","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 20:03:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aade28784bc14195baed9e2975b958e3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0969d267cd3b46f5bfab98c2840b451e : SUCCESS in 9m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d6f56cf8d674463ba655096db69623f : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef44593189624ac1883dadb1ac453a1a : SUCCESS in 5m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8a080b23374d4b3980a1872bde2c491a : SUCCESS in 5m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/30c22f125db847a1ac68b0c12b71f129 : SUCCESS in 8m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e69c848d88843c7bd7bdf8f93e5c46d : SUCCESS in 13m 13s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/a18d6ffac11b4918b2acd4609378e41e : SUCCESS in 4m 38s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/82252b86e7cb4c6f8f2219b8035b4df5 : SUCCESS in 7m 40s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/cacfb755a2b2444bb0c3470d99579ebb : SUCCESS in 11m 06s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/01aa44e400e44c26a6c3187da84f4139 : SUCCESS in 4m 43s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/65e9496ec7f24303aaee317d378f2315 : SUCCESS in 8m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4a2e7bed02a8425882697665f4db197a : SUCCESS in 6m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/471b262504bc4f51b84ec4e93ee095af : SUCCESS in 7m 44s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/825a095b223648cbba1c7d8e6b737866 : SUCCESS in 1h 11m 49s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/3d9f722c66ca4205a5cfae3c42f5b063 : SUCCESS in 2h 06m 21s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/c91f3b207b0242369508ef29baba30ab : SUCCESS in 1h 26m 51s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/cedb93383d724edb960ce9d79733e68b : SUCCESS in 1h 47m 59s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f77d0dbb5f6442c2bfa9aca0c56aef08 : SUCCESS in 1h 11m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/624b18d2ada3433fae6803403d706707 : SUCCESS in 45m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6ff062c504474eb78611e095ba89c10b : SUCCESS in 53m 18s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/bc932dca088548c0b5321df291d2a586 : SUCCESS in 1h 03m 28s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/898cf01a6aff461aa386f4f0a1a04a97 : SUCCESS in 50m 28s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0aabfb10054e4bfdb7b0925525a7a214 : TIMED_OUT in 2h 32m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d9506678e969464192c9764dcd0f17cb : SUCCESS in 2h 52m 03s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5cb1437fb9234903b0af811539a34c03 : FAILURE in 36m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/cdd6e14e26d848b993a21696b5c4d4e2 : TIMED_OUT in 2h 34m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0f7087abce76d7471e3fc9b1edb15e11ca746e3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-08-22 22:42:55.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"93cffcd378f75c07c49e4229e79a54c27ecf5cc2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-08-22 22:43:11.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a481bed05b7c9df12ee69a592f01d1ace26220a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 01:32:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/865fc266b1c747aabf13093ce6aeb835\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34490dde3b294807b7da5a15e22cefd7 : SUCCESS in 9m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c01332b1271b4a65b6e23a81345883f7 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4879a1a68a514b95ad36c2feacdc0b56 : SUCCESS in 4m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2c3f5d4e61de4ed886b86890d07cacac : SUCCESS in 4m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9c9d6642a0ef4c3aa856ff7f5b66959e : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a278f3cc295343ef9ea86b4f0ba1dd80 : SUCCESS in 11m 13s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/8a29e044e9944c7facf49a4d552a18f6 : SUCCESS in 5m 13s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/c42350e4ee7b4d259ef48227966e3625 : SUCCESS in 7m 02s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4c3b73624db64a36b5e122071d667936 : SUCCESS in 3m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ce577f3d7d6d44c5a5a0cedf6ec38f41 : SUCCESS in 8m 07s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5e35f6e60d01493599c751ff4a828107 : SUCCESS in 6m 51s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0dc8eecc39e34e389675102721313d7f : SUCCESS in 5m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/89b778ba6849449b8d7efa1761cee8db : SUCCESS in 1h 28m 00s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/850f5a32295c4e798bf3eaff5b66c889 : SUCCESS in 2h 09m 17s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/d8457a69bc894a3b8366e12ce143bbb9 : SUCCESS in 59m 21s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f7413e3f69674af9a5778daee88ce43c : SUCCESS in 1h 50m 54s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a676db645a65482081ff951a49d0649d : SUCCESS in 54m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9a1823a3ba5f4d0a8f6825fa68209543 : SUCCESS in 33m 11s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5a5f6ed0d531439c9112a571d3ccb2de : SUCCESS in 41m 27s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/cc4c6366feb44a229dc46221b94dd684 : SUCCESS in 57m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/948e11cf9b744e638b60f68526cfa1f7 : SUCCESS in 32m 01s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6371f194b38c430ab8b9485757906c31 : SUCCESS in 2h 27m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a545f5c7c886410786a70c1cae96295a : SUCCESS in 2h 46m 48s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4a5d9077582d48a69b8df6902ae4396f : TIMED_OUT in 2h 32m 40s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/59d0076a8ef546b1984d44b400d3ad41 : TIMED_OUT in 2h 33m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b77c6271040c12c3d88836f8c0068b469771224a","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2024-09-02 16:12:53.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"6dc902b15c57135ce1202e7c3f735e27d7b3446a","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-01-23 21:45:53.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"13a646314033bdd7631cb98a1237f51dc9f663b2","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2026-02-17 09:51:27.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ca88dc391e9d0302e0dc9b899889ecb3a8cd6acf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 09:51:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e6f80c5fc86f0e6648c9124532ff31fe0bb25262","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 12:12:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd27134997f84ef398cb3e4f1a02197b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/746517b279b84af1a7aae1b63179dad9 : SUCCESS in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/04d1bbb0de4c4630888559e39dd5bcc6 : SUCCESS in 2m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6b7a9b61714e4384be5a656822c91477 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37cbb89c199c487d95a259407247fe25 : SUCCESS in 9m 58s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/46a7eb0e325546ec8b7e458f1394460a : SUCCESS in 2m 16s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f28fa757571e4099b90c554956aa9adc : SUCCESS in 7m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/51488290f9e144baadf9f50e02820044 : SUCCESS in 1h 35m 35s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/c0892eac2dde436cb1351f15cfb1569a : SUCCESS in 1h 47m 50s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/4120f3edf32c411c8441e6d91f2da6a5 : SUCCESS in 1h 13m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/89a177c373f84d25aded49c4e074e405 : SUCCESS in 36m 45s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c92183597a60471aa623aa0f412201b4 : SUCCESS in 45m 14s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/b99caea3a0f14f7b8f271deebfb8ac92 : SUCCESS in 44m 18s","accounts_in_message":[],"_revision_number":2},{"id":"2f96a3f63927d39e9920508a4b400f249238597f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 12:12:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"8773a0dab18dae1088c38474b377b650593d70fb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-17 12:13:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4bf4e275011b41be8cffc95050918bf9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a21f754739d14cd7955f1693223f4190 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d9ddfcbcf0ac81419d53a1f2093a4f6db0ac577c","revisions":{"c95f47a9a3341581f20b31f54516cad091b265d6":{"kind":"REWORK","_number":1,"created":"2024-04-17 16:33:28.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/24/916124/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/24/916124/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/24/916124/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/24/916124/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/24/916124/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/24/916124/1"}}},"commit":{"parents":[{"commit":"603d437aefc5234af5790933185c135c50553080","subject":"Merge \"Add pyupgrade hook to pre-commit config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/603d437aefc5234af5790933185c135c50553080"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-04-17 16:31:41.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-04-17 16:31:41.000000000","tz":120},"subject":"Delete Vagrant related files for devstack","message":"Delete Vagrant related files for devstack\n\nChange-Id: Ic19bc7d2146433f0f7a502e384c689a74fb74bf6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c95f47a9a3341581f20b31f54516cad091b265d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c95f47a9a3341581f20b31f54516cad091b265d6"}]},"branch":"refs/heads/master"},"d9ddfcbcf0ac81419d53a1f2093a4f6db0ac577c":{"kind":"REWORK","_number":2,"created":"2024-08-22 22:42:55.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/24/916124/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/24/916124/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/24/916124/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/24/916124/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/24/916124/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/24/916124/2"}}},"commit":{"parents":[{"commit":"b71e752c0fff2939e4347a22e55c7cbb3e9c7f9a","subject":"Merge \"Fix test_update_load_balancer_status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b71e752c0fff2939e4347a22e55c7cbb3e9c7f9a"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-04-17 16:31:41.000000000","tz":120},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-08-22 22:42:54.000000000","tz":0},"subject":"Delete Vagrant related files for devstack","message":"Delete Vagrant related files for devstack\n\nChange-Id: Ic19bc7d2146433f0f7a502e384c689a74fb74bf6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d9ddfcbcf0ac81419d53a1f2093a4f6db0ac577c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d9ddfcbcf0ac81419d53a1f2093a4f6db0ac577c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
