)]}'
{"id":"openstack%2Foctavia~916761","triplet_id":"openstack%2Foctavia~stable%2F2023.1~Ia106d81c1b2588e5d938d2238c8a2f6660bf5ef1","project":"openstack/octavia","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-04-25 13:06:05.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"Ia106d81c1b2588e5d938d2238c8a2f6660bf5ef1","subject":"Fix fully-populated API with allowed_cidrs","status":"MERGED","created":"2024-04-23 08:27:03.000000000","updated":"2024-04-25 15:13:42.000000000","submitted":"2024-04-25 15:12:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"916761","cherry_pick_of_change":916100,"cherry_pick_of_patch_set":1,"meta_rev_id":"390529f1890f05c5d21aa756b6562180f1f6f24a","_number":916761,"virtual_id_number":916761,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-25 15:12:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2024-04-25 13:06:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-04-23 09:16:53.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":1,"date":"2024-04-25 13:06:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-23 09:16:53.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2024-04-23 10:47:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"42e1f1a388ae4024d231b538d6ae9cc297cda16f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-04-23 08:27:03.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2023.2.","accounts_in_message":[],"_revision_number":1},{"id":"f36f741797aa29a4ea2ec096dd4b48fa8ee06689","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-23 09:16:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dfc7b2557eb8aa703cc2d7a921af8146ca77ed41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 10:47:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2aa59fea169f45ba950fb7ec0e707531\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/619621afec004ec5a69103ed100ddce3 : SUCCESS in 11m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dd54341462a4f54945e07ce3c0fdc36 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7941fc84605748a091a3dd1f4d67c594 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f00fb80ded544454a3a336cd482d0030 : SUCCESS in 10m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7bd1f66e74a34ffdacc435feedfb82b5 : SUCCESS in 4m 58s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/b2be575d9ebb4d379ff4f3ddd420de52 : SUCCESS in 5m 19s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/3d4571ff82f545fda1cb4091bf21ebcb : SUCCESS in 7m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f53382d32f194be7a5f60c14885ebf4f : SUCCESS in 4m 29s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ff580409db024cac8953c9ca677f55e8 : SUCCESS in 7m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9db042272b6d41a486f7cac3fab3082d : SUCCESS in 7m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e9b42d0c57a843efaa4a841931723295 : SUCCESS in 8m 53s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/121780cce5ea46afb39fcb7999a62b8e : SUCCESS in 1h 26m 19s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/19bee02edca8467b99da78402e7331d5 : SUCCESS in 1h 18m 46s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/a6da635531f440a0ba42cb4e3288de47 : SUCCESS in 1h 04m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d4d41a49ac10488e8207f68652e68479 : SUCCESS in 32m 40s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/79962cdfb21845d8aaee63045d3d6adc : SUCCESS in 37m 51s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/0895dafe870e4c93bdd72104e2b72497 : SUCCESS in 38m 10s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/f684d5ef9b964345a98d957a131b96b9 : SUCCESS in 1h 03m 31s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d895db7b86c2449e83a4db9f29d9300f : SUCCESS in 30m 49s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e27696dbbc8f453cbd95e2144b63afea : SUCCESS in 1h 42m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c1a465557da3476f86f73efecc09ade7 : SUCCESS in 2h 18m 07s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/aa44efd7c74f4d0fbf29280b594d4277 : FAILURE in 24m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/343c86812eb24fcd95adc4f94c7d7a1a : SUCCESS in 1h 46m 49s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/4d4122592ac14b319f79bd460ae5daa6 : SUCCESS in 1h 52m 56s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/4ee30c551c56455590fb8300b60511dc : SUCCESS in 1h 55m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fda285d6303fde1cd913dc2b566daf81dc212d38","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-04-25 13:06:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"52219450456eb6b7ee2ae5c6fb4f66e4e36669c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 13:06:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0fb3dcea48709796f150914c8be4668b036c0e6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 15:12:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c836bdfc0908427eaa243d7654275033\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13ef0dcd6db341d1bbbe76c265e32e0a : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27a774e8764b429f9485bd0e5f2b585a : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dab095ca8e741248298b239220d9f95 : SUCCESS in 11m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/abf6cb17bed340b589e40ed364290366 : SUCCESS in 5m 45s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a64bd509eb4b40c8a22c830dcff1b5c1 : SUCCESS in 4m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9b7f6f769f2a49dbaa8dc790321bbbf0 : SUCCESS in 6m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/664d4905b6784f42952dd163a9cbd7ba : SUCCESS in 1h 35m 24s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/941a94a3d7a040c091cff739f556c06f : SUCCESS in 2h 04m 09s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ecde86cc80744e818cd48f2f0b57a88b : SUCCESS in 1h 29m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9bd4adf457264e9cb733b4e5a6645bcd : SUCCESS in 1h 04m 05s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/473aa3377f12411684a25442e534cc74 : SUCCESS in 1h 02m 05s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/1572b784d6274488b14de541b57df767 : SUCCESS in 56m 35s","accounts_in_message":[],"_revision_number":1},{"id":"7736dd76e74fdf049d0550f3bd7277eb2a86826a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 15:12:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"390529f1890f05c5d21aa756b6562180f1f6f24a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 15:13:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2aaaff1c3fe4d9faa91d633fd4561e3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b0698c82be14dee99ee2787f28eb60a : SUCCESS in 1m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e5fdd6267894d3e8891f50782e115ed : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d943336c0eb7461838a040e263355003682df960","revisions":{"d943336c0eb7461838a040e263355003682df960":{"kind":"REWORK","_number":1,"created":"2024-04-23 08:27:03.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/61/916761/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/61/916761/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/61/916761/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/61/916761/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/61/916761/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/61/916761/1"}}},"commit":{"parents":[{"commit":"41e77076454850df542f9c30243599f05074257c","subject":"Merge \"Handle empty delay on update healthmonitor\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/41e77076454850df542f9c30243599f05074257c"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-03-13 08:31:17.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-04-23 08:27:03.000000000","tz":0},"subject":"Fix fully-populated API with allowed_cidrs","message":"Fix fully-populated API with allowed_cidrs\n\nWhen creating a LB + a listener with an allowed_cidr with the\nfully-populated API, an issue happened when Octavia validated that the\nallowed_cidrs and the VIP ip address have the same IP version. The\nvip.ip_address value was not updated in the load balancer object,\nforcing the expiration of the DB object before entering _graph_create\nfixes this issue.\n\nNote: there\u0027s no change in the tests, the test function for this feature\nexists, looks correct, and passes successfully, the bug is only\nreproducible in octavia-api.\n\nCloses-Bug: 2057751\nChange-Id: Ia106d81c1b2588e5d938d2238c8a2f6660bf5ef1\n(cherry picked from commit 5d9b23c6021ad961fafc17866c6b8ea50f2dc824)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d943336c0eb7461838a040e263355003682df960"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d943336c0eb7461838a040e263355003682df960"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
