)]}'
{"id":"openstack%2Foctavia~916983","triplet_id":"openstack%2Foctavia~stable%2F2023.1~I40797443afe789021f3697fd779ce9f3aa3d4815","project":"openstack/octavia","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-05-28 17:06:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I40797443afe789021f3697fd779ce9f3aa3d4815","subject":"Fix quota error messages","status":"MERGED","created":"2024-04-25 07:13:42.000000000","updated":"2024-05-28 17:06:45.000000000","submitted":"2024-05-28 17:06:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"916983","cherry_pick_of_change":879278,"cherry_pick_of_patch_set":1,"meta_rev_id":"d1b6487417dc4cfeaaad227922d2687009ef1d46","_number":916983,"virtual_id_number":916983,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-28 17:06:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"all":[{"value":1,"date":"2024-05-16 06:47:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":1,"date":"2024-04-26 07:45:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":2,"date":"2024-05-02 15:57:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-28 14:58:20.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-05-28 14:58:20.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"CC":[{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-25 10:20:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-26 07:45:58.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"},{"updated":"2024-05-02 15:57:48.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-05-08 13:59:27.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"CC"},{"updated":"2024-05-16 06:47:34.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"state":"REVIEWER"},{"updated":"2024-05-28 14:58:20.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"bd660e13345598a108ad2ebfa6e9c064cecb0a99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-04-25 07:13:42.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f82148748c3f106eb8457a319edaeaada93ac329","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 10:20:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c1dc172cf67482f8df45ad89cf35bf3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ecfb63e299344d9ab4345e11d50a0b3d : SUCCESS in 9m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f43da2e033144bf18f7eff36ad580349 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/67c18fa9733f49f78f6c5f842e44b03f : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b40e84b174c0489c90820d69e28f27a8 : SUCCESS in 11m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e1a0e178455461aa0bce8ad93f6c8dc : SUCCESS in 4m 50s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/ac578e77e17545ccb1ceb38bbd835c0f : SUCCESS in 4m 42s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/896f8b63604842d194b72847fd461696 : SUCCESS in 7m 49s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/594b7556796d47b7a71aa6ae5cfa4d85 : SUCCESS in 4m 08s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/60ec049df6834b1e80dc8ab9579c9aba : SUCCESS in 7m 53s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/259cc08ca1244bc8a5fdb4c1eb63f7fe : SUCCESS in 6m 09s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/89c06167753f4d3eb3bdabb89b8ef64e : SUCCESS in 7m 49s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/27734b9f1fc94aedbeb485d0f62b390a : SUCCESS in 1h 56m 57s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/d28ef63ea7dd49e2bba11e966fae6613 : SUCCESS in 2h 02m 23s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/14259d44829541af91e6cbc53873291f : SUCCESS in 1h 26m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2433ce28cb89468da19b94ecd9e462f5 : SUCCESS in 52m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fab975a045ed449c9c4749773b03c041 : SUCCESS in 1h 03m 55s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/285e26d8a05149cba0da61a03981184e : SUCCESS in 1h 00m 40s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/12f64558ea5c472ebdc7441ae8732ffd : SUCCESS in 1h 07m 04s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/78b0136037f64552928a1120cab4be91 : SUCCESS in 48m 21s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/239b394536554715a3cbcb8538e23990 : SUCCESS in 2h 08m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c9814858e6e145c6a4b8266bb901c924 : TIMED_OUT in 3h 04m 17s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6629ffc95dd3407781017b10f883c416 : FAILURE in 31m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/46a0fcc91d50413eb9d033553e7c780d : TIMED_OUT in 2h 33m 41s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/21b20c3967bb4975b0fcbf754521c21f : TIMED_OUT in 2h 34m 29s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/59f65afed22d49179797fda5979e3063 : SUCCESS in 2h 24m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d3653e37701961f14695f690dc44baf8b595ad0c","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2024-04-26 07:45:58.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"db271c0dc0b04dacc6d40faa766162ad1845d933","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-05-02 15:57:48.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a1f6e0fadd1e26c992d665ae73f2494df0d09c05","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-08 13:59:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d6c7b1ec890002e119ba5926166e599aead81d6e","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-05-16 06:47:34.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"80db85dd52d28e4ddf56ebe3e88e8785035d8b33","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-05-28 14:58:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"429fae9ac70600c905382bb8a7fffd360020938f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 14:59:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1aa3d2e6cd24a2102051273fa22eb4dc84978912","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 17:06:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45c5b0861bf74566be45ede196c50649\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efcd087eeff745d0a904748571aa6778 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6237be3825404dbf849203def8d709fe : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5223bd39f4bf42a8a2904ee6d14fa5f4 : SUCCESS in 12m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5772a8937594443abd1dbf7a6f2e1c7f : SUCCESS in 4m 35s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/347c840b86b4426788695fb27fdd8420 : SUCCESS in 5m 15s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b82bab662571425baa6fdcf091d61482 : SUCCESS in 7m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1be500131eec468eb0a27ad57b4563fb : SUCCESS in 1h 29m 20s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/f5f08c76d6a5487e97e96c363588f1ed : SUCCESS in 1h 19m 40s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/54f801628338406e94f20ebf91448d68 : SUCCESS in 1h 28m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8422fff29a604acaa0d7396342046ea2 : SUCCESS in 32m 53s","accounts_in_message":[],"_revision_number":1},{"id":"d1b6487417dc4cfeaaad227922d2687009ef1d46","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 17:06:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f65068f4fa1f5165c382681036d31cc217d0ee36","revisions":{"f65068f4fa1f5165c382681036d31cc217d0ee36":{"kind":"REWORK","_number":1,"created":"2024-04-25 07:13:42.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/83/916983/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/83/916983/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/916983/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/916983/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/916983/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/83/916983/1"}}},"commit":{"parents":[{"commit":"41e77076454850df542f9c30243599f05074257c","subject":"Merge \"Handle empty delay on update healthmonitor\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/41e77076454850df542f9c30243599f05074257c"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-04-03 08:10:35.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-04-25 07:13:42.000000000","tz":0},"subject":"Fix quota error messages","message":"Fix quota error messages\n\nThe quota error/warning messages displayed type(_class) which was\ntranslated to \u003cclass \u0027type\u0027\u003e. Now the messages are fixed and display\nhelpful logs:\n  Quota decrement on \u003cclass \u0027octavia.common.data_models.L7Policy\u0027\u003e called\n  on project\n\nRelated-Bug: #2015050\n\nChange-Id: I40797443afe789021f3697fd779ce9f3aa3d4815\n(cherry picked from commit 84a4c071eee4ebf16b53831f9798e8d4e649c74b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f65068f4fa1f5165c382681036d31cc217d0ee36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f65068f4fa1f5165c382681036d31cc217d0ee36"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
