)]}'
{"id":"openstack%2Foctavia~917532","triplet_id":"openstack%2Foctavia~master~I1ba4c9ea5f3a60b53063dfaa73ffd27c35416f0f","project":"openstack/octavia","branch":"master","topic":"excludes-cleanup","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-05-04 17:06:20.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-11-16 06:52:52.000000000","reason":"Change was abandoned"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2024-04-30 13:38:30.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I1ba4c9ea5f3a60b53063dfaa73ffd27c35416f0f","subject":"Remove old excludes","status":"ABANDONED","created":"2024-04-30 06:49:28.000000000","updated":"2024-11-16 06:52:52.000000000","total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a1e063cd340c41780db8a66d8e89ca90d03c512b","_number":917532,"virtual_id_number":917532,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"rejected":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":-2,"date":"2024-04-30 15:12:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":-1,"date":"2024-05-01 16:44:52.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"date":"2024-05-02 15:11:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-04-30 13:38:30.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"date":"2024-05-02 15:11:57.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-30 09:38:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-30 09:57:46.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2024-04-30 15:12:10.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-05-01 16:44:52.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"beec0659fd7a46a444fb7998e6a5a820d309839e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-30 06:49:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f7d064cd8c84d51677bd11d90901b90124245cad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 09:38:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/117f29651a124bcab7239141216254ed\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3fe169688c3f4fb9821adb1493067143 : SUCCESS in 8m 20s\n- requirements-check https://zuul.opendev.org/t/openstack/build/724b7fe39ae1461f9478afae7ac3ce3a : FAILURE in 4m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/693de462559945eeb06936f7394e96bb : SUCCESS in 8m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0c33255ba1648e2b99429c55411b346 : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f69e8be30b2445f89e1aa79f702b0215 : SUCCESS in 4m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2ec0948d5b164534a1271ae4ca4ebf33 : SUCCESS in 6m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6ee27967ab9d46628e8ad2295e90e30d : SUCCESS in 8m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb1c050e097544ac876ef18a5415e5ca : SUCCESS in 14m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc37e4aa130442f2a74bd0050c3fe074 : SUCCESS in 5m 31s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/33ef3f1347d948a6818cdb310472bc69 : SUCCESS in 4m 08s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/fd9cdaba5f5c4cdab937285c7b1cb1be : SUCCESS in 5m 11s\n- octavia-tox-functional-py311-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/e7912d02edac4446a283f56a3c68add4 : SUCCESS in 5m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ce4a359a36e143f0a032b3066892a197 : SUCCESS in 4m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/49a8fcac4d2a44f38e25bc43dfce5aa6 : SUCCESS in 8m 19s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/50571ba32d924c3c8cb785b8362545cf : SUCCESS in 5m 42s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/bb1f5dee18d7496e98805b86ac0cd3e3 : SUCCESS in 4m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/310a1a89e23245bb98e0a6af1107d450 : SUCCESS in 1h 40m 37s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/d448b49ce6c2470cb825fdd98238a1c8 : SUCCESS in 1h 51m 31s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/80d01f31ed844a1da4a6e3ed8785a86c : SUCCESS in 1h 06m 44s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/1aa2c808faac451fa6b024c60685cb84 : SUCCESS in 1h 55m 09s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/bd8c401faa33427db25caa8b3f53fc1c : SUCCESS in 1h 03m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/78155f4802c74fd981f978df32035a3f : SUCCESS in 47m 07s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/aa98e82e348e4dd598c54e7d73fcc814 : SUCCESS in 1h 02m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c1de60ac46724c14ab4979dbc7994ef3 : SUCCESS in 44m 55s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5084c7abe76344699ea4de7f1a2486af : SUCCESS in 2h 24m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9f70a2198b5f48e09323b9620deb1a76 : SUCCESS in 2h 18m 10s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1f389fbb2a0d4499bb21c4f2b5ce1b0f : FAILURE in 31m 29s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/a8f53dc7dddc4f24b4e22b40c06b77e0 : SUCCESS in 2h 35m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bb99c62d9ba8a6acc78d5b341b6f7ad49cf470df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-30 09:40:28.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"2bdfba7d87e5e79e7f4d84f0de106e828e05f7e7","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-30 09:57:46.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ce5154041ae6af20d0bbf3296782f239e03c9f46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-30 10:00:25.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"fceb46fe8805a914b95ebb5233b651b7cd17fc5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-30 10:03:50.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"d8950a8b59bcb55f130109014693b99b93783518","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 10:05:07.000000000","message":"Patch Set 4: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/917607 is needed.","accounts_in_message":[],"_revision_number":4},{"id":"88b1a339adce62569eec4cb9b54605e570c3a861","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-30 10:05:55.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"72118a2d37f86dc601699aa9e20560d14bb4ff80","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 12:41:20.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95d02a5f0fec4140992026620141c61d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e3390742341540a4b15e29839bed78d0 : SUCCESS in 5m 08s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b13a3eeb6d554ec387d1b984bd8e0ee4 : SUCCESS in 2m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c9fd79f92b641cfbf3b5c5a998fbb62 : SUCCESS in 10m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f3fb6e631b44528828bd324b5bd7f62 : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/38f543594c3840a2ae1a28d885be147f : SUCCESS in 5m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1c36bec526e14f8aa9b1812dada900f9 : SUCCESS in 4m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a80b0c32dc604c2bb81550560dcec86d : SUCCESS in 7m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0afdf8cbf5c64d11a69cbe5662f15935 : SUCCESS in 14m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ea75cb530444ae8aab38999d30b42c6 : SUCCESS in 5m 40s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/ea6ca9c01816461b9d6398b6c1a79326 : SUCCESS in 5m 55s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/dea14b969a5e445f84d5c86b68c44530 : SUCCESS in 4m 43s\n- octavia-tox-functional-py311-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/80bf135aac904f71a3ed7e3827afcacd : SUCCESS in 7m 05s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a36cdf9ef6ed46f3859ae2ccc48dc74e : SUCCESS in 4m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f85b926327024430812ed9a9859c2e9a : SUCCESS in 9m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/31bde1971ac24b02855206dbc56cad4e : SUCCESS in 7m 21s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/16d6c790447c4873a3e4be0fa5adb53e : SUCCESS in 6m 33s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1fe35c426a4a49ccb719284533c3aba5 : SUCCESS in 1h 17m 21s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/48c69e9399b84bf582b6af47ce0edf5d : SUCCESS in 2h 29m 06s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ed924f460a704d65bbe3d6cc57c4442f : SUCCESS in 1h 24m 06s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e981c309909449b0ac94a0d439f7b91d : SUCCESS in 2h 18m 16s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/6c0900c6a7a749dbb2c214fa4640827e : SUCCESS in 1h 32m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/284003be04174c878e738da3901db25e : SUCCESS in 50m 52s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/2e832202842e488fb49b71db5d44dc70 : SUCCESS in 1h 15m 07s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e6b620d93f43454e848baacdfdbd63ba : SUCCESS in 45m 14s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5086794bcdf641b18f6663c15308ce4d : TIMED_OUT in 2h 34m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6b04a08b0b26440bbd7f46578be9975d : SUCCESS in 2h 32m 25s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4589c815934349cf92a573e44f2f1650 : FAILURE in 20m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/0fd47969105f42f6b68cfd417892e9f7 : TIMED_OUT in 2h 33m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"cdd12718568c38363d069274a80d5c8cc72849ae","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-30 13:38:30.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"680d79621609765ca4adc8c83c458c158b7cec3d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-04-30 15:12:10.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"bec063167dc719462a076d6004133bc74ce32f83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 15:12:23.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0e36ce1472254d8e8cf5fe3995a54fe7\n\n\nWarning:\n  Change 917602 in project openstack/requirements does not share a change queue with 917532 in project openstack/octavia\n  Change 917607 in project openstack/requirements does not share a change queue with 917532 in project openstack/octavia","accounts_in_message":[],"_revision_number":4},{"id":"2fd1ae3c1a08f0866a92be6ca7ffb3454bc57a32","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-05-01 16:44:52.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"10c8f7e0d827e3fabe754d2fe4e41dbba417eef5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-05-02 15:11:50.000000000","message":"Patch Set 4: -Workflow","accounts_in_message":[],"_revision_number":4},{"id":"676b05b1200954e0e57f2dfcdd8d95703d4aca9e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-05-02 15:11:57.000000000","message":"Patch Set 4: -Code-Review","accounts_in_message":[],"_revision_number":4},{"id":"34b56ef09e7317bd88263375bc8ee133a1ae4658","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-02 15:19:38.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1ec8cb5890c6b727803168895e0814a280101da3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-05-03 18:51:55.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9a26255cceed7ceb1ff77bfd307aeb6a48af3cc9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-04 17:06:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a1e063cd340c41780db8a66d8e89ca90d03c512b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-16 06:52:52.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"5f8ed0c40f9d692ca0edf98556eb58c04b677dd9","revisions":{"661362271e2207f7e7dc40aaddd2446d8c79b7cf":{"kind":"REWORK","_number":1,"created":"2024-04-30 06:49:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/917532/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/917532/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/917532/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/917532/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/917532/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/917532/1"}}},"commit":{"parents":[{"commit":"5898d54b137f5fb7a1d45bf8d620ef8ab8133259","subject":"Merge \"Drop redundant envdir\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5898d54b137f5fb7a1d45bf8d620ef8ab8133259"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-30 06:48:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-30 06:48:54.000000000","tz":540},"subject":"Remove old excludes","message":"Remove old excludes\n\nThese are detected as errors since the clean up was done[1] in\nthe requirements repository.\n\n[1] 314734e938f107cbd5ebcc7af4d9167c11347406\n\nAlso remove the note about old pip\u0027s behavior because the resolver\nin recent pip no longer requires specific order.\n\nChange-Id: I1ba4c9ea5f3a60b53063dfaa73ffd27c35416f0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/661362271e2207f7e7dc40aaddd2446d8c79b7cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/661362271e2207f7e7dc40aaddd2446d8c79b7cf"}]},"branch":"refs/heads/master"},"0af8d9175d60fecb69c394efae16e183ef2b6877":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-04-30 09:40:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/917532/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/917532/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/917532/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/917532/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/917532/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/917532/2"}}},"commit":{"parents":[{"commit":"5898d54b137f5fb7a1d45bf8d620ef8ab8133259","subject":"Merge \"Drop redundant envdir\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5898d54b137f5fb7a1d45bf8d620ef8ab8133259"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-30 06:48:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-30 09:40:10.000000000","tz":540},"subject":"Remove old excludes","message":"Remove old excludes\n\nThese are detected as errors since the clean up was done[1] in\nthe requirements repository.\n\n[1] 314734e938f107cbd5ebcc7af4d9167c11347406\n\nAlso remove the note about old pip\u0027s behavior because the resolver\nin recent pip no longer requires specific order.\n\nDepends-on: https://review.opendev.org/c/openstack/requirements/+/917602\nChange-Id: I1ba4c9ea5f3a60b53063dfaa73ffd27c35416f0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0af8d9175d60fecb69c394efae16e183ef2b6877"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0af8d9175d60fecb69c394efae16e183ef2b6877"}]},"branch":"refs/heads/master"},"bb873e20a09660c323e3ea016e36c0665a8f9258":{"kind":"REWORK","_number":3,"created":"2024-04-30 10:00:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/917532/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/917532/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/917532/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/917532/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/917532/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/917532/3"}}},"commit":{"parents":[{"commit":"5898d54b137f5fb7a1d45bf8d620ef8ab8133259","subject":"Merge \"Drop redundant envdir\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5898d54b137f5fb7a1d45bf8d620ef8ab8133259"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-30 06:48:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-30 10:00:16.000000000","tz":540},"subject":"Remove old excludes","message":"Remove old excludes\n\nThese are detected as errors since the clean up was done[1] in\nthe requirements repository.\n\n[1] 314734e938f107cbd5ebcc7af4d9167c11347406\n\nAlso remove the note about old pip\u0027s behavior because the resolver\nin recent pip no longer requires specific order.\n\nDepends-on: https://review.opendev.org/c/openstack/requirements/+/917602\nChange-Id: I1ba4c9ea5f3a60b53063dfaa73ffd27c35416f0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bb873e20a09660c323e3ea016e36c0665a8f9258"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bb873e20a09660c323e3ea016e36c0665a8f9258"}]},"branch":"refs/heads/master"},"5f8ed0c40f9d692ca0edf98556eb58c04b677dd9":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2024-04-30 10:03:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/917532/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/917532/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/917532/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/917532/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/917532/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/917532/4"}}},"commit":{"parents":[{"commit":"5898d54b137f5fb7a1d45bf8d620ef8ab8133259","subject":"Merge \"Drop redundant envdir\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5898d54b137f5fb7a1d45bf8d620ef8ab8133259"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-30 06:48:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-30 10:03:35.000000000","tz":540},"subject":"Remove old excludes","message":"Remove old excludes\n\nThese are detected as errors since the clean up was done[1] in\nthe requirements repository.\n\n[1] 314734e938f107cbd5ebcc7af4d9167c11347406\n\nAlso remove the note about old pip\u0027s behavior because the resolver\nin recent pip no longer requires specific order.\n\nDepends-on: https://review.opendev.org/c/openstack/requirements/+/917602\nDepends-on: https://review.opendev.org/c/openstack/requirements/+/917607\nChange-Id: I1ba4c9ea5f3a60b53063dfaa73ffd27c35416f0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5f8ed0c40f9d692ca0edf98556eb58c04b677dd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5f8ed0c40f9d692ca0edf98556eb58c04b677dd9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Verified\u003dMIN"],"failing_atoms":["label:Verified\u003dMAX"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
