)]}'
{"id":"openstack%2Foctavia~917675","triplet_id":"openstack%2Foctavia~stable%2F2023.1~Iad78de1764bc3a3f699b5feef9e58999d2efe613","project":"openstack/octavia","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-06-06 12:57:19.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"Iad78de1764bc3a3f699b5feef9e58999d2efe613","subject":"Fix incorrect masquerade rules in multivip LBs","status":"MERGED","created":"2024-04-30 11:40:55.000000000","updated":"2024-06-06 17:32:08.000000000","submitted":"2024-06-06 17:30:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"917675","meta_rev_id":"3207981648db49a1ca0572f941c30f00b15a3d44","_number":917675,"virtual_id_number":917675,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-06 17:30:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2024-05-08 13:53:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":2,"date":"2024-05-02 16:48:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-14 11:55:34.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-05-14 11:55:34.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-30 14:13:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-02 16:48:05.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-05-08 13:53:45.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2024-05-14 11:55:34.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"35074540de71f702301eed5741e43c2f5669925f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-04-30 11:40:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7064f1acdd3d9ea9eadd879241a4c29bd9566611","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 14:13:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf490af10dd949829dfe8fabb7968217\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ffae4234dc314eeda2afc0ad9bb2a9bd : SUCCESS in 9m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d61af301c9f048d9bd0f806afb868f73 : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75b1d30a2d8446368016bcf88f5f53a3 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bab58b84f2d440d819065c6c4cc7c5a : SUCCESS in 11m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e2bea20dabb42659055bcf0b1cb91a6 : SUCCESS in 5m 47s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/e6929a495a9e4ac09fd364d457d4e356 : SUCCESS in 6m 56s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/41275148092f404f8e752b5f1882602a : SUCCESS in 8m 36s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/38870f26058748728ea536e4eb00ccbe : SUCCESS in 4m 57s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/94407e2c48eb4731ae2a503efcedf984 : SUCCESS in 10m 01s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/32febaa6d26845d0aa434b3bcc534066 : SUCCESS in 7m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/79a7c6f6056b4ec6a1c2c7ace255a908 : SUCCESS in 7m 46s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e6c0de6b38d0427796f651792da3a6ea : SUCCESS in 1h 52m 42s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/bd86fa74a80d42d39ee63ba17581f3f1 : SUCCESS in 1h 19m 59s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/c86fd2682b4c46509f70428d574aef71 : SUCCESS in 57m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ffadbd8883ab4ff9aec69b1f8e467342 : SUCCESS in 34m 25s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/db7a42ee7336417e8ef0b40cd54cfbec : SUCCESS in 52m 44s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/7462d4841c4249a081695fb7b27fba48 : SUCCESS in 55m 12s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ea7e6e43c10942aebdf0427e60699968 : SUCCESS in 1h 07m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/221214a9da9a4cc697f08ac9e278638f : SUCCESS in 30m 12s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2cf81c8f79f94d21b3d4a43cd0efc44b : SUCCESS in 1h 35m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3c3152a1647440ca9b6ad7fadaad8cbd : SUCCESS in 2h 18m 32s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d4a74df7481c4630a2d3c4fbcd6ee6ad : FAILURE in 25m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/5b8f34e2f0464449a8986cea94a8f212 : SUCCESS in 2h 16m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/30433e56b7b74af8bca177685f57ff2c : SUCCESS in 1h 55m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/f53429ebbf084350be4ce0b64b28d814 : SUCCESS in 2h 30m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4e3d8da0a472ac8665b8d2ed7f9be9805f89f1f6","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-05-02 16:48:05.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"03ca921422847740bdab5449c26b04a821584a09","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-08 13:53:45.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b6ae40208808e3a3726dfce0490d0896c5f2e695","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-05-14 11:55:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"14ddc5a61e81df52912ea00c80974b5a12c56867","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 11:56:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7eecaf7609c87945e0a6c704b42b5b3abc6f1af5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 13:32:12.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f36c4c1bdea24c68b06bc88897cbbe7f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0f4d92687ec4dfcb8ec8f021a29172e : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1cd9694e118a4dd5a517d27b2c5bdb68 : SUCCESS in 6m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f840b27b998045cca629f90e1498d357 : SUCCESS in 12m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d11e829e5711493bbc7ded1f02cf7b66 : SUCCESS in 6m 37s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6bce86489453400d9514ab57cdfb06e2 : SUCCESS in 5m 17s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9bdbef98320f4d6cb55d8575d81df5d4 : SUCCESS in 9m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8c5bc04f3a8f431e95a9ea3cc761fb5a : CANCELED in 29m 44s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/11f473e88ac24ed9801b2ff5712f9a35 : CANCELED in 30m 21s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/79af4197f84c4f27abb44f20d4277ccb : CANCELED in 30m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/861cbaa61b64458d9944a88d21fcc0d8 : CANCELED in 30m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/82efbc9377ff454fb4d4fe55381d02a9 : FAILURE in 29m 47s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/258142495c3c43be8f5b1a24fab9b380 : CANCELED in 29m 52s","accounts_in_message":[],"_revision_number":1},{"id":"e3a0172fe1b776ddcea7567f286d5a3a84c775ea","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-06-06 09:39:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1f5f66213f9e85dffbe169a38420e8248ab67cca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-06 12:15:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0d2a0c006757482bbbed1a6ededb8925\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7969bde77b58456897615052673b8826 : SUCCESS in 10m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f51514395c14a1d9b6dc79bcb8198ae : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed3a50aceba64132ae078d20fd8f03d6 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17e61bff95864fc8997e627afbccd371 : SUCCESS in 10m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50db47898f02470aabb49a0f52b278fe : SUCCESS in 6m 15s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/3b887f57171c4d989dfb2bb6de5f32e9 : SUCCESS in 5m 05s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/355d6d73375941149af9340ba6a04182 : SUCCESS in 8m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5cf7275c745f436b96d4e72c0e6a2bcc : SUCCESS in 4m 10s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/05e0d53d945a413aacfc72ae0624f7da : SUCCESS in 6m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/dcd3ce517d1c48eebea80e966e03deba : SUCCESS in 8m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/98b821d5ad684b93b12f1b4d0b18f82e : SUCCESS in 7m 25s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c85bcc96b2464348a2da1419ec4de4c4 : SUCCESS in 1h 33m 25s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/e9cae78f2f2e4aacafc55ca738aa8be3 : FAILURE in 1h 28m 26s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/305d7b7c4bb743c9ba187034c06263c9 : SUCCESS in 1h 03m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4d965714140946d1864abdaa9398be76 : SUCCESS in 40m 45s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/496b151d7eaa436f89ddb0643e7bcde3 : SUCCESS in 1h 04m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/768036fdf637457891f47f5e5d688b4d : SUCCESS in 36m 02s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f5eea5b73b864448bd5f35d0c51593e3 : SUCCESS in 1h 52m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/998d4e40ed674f828ef14db786d7809e : SUCCESS in 2h 00m 31s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b727596824044163b7e859afdf900a7c : FAILURE in 23m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b505d05e8e9c4980ad77329e994a4f68 : SUCCESS in 1h 59m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/283a863c6fcf4a288909845d7d88acb2 : SUCCESS in 2h 06m 35s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/8dd1ea4f65334568a626ea3dd29bf73e : TIMED_OUT in 2h 33m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d425ad3430e8ae56c63a6274a8fa6f2860d25ee3","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-06-06 12:57:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"185d091f48d34b357a15b998265dc5ba32a9386a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-06 15:26:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01925ecaf61140dba443d1074bcc29f5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d787efc8fcde4a1a8378e21a1dc5876a : SUCCESS in 10m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bf1a14855de49099c0945dbfa490505 : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e131da79fc54c27ba9b1ff21e7425c5 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ea226f35412433dabd2ce6765bb0586 : SUCCESS in 11m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e53c206ef4af460d973738e2f7e08362 : SUCCESS in 5m 10s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/91fd6388c0aa4b82982fd6fccd6fb3a3 : SUCCESS in 5m 11s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/4e08af8556d44a07b66a347c5ee69b82 : SUCCESS in 7m 42s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e7d31234e0ea48c8a5ee0a0c3088ef54 : SUCCESS in 4m 15s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b1a3a8baadf14935a229330543f27291 : SUCCESS in 7m 35s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/05a66a310698432695caf3eda94b4b3e : SUCCESS in 7m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/99bcac34cadf4f5ca414811fe8693f60 : SUCCESS in 7m 28s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f044dcc747374e7a862d499d8e2e33be : SUCCESS in 1h 43m 28s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/d498252f35ab41cfb1bb6a988086e977 : SUCCESS in 1h 49m 47s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/83ec77587fad4043bb4422408986c924 : SUCCESS in 1h 23m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fe14d14f08a8493ca5fc483820aa8b73 : SUCCESS in 48m 43s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/f635fd173d914c8fac70abe13be8ec59 : SUCCESS in 1h 17m 36s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/73b9d087b6cf4a76a7673548a79bb386 : SUCCESS in 29m 32s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/00d0074b828a46b69da35fa5b6490248 : SUCCESS in 1h 48m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b22945ed92894cd8bde29705be1e26dd : SUCCESS in 2h 27m 53s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1cb2a093700f4926a6fc9f2c44359e97 : FAILURE in 22m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b0c49d3170784b4da2a51d2e7c28ac0c : SUCCESS in 2h 06m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/197398e85e414e1c8729b6e8cc9ec374 : SUCCESS in 1h 43m 18s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/195274a93fec4f18a61274827d2cde11 : SUCCESS in 2h 11m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6326b0a1a2070fd1d2d4c016feac474c486fc727","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-06 15:26:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"78c9bc5e70d93c77889c321c569af7b532462c64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-06 17:30:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fec64842e052426dac24f33f838fae88\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fd5eadce113450b9f1ff3cf1b0a835a : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f8d68190a0c41c1bf864a7484e83f75 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5de019ac136e4ba2ba355b6993ad1898 : SUCCESS in 10m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/617c7605ba32479aaaf134efa26ca7cd : SUCCESS in 5m 30s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3c3d3360f6e7411e900d7d2b9adb0d2d : SUCCESS in 5m 51s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/23a6958ccd4d496fa8e17f682cc6bb93 : SUCCESS in 6m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/63ee9c8e1d0d44f3a9b99af99964ead8 : SUCCESS in 1h 41m 37s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/3dceaefe0f824a8db14959c134b18d02 : SUCCESS in 1h 49m 43s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/9b6dbcb670e0401e82eef5f3a9606cb7 : SUCCESS in 1h 30m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/22b2140af2624bb1a9666ac3658a6ca3 : SUCCESS in 42m 39s","accounts_in_message":[],"_revision_number":1},{"id":"800efe5316668543e0d0c04a9c01efa28bd294a6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-06 17:30:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3207981648db49a1ca0572f941c30f00b15a3d44","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-06 17:32:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed2d27d0929a4b8fa211eb474c552627\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db8fde4a4cd34595967bc8189d084323 : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dbf83a9389e44ab8881885b3c73b12b1 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5704cd977ad6e99dd1385ee820c10bf3ba009bf4","revisions":{"5704cd977ad6e99dd1385ee820c10bf3ba009bf4":{"kind":"REWORK","_number":1,"created":"2024-04-30 11:40:55.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/75/917675/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/75/917675/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/917675/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/917675/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/917675/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/75/917675/1"}}},"commit":{"parents":[{"commit":"d943336c0eb7461838a040e263355003682df960","subject":"Fix fully-populated API with allowed_cidrs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d943336c0eb7461838a040e263355003682df960"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-02 07:11:29.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-04-30 11:40:50.000000000","tz":120},"subject":"Fix incorrect masquerade rules in multivip LBs","message":"Fix incorrect masquerade rules in multivip LBs\n\nThe up scripts of the interface files were called only when the\ninterface moved from down to up, it means that they were not called\nduring the update of the configuration of an interface. So if during an\nupdate, if an ipv6 subnet was plugged while the ipv4 subnet was already\nthere, the up script that sets the masquerade rules was not called.\n\nIt broke connectivity for either ipv4 or ipv6 between the client and the\nmembers in UDP listeners in multivip load balancers.\n\nCloses-Bug: #2037943\n\nChange-Id: Iad78de1764bc3a3f699b5feef9e58999d2efe613\n(cherry picked from commit 5c051a6ab6604c0cc38d59e1fdc0e6b199ed9932)\n(cherry picked from commit 1f38d29b20f0769774517ae95d7d46c2eaff3265)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5704cd977ad6e99dd1385ee820c10bf3ba009bf4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5704cd977ad6e99dd1385ee820c10bf3ba009bf4"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
