)]}'
{"id":"openstack%2Foctavia~917679","triplet_id":"openstack%2Foctavia~stable%2F2023.2~I85c8a615c4f2e08e28939805ae0e9b2028dadaed","project":"openstack/octavia","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-05-14 15:06:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I85c8a615c4f2e08e28939805ae0e9b2028dadaed","subject":"When we failed to load pkcs12 cert print warning","status":"MERGED","created":"2024-04-30 11:45:52.000000000","updated":"2024-05-14 15:07:44.000000000","submitted":"2024-05-14 15:06:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"917679","meta_rev_id":"9186f7c4a67d9c45a4574239045bf3cce60b403d","_number":917679,"virtual_id_number":917679,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-14 15:06:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2024-05-08 13:40:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":2,"date":"2024-05-02 16:51:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-14 12:00:01.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-05-14 12:00:01.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"CC":[{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-30 11:45:52.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":14525,"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","username":"vsaienko"},"state":"CC"},{"updated":"2024-04-30 14:51:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-02 16:51:46.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-05-08 13:40:04.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2024-05-14 12:00:01.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"ce405d9a2f26f72fdb7374e438c883b4c5f8ef7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-04-30 11:45:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c919ee3f7eefa97022ad12c806f2b260a06ea2c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 14:51:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e6946aae3834ac787e73ac2fb1d04e3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/07e9493d18044f8f94c4ec723b611042 : SUCCESS in 8m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/487ff7cbdda94e3e9c63f6f2ca22f654 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b81f672b20b442d6be5a9c0d9c18f9cf : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d941656876e41c2840b6ca8d72dd815 : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f6fc304e9aa648ab9ca38a29ca4307dc : SUCCESS in 5m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0e4bec6426104f2295d5bf8df1c394ba : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3e8d5e789ea4df5a9a5b157780e6a64 : SUCCESS in 10m 09s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/49dc8da52a53425bab496d397ccac203 : SUCCESS in 5m 08s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/96712e5dbde24e53a833679376d062fb : SUCCESS in 7m 33s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2cdd6ce6b10946b392301e3873e15a58 : SUCCESS in 4m 57s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8cae71cd60ee413fb8a88a040c478d5e : SUCCESS in 7m 05s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e7853a2cf7184a39ae1e3f22f9c729f6 : SUCCESS in 7m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/10f104f78cff4e13b44d7002a43d6d66 : SUCCESS in 9m 08s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2592cc53b3d04cbcb0c4f9dee76ff69d : SUCCESS in 1h 37m 37s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/624a40f36ec64534915b6134d7adcbb9 : SUCCESS in 1h 52m 54s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/25528e138af7405499515045e8136ba5 : SUCCESS in 1h 28m 18s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a9bc6f61822c4099a4076b9f053fa88d : SUCCESS in 2h 04m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/75fafe37b779459b8f5a72dc98a4c21b : SUCCESS in 1h 20m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bfa640276ad04749b90bcf4f0cf53a16 : SUCCESS in 55m 01s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/005bd968fcbf46afb90110ac8c1b03a9 : SUCCESS in 54m 03s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/5b61417e78854673bda5181b633f4f34 : SUCCESS in 1h 10m 20s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/74b29cbf583042159b94fe6f1a778d27 : SUCCESS in 43m 07s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ce11b14ea5c04793b747a2d2788b03f1 : SUCCESS in 1h 37m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ec7ac7dab38e49e4b37c8c4fa6574091 : SUCCESS in 2h 59m 11s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2220d8c1be6e42f2a6bb641be0f4520a : FAILURE in 31m 38s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/60192ddb1cd4460b95395425085a6b81 : SUCCESS in 2h 10m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"000658a70822132c8ed917739ff4d444ef432a5f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-05-02 16:51:46.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"142cce03fbb12df8808979401cb5bec1b95af643","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-08 13:40:04.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8f6beeda40ac2ee97ab212ef5ce8aefb9f5cc361","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-05-14 12:00:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"43e3f6b9fc1884d2eee108370ae151475b24a6dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 12:00:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"25fe5b108792d6de17803bbeb9c22a89d0b00710","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 15:06:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13d56afb5d864745bae3dfae683f3a5b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/680c8daa9c4d426496fdc19c8682dcc8 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54357459a9554fee86faaea287011cef : SUCCESS in 5m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec862b3d123a492aa7164dcf439b2cb3 : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/013b05faac0a41c489a2d884ac7b2403 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d25f67bb2a1645148e9d351e15c3fff9 : SUCCESS in 11m 18s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1b6dde3aaab243bd8badaf90e938543f : SUCCESS in 4m 57s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/17bd789bbb9b4c6485aaf8a06e16fc55 : SUCCESS in 7m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b01d08e4203d4b569ae93a2b1b46e3af : SUCCESS in 1h 41m 22s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/3b146cf433ef4401a96b075a959a5468 : SUCCESS in 1h 51m 51s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/eb919359733c4adc93ceb4bc8d81c2fa : SUCCESS in 1h 16m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/949ff54943d640e98b9b82833dbd9bbc : SUCCESS in 39m 24s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/43fdd7fe81da4e05914edf23072317be : SUCCESS in 53m 23s","accounts_in_message":[],"_revision_number":1},{"id":"556a9a6aa23886f98bf07646132aec818ae4cc2d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 15:06:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9186f7c4a67d9c45a4574239045bf3cce60b403d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 15:07:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca4b1abab2a448d99815474a1752e2db\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c7abb83ec1e4885b7b62fb51acd21f9 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dbe59b8b46b3457808035abc5aaadac1f9967111","revisions":{"dbe59b8b46b3457808035abc5aaadac1f9967111":{"kind":"REWORK","_number":1,"created":"2024-04-30 11:45:52.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/79/917679/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/79/917679/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/917679/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/917679/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/79/917679/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/79/917679/1"}}},"commit":{"parents":[{"commit":"56e45454fc8968f7a4513141a3ff5eb48d454ea6","subject":"Merge \"Fix Amphora Configure API call\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/56e45454fc8968f7a4513141a3ff5eb48d454ea6"}]}],"author":{"name":"Vasyl Saienko","email":"vsaienko@mirantis.com","date":"2024-02-19 16:05:25.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-04-30 11:45:44.000000000","tz":120},"subject":"When we failed to load pkcs12 cert print warning","message":"When we failed to load pkcs12 cert print warning\n\nPrint actual error when we failed to load pkcs12 cert and\nfalling back to the default implemntation, as exception may\nnot be related to certificate or its format like an issue\nwith wrong methods during cryptography version mismatch\n\n*** AttributeError: module \u0027OpenSSL.crypto\u0027 has no attribute \u0027load_pkcs12\u0027\n\nRelated-Prod: PRODX-39931\nChange-Id: I85c8a615c4f2e08e28939805ae0e9b2028dadaed\n(cherry picked from commit 96846e7b66f50ed0af7ce99f547e7d2e9cb2cda6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dbe59b8b46b3457808035abc5aaadac1f9967111"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dbe59b8b46b3457808035abc5aaadac1f9967111"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
