)]}'
{"id":"openstack%2Foctavia~917680","triplet_id":"openstack%2Foctavia~stable%2F2023.2~I070d27df8acc069836a8c7d4b0cf492c5b6a6168","project":"openstack/octavia","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2025-04-14 17:43:23.000000000","reason":"Change was submitted"},"27442":{"account":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"last_update":"2025-04-14 17:43:23.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2025-04-14 17:43:23.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-04-14 15:09:12.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2025-04-14 17:43:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I070d27df8acc069836a8c7d4b0cf492c5b6a6168","subject":"Cleanup duplicate DB sessions for quotas","status":"MERGED","created":"2024-04-30 11:48:46.000000000","updated":"2025-04-14 17:44:47.000000000","submitted":"2025-04-14 17:43:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"917680","meta_rev_id":"4b3919f155b7bc9c28a76770b08fdebd65ea14d0","_number":917680,"virtual_id_number":917680,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-14 17:43:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":2,"date":"2025-04-14 15:09:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":1,"date":"2025-04-14 15:09:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-30 14:47:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-02 16:52:52.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2024-05-08 13:36:18.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2024-05-14 12:02:39.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2024-08-02 12:34:19.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"},{"updated":"2024-11-20 19:04:26.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"024409d8dd9ae10a437eb4e6c6b0c7b66968c30f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-04-30 11:48:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9a6170296a4fe85f9660f94ae4a34ac1492be34c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 14:47:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/258fc2427ad845c2b96b98dee78cd9c6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c9520e499064e5a8d91ff8be993b482 : SUCCESS in 9m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6ce446aa0bd4ab1be9e41dea7c80a98 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4fc66b8ddcb7473386c785d7e08c45e9 : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8bbcee781cc44cb5aed95b898e654961 : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/00e6fdb66c4148e894e84e8657627a04 : SUCCESS in 5m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9ad885739856418ab6e9e50b6ff30a46 : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33fd60b189c548d4a6945bae40ff6c1a : SUCCESS in 9m 45s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/0ba3d4a1273945bebdb59082e6e8d7ee : SUCCESS in 5m 38s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/667f2aea7ad74b619f80af6206c6ac5e : SUCCESS in 7m 57s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/13ace686047642b984e616fa43db6164 : SUCCESS in 4m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/649316a06eb340b6990852c29a94a2be : SUCCESS in 7m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b2bf52862d6c49f39423eb3a288eccca : SUCCESS in 7m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a9b0291e55924266be5c539aa7c22850 : SUCCESS in 8m 13s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4b2b7c514e574e3abb836ed2bc9088f5 : SUCCESS in 1h 52m 17s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/3248225ef37e439fafc0ebfa99819263 : SUCCESS in 1h 47m 42s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/02d310cc0e0e461ea8073e727fc23308 : SUCCESS in 1h 07m 02s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a5627f8dbc974a1a944ae37ce5a00dfd : SUCCESS in 1h 57m 33s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/1ec67222b0064c299202f02abb7fd999 : SUCCESS in 1h 26m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c2ed462c377747fb8b0a372627e3002d : SUCCESS in 52m 50s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6656094e923c4b99b080e80ef6efabc3 : FAILURE in 25m 52s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/2e9b177e88cf4cbdbc108f4317117cea : SUCCESS in 1h 11m 21s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/04655fb80a0b4881b44fa5b3d9924fba : SUCCESS in 47m 18s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/90888194e5d14d67bebaaf836ba3d8e0 : SUCCESS in 2h 20m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/00aa414c09eb43fcb51d5f30863c2629 : SUCCESS in 2h 51m 34s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f65c724a0562446496826f8bfaa68c4a : FAILURE in 31m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/31240edc3aff42cc9d1d0f8b120b6577 : SUCCESS in 2h 10m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c74861dcaa5c5663135433eddc215821713fef0b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-05-02 16:52:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"df8e07317571c83942e0b2b3d6a15fe24c9efc16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 19:37:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4c0ff0ad4784f50b568755efb7000da\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/74ea51ae454b43fc856c08ed71887654 : SUCCESS in 9m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b351750a4e224e09b4f3e3ae28bb0dea : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5dc82d58eaa749bbb8f09c76e2d11c79 : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c6992d80109348af8029e5ac04076b80 : SUCCESS in 6m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/db89457ba88a4622bc4906540d498b7d : SUCCESS in 6m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/770dfaa1e6d6470087c360534b8afadd : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c37b3f8e214744a9a682814d2787ec9b : SUCCESS in 11m 55s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/ae3184aed22a4148a6a39529f595da90 : SUCCESS in 5m 46s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/c7c66904b503452b80efb2d200caeb0c : SUCCESS in 7m 56s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/69a74afee4a34d94b31487305b313276 : SUCCESS in 4m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/433b4d1dba1f4d7da706b83496eaa432 : SUCCESS in 6m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/df6affc6f2194f858c8d5f0910d75276 : SUCCESS in 7m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0fa9b655ee81456794bdc4f71bea7aac : SUCCESS in 8m 07s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9f6edd7c714f4316b5bc1aacead971d6 : SUCCESS in 1h 26m 48s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/1cb501e6a45447d5be3387c11680d74a : SUCCESS in 1h 42m 39s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/429b9bf5280246fabebc29e28a0895b4 : SUCCESS in 1h 15m 29s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/10f84c51e0f542449a9d26130cbf543c : SUCCESS in 2h 06m 50s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/2881dcb86dbe4b459c806e11f2546477 : SUCCESS in 1h 25m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d9e3a6bf17ae49da8f6c49c9a988969e : SUCCESS in 53m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5fd84145b6804ad8b6ec7a4f4d887dc2 : SUCCESS in 57m 29s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/2c333b7faf6141bbbfa8aba3c614e81a : SUCCESS in 1h 12m 07s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f54a3fe93eeb421d93844289142b2ced : SUCCESS in 39m 04s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ab828dabc8204c4cb21c439e6e6fb47a : SUCCESS in 1h 55m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/81f23bccd3a9495388af945e73ff9cc9 : SUCCESS in 2h 40m 56s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7c3ecd7079f8485a82b96848676c6b8c : FAILURE in 21m 23s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/9cfa150f30d84bad8ab8a0ea8e0e4cda : SUCCESS in 2h 34m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fc3a2b5445dc8a742ca204768474abff78c1d355","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-08 13:36:18.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"53bf8b3150a18c47c205e0f14604e69a85a5bc72","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-05-14 12:02:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f5af79657178c8b3392bf85893313f6f998a9629","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2024-08-02 12:34:19.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"eab0f7f996b682e0c0fcafa871fd96ac70d41d3f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-11-20 19:04:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"93744ad748e9698835f85dd435db35e169b9f0ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 19:06:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0de0634e46621a1c68ecdec4f52997c355e1e753","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 20:50:04.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ad1c8ae9b1ec46448efba26d66ddc6b1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e65655329e9e4d258618b91007c9b16b : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c0b858be8ed43878b2a4b1c413528ba : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/45af18253ca14cb7a17e159ab29078bc : CANCELED in 2m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f942afe149c9493da462f292992c5bf3 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/335c7eba73ff4f4f9ccfc68b7d98be13 : CANCELED in 3m 23s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6df1bc74522c4a48ba7a89ae3486fe8e : CANCELED in 2m 58s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ea981b9d8710447db338862a2561a6af : CANCELED in 2m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9f9ebd58942a486ab8a6fa5a3f64226f : CANCELED in 7m 49s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/96f24112d8cc4c2081c51b14f97874ac : CANCELED in 7m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/dd30527dae1048e7819635048704f986 : CANCELED in 7m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bd5b9b0756724927b92d5e80f120b7fc : CANCELED in 6m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6a15543feec24bf597e029bdc8277735 : FAILURE in 7m 45s","accounts_in_message":[],"_revision_number":1},{"id":"e5db0adb4e427f4304ad2037921dd25ca852d9fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"real_author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-04-11 22:16:20.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_29244\u003e\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"_revision_number":2},{"id":"4083b517fa8afb55c6a927df30244c9f20cf2b47","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-04-11 22:16:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8f95d42ce36953c2089516554c8139d39343b180","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-12 00:55:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/731c52cf2bc441dab6e16400644ba630\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cbd9e024b27e44ca8168f706c8f8f311 : FAILURE in 9m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a3ff39813484cbb8cc220443bc54291 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c33d302eda7e4a549a84e8da18541ce7 : SUCCESS in 2m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6b79962976a24c1b89c8c88e08670cb9 : SUCCESS in 2m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f86744fbf59048e8853abdb5352a1bd1 : SUCCESS in 4m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/782e3202aadc48d2aea93523d3fb6b12 : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67eaf445fd5d44c7bc1544817513ff64 : SUCCESS in 10m 29s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f3bafacbd83442ed9a0898fcafa44efd : SUCCESS in 3m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bcc48eeb645849d08945094be3e5420d : FAILURE in 7m 56s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7d7acfe8c1ab4fb1a068f5ff9b93a890 : FAILURE in 6m 28s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0d5b356ed06e494e9138d7893dab2be4 : FAILURE in 6m 21s (non-voting)\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/1a482bacb15641d2b7d2e14b4497ac11 : SUCCESS in 56m 38s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/2378dbd5217d44a69ee120806b34f447 : SUCCESS in 1h 25m 48s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/42fca2b4db0a436381236b2b69323f49 : SUCCESS in 1h 24m 36s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-stable-2023-2 https://zuul.opendev.org/t/openstack/build/03949420c21c48759e657dc4159e96d8 : SUCCESS in 1h 28m 36s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-stable-2023-2 https://zuul.opendev.org/t/openstack/build/53f61377b3f849328e927ce78f2eec6c : SUCCESS in 1h 06m 22s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/d03bd8bc47a24cd2bc0618f4714b0be0 : SUCCESS in 40m 24s\n- octavia-grenade-stable-2023-2 https://zuul.opendev.org/t/openstack/build/9b115777021b4570a29104fbaf40781f : SUCCESS in 55m 55s\n- octavia-v2-dsvm-tls-barbican-fips-stable-2023-2 https://zuul.opendev.org/t/openstack/build/b043241d33f04e9093a735df149ec2e6 : FAILURE in 33m 41s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-2 https://zuul.opendev.org/t/openstack/build/dad4df646fff4142aa75dc14ec8f1f31 : SUCCESS in 35m 29s (non-voting)\n- octavia-v2-dsvm-cinder-amphora-stable-2023-2 https://zuul.opendev.org/t/openstack/build/12fdbbd221f94e4a81b6ecb874541f90 : SUCCESS in 1h 56m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node-stable-2023-2 https://zuul.opendev.org/t/openstack/build/75f7788adccf47998fe93ac2e91444da : RETRY_LIMIT in 4m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only-stable-2023-2 https://zuul.opendev.org/t/openstack/build/48265c8f3e0144b0865e48da6ce41ce0 : FAILURE in 25m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-stable-2023-2 https://zuul.opendev.org/t/openstack/build/8ae65306f43e4aa1bd1334c740b66736 : TIMED_OUT in 2h 32m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"31655a7465d9ab50543e04f3831e3335037ec85f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-04-14 07:58:35.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"e3eee0464db3e97c39c7912afc27385a0629909d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-14 10:37:52.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7d1315e42cdb4981b02b536a067c8038\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef71a7e3781b4840b2158ac68207e20a : SUCCESS in 11m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6495605d981c4312b1e1fe1c8aa3bbfc : FAILURE in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/faafc7586dea426bb4123b1438a17fde : SUCCESS in 8m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98c6f54ea53043438b1afa1a232aae08 : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/546b7e9fc8814960b5bc5485387b5211 : SUCCESS in 4m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0df5c2d2e47e49119aef2e691c5d4c79 : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4b5bcba1d8146018eb521c0c9932f85 : SUCCESS in 11m 24s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0fdd627948904e7ea6829c1caef0c4d3 : SUCCESS in 3m 18s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/24ad51bcfca1414cbc0836c430a44298 : SUCCESS in 7m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cbf1868a7a394b0db29f49c85aef23e4 : SUCCESS in 6m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/24441ac1a8934694ab566f135db453b8 : SUCCESS in 6m 34s (non-voting)\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/9d9fc03869e84520a3da50a765364eb2 : SUCCESS in 33m 56s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/244a700a4e0b422491eae54d97df58bd : SUCCESS in 1h 58m 05s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/ecfadd7874f2400e8cfa6fb546dca928 : SUCCESS in 1h 20m 45s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-stable-2023-2 https://zuul.opendev.org/t/openstack/build/03516b227e4f470bb43860878ca2c9a8 : SUCCESS in 2h 01m 32s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-stable-2023-2 https://zuul.opendev.org/t/openstack/build/6eb5a76276404017bc237fd70677e807 : SUCCESS in 1h 21m 43s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/60da7ff1e4d540199619557415d90cf4 : SUCCESS in 45m 27s\n- octavia-grenade-stable-2023-2 https://zuul.opendev.org/t/openstack/build/e1196fc9c6d54e3f9bf422a06c4fa04b : SUCCESS in 52m 38s\n- octavia-v2-dsvm-tls-barbican-fips-stable-2023-2 https://zuul.opendev.org/t/openstack/build/74b13e49602d48078845626b00e34723 : FAILURE in 1h 12m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-2 https://zuul.opendev.org/t/openstack/build/ef9beb04aced44e6bece9e3ca788b81c : SUCCESS in 40m 06s (non-voting)\n- octavia-v2-dsvm-cinder-amphora-stable-2023-2 https://zuul.opendev.org/t/openstack/build/21b94ea7d192482d859b1f96fde27942 : TIMED_OUT in 2h 33m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node-stable-2023-2 https://zuul.opendev.org/t/openstack/build/18a95ab56f714a848ed36f0abf717990 : RETRY_LIMIT in 3m 44s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only-stable-2023-2 https://zuul.opendev.org/t/openstack/build/fb200baea1e54884ba70756a7ab7dbf6 : FAILURE in 32m 59s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-stable-2023-2 https://zuul.opendev.org/t/openstack/build/edee471918ed405396c3783fbccd48f7 : TIMED_OUT in 2h 33m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a0b16c8ce4632b906ac1660e935b65d9b5f3b627","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-04-14 13:13:15.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"27b845f57bc574079587df6a508d96beab864f19","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-04-14 15:09:12.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"09ed9e7f5d76e94d9db12493a06ef74c14aefca1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-14 15:50:57.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfa56abdcab14bed8d787f1714bb47af\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f92c2ee8460045f2a1c9d7609ee8060e : SUCCESS in 9m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70470ca131664055a670601f4bb99a7e : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/054a6b0202d94b8f83d94b3592f0695d : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c327d8e0b3340c3b624b03b09e88a12 : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bceb13e7f0994a6f98936403bd1fc727 : SUCCESS in 4m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/adea754288ff47e9beb5dc1977bd98ed : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da145c45421c4bfd919521b7cdd2414c : SUCCESS in 10m 30s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/402608b5657f41d0a8c8e2f9284857c0 : SUCCESS in 2m 19s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0db9704d0b6e4b1c9fbf47eaa4cc9076 : SUCCESS in 8m 07s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1947fd8dddea4cdb9bf29a3f74756a0c : SUCCESS in 3m 46s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/81aa707dfb274ed89a8a7a0535ab3afd : SUCCESS in 8m 06s (non-voting)\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/8d7eeb276a194af28fafcf681ac036e0 : SUCCESS in 1h 00m 16s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/96e64305fc5d45399ec93b6e3fcecb6d : SUCCESS in 1h 17m 02s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/e78c0359f06d4bfe82de69fb942dc0bb : SUCCESS in 51m 43s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard-stable-2023-2 https://zuul.opendev.org/t/openstack/build/f9d79c6800f74b74ac36025e23de3506 : SUCCESS in 1h 19m 31s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard-stable-2023-2 https://zuul.opendev.org/t/openstack/build/1b624b6ccd704dc48016796bb3a39219 : SUCCESS in 1h 01m 26s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/52f609793d254716af5583d9759b26e5 : SUCCESS in 32m 49s\n- octavia-grenade-stable-2023-2 https://zuul.opendev.org/t/openstack/build/a275cff7aab74954bfc9e1935228af70 : SUCCESS in 35m 05s\n- octavia-v2-dsvm-tls-barbican-fips-stable-2023-2 https://zuul.opendev.org/t/openstack/build/fc2fb694bd88442c9368c7a4ada29562 : FAILURE in 1h 09m 28s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario-stable-2023-2 https://zuul.opendev.org/t/openstack/build/a62ce1c4a7834705b2928a7a1f353dc4 : SUCCESS in 43m 46s (non-voting)\n- octavia-v2-dsvm-cinder-amphora-stable-2023-2 https://zuul.opendev.org/t/openstack/build/45e79d9ab1a44601abadbf7724beea2b : TIMED_OUT in 2h 34m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node-stable-2023-2 https://zuul.opendev.org/t/openstack/build/c31efdc06d5b437280f505537f50833c : RETRY_LIMIT in 5m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only-stable-2023-2 https://zuul.opendev.org/t/openstack/build/babc61a0137041d6bb458380f93e2268 : FAILURE in 12m 10s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream-stable-2023-2 https://zuul.opendev.org/t/openstack/build/2367d526b774448d988f8b2adad9ac5a : TIMED_OUT in 2h 33m 07s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a04fa2ce57d00d9fc5bfc17d58d528c2fa574302","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-14 15:51:44.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c33f58d0c2f52f68784d87a2ce891f5847ac6847","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-14 17:43:23.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be09c67f87964c38b2efc48e56c236c0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25a44a27f0a0443794a58d6e1b57e9b6 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f1bfb984a9141ada5432b2cd89e7d24 : SUCCESS in 7m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd5165cfaec848fba74266ec3d8ca847 : SUCCESS in 4m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2bb256579d1846318bda297f02ccaf3a : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42ff538e17f74e8b910c24abab5c3dac : SUCCESS in 11m 58s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/26af844845844110a2a54360ae19e64c : SUCCESS in 5m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/eefff785bd0747319cbee80f1dfcdecc : SUCCESS in 6m 53s\n- octavia-v2-dsvm-noop-api-stable-2023-2 https://zuul.opendev.org/t/openstack/build/1702778e9c614789b8c810b6c124d452 : SUCCESS in 1h 45m 14s\n- octavia-v2-dsvm-scenario-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/85a4510956274e18b59ffd9c5d96b2d3 : SUCCESS in 1h 37m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops-stable-2023-2 https://zuul.opendev.org/t/openstack/build/e3bc551b1f3a4637a4886ba5f60ff766 : SUCCESS in 1h 17m 59s\n- octavia-v2-dsvm-tls-barbican-stable-2023-2 https://zuul.opendev.org/t/openstack/build/8925fb846bf14bc395cc1d0b396c9499 : SUCCESS in 43m 44s\n- octavia-grenade-stable-2023-2 https://zuul.opendev.org/t/openstack/build/bac1a1ea482e4b8f848e929477cd4f9d : SUCCESS in 48m 08s","accounts_in_message":[],"_revision_number":4},{"id":"7a3b004beb28599b6fd69a0d60045ef73642405d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-14 17:43:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"4b3919f155b7bc9c28a76770b08fdebd65ea14d0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-14 17:44:47.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af8eaa702fea46e89328a51d6d27a553\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/635e425272994284bdd2e983a31427dc : SUCCESS in 57s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"3142d13ab4e422e3ca92e273ec0e660d3c4acc95","revisions":{"ed705496087a6b5e8e4ffe60503bbd45fd51ec9b":{"kind":"REWORK","_number":1,"created":"2024-04-30 11:48:46.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/80/917680/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/80/917680/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/917680/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/917680/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/917680/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/80/917680/1"}}},"commit":{"parents":[{"commit":"56e45454fc8968f7a4513141a3ff5eb48d454ea6","subject":"Merge \"Fix Amphora Configure API call\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/56e45454fc8968f7a4513141a3ff5eb48d454ea6"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-31 09:59:22.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-04-30 11:48:15.000000000","tz":120},"subject":"Cleanup duplicate DB sessions for quotas","message":"Cleanup duplicate DB sessions for quotas\n\nSome functions in Octavia receive 2 DB session parameter, one is a\nsession in \"autocommit\" mode, the other without \"autocommit\" mode.\n\nNow that the autocommit feature is removed from sqlalchemy, we can clean\nup some leftovers of this migration, 2 session objects were passed to\nthese functions (and sometime the same object), only one is needed.\n\nRelated-Bug: #2038798\nChange-Id: I070d27df8acc069836a8c7d4b0cf492c5b6a6168\n(cherry picked from commit 2a1c4709a0f4a14510ef48358f4063034c28027a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ed705496087a6b5e8e4ffe60503bbd45fd51ec9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ed705496087a6b5e8e4ffe60503bbd45fd51ec9b"}]},"branch":"refs/heads/stable/2023.2"},"7e3a271207967ce3ab0b22c07b84abf65fc1b7c1":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-04-11 22:16:20.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"real_uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/80/917680/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/80/917680/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/917680/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/917680/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/917680/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/80/917680/2"}}},"commit":{"parents":[{"commit":"ce0066f03fb605fca20fea06f1d0967e5957f10e","subject":"Handle undefined protocol field in security group rules correctly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ce0066f03fb605fca20fea06f1d0967e5957f10e"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-31 09:59:22.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-04-11 22:16:20.000000000","tz":0},"subject":"Cleanup duplicate DB sessions for quotas","message":"Cleanup duplicate DB sessions for quotas\n\nSome functions in Octavia receive 2 DB session parameter, one is a\nsession in \"autocommit\" mode, the other without \"autocommit\" mode.\n\nNow that the autocommit feature is removed from sqlalchemy, we can clean\nup some leftovers of this migration, 2 session objects were passed to\nthese functions (and sometime the same object), only one is needed.\n\nRelated-Bug: #2038798\nChange-Id: I070d27df8acc069836a8c7d4b0cf492c5b6a6168\n(cherry picked from commit 2a1c4709a0f4a14510ef48358f4063034c28027a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7e3a271207967ce3ab0b22c07b84abf65fc1b7c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7e3a271207967ce3ab0b22c07b84abf65fc1b7c1"}]},"branch":"refs/heads/stable/2023.2","description":"Rebase"},"b6ef20a58de77c836f7cfea68ea89f37ac731717":{"kind":"REWORK","_number":3,"created":"2025-04-14 07:58:35.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/80/917680/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/80/917680/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/917680/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/917680/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/917680/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/80/917680/3"}}},"commit":{"parents":[{"commit":"ce0066f03fb605fca20fea06f1d0967e5957f10e","subject":"Handle undefined protocol field in security group rules correctly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ce0066f03fb605fca20fea06f1d0967e5957f10e"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-31 09:59:22.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-04-14 07:58:19.000000000","tz":120},"subject":"Cleanup duplicate DB sessions for quotas","message":"Cleanup duplicate DB sessions for quotas\n\nSome functions in Octavia receive 2 DB session parameter, one is a\nsession in \"autocommit\" mode, the other without \"autocommit\" mode.\n\nNow that the autocommit feature is removed from sqlalchemy, we can clean\nup some leftovers of this migration, 2 session objects were passed to\nthese functions (and sometime the same object), only one is needed.\n\nRelated-Bug: #2038798\nChange-Id: I070d27df8acc069836a8c7d4b0cf492c5b6a6168\n(cherry picked from commit 2a1c4709a0f4a14510ef48358f4063034c28027a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b6ef20a58de77c836f7cfea68ea89f37ac731717"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b6ef20a58de77c836f7cfea68ea89f37ac731717"}]},"branch":"refs/heads/stable/2023.2"},"3142d13ab4e422e3ca92e273ec0e660d3c4acc95":{"kind":"REWORK","_number":4,"created":"2025-04-14 13:13:15.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/80/917680/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/80/917680/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/917680/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/917680/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/80/917680/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/80/917680/4"}}},"commit":{"parents":[{"commit":"ce0066f03fb605fca20fea06f1d0967e5957f10e","subject":"Handle undefined protocol field in security group rules correctly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ce0066f03fb605fca20fea06f1d0967e5957f10e"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-31 09:59:22.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-04-14 13:12:35.000000000","tz":120},"subject":"Cleanup duplicate DB sessions for quotas","message":"Cleanup duplicate DB sessions for quotas\n\nSome functions in Octavia receive 2 DB session parameter, one is a\nsession in \"autocommit\" mode, the other without \"autocommit\" mode.\n\nNow that the autocommit feature is removed from sqlalchemy, we can clean\nup some leftovers of this migration, 2 session objects were passed to\nthese functions (and sometime the same object), only one is needed.\n\nRelated-Bug: #2038798\nChange-Id: I070d27df8acc069836a8c7d4b0cf492c5b6a6168\n(cherry picked from commit 2a1c4709a0f4a14510ef48358f4063034c28027a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3142d13ab4e422e3ca92e273ec0e660d3c4acc95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3142d13ab4e422e3ca92e273ec0e660d3c4acc95"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
