)]}'
{"id":"openstack%2Foctavia~918201","triplet_id":"openstack%2Foctavia~master~I2fbbfc297dcf8c33de81d883a55bb7bac9dfc4ca","project":"openstack/octavia","branch":"master","topic":"opts-cleanup-dalmatian","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-08-13 17:25:50.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-08-13 17:25:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2fbbfc297dcf8c33de81d883a55bb7bac9dfc4ca","subject":"Remove deprecated [amphora_agent] agent_server_network_file","status":"MERGED","created":"2024-05-05 06:52:23.000000000","updated":"2024-08-13 17:26:56.000000000","submitted":"2024-08-13 17:25:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"918201-opts-cleanup-dalmatian","meta_rev_id":"0632dcee59b3c7902257a1fb911fd514d6511b5e","_number":918201,"virtual_id_number":918201,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-13 17:25:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2024-08-13 15:34:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2024-06-10 14:42:23.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2024-08-13 15:34:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-05 09:40:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-07 15:55:21.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-06-10 14:42:23.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2024-08-13 15:34:41.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"97391a01ca0354eacc76fbd4e3bfddee92a9eb5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-05 06:52:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c04baf68ab31e95600440e29cb02b53a98e48eb0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-05 06:57:49.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"d6ca3fac83c791d6a29cadff36f797707051e733","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-05 07:01:30.000000000","message":"Topic dep-opts-cleanup-20242 removed","accounts_in_message":[],"_revision_number":2},{"id":"e02501ca680a4cc0b072a97c79443df9a8a682fc","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-05 07:01:46.000000000","message":"Topic set to opts-cleanup-dalmatian","accounts_in_message":[],"_revision_number":2},{"id":"d2fac52fdd37f50fb52b78f0e8c37cff55f8d34b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-05 09:40:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1dceb14ffc24302aa1bc75b3f9d70fd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e5ff72768a54a958c86f382ed9807c5 : SUCCESS in 10m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73a49108d61f4c7886ecf60417dc10f9 : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c4c166a6115b4df6845d22fef3aa33d7 : SUCCESS in 5m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/721dfdcaa845431786193e4f4b141079 : SUCCESS in 5m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f1719b28072344c2b16ca4d2604b15e8 : SUCCESS in 9m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7e1efb5911a48e3a27a73830f0ecc33 : SUCCESS in 11m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8eaf9b8b9790462780f4f925649700e7 : SUCCESS in 4m 42s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/26ffcb9db4b34fceaaf7cb5576b0beda : SUCCESS in 5m 17s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/55d1f17779d5465cae982220aa48d705 : SUCCESS in 5m 42s\n- octavia-tox-functional-py311-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/a62f3887fe454301af2a16e2e65ed6bd : SUCCESS in 9m 50s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4d79fd39e92a42e98d3252a8d0499cc6 : SUCCESS in 4m 13s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5789a0acd6c048879ead799fb9b514ee : SUCCESS in 7m 24s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7d5316cab2114cc9b142cf4432f21de3 : SUCCESS in 9m 07s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e954f9dd30574641961b0ca978cbf908 : SUCCESS in 4m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1fce43757375433c9736e65e709bee37 : SUCCESS in 1h 22m 39s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/18888319e942494f8757ec7b977b4ebc : SUCCESS in 1h 49m 51s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/8b726b22db2c47ada36892026503a39b : SUCCESS in 1h 15m 16s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/0c4dc83937064e139c121578c9cbcc5a : SUCCESS in 1h 53m 29s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b1c96eac812b4d51a206809de210a70d : SUCCESS in 1h 03m 45s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/41ee98dfe85248e787574d1f3cf361e0 : SUCCESS in 41m 45s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5ee4aee3408a445dac2386384e5a40b1 : SUCCESS in 50m 41s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/a514010eb7c44aa89a37f17bc762bc1b : SUCCESS in 1h 05m 04s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f1c8907e85e149b8997ad0ffc015f7c8 : SUCCESS in 45m 57s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/30cc9f74cc6e431e910a8dae04ac4f6e : TIMED_OUT in 2h 32m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c1c0cb36149c404fab7edc3389253535 : SUCCESS in 2h 22m 12s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f2183d27136b4994a2267d79986e3126 : FAILURE in 36m 40s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/a4e3508232894639af7200361d1cc66f : SUCCESS in 2h 28m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e52019d482c3a7869ced931c0ad5d85a8296d806","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-05-07 15:55:21.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3fc1ab0c7369337e5cd8c0e28a3beb50b4021c26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-06-08 08:36:16.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"00d93e90e75f43ac2e714ff90adca68a936722f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-08 11:23:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/098c846df9f3464c859a67e78c7d906b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4bb2d622b7f14ce2ae837745194cdc91 : SUCCESS in 11m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccd35a817ebb44e4bc895d7dc26ebb47 : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b5efa81c03c144e9b5f05b75bfc2c93a : SUCCESS in 4m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d640adb26437430fb4f5822cecfca416 : SUCCESS in 4m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/eab9c3d92f104e9f8d826dda42c76107 : SUCCESS in 8m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/812587a8b71c4517b5caa96e31f84fdc : SUCCESS in 11m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bccf33ddf36b4e4ea117a7595abace56 : SUCCESS in 4m 52s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/6b6260ed43d745ada5053f2a4329055c : SUCCESS in 4m 55s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/de6d29e835be450394c3a6a3e9b258aa : SUCCESS in 4m 59s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0c8e694f9b0f474fbb417e7927687278 : SUCCESS in 4m 40s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/37ab7cff42df4ab7b3ea423a25af355f : SUCCESS in 7m 11s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/45d70ec00e784b2c84ed0a09ce6e6301 : SUCCESS in 7m 24s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/494639d0b6ca491b8a4af69accf9efd2 : SUCCESS in 4m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4257a50a42744ef4a0f02016273eea79 : SUCCESS in 1h 25m 20s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/a159d741e43c427abedc95d40a980ecf : SUCCESS in 1h 58m 22s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/fdfadf7299bf4b40b522fa1714a0a593 : SUCCESS in 1h 19m 30s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/64bc0e6b44c44d11a272f2afac5be867 : SUCCESS in 2h 07m 51s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/19a2f54975314e33aca3e2104e860bd0 : SUCCESS in 1h 12m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/89dae62d77744a7a8c18faf8ed868471 : SUCCESS in 47m 06s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2531a2e50f9a43ccb9de703ae480889b : SUCCESS in 50m 04s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/9ac97001cf1e4d32aceb048934db3860 : SUCCESS in 1h 04m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a400b701b4724a7cb3f333bf1dab75b1 : SUCCESS in 44m 52s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/067be50e567244e68b48517b5b5d4d18 : TIMED_OUT in 2h 33m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/540aa9ec2d2042f7b3a58e10838bc323 : SUCCESS in 2h 44m 06s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f0d84eecd6334b78823b1e42f6ade4c7 : FAILURE in 31m 58s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/e3370e40c1ff4873b3348ac5762c70ef : TIMED_OUT in 2h 33m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"65a1092510ffe2e608892b58d28b077085c1d078","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-06-10 14:42:23.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8a61cb9eaa30158477185088434e464801162251","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-08-13 15:34:41.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"bc852351c69513896a69719ff7edc38e063d27a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-13 15:35:17.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c48db77498806257df603f604c43c852f7f0ee57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-13 17:25:49.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06e659e05b1b4c1c9bef77d1c3c41f2f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b50a66ee4c34ae2858c909481c530b0 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2fed29a4c4ee44bab7c3c9d056fcc5e6 : SUCCESS in 4m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/efc9b2de9781409db4565c028286adb8 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cb4b88a8c3a4adc999752a0c7616b97 : SUCCESS in 8m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2449ec8b6014c1d9431f5e71f275e28 : SUCCESS in 5m 28s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a31500ad80a54d66aff3af2c416c7215 : SUCCESS in 4m 02s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/26a7a14e778846d7ac970c6b5503d348 : SUCCESS in 7m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1e271137b9e547adaa8878951d759b1b : SUCCESS in 1h 28m 51s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/a531c5bb50c1445ab175bf2d5aed4185 : SUCCESS in 1h 48m 04s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/776849e791bc46a1ba970f4162891a0e : SUCCESS in 59m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5ef24e5c6b8d4e22b5774b8f8ccfd1d7 : SUCCESS in 37m 26s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f692718661554a59ba54ed257a39f638 : SUCCESS in 42m 25s","accounts_in_message":[],"_revision_number":3},{"id":"d5638a712d203c23570de0f49cff9026406ab55b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-13 17:25:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"0632dcee59b3c7902257a1fb911fd514d6511b5e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-13 17:26:56.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/410cdc24f9bd445a8f077a6c7378034b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fdfd1d6a93040ec919fc0fb80b6b011 : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c34adec66bac4e6986211b4004f048c6 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"fc7c61993ee7f89550889139904122037d1cb03c","revisions":{"3f6174f9037e8d40c4e5da683da76aff1cc5a3d8":{"kind":"REWORK","_number":1,"created":"2024-05-05 06:52:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/01/918201/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/01/918201/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/918201/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/918201/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/918201/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/01/918201/1"}}},"commit":{"parents":[{"commit":"441686799752e3a0e5ef5b2bdfebc19069bbcfc7","subject":"Remove deprecated [health_manager] health_update_driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/441686799752e3a0e5ef5b2bdfebc19069bbcfc7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-05 06:51:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-05 06:51:06.000000000","tz":540},"subject":"Remove deprecated [amphora_agent] agent_server_network_file","message":"Remove deprecated [amphora_agent] agent_server_network_file\n\nThe option was deprecated in Xena release[1] and has had no effect\nsince then.\n\n[1] 5dd7ad9ad8998c5285fe60d9ee4210f87ed45ccd\n\nChange-Id: I2fbbfc297dcf8c33de81d883a55bb7bac9dfc4ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f6174f9037e8d40c4e5da683da76aff1cc5a3d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3f6174f9037e8d40c4e5da683da76aff1cc5a3d8"}]},"branch":"refs/heads/master"},"bc7eadb20ada741ee4138173154b3fb012939456":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-05-05 06:57:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/01/918201/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/01/918201/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/918201/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/918201/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/918201/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/01/918201/2"}}},"commit":{"parents":[{"commit":"874ec0691f57f58b690004cb246ce066604f0ba3","subject":"Remove deprecated [health_manager] health_update_driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/874ec0691f57f58b690004cb246ce066604f0ba3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-05 06:51:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-05 06:57:40.000000000","tz":540},"subject":"Remove deprecated [amphora_agent] agent_server_network_file","message":"Remove deprecated [amphora_agent] agent_server_network_file\n\nThe option was deprecated in Xena release[1] and has had no effect\nsince then.\n\n[1] 5dd7ad9ad8998c5285fe60d9ee4210f87ed45ccd\n\nChange-Id: I2fbbfc297dcf8c33de81d883a55bb7bac9dfc4ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bc7eadb20ada741ee4138173154b3fb012939456"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bc7eadb20ada741ee4138173154b3fb012939456"}]},"branch":"refs/heads/master"},"fc7c61993ee7f89550889139904122037d1cb03c":{"kind":"REWORK","_number":3,"created":"2024-06-08 08:36:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/01/918201/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/01/918201/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/918201/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/918201/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/918201/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/01/918201/3"}}},"commit":{"parents":[{"commit":"fcc61e9ca8e79674cf641eefcb65972e60a25e7d","subject":"Merge \"Replace use of deprecated cert.not_valid_after\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fcc61e9ca8e79674cf641eefcb65972e60a25e7d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-05 06:51:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-06-08 08:36:06.000000000","tz":540},"subject":"Remove deprecated [amphora_agent] agent_server_network_file","message":"Remove deprecated [amphora_agent] agent_server_network_file\n\nThe option was deprecated in Xena release[1] and has had no effect\nsince then.\n\n[1] 5dd7ad9ad8998c5285fe60d9ee4210f87ed45ccd\n\nChange-Id: I2fbbfc297dcf8c33de81d883a55bb7bac9dfc4ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fc7c61993ee7f89550889139904122037d1cb03c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fc7c61993ee7f89550889139904122037d1cb03c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
