)]}'
{"id":"openstack%2Foctavia~919587","triplet_id":"openstack%2Foctavia~master~Ibfa282e7311a06ed69797cb9969aa66459372d79","project":"openstack/octavia","branch":"master","topic":"fix-pylint","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-05-15 11:25:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibfa282e7311a06ed69797cb9969aa66459372d79","subject":"Fix new pylint errors","status":"MERGED","created":"2024-05-14 14:39:33.000000000","updated":"2024-05-15 11:26:28.000000000","submitted":"2024-05-15 11:25:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"919587-fix-pylint","meta_rev_id":"88cc927e559ffaadbbeb251bee1989472d12fc25","_number":919587,"virtual_id_number":919587,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-15 11:25:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2024-05-14 16:05:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2024-05-15 09:38:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2024-05-15 09:38:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-14 16:05:33.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-05-14 17:53:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-15 09:38:15.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"2c2ba8578bcec09a9f9e6e018d40ddf54d48d01c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-14 14:39:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d7aac2aad8b4434ba911f8322d10b1bdfe5e2f49","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-05-14 16:05:33.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e0fd5fa2e40c002c5a3a131dc68d4012ff6ed644","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 17:53:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9052da93f294933adf21b634f2df66a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d5027ae4abf348fbb4e57b9e2daf7807 : SUCCESS in 12m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/280df56239154a96bc44a8c7d60d53fe : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/73fea67555c34b91a8b475612b890f99 : SUCCESS in 5m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/313af0e6ae8c46358f5acd597d62bc01 : SUCCESS in 4m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bb4baeb5db524faab062d0abeb466037 : SUCCESS in 9m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/052ad0d6ee694b45875ad1066cf3d34a : SUCCESS in 12m 28s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/cd79c0aae9a94f41b26eb0531e955e2e : SUCCESS in 5m 35s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/4cb3d38d9ab24e32923b564ced103571 : SUCCESS in 5m 46s\n- octavia-tox-functional-py311-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/8fdad159c463482f8dd8b26d37247738 : SUCCESS in 7m 22s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3c7c3d93b29f4c128a9188275cd15105 : SUCCESS in 4m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/16f4ec1483724255b36515c96ab2015e : SUCCESS in 7m 28s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/184bd3f29ac14144868e95c9cfb0fb86 : SUCCESS in 7m 27s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/24d4b9eb313f482ea97e3f44f5014280 : SUCCESS in 4m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1e0c85932cb14be980e03b34f68d953a : SUCCESS in 1h 39m 44s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/32f3c60d55364319a52a9f96a8d1ce29 : SUCCESS in 1h 42m 47s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/baa697b3406b4770bdd5cfd510daf82d : SUCCESS in 1h 23m 15s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/957df7d0730f486f8eaebfb07ea1fe92 : SUCCESS in 2h 19m 30s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/eafb22804d014e839a41c1c4c019e803 : SUCCESS in 1h 18m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/23447d8d043a43e689dc9b0cbad1a1b2 : SUCCESS in 54m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/19d94c102c2a4fae8611a791cd788900 : SUCCESS in 1h 05m 03s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c3968a3f41fa48cb9cd00dfbe202d814 : SUCCESS in 1h 13m 17s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/79da0ca707fe49f2a2fa955619e7bec7 : SUCCESS in 45m 47s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4ed91e3535404c0092911bb312215c57 : TIMED_OUT in 2h 33m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/48181200a45a459d8af5c0ba6cb60dca : TIMED_OUT in 3h 03m 09s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/231779216a854b49818a2128f06189e4 : FAILURE in 31m 50s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/f4d54e35f5114deb9b85b354bc6f8e40 : TIMED_OUT in 2h 34m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9dbe00c64eec57ea034ca6616c981cc7fb476434","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-05-15 09:38:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2adc1c767a5425b4bafb368844ecb4dd44b8a000","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 09:39:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bcc3f3d5c7fd22089100812963d01d8f9d280b2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 11:25:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35e9adb3a20d4abf8701c69e322e77a2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf392c2fe65a4f46a53013e48894c142 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c92f5282c5e744779edade5bc231e6d4 : SUCCESS in 5m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/809be855ec5647f6b9202966abb1a42a : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fb4dc6858c64e2cb82d3b82fea4c0cc : SUCCESS in 10m 07s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/235a8182176d4c7ab6fdeb69d1c266d0 : SUCCESS in 4m 27s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/acdbab7f6ef34ddabd89c70c18ce1e84 : SUCCESS in 4m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/21a4b9ec4418444a8ac50d810a22191b : SUCCESS in 1h 32m 00s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/4e1e2f5a17094231a31c41cfc4cf7513 : SUCCESS in 1h 45m 10s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/895f9597783d423b83cbfcd32a56cd29 : SUCCESS in 1h 08m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0272ddad65c24db5ab2a74118135afb6 : SUCCESS in 40m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7352cabff8cb4a1ca881eb422cfd9e02 : SUCCESS in 43m 58s","accounts_in_message":[],"_revision_number":1},{"id":"eed7ecb6b2811e9c045ba6ed981a5ebc24489b65","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 11:25:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"88cc927e559ffaadbbeb251bee1989472d12fc25","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 11:26:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fcdf202d659c46eb998e5a13e21255c5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d23c0c4a46504724b4b406a6a18c24bb : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bc259c0bf06a9ff6df2362239ec13e3e028253b0","revisions":{"bc259c0bf06a9ff6df2362239ec13e3e028253b0":{"kind":"REWORK","_number":1,"created":"2024-05-14 14:39:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/87/919587/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/919587/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/919587/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/919587/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/919587/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/919587/1"}}},"commit":{"parents":[{"commit":"5898d54b137f5fb7a1d45bf8d620ef8ab8133259","subject":"Merge \"Drop redundant envdir\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5898d54b137f5fb7a1d45bf8d620ef8ab8133259"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-14 14:37:18.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-14 14:39:14.000000000","tz":540},"subject":"Fix new pylint errors","message":"Fix new pylint errors\n\nFix the following new errors detected by pylint 3.2.0.\n - possibly-used-before-assignment\n - used-before-assignment\n - useless-return\n\nChange-Id: Ibfa282e7311a06ed69797cb9969aa66459372d79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bc259c0bf06a9ff6df2362239ec13e3e028253b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bc259c0bf06a9ff6df2362239ec13e3e028253b0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
