)]}'
{"id":"openstack%2Foctavia~919974","triplet_id":"openstack%2Foctavia~master~I9150da04214f470060020123813bb71235acbc20","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-08-22 22:35:09.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-07-31 08:01:44.000000000","reason":"removed on reply"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2024-08-23 04:06:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9150da04214f470060020123813bb71235acbc20","subject":"Fix failover when using SRIOV VIP","status":"MERGED","created":"2024-05-17 16:49:38.000000000","updated":"2024-08-23 04:08:18.000000000","submitted":"2024-08-23 04:06:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"919974","meta_rev_id":"f805d317f4c3c1546b0e76fabfad70d0ddb77a03","_number":919974,"virtual_id_number":919974,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-23 04:06:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2024-08-22 22:30:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-08-22 22:30:08.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2024-08-22 22:35:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2024-08-22 22:35:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-17 19:48:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-05 16:47:09.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2024-06-05 16:54:19.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"CC"},{"updated":"2024-08-21 16:08:08.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"26e8df90501ed2142dcfd56cbc085b4af27ab9f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-05-17 16:49:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"23af094d67e678639a6fb9501f00bae3e7fd0e74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-17 19:48:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16900faf7d3945f4b4f65eb13c46a81f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b07cd7016f114035b1a9be81f78f6c7b : SUCCESS in 7m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbc2b6dfee1944f588f5b7bbace5268a : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/affcada688b04792b570f205dac601fe : SUCCESS in 4m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0be9128a503b4c8295e2e2b81fb4d12d : SUCCESS in 4m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8b7859c91c9e463bb54ed42dd1a6bc79 : SUCCESS in 7m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e70edd2161714c61ba7c23b840bb153f : SUCCESS in 9m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8552750e42e4e5cb033e8730a18b908 : SUCCESS in 4m 36s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/c3fb10ad7af34d1ab8d5340a71a883bc : SUCCESS in 5m 31s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/b0b8386f17ed401d94daaadd148ee3bb : SUCCESS in 4m 37s\n- octavia-tox-functional-py311-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/31edf106f53e4c1da980931d2542470c : SUCCESS in 5m 24s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/caee7a190eb142c8a4772ba59128430e : SUCCESS in 4m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/aedc3c531d874f7eb057e14cb5515695 : SUCCESS in 6m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5381c557bd544fec8377ca272cd68967 : SUCCESS in 6m 31s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a47a3846ee1f45adbd8cee8e45691fb6 : SUCCESS in 5m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9444c4dddcdb4cd7a68c14c050a43011 : SUCCESS in 1h 38m 57s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/ba4f6e8dd44741f29033862c26bff5b2 : SUCCESS in 2h 03m 37s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/fd34b97725f84f2f8fd06e19ca48c456 : SUCCESS in 55m 05s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/0a94bbe95e5b44d9aa8b71576ff32726 : SUCCESS in 1h 47m 18s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b54a8fb8412448fca6ff2e6a8a8bd7f9 : SUCCESS in 1h 06m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6985de9d7ded43318b7555982d83369e : SUCCESS in 45m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1fa4bc2fa77445409ed4b9d9eea79f49 : SUCCESS in 38m 05s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/e1ac7f6fd2054e74b834bafd811cd6dc : SUCCESS in 1h 02m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/43a2ddb9db0a41c18faf061595eb6594 : SUCCESS in 34m 23s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5686983cd6804083ae08046d883e9ffb : TIMED_OUT in 2h 32m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/82b9d4dcb12e4700b33ccaa8a1177889 : SUCCESS in 2h 57m 24s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f304dc86b0a6435085f59d5add16fa2b : FAILURE in 20m 21s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/e1ca4bdb9d534630a8015c4e2c032b63 : TIMED_OUT in 2h 33m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a078c67f2750f08b521d0430a43b5d07604b199e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-05-17 21:29:52.000000000","message":"Patch Set 1: Backport-Candidate+2","accounts_in_message":[],"_revision_number":1},{"id":"4979b9d12e1c2c94bc7338a4517525fa3c3731ff","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-06-05 16:54:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6c7cf78f0f93ccb2ed6dbc5b82723337dd36bf83","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-06-05 19:54:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"97f4d6604b9fe6b0ad4962bcb986376bd4964757","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-06-11 07:36:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"76dbc5fddd776ed52f807992d51f267a3fae9df2","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-07-31 08:01:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5fcdf8f78911368fe33dafcffe60d359762d3eb4","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-08-21 16:08:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7b438f9e05cf36ff301971027ce6dc13ac576903","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 16:08:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"07253cc19a4a20dbecd0cb021779113e7b873b6e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 16:35:57.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/47e561f84e83401f8b2e838d73d41a37\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef9de74623ff43f1aaec45c4041436a1 : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/82d7889e15a2401c9ef759b59095b27e : SUCCESS in 3m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fa2676ede7ed491eb8fc8e8380733175 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55a61b288f3843749950173064c38b44 : SUCCESS in 11m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5be3806b1a4a4b2bb1e7cba3f161893d : RETRY_LIMIT in 5m 11s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6dcb8bf3a3f34250a3b56b6fd12b562e : SUCCESS in 4m 31s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/6caf0fc93b714e2885c1631381532b53 : SUCCESS in 6m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e7a927bed126462ebc9988554ea34d22 : CANCELED\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/ddb6d4cb0343411a900be10129456c87 : CANCELED\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/f88d74bbe09f4c29b3593c325aac77eb : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/894d6b5318054bb085b7b829fee48407 : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/48a6f90f9c7f4f9689dcd6977d591aa1 : CANCELED","accounts_in_message":[],"_revision_number":1},{"id":"11bf1a8120390ca47116e26db999019e7d2b4e68","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-08-21 23:46:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f36d2e7a970857e2993950d5720fae5ed8d45a6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 02:57:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b5c7e56f26eb424692dedd1826224c75\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4cf3f12d90c6454d863341425d2b2283 : FAILURE in 9m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eda9b010162341e9b04e801bdde8e9e6 : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/260d2c1829c7416dbf53f246a38a12cc : SUCCESS in 5m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1df7d9e27be04c17b9fbf1b86ce2da0e : SUCCESS in 4m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cda7b16bfdfa407b83d18fc5b22dde4a : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e42fb51120e4229886fc46ab9856144 : SUCCESS in 8m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ab56cbe114d4eb1a2414d6b90cf0cac : RETRY_LIMIT in 4m 46s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/8cf145ee4ba54122b1baedb345d6c5d3 : SUCCESS in 5m 58s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/93bb9a7d237947cc815e4c5d543b5b1b : FAILURE in 9m 32s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/93de5ab494044cf7baadf3e5ca8245e0 : SUCCESS in 3m 26s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/02220a280ce74de4b5e9bda05fec2a28 : SUCCESS in 10m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2e2e81c72dda4cdaa38d67b2a60bf4fa : SUCCESS in 6m 19s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d10048ad97fb43208aaf931a5cb97dfa : SUCCESS in 7m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4c98b8dcb8b049cf9191c13c1b58ca06 : SUCCESS in 1h 22m 15s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/97525df380c84af7b6f875cec1e311ae : SUCCESS in 1h 58m 11s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/fe5ebf0ce4044fd9be224ca54ad18616 : SUCCESS in 1h 00m 38s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/9b53e2507bdc4f6889af3b9107c9e42a : SUCCESS in 2h 27m 30s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/81e02f9925e344c0abee5925e8a7f38d : SUCCESS in 1h 12m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5a56245df2ae442aafd08c75ce5a7495 : SUCCESS in 50m 42s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/aaa3ddde92a84889b699979c86697279 : SUCCESS in 40m 29s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/78bd3ea1dc4f4aa3a8d359129718b5d2 : SUCCESS in 1h 10m 26s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/712de7b6ecd94d21b2cadf527f33b777 : SUCCESS in 39m 05s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5084b0fd17e341f197be6ddc9759f2bc : SUCCESS in 2h 18m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4d6b620ac77f419fa63435899aa70178 : TIMED_OUT in 3h 03m 04s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3e6807e9f1b14b419c4ce2cbceb64650 : SUCCESS in 2h 15m 15s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/e8ee0158677041e5a8c7e9805527d922 : TIMED_OUT in 2h 34m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8417180e7ab936191865313fc601f4330f60a532","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-08-22 22:30:08.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Backport-Candidate+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE**\")\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"b5f27e97e97210ecc9952167c0ba35c7f98a3014","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-08-22 22:35:09.000000000","message":"Patch Set 2: Backport-Candidate+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6b1f8cad06f4177b9fed626f337e029f34a5c5db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 01:34:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b1e7845b62844cdbc24b96c2d80ce8c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c1580035ba64847b94b7165f8e5e717 : SUCCESS in 8m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54b9d9e1fb4e4f72b93a2c04cac09d67 : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e188091644bd4582ba7df6e3f5782b9b : SUCCESS in 4m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ac20bb5914e747809ff120753ce7df9b : SUCCESS in 5m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/980c910765bb473d8bc90563e163e8e8 : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e6782ebb523444da9ef7279e849a03f : SUCCESS in 10m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90dd84f8908a4fb1be7d671f9f8ec2cd : SUCCESS in 6m 54s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/13466d481bb34fd290cfb361ab2295cf : SUCCESS in 4m 54s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/77dea3caa2ae4fbca40a9aa213012ab2 : SUCCESS in 6m 56s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0410b3d7cbb94864af12f81e4749a9bc : SUCCESS in 6m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/aa4ed78340b4437f98036565bf48055b : SUCCESS in 7m 34s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9737f2b4555241e6bb2f3027ca960501 : SUCCESS in 7m 27s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/6ef946b94c5f4d7ba671664ab14a0f02 : SUCCESS in 10m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1c43ebc1761642c5a184191e5a9abfe4 : SUCCESS in 1h 20m 24s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/af92908437ba49f2a205cfb24ee11bc5 : SUCCESS in 2h 20m 29s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/d6bdb16413c747dfa2809feaafd8f5c9 : SUCCESS in 1h 25m 24s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/1af3d240105d4312a98863f13f87eaf1 : SUCCESS in 1h 56m 01s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/dc7aed24e69c418692c10fb4408688c1 : SUCCESS in 1h 06m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/350d74d81e14458fb407a1d9838dd126 : SUCCESS in 40m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3c00178b2f7042f8b24dcbcf8ba88da5 : SUCCESS in 46m 17s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/03907977bcce41289b9e5cc7d3214f64 : SUCCESS in 1h 04m 17s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ad45994af67a47928419492775476445 : SUCCESS in 38m 53s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/88462aaec7114bd38be5105b989d46c2 : TIMED_OUT in 2h 32m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b504aa7cca60496cb7b0fd70f5664a66 : TIMED_OUT in 3h 02m 49s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/75013c4a30b54a0494c9b368b1ea8af7 : TIMED_OUT in 2h 32m 30s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/2391bd5681934c3299d7db2c1790cebd : TIMED_OUT in 2h 34m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7fdd0845fa8e2ae2236e2fef6a969cdb5b38dc62","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 01:34:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4a68eff16bcc21452bcf7e653668a5d89498e5d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 04:06:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5548a81afe0643f9bc4fce0f395a2fa6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dcc27eeabe04e7ab378a2ab657c0067 : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/adb1e9f04a3549d8b36e01097e33b0cc : SUCCESS in 5m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aa2086de5d3b4f8ab2122e7a7811218f : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d12239a5adc412588b693e2aaa8158a : SUCCESS in 12m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/597a11d364294cd5bb4b23140fbbbf24 : SUCCESS in 5m 25s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e58a2c487fbf4a179e8f9f386ed9b044 : SUCCESS in 4m 54s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b0f50d1b1f1449b5bfa6b02eafa24b33 : SUCCESS in 7m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/762c925e1ed94a66896fcdc69842e05c : SUCCESS in 1h 21m 09s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/79d90193bccd469d98d5867ad933b0e4 : SUCCESS in 1h 41m 31s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/c2441b7a33bb49a59fcca00d17302aa4 : SUCCESS in 1h 09m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4335663108314452bf60ce628cdca04a : SUCCESS in 32m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c1eb0fcd66484a6fbb85a24883e4cdc6 : SUCCESS in 55m 00s","accounts_in_message":[],"_revision_number":2},{"id":"64ffdd27cfb8d00b0cab9ee4bc5bd0825bc0cea8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 04:06:56.000000000","message":"Change has been successfully merged\n\n1 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":2},{"id":"f805d317f4c3c1546b0e76fabfad70d0ddb77a03","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 04:08:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8f934696b7c44fb83306ee328b272b3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/505b2358108b4f1d824027437c31a31e : SUCCESS in 1m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/556ad2ec5f394dc58b475b324e8ef431 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a4317a80d61ce2d8e699ea7ca17f161ecc4db8ee","revisions":{"c7ae72fe56d6660768ca479d6278879a4734f0b1":{"kind":"REWORK","_number":1,"created":"2024-05-17 16:49:38.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/74/919974/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/74/919974/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/919974/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/919974/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/919974/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/74/919974/1"}}},"commit":{"parents":[{"commit":"b43620cfa8e509647f99835e6eb43416a7f214f8","subject":"Merge \"Fix listener update when using SRIOV VIP\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b43620cfa8e509647f99835e6eb43416a7f214f8"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-05-17 16:46:33.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-05-17 16:49:36.000000000","tz":0},"subject":"Fix failover when using SRIOV VIP","message":"Fix failover when using SRIOV VIP\n\nThe previous patch[1] that added a check for Amphora status missed\nadding a task that generates the Amphora status information when a\nfailvoer is triggered.\nThis patch corrects that by adding the AmphoraeGetConnectivityStatus\ntask to the get amphora for lb failover subflow.\n\n[1] https://review.opendev.org/c/openstack/octavia/+/912599\n\nChange-Id: I9150da04214f470060020123813bb71235acbc20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c7ae72fe56d6660768ca479d6278879a4734f0b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c7ae72fe56d6660768ca479d6278879a4734f0b1"}]},"branch":"refs/heads/master"},"a4317a80d61ce2d8e699ea7ca17f161ecc4db8ee":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-08-22 22:30:08.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/74/919974/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/74/919974/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/919974/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/919974/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/919974/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/74/919974/2"}}},"commit":{"parents":[{"commit":"b71e752c0fff2939e4347a22e55c7cbb3e9c7f9a","subject":"Merge \"Fix test_update_load_balancer_status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b71e752c0fff2939e4347a22e55c7cbb3e9c7f9a"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-05-17 16:46:33.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-08-22 22:30:08.000000000","tz":0},"subject":"Fix failover when using SRIOV VIP","message":"Fix failover when using SRIOV VIP\n\nThe previous patch[1] that added a check for Amphora status missed\nadding a task that generates the Amphora status information when a\nfailvoer is triggered.\nThis patch corrects that by adding the AmphoraeGetConnectivityStatus\ntask to the get amphora for lb failover subflow.\n\n[1] https://review.opendev.org/c/openstack/octavia/+/912599\n\nChange-Id: I9150da04214f470060020123813bb71235acbc20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a4317a80d61ce2d8e699ea7ca17f161ecc4db8ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a4317a80d61ce2d8e699ea7ca17f161ecc4db8ee"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
