)]}'
{"id":"openstack%2Foctavia~920135","triplet_id":"openstack%2Foctavia~stable%2F2023.1~I914aaa2eb57cf4c7206909e8ea17af1033c54965","project":"openstack/octavia","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-05-28 06:54:57.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"I914aaa2eb57cf4c7206909e8ea17af1033c54965","subject":"Fix example policy file system-reader role","status":"MERGED","created":"2024-05-22 15:39:13.000000000","updated":"2024-05-28 09:47:42.000000000","submitted":"2024-05-28 09:46:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"920135","cherry_pick_of_change":874951,"cherry_pick_of_patch_set":1,"meta_rev_id":"11db02f03c912e6cef2a000034664d7e59e9b4ae","_number":920135,"virtual_id_number":920135,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-28 09:46:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2024-05-22 15:42:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2024-05-28 06:54:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2024-05-28 06:54:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-22 15:39:13.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2024-05-22 15:42:26.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-05-22 18:14:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"86b64897fb3ea3ed28001963f845da102fbe1f29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-05-22 15:39:13.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"66e5f969905c90822825de2199052d51be9df90f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-05-22 15:42:26.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ebdfb3b8d96b9321ab643579f827df1851aed017","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-22 18:14:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/65296579294c47709e4481ef3af8cb86\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f68344a0ff684348a42de266527a2d4c : SUCCESS in 13m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5719eefb860349f1be9ad6f452210e9b : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a23131534d44fe6b2105a0fc4c03ca7 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/281f1625f12b4bd0a0547cae3ca31f67 : SUCCESS in 11m 24s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/3110c2fb45ae491c9efb5d68c06ef314 : SUCCESS in 4m 47s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/6766e87cb958449fab1797ea65211517 : SUCCESS in 11m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/24690ba9b8b64c9198af7f80c1cbae4a : SUCCESS in 1h 38m 54s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/34a4466a9ae647a9a76a7bb8ed258945 : SUCCESS in 1h 19m 29s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/bb2d27b031564ea79413dc63b3ce8c7e : SUCCESS in 54m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b284b13b9fe945d0bcbd70cf74954b97 : SUCCESS in 46m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/350cc75e2f6146d59a030391cf4848db : FAILURE in 32m 33s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/c75a0618c72d4e70a7ed326cff95b76d : SUCCESS in 52m 41s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/448a86d1f0d944c4ada593a4bf984db9 : SUCCESS in 1h 08m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/25a63f8a782d49b398dba340eca80edd : SUCCESS in 35m 14s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/448d7f2538f64e30a57fd45ea8131c57 : SUCCESS in 1h 59m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6e2832a864b14fb2aae91a9ac28ae7ba : SUCCESS in 2h 12m 13s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/85b186175e45450cb3744ffcea0bef32 : FAILURE in 32m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e9a6c316bbab4507ad157d23249918b1 : SUCCESS in 2h 03m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/83ccc9ce5c4742a08860dc08001283fc : TIMED_OUT in 2h 32m 43s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/216a185de58a4599bdf7e4610199fd5a : SUCCESS in 2h 10m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"10d893fb00faf3cab859034a36da4877214210d4","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-05-28 06:54:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dc31b663e4aee6b185e6ac6a45ae831e2d451954","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-05-28 08:08:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c53ee3eaad5d6e52a8ee002ba47a09c7fb1f16b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 09:29:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35218528068a4e028bfcff44f9daefe4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a29e5b5ef82c43299dc5c9f6024c53ed : SUCCESS in 11m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/039ec187ea104d929913883c947143cb : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/067d7974f5b745a5993d0f9d2bb85d56 : SUCCESS in 7m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f112d1aa5c74fdbb6886d9a07ed7422 : SUCCESS in 11m 55s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/6fea5fe56cdc4b269a6fd710d9ca9a59 : SUCCESS in 5m 18s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/57077c9255684fd4b0526db99469ec7b : SUCCESS in 9m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a992773716bb403992822031de3c45fb : SUCCESS in 1h 32m 07s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/aca1b062c72f443196175c4c67420a43 : SUCCESS in 1h 21m 19s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/184ca2710bd546e097e3c0cedaad61bf : SUCCESS in 55m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ebed57e6f7b940f08e61593b1b8f20f9 : SUCCESS in 34m 46s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/b2565845b5d246a0a8441453674ab7be : SUCCESS in 1h 01m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b364871d51bd450fb2fb2d3ac25f1d97 : SUCCESS in 33m 06s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/03b2b21944a9488195f870566527459e : SUCCESS in 1h 40m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d2e830179a3742b9a7a6b828b638b336 : SUCCESS in 2h 16m 06s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/452ef06e558d45cd84a4ed9eccd5fa69 : FAILURE in 25m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/1dbaeae784044f6694335fe3a77901d9 : SUCCESS in 2h 32m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/856e970300b148f28a399e9bb7b87c64 : SUCCESS in 2h 03m 37s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/9363b124342a40eab6526c1418b77e16 : SUCCESS in 2h 27m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4e097d2dd39339ab72778e6424640162a0f663e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 09:29:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"57f8f18e8dca3eed614dd57de43c2e8ac08a013b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 09:46:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf6cb9d20d7b478f826a2f823594c1e0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03558ae6d63a46e2956a5104d3a47b61 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/288eb75b373644129484ff09b264e84f : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d24a88dd81f4fddb64d265fd43fce08 : SUCCESS in 12m 31s","accounts_in_message":[],"_revision_number":1},{"id":"5f7c56bd28c8cdc67261568cc0dbf7e211caec93","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 09:46:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"11db02f03c912e6cef2a000034664d7e59e9b4ae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 09:47:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe33f60bdd844d5caa2d091f9033db77\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a57d8b2be914160bc3f5d3fbb78e9b2 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"540e3d349960db81311531b385cf9be1d09e6934","revisions":{"540e3d349960db81311531b385cf9be1d09e6934":{"kind":"REWORK","_number":1,"created":"2024-05-22 15:39:13.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/35/920135/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/35/920135/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/35/920135/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/35/920135/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/35/920135/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/35/920135/1"}}},"commit":{"parents":[{"commit":"06251e9716719dd8f62144ecd41d26360a539f7d","subject":"Merge \"Fix TLS-HELLO healthmonitors in the amphora-driver\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/06251e9716719dd8f62144ecd41d26360a539f7d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-02-23 18:22:52.000000000","tz":0},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-05-22 15:39:13.000000000","tz":0},"subject":"Fix example policy file system-reader role","message":"Fix example policy file system-reader role\n\nThis patch fixes the sample keystone_default_roles-policy.yaml file to use admin instead of reader for the system-reader role.\n\nChange-Id: I914aaa2eb57cf4c7206909e8ea17af1033c54965\n(cherry picked from commit 32988c0ae82ac8b52a95d5bf5b2f087d12230642)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/540e3d349960db81311531b385cf9be1d09e6934"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/540e3d349960db81311531b385cf9be1d09e6934"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
