)]}'
{"id":"openstack%2Foctavia~920156","triplet_id":"openstack%2Foctavia~master~I737e3228eccacccf5370a666c793a0714dbe0b24","project":"openstack/octavia","branch":"master","topic":"octavia-sriov","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-02-26 21:01:14.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2025-02-26 17:36:52.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2025-02-26 13:10:37.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"I737e3228eccacccf5370a666c793a0714dbe0b24","subject":"Add support for SR-IOV member ports","status":"MERGED","created":"2024-05-22 00:13:03.000000000","updated":"2025-02-26 21:02:35.000000000","submitted":"2025-02-26 21:01:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":25,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"920156-octavia-sriov","meta_rev_id":"554fb2736cdf3daac974a5245ce0f837c9a82b7e","_number":920156,"virtual_id_number":920156,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-26 21:01:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-02-26 17:36:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2025-02-26 16:12:11.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-02-26 17:36:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-22 02:54:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-14 12:04:16.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2024-08-22 06:20:25.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2024-08-26 13:14:16.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"cd0c0640a8c561a4aa0fa8b039b55ed8f071939c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-05-22 00:13:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"087843c622f97f5a2367b70d0f0385776c4b99ce","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-05-22 00:13:26.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b722ea3ca22c9db5dc914e67ea56ddce5a75c872","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-22 02:54:40.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/56d5b2e69ed947b7a03bd6eba50fc3df\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7cd72b82925045439110f79fd7a97a3b : SUCCESS in 4m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f38c62e51f3d41b09f6732f1ca0162da : FAILURE in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/505cb5720aee447da686786ada398455 : FAILURE in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/71896e510ba04d34a14a6a0d298ff2bc : FAILURE in 4m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/555e7340e0a3474fac80193d6546a00d : FAILURE in 5m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/657c6acaff57423aa3e845e27eaadf4a : FAILURE in 8m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c53e4235672645699e458d6bc3d7f72c : SUCCESS in 11m 33s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/9e0dc8427f9b40c7b089417d79b69c6e : SUCCESS in 5m 12s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/5ce5efba57874313bbf32c461d0c5330 : SUCCESS in 5m 15s\n- octavia-tox-functional-py311-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/5419c1865f9f4ef383243b738a0ac10b : FAILURE in 6m 04s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c2a3280dc18040f5a64944da8de257b5 : SUCCESS in 3m 46s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0dc03dafa1754c1bacac68a5566f0fa6 : FAILURE in 6m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/25eee44d194c4f3b8bf171324df3bf09 : FAILURE in 6m 30s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/66db8220b2eb44d6a4552cbd836046f7 : FAILURE in 4m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d5f2037e88c64d848f57a88446d78a3e : FAILURE in 1h 32m 18s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/fc7d8f08532d4cf78eccdee78cebd13c : SUCCESS in 2h 05m 23s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/16da2c1d498244fc81f75dada0f6cffd : SUCCESS in 1h 13m 58s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/cd03d5a74be742149fa1c2f92847399f : SUCCESS in 2h 09m 33s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a809a396773b404bbc754701a627e26a : SUCCESS in 1h 04m 47s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e63cf7e7aaf04feabb822be178b2e24e : SUCCESS in 47m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bcab4ea4fb98411cbd502925af3452ec : SUCCESS in 45m 30s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/9f66c3a2521544bfb0bd6142b67969ea : SUCCESS in 58m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7a9af1f6cafa4296b41fbba5b6aac057 : FAILURE in 36m 45s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2a403563ccf646b8b85d3870026a72c4 : TIMED_OUT in 2h 32m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a3f985e6e89144468f90db7296361d1b : SUCCESS in 2h 17m 50s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5d66b6394664458d8baba3c60843b079 : FAILURE in 27m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/4a79243ac0ff4043b1c260dc18d83f23 : TIMED_OUT in 2h 33m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"25aa75ad24fc2e918df07c673a02d501a605f91f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-05-22 15:54:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5a2cecfa17f6fe52a77d9145c92cf51540cfe7bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-22 18:36:23.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1070786ac1604bfe81ab35b44d85180e\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1a53b43072fa4e06a107a86672c3cdad : SUCCESS in 4m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c28508ed938b451ba1b9db31013961f9 : FAILURE in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dbbf01736774a3e89c6e61d1634635d : FAILURE in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e8a66de3bc684709a45e32bff0187bdb : FAILURE in 4m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9892a5345fe746d080e125687ec6b530 : FAILURE in 5m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bf3e84215b52482d8398f4e6b6750597 : FAILURE in 8m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b03dcca427d24369a77bf605ed9de1de : SUCCESS in 11m 47s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/c8da75abec6648eca2f864aa1752196e : SUCCESS in 5m 22s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/3ecf9e4b8a3443729ebb493b10883f6e : SUCCESS in 5m 24s\n- octavia-tox-functional-py311-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/92902644a50e4e4491389bab3680a857 : FAILURE in 5m 42s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/baf756ba620e4342a77157515e3200c5 : SUCCESS in 4m 46s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/60b95c445b2c4bbba3806ee6b2de6b09 : FAILURE in 8m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0ded986fd7364bd08dbc26cc5aa6c81f : FAILURE in 7m 43s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e8989cadcea346da85105c26a047a232 : FAILURE in 5m 15s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/df5745e2730648f7991efd5830f30278 : FAILURE in 2h 05m 26s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/fd289f861bff4c3ba216bc926a06a3ed : SUCCESS in 2h 18m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ec382531843c4838a06a3dbf3d37cb4a : SUCCESS in 1h 18m 11s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b9ebedee99ef440ba0119fc9ff811c05 : SUCCESS in 2h 28m 29s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/8061a030f4f34d26be4d087566309088 : SUCCESS in 1h 22m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0c27768baf0d4ee68e2aea65506bff2b : SUCCESS in 48m 02s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fbe1d6383b4e41049641a34aaec5cf96 : SUCCESS in 58m 11s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/5b179857d3f74fa0bd5ea05cfd3a1562 : SUCCESS in 1h 16m 02s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6643761b09344ae6bc9aaa8d618446b7 : FAILURE in 38m 36s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/64908a29f9584810b4be5a689d7f5bc2 : SUCCESS in 2h 11m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8d12b06af1124f749ff557769c6b1e39 : SUCCESS in 2h 24m 53s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3ac370636261400aa028029d38047490 : FAILURE in 23m 31s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/a749af76ea324092b7844966a248b4d9 : TIMED_OUT in 2h 34m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"70c89e339169cd96d444907e23e28309f7b6b16f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-05-30 04:16:06.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"36b3c1288583bd78a0e880ae00011fcbb24e3c50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-30 07:29:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/45b16f7f1bd54f76abcd5f3479249752\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5af4eac685674806b0d5fe40a09f9a85 : SUCCESS in 4m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/803e4b15b8904b938a3f6906f55999d4 : FAILURE in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/553b8ddd478e4165aafe1ddcddb9e747 : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3fa902b8ddd84eda80ef311f0e00c29a : FAILURE in 4m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e59f1d1125594a46b42cb40b5684a0c7 : FAILURE in 5m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bd13a85b97fc44009bdb5a01f892007a : FAILURE in 9m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97528e5436c24230b51a2fcdbd953831 : SUCCESS in 9m 27s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/6a20a68dd5be4aa2b0ffea950f72aeba : SUCCESS in 7m 20s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/efde893e586c43fe9453cf532cb8f229 : SUCCESS in 6m 26s\n- octavia-tox-functional-py311-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/144cbbfcfa944d168acbc174fe76ca05 : FAILURE in 7m 39s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/168e31f9163844508bf4641489c067d0 : SUCCESS in 5m 36s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1f59c12ac8ad4398a30453a91eb313b0 : FAILURE in 11m 09s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/117cd9c0845a4f7fa087acb6b834d605 : FAILURE in 9m 06s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/571970b3f65f4700a8d7427a8e0ef5a0 : FAILURE in 4m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1c0132fead794ea8a9fd1ca193bb594d : FAILURE in 1h 51m 17s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/4611274be5cf458c939b9024dd809d9f : SUCCESS in 2h 11m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/75a9eac7bb664a2093d7b75b93c2d65f : SUCCESS in 1h 18m 54s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/edb458f182ec46bf816674e33beb4749 : SUCCESS in 2h 17m 54s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e23c6f5c36e446a6a85863601f394f8b : SUCCESS in 1h 24m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ecc2b4f974a04ae48a2995977c2c609b : SUCCESS in 46m 55s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5a4193b6162740d580cc55ff1b1e7c19 : SUCCESS in 57m 56s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c83dab07c72b4196a8115a9265bd88c3 : SUCCESS in 58m 31s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/415408e57d834c609bc5a7c25f58d279 : FAILURE in 46m 46s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3ec93395875f401e8d55b3bf3c0c7956 : TIMED_OUT in 2h 33m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7e1af4d2eddc453a95d3955cfbb564da : TIMED_OUT in 3h 03m 02s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f40654b45073417c8083dddb4864a366 : FAILURE in 28m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/d499212493ef45cd975db9fad4f610a3 : TIMED_OUT in 2h 33m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"85be705cc389ad833fa0510bb5fd6b4302236082","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-06-04 15:48:48.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"56d1405a6be0ecb08f8cf7634d1f1fcc72f3165e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 18:55:57.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e4912a212e784a6eb87ea7b382f9cbaf\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/430a63d80c3c48f59faa25f680fbc28e : SUCCESS in 4m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/693f9333e43e487282b554db348163e4 : FAILURE in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6332c20440f424a90bd4b2042718910 : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9bcd77e421de473b8096d86a2c0ced7e : FAILURE in 4m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7b0ce5d0acb44fb99628dc51794d84f3 : FAILURE in 5m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/17eb75a43c944f839f1d7c7c63fd1403 : FAILURE in 8m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebe641ea9ced4a13a95685a9e8038875 : SUCCESS in 11m 59s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/50a70bf7ba134a92825fde78a61085e2 : SUCCESS in 5m 49s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/5352390e760c47f5be7ddf1d976dabc0 : SUCCESS in 5m 19s\n- octavia-tox-functional-py311-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/2723ddc7e96a4c3e9ea3b80a894225a8 : FAILURE in 7m 04s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/83836c986fda4434a731b256e207f19d : SUCCESS in 5m 39s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fbb4897048c54142ad6d3176b0b8f9d8 : FAILURE in 8m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d1c33edfdb004686964ae5f08373bc8c : FAILURE in 9m 14s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/356eb5ddbb6c478ba3db4ac41b5e2cf8 : FAILURE in 6m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/59ebf44d10544850bf6cdfd9ce5ffc0d : FAILURE in 1h 47m 37s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/48fc4ba09e044198a51c3d858e5c754a : TIMED_OUT in 2h 33m 58s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/d227d86258504a349f896accdc097dec : SUCCESS in 1h 18m 56s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/6f9fb878aded4a1ba533055af3f27b93 : TIMED_OUT in 2h 34m 29s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/241d0cb1d7be47b8a5e767f36db3148f : SUCCESS in 1h 05m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/579e7bbe59764937a42154908a999e33 : SUCCESS in 53m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a15df91e3541481d8247e3b9a0f46b35 : SUCCESS in 44m 02s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/e56aba09883b4c9280e3e7db21a9f973 : SUCCESS in 1h 07m 59s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7f69127569a74a87ada78783f5f8a086 : FAILURE in 42m 05s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b16d5816e35c43afbf08ba901addce69 : SUCCESS in 2h 29m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f6ae9010a20b41c0ace2975bb4cff374 : TIMED_OUT in 3h 03m 18s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/aa37e01bbe644078814ffcdc665c09cc : FAILURE in 22m 06s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/d3e9b771cd14470093dedf49bfcde65b : TIMED_OUT in 2h 33m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"335d7c85c2102033f0cc9008023c017dbc1d6977","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-08-13 14:04:56.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"4ea70b551d0f43ca86146266488d86a40852a56e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-13 17:06:30.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1c6ef871a339405bab99c9266b29c83b\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0af2f483273c4198b27234d59838ff84 : SUCCESS in 4m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c54ee6ad17164d62a0b5a6b54703356f : FAILURE in 6m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72b9f239d0e64b01b3669c60ba26c316 : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2eb1999301654fff9822dbc29a091689 : FAILURE in 4m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/af20114caf174dc5a923ac1aa0e8c25d : FAILURE in 4m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2c70353aa1ea44ebb6452efddb3c4852 : FAILURE in 10m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/699cd2283bdc421188e9307f818985a3 : SUCCESS in 9m 15s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/fe509e0b719446e39855dc2e7dd7e408 : SUCCESS in 4m 54s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/e25b4a2149f1448e9e887aa6a16ed4e3 : SUCCESS in 7m 44s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d277c3e44957450ba3ce8d4413ba9fd4 : SUCCESS in 4m 12s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c457ce8a0a1a4e71a0012c69b0ecc1aa : FAILURE in 7m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4c8b19a9fced437d8c0c66d53884fe32 : FAILURE in 9m 20s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c258c964e8f740fdaeceb6ea909ace79 : FAILURE in 7m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/56842ebecfec4c319b2ee82bcd530844 : FAILURE in 2h 33m 02s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/38efee168b224e609e6c1f3492956077 : SUCCESS in 1h 50m 35s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/7490969cfddc4a87b3d60b9089ceb0c0 : SUCCESS in 1h 01m 27s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/79fd2c6585974c68a36a148f41694929 : SUCCESS in 2h 35m 11s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/601e50bdf42d4606bdc20163ee416107 : SUCCESS in 1h 12m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e39daa534c19478580ce1fbabb2a7654 : SUCCESS in 44m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/879f783a0b8d4699b3c0e584987393cf : SUCCESS in 1h 00m 25s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/13df10caccc144c78eb0840b65a2ef7d : FAILURE in 50m 09s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2dbf4be797724186abe2c8b13c06482f : FAILURE in 40m 58s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9a2e8f69f10d41829857ce1e57e3dc44 : TIMED_OUT in 2h 34m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b5c5e81c0eab46ab8b34c3a1e165fc8f : SUCCESS in 2h 44m 26s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/844e0aaa9d264a0fad4ab8cfc10aa21e : SUCCESS in 2h 30m 53s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/32925d67631b4faeafe8b2d992d24b24 : FAILURE in 30m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d340207696e058e65449c22f3bb33ce8e8b21c66","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-08-14 12:04:16.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"33e01e2376289de164ada91e91414221c63a98bf","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-08-14 22:43:25.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"94e40f106dcc74ea7c492f217e2694306e334e85","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-08-14 22:43:36.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a5d4e7c8fd07174d587470fc0d52a4cd48c1b549","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-08-14 22:44:56.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"63e9f680654c0756f909aa76ae5db33fd39c4b55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-15 01:24:58.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8027f9e2d68e4b06915acfff16b56bd1\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d0935027b3754af6a87c5ceef91dbc11 : SUCCESS in 3m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf1d14d8a41848939b8bf1785933713e : FAILURE in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f5030c9d6a7426fba8a630559873353 : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/09ae5ca25ed749a1b96d6c2b4b443e45 : FAILURE in 3m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7e8010870aec490dadf6af061fd2feb3 : FAILURE in 4m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/98078f6f9c7e44f085fa54ab057d8715 : FAILURE in 6m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d1801e41c9f47988d43c00b72b82764 : SUCCESS in 8m 08s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/4baa7a15e2f8430b9cd7bfa84e6a603e : SUCCESS in 4m 32s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/700001d84acc49b4b2fdf208cb140cd4 : SUCCESS in 6m 44s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/18b96395b3af4d8fbe6e4459ed6213c9 : SUCCESS in 3m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a9f8608778eb4108b968012f4b15dad5 : FAILURE in 6m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8fbc55f753c14d598a662f7dc7aef88f : FAILURE in 7m 42s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/740e75722160409ba056c1276ca7eaa6 : FAILURE in 7m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/95e611582a9640239d4beffa23f39b53 : FAILURE in 1h 42m 00s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/250da20d2ed940e0b1c9dce4064f2b15 : SUCCESS in 1h 46m 53s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/6d5c3cf3b21f4415904a2f33c37d2053 : SUCCESS in 1h 04m 07s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/fda8431461f348eaab4db52b84281357 : SUCCESS in 1h 51m 36s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/0dbd4d372ea74e9e8fbee77687191c1b : SUCCESS in 57m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/aeefb6a632ed401abc1eff08e6739f6a : SUCCESS in 34m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7831bcfc2fda47cea5fa9daf9be7191c : SUCCESS in 48m 23s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/a325cbad5af8487ca123e9b185f4602c : SUCCESS in 55m 03s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5bd4d1ad2dcb40918e5e89c6e1c8e7aa : FAILURE in 38m 45s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1fc9ff3b12b84627813040e5f06946e0 : SUCCESS in 2h 13m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d5c48a2ef85f4808b4669fcfca5992f0 : SUCCESS in 2h 30m 51s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8e1695b788074a15aeaf2e665cbb5a43 : SUCCESS in 2h 03m 40s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/9ce144de03a149c2b989394243daad0f : TIMED_OUT in 2h 33m 09s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c3e0af66b03b395325f5be31a47bf762f567d89d","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-08-19 16:40:40.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"cb4a2676625181bd3bda1a51bb680b8d77952b23","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-08-21 23:50:36.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"83cafd5cbe51a9b4d0f45409add24ac8bf22ffb3","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-08-22 06:20:25.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7a239a6094fd23ba3595b20eb68a22bd1117fac0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-08-22 22:12:06.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"d9015924402311e8dcd9eea2bf15dbb347736fd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 01:09:20.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/447db80afa6744189833b19e1a60edba\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/59cd7fbbd64e4ccc9b107a594029e26a : SUCCESS in 4m 07s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6b0a0e6fc3bd4655b993cb215e571cec : SUCCESS in 1m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac5e433f52eb42c1b165c0780246188d : SUCCESS in 10m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13876f140d974f069b8a4462fe0b74e5 : SUCCESS in 6m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a40809aafe045eeac5d3f088efe1bbd : SUCCESS in 4m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/14611fec8d974a7da87a8875e827ea14 : SUCCESS in 6m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/07d6d8e6a72447b3a9b183b662ad832e : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5ff32db679b49a394b2d9b7a705fb76 : SUCCESS in 10m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10fb622a7a3040aa913013abbb97c1ff : SUCCESS in 7m 02s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/b7af23a80566449caddea19e8fc3aca6 : SUCCESS in 4m 56s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/866cd7a437924a42aebe7dfb5ce4f552 : SUCCESS in 8m 03s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e1799041b5cc45408945c7472e795809 : SUCCESS in 4m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/964dbc0326d74f15b052f1fc8338275a : SUCCESS in 6m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/31608419b8d14fb698bad65998535832 : SUCCESS in 8m 50s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/80f7e965e07648df9fc424c2f825caad : SUCCESS in 6m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c924ff5adffc47fdb5dab2bf1f199d27 : FAILURE in 1h 54m 18s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/85c66cb127a24bd0bd5205f9274c3a6a : SUCCESS in 1h 46m 21s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/330dd514a92b442fa37d7148e4719d2f : SUCCESS in 1h 01m 46s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/7dfc4b74fa734e80ab5e852cea5c6471 : SUCCESS in 1h 49m 29s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/022ea1ce77bb4cd98ce6b82d434f3ba5 : SUCCESS in 55m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/48fb5ad4b4e9439b9c3481ac35cdfe18 : SUCCESS in 33m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1da3cf94231a4449a16fa40080a4bc04 : SUCCESS in 39m 44s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/776193418e374715b5983828b9c6d952 : SUCCESS in 1h 06m 14s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2300a6ef8d874a9ea0e1a0fc4739a069 : FAILURE in 44m 00s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5b840b6a0c914853930616ed6dbecac9 : SUCCESS in 2h 24m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9d7d143a879e4808ac1158445b902cdb : SUCCESS in 2h 54m 43s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/08f6ac0c6f6045e5866d66d0a15af5f6 : SUCCESS in 2h 18m 17s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/96861802478c426e9a970ba44f68ea0d : TIMED_OUT in 2h 33m 42s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"0c699c3d49c7cbaaf0076120aaf39462bc8df274","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-08-26 12:40:10.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"cee27c8cf9aa0d844604777b59fd6de799e6eed8","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-08-26 13:14:16.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a3707446bb2b24bd7dbfd4c860911ff8aa5c3f7e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-01-15 04:06:14.000000000","message":"Patch Set 6: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"cefc84f8389de38d82e68ec94949d212265be53d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-01-15 04:06:25.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":7},{"id":"10db3836b609ce7853db3efb96ca47435d3acc6d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-01-15 04:06:40.000000000","message":"Patch Set 7: Workflow-1","accounts_in_message":[],"_revision_number":7},{"id":"a269d3a5c66a7834f3db02b0718c0c56bd3c94ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 06:55:33.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/247181ffb3074edb863339b0332225ad\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/70b0e63d8752441488041cf4706e051b : SUCCESS in 4m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/881f7c7b5abe4eefa0ed683a72f529c1 : SUCCESS in 2m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd537d37f09b48a3b02b635c0a83c29a : SUCCESS in 10m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7375a4c733c245da8caeff750a8f00a2 : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df2408ea7af745958f9f42ee4fcdd7fc : SUCCESS in 4m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/42ad1f8f64d24c489286f661b432fc96 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0663171b9834d46a28b44be1934d480 : SUCCESS in 10m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c330b02a67045a5a2f94d291cd2d0de : SUCCESS in 5m 10s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/b936ddfe961d466fab7161b7fd51580c : SUCCESS in 4m 20s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/439ecc1552904563b05fe3dfb70db56d : SUCCESS in 7m 11s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d80f70947caa4da487862672461bbaff : SUCCESS in 4m 49s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cee2621ef07747b297bf518ee9ffdc6c : SUCCESS in 9m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7b9d4507e9cb4b8ab8889c1566e6c8fd : SUCCESS in 6m 46s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/bc19744cf4bc4b23a9a3cbc7dd8b7c54 : SUCCESS in 7m 30s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ed09a7b37fa64d15be877bfc1a5595b1 : FAILURE in 1h 35m 03s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/7be7ecc4521f4aa992aaa8d9604f1a90 : SUCCESS in 1h 34m 23s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/2180e2ae5424434081b19593e69e0423 : SUCCESS in 1h 01m 16s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5447f05167f847deab401cb306ec3726 : SUCCESS in 1h 53m 32s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/341f0f1ef9eb488dbbf19d003a7ed0a1 : SUCCESS in 49m 50s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/bc659da6c7c945ae985e7b6570e4877a : SUCCESS in 1h 59m 45s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/59a63db25ac04a42adcc4733e3545b22 : SUCCESS in 1h 01m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a0d1f016681e4af2b33871a76ce3c327 : SUCCESS in 37m 52s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ba6c09af8fc64086a77015bcab36269f : SUCCESS in 35m 13s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/18897c17867144419929135f4d9020e0 : SUCCESS in 55m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/17e16bffb4594934868956d0096ddf2f : FAILURE in 35m 18s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/de00461ff773490397f1fa7836abd684 : SUCCESS in 2h 14m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/200bab2164b3440981ac4920e441777d : SUCCESS in 2h 34m 10s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9a781323608348c392227013beb2aa35 : TIMED_OUT in 2h 35m 12s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/a55a25c7a644452baa0706c3cfc17009 : SUCCESS in 2h 22m 58s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"7005ca93e71677e8fb01a89aa4d8cd76262256eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-01-27 22:12:04.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":8},{"id":"318be49a0b8ff055d78cf9592aea7765c7c552d9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-01-27 22:12:31.000000000","message":"Patch Set 8: Workflow-1","accounts_in_message":[],"_revision_number":8},{"id":"f03ec8d7fc2fc14105fcb1297c19d6530bb52455","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-28 01:24:08.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1460c6a2a97543bea0d3517602529ca9\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/638614a4568c46508336ac43f3ac0ba8 : SUCCESS in 4m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d6581e503ea648ddb608fd69212ab507 : SUCCESS in 2m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a9308c4357b14aa58675ce0e31d24d03 : FAILURE in 12m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df405dfa2f6b4db4a10ba4dcea13daab : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8dd1105c1de24c4580ca6935451a49de : SUCCESS in 3m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5147df991a744f8189d69793ca7c5f00 : SUCCESS in 6m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59d9a8767b4048e58271b9159822e1f6 : FAILURE in 4m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2cdb3b56ec6d42a68aa8d6711f8dd51d : SUCCESS in 6m 21s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/cb29ffe9a5434130b0088f7237a7e54a : SUCCESS in 4m 16s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/64219cbf78c944b1817e41f9987f02cd : FAILURE in 7m 13s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/be4b640188da4bbaba0fca7be99dd991 : SUCCESS in 3m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/eef41ce8818b443f8975dc2b7808c14d : FAILURE in 6m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/98b29421104644c3923f0b5c1ff356ae : FAILURE in 6m 32s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/737d092b0b83439db9323c9697a9f0b5 : FAILURE in 7m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f705831d716b4e65ad35791730acb6b4 : FAILURE in 1h 52m 53s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/ca9803a4a9d94cf6964316d71e3bd5cd : SUCCESS in 1h 58m 53s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/7075a0665a504b31bb2f5490a14113c1 : SUCCESS in 51m 07s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/81430059e3b34074978bcd9777f15c17 : SUCCESS in 2h 22m 44s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e9d3ae9d0d9f4439ac7778c1d819687e : SUCCESS in 53m 17s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/58b7900e905e4afdb0d91cf33df09ed4 : SUCCESS in 2h 02m 12s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/91e705a8344d4a99ae186fd911934b19 : SUCCESS in 1h 40m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6d4f5ee313344b8fb22c5af830fee7ca : SUCCESS in 40m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/87f2989cef6347ca8770ce738528d48d : SUCCESS in 33m 32s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/0b5b1ee7b486467ca7c67f0198720830 : SUCCESS in 56m 19s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d112d0b1f687425a91ef720a334e4d2f : SUCCESS in 43m 43s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/29012362b1374b86a495ed316699e100 : TIMED_OUT in 2h 34m 36s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cd882d0b9b654f3aa8f04dc4a8b62431 : FAILURE in 1h 55m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1e2eff19a85c4f16852d67f7c8c4bed4 : TIMED_OUT in 2h 33m 17s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/2d60cb2687e748949c043737bebdc534 : TIMED_OUT in 2h 35m 14s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"1e0fbf56e5e79240126c12eca2cda607783f3f76","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-01-28 01:33:55.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":9},{"id":"d2d3aad20d38e7a414735870239a028a996dbdc5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-01-28 01:45:53.000000000","message":"Patch Set 9: Workflow-1","accounts_in_message":[],"_revision_number":9},{"id":"f6ef1008ea6e6de9f81c4d3fc14033bac95f706a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-28 04:38:23.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c48c520a4715445c8fd13867503daf16\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9cf49d5c048a462aba31904148f22c96 : SUCCESS in 3m 59s\n- requirements-check https://zuul.opendev.org/t/openstack/build/11b7ebdbf9b54ab7bc7ec909ca062b1a : SUCCESS in 2m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25af57d3e0a740bd9e7afb95ba2de3ba : SUCCESS in 9m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1909b6ab02e42928a5f8f93e0ce6974 : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e2d721f431734dc3a29057c5714648fc : SUCCESS in 3m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6b908088d9f240ef8e6d10336986e103 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/255ae7e0698f47de938549ac7f83f309 : SUCCESS in 13m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee82426e070b45ee9d6e54c4d6607dc7 : SUCCESS in 4m 17s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/e2392a93927b432cae9093f8b5167c26 : SUCCESS in 4m 15s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/a8f6d1ff0e184fc89fc9dde14a1fa3a8 : SUCCESS in 7m 34s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/78ec5aa1fdcd49548ae4dea21f72915e : SUCCESS in 4m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a48699e240e543ea91a233d6e365ae37 : SUCCESS in 5m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ff47b34f147948209296bb0fb15c2828 : SUCCESS in 8m 22s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9613f7e5b9174ad58fc23a2e8b673231 : SUCCESS in 8m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7ed2f5e5e7ad4f8ea9d29dccf7423e1c : FAILURE in 1h 38m 00s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/4891712ded43457e8360bb7a63e8bf87 : SUCCESS in 2h 06m 16s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/687def5e4f914fbf9a9b6061f54b058e : SUCCESS in 1h 07m 25s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/02378340ae1b4098adad8aee67c7b0e0 : SUCCESS in 2h 02m 57s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c7c528b2f3c34e94bf9b08c27489184f : SUCCESS in 1h 08m 53s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/4764c9acbe12499cba67dee911ec69ad : SUCCESS in 2h 12m 16s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/7190e130ca574e1c92285ae1d81e5b1b : SUCCESS in 1h 19m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d6e9e77bd9fc4357ae3c69bc8d51aea0 : SUCCESS in 36m 14s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a66cb23377364fe9908513a3653d6681 : SUCCESS in 48m 20s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/e039b6199994455aab9ec07e9a719ebe : SUCCESS in 54m 21s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d39ab1a7221b4bacae0d74c3fab846d7 : SUCCESS in 35m 40s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ae96978396314e0f82d2d15f3939ff00 : TIMED_OUT in 2h 33m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e9d43c5223cf45c38bf43e369911423e : TIMED_OUT in 3h 03m 20s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4e497730e4704c5ab100296926a73074 : SUCCESS in 2h 30m 14s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/10ad95965f40400aa83af1b4b3683e1a : SUCCESS in 2h 27m 10s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"02dbe54dc2d08d3bb329f1b4694b759a721fe13a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-01-30 21:06:19.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":10},{"id":"294ef5a3a5e9063ed377f433c1083315240bf93f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 00:01:27.000000000","message":"Patch Set 10: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/21e201940e4a41779fd02f142fbdfefb\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0a35a66b52c14d2d8516e7451631b2a0 : SUCCESS in 4m 48s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2d75ef54d0744ccaaa1d87aa510f18d7 : SUCCESS in 2m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd17dba48de54ae983debc67f0dd0fcc : FAILURE in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a5bfb39223e4d7882c91b7504ffbd2e : FAILURE in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/afeb4e2a46b643c08f849a91a8e023a7 : FAILURE in 4m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6d3c9380d1b84ed3b03b493a151b0ef9 : FAILURE in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b92fcc9b8b3f45e1ad47eb15cf1c7dcf : SUCCESS in 8m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2dc361b2ef804d58a6f93a43d104b253 : SUCCESS in 4m 30s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/c61d7e22244947dca5b3744fa3595926 : FAILURE in 4m 30s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/5563928575424fa3b9847352f5266cdc : SUCCESS in 5m 07s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d54c7930f4f24956a5a1ffb1cf904459 : SUCCESS in 2m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f3820845fd2f4cbfb1c945b3f4914b59 : SUCCESS in 7m 49s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/eaec523342534993813485dbfe238a6e : SUCCESS in 5m 38s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e0540886cdf848eb97a8de98537e54af : SUCCESS in 6m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6b64ac5416fc4e5284c4431d6fc7004e : FAILURE in 1h 46m 09s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/21316c0e24c74c9db158ab4ffeacdf29 : SUCCESS in 2h 19m 04s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/9514495e65a34906a3b09088569f8a21 : SUCCESS in 1h 13m 16s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bcb15484377d47b1a8d4a96d432899f4 : SUCCESS in 2h 04m 15s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/04dde023361244e28aade7ef5676b8e6 : SUCCESS in 1h 09m 47s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/2657002d820a460b9bb7dc6dbee56f63 : SUCCESS in 2h 14m 56s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/449dcc23b1a44b7697a3af1635042325 : SUCCESS in 1h 25m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/82af6e68b7754983b48e31fcd5e4157d : SUCCESS in 49m 34s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1e9ae5b3c1b64a2088bb0557385569c8 : SUCCESS in 1h 01m 19s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/f980879c35d14391b51e2f1fdaca6ac7 : SUCCESS in 1h 06m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0d05ab22617049839f2180662b45d125 : SUCCESS in 45m 32s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/57d266cd1bf04a35bd82c130ed50de87 : SUCCESS in 2h 21m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/60b44be28032487080eb1d34df26d5ff : SUCCESS in 2h 52m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f53cd63e71344b99a784e401c1648802 : SUCCESS in 2h 29m 48s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/ded534858d9942c49c8984f9f5cdcc55 : SUCCESS in 2h 29m 35s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"4b7c97136e78cf848330a934ce65569e084ef5c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-02-05 16:16:39.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"530247c8ec01fa6cad15cd1da9aee8428b4ebcca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 19:24:08.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8ad6382c6b30412083baaa1ef0dd45a3\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9fb46afceb134c34ab877b40fc654cd1 : SUCCESS in 4m 22s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d9474e9496a2441fb3eb2f6b90a2d748 : SUCCESS in 2m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/75a05c9e74444680b4e18abfdc961af5 : SUCCESS in 10m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4131d304014349e3ab4d5bc2f21f6aca : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f95fbc73b4cc4ad7892faae1796edd19 : SUCCESS in 6m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/90c01297ed074bebab3de14eb3efc86c : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a2f6d2af55d4aa589a06541ad677b73 : SUCCESS in 10m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7fae133e061c4dae812aa3721fd5ac07 : SUCCESS in 4m 26s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/d8b5bd56837b4c22a6d588d94f5a17a9 : SUCCESS in 4m 08s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/d6ba1bd90d6e4504a0cee33537500c35 : SUCCESS in 4m 23s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5b43d6cd29344da69f499a5e10c644ed : SUCCESS in 4m 46s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0c3f04c17d1348a9a1787262f0c5dab4 : SUCCESS in 8m 25s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f297302327654a81b468f8206d18d3ba : SUCCESS in 7m 48s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e104de9074b745819df629e955fb1178 : SUCCESS in 5m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/89645bbcf3524f0a80bf46d7f8df5ba0 : FAILURE in 1h 22m 34s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/e6b0f4bdfeb04739928fe21719bacf61 : SUCCESS in 1h 52m 31s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ac5f7ee41a6143ae8a9c76b94dca1032 : SUCCESS in 1h 00m 59s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5edc3bfff4e742fdad0188d97cdd764f : SUCCESS in 1h 49m 07s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/59dc4d0ad12c4af49007c4f53da354b0 : SUCCESS in 1h 10m 23s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/52836eed0fae442da880139ab0e23565 : SUCCESS in 2h 13m 25s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a3d12983305d41b687e70ee41ed5c472 : SUCCESS in 1h 17m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/10ab02b4a85149ffbe660dbaf71ab6cc : SUCCESS in 58m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6b5bf70c872142dca07f30c26223bf2e : SUCCESS in 47m 44s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/3b2f8c123a1c420e85c1d0198de2d00d : SUCCESS in 1h 04m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4840829a93244d84b064868be593202a : SUCCESS in 39m 37s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e5e30b6ef58b4c2b9085d58787cdca40 : TIMED_OUT in 2h 33m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/be6f91777e0f4374963f5b34d2fb321f : TIMED_OUT in 3h 04m 36s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b2d7b2c07aa94840ade033130a7fec6d : TIMED_OUT in 2h 33m 04s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/4dc48681848d4894ac20324ef7cbb411 : SUCCESS in 2h 10m 34s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"796b026eefeb0561fb47804b34fc2cec5ef644c7","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-02-06 10:40:20.000000000","message":"Patch Set 10:\n\n(4 comments)","accounts_in_message":[],"_revision_number":10},{"id":"5e413d23d5899b08371e5a8135d9d6881c7f9a13","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-02-06 12:25:20.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"d75692bd8d8ef885fc5df72f4cac66dbaae64faf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-02-07 16:52:21.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"5351878037dca04f1952ff09d71a8cd47dede4e9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-02-07 16:52:34.000000000","message":"Patch Set 12: Workflow-1","accounts_in_message":[],"_revision_number":12},{"id":"f49d436b64229ff9d142613be61d15c61d8e8b48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 20:00:09.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1b2e75755f6449f8ae3505db80e7c7c6\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4b2dcf68657b401b8a40559666119a9f : SUCCESS in 4m 25s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c22b20a8fe6e461d87d21516dc59a052 : SUCCESS in 2m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54d21266ef944ef69db694b58e906f3a : SUCCESS in 12m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad24b68730a04310ac638636ee12c124 : SUCCESS in 7m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb7d3889922a4aa48c2115c75026b1c2 : SUCCESS in 6m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e28b8e4923924c92aabaabdcac8d2235 : SUCCESS in 6m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da3ec8bc49f84aec8b92260ab4f8219d : SUCCESS in 17m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ec3284ef3904a0b9a6871e67a74ef86 : SUCCESS in 4m 24s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/2819c4090a0b430e8a4c11484afde247 : SUCCESS in 6m 46s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/5bef89dbc90a4bee9cd73afd45cc6f68 : SUCCESS in 6m 30s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7d30593fb5054391ba9165f9ecfd093a : SUCCESS in 4m 58s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bf5f00dfa32b423a9864d810db2594bd : SUCCESS in 9m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c77189190b73453983800099e2f22110 : SUCCESS in 7m 57s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/152a2fd7e87d4ac1ad264dfea9b3fe6c : SUCCESS in 8m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7ef70d9d80f94c9ba3169b5c497b33fe : FAILURE in 1h 41m 56s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/53e6a68304574ff69d009e74155e683e : SUCCESS in 2h 18m 02s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/8e74f77c01624dc4921f29d8e6360ee2 : SUCCESS in 1h 25m 05s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0a507d3f69cb4e44bff342807ea00de2 : SUCCESS in 2h 32m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c3007dab46994647a0c59f65e0f6d3cc : SUCCESS in 1h 01m 07s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f63957e462554136ad734e24397836f7 : SUCCESS in 2h 18m 52s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a6581dcb9684436b9e8d7f889b27a2ed : SUCCESS in 1h 28m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d57a2acf71894c48af227a327a706a72 : FAILURE in 13m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0c02bb7980654101b3e324bb57e74620 : SUCCESS in 1h 01m 53s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/63e9915ee8374602a35d4b271d672bb4 : SUCCESS in 1h 10m 56s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2ed4318b0a83455a9cfc6eee80dab88f : SUCCESS in 36m 19s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2290f734e92e4e0980c2e31a150eca23 : SUCCESS in 2h 30m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c591136ae8114ff7bfe2fd5e61da6a9a : TIMED_OUT in 3h 03m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/56aa34a016314dafa0e24f206b79b2dd : SUCCESS in 2h 24m 33s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/450215ad7a264620b4414111cc0d1378 : TIMED_OUT in 2h 33m 40s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"3e841b43d58ab3014882d03ed22635a22b263723","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-02-07 23:14:23.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":13},{"id":"ec3e1a7f90934cd58f55470bcd4c31bc639c5b20","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-02-07 23:14:30.000000000","message":"Patch Set 13: Workflow-1","accounts_in_message":[],"_revision_number":13},{"id":"decf95115d90ec70990114113e4710a16006c8dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-02-08 01:48:55.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Workflow-1\n","accounts_in_message":[],"_revision_number":14},{"id":"47e20016365bc1d72f95b4037af11c7e5a6f2d02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-08 04:58:12.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/554a61a15f76413789c11701864ceab8\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ce85a9dcb1294d4d97054cc2e8a8d9e8 : SUCCESS in 7m 10s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ecec58e48e5541afbe8d2ae948095fe2 : SUCCESS in 5m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f989219a20ea4f4a8fbd2745ab49896c : FAILURE in 8m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad2820219d3b4106a7b20dcddc92fe98 : SUCCESS in 7m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e0749a7904234aada7b1d81b0d4667ad : FAILURE in 9m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e5f65914b76b40b687d96f4cf34d4d77 : FAILURE in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2a861088d334b2591ead154f0e7b752 : SUCCESS in 9m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a68a973825cc4394b8d3f5675223f332 : SUCCESS in 4m 41s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/feb9d1e1b9224cc1a049d2837512d20d : FAILURE in 4m 29s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/09a1663fcbf7427eaf4adfc7a16ac247 : SUCCESS in 8m 17s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/db845a46b3ff46849c7f087bdaadf0ee : SUCCESS in 3m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/685003a2373c43348043333dd36ec0f5 : SUCCESS in 6m 49s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/baf2b287068c42d79eb94df064ebd359 : SUCCESS in 7m 08s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f7334d0d639040b0959d38649b777353 : SUCCESS in 7m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a51f114ea3b34913ab25dd678e363a4b : FAILURE in 1h 08m 32s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/475a6653b90449888efe625e60a87ce5 : SUCCESS in 1h 45m 08s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/038c096c930d48bfa16e8d87a374fb0e : SUCCESS in 56m 48s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/12a457a81aca467fa98d514f4340aad8 : SUCCESS in 2h 15m 08s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5ef468d8c3d843f0b40d8200a6cc6b59 : SUCCESS in 1h 36m 03s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/6cf330432e1543b980c8d8b2e866aa1a : TIMED_OUT in 2h 34m 19s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/33d0ab5f09f341f3b1fcb78b66186982 : SUCCESS in 52m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bd69dcbbd17d41369c38133f6f6cc27c : SUCCESS in 59m 11s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e5ac31a4b53c404eb3f4b4c08a3f0bb0 : SUCCESS in 1h 12m 25s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/1b8ddaf3832d4e44a1b9858bbb251e33 : SUCCESS in 1h 13m 25s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a92ed5f6702c43fc8105bdb3b027aa70 : SUCCESS in 57m 23s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9cacb420c89e4dfdbc663586feb8fabd : SUCCESS in 2h 20m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d08576e792cb4578a569407c36f9ac83 : TIMED_OUT in 3h 04m 25s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/eb3e1640a9774503acce0c3c903eb04d : TIMED_OUT in 2h 32m 58s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/ba580054974148d1ae772239578deabc : TIMED_OUT in 2h 38m 24s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"d1ec98133b8695b921e10f3be7460507328cc99f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-02-10 18:05:38.000000000","message":"Patch Set 15: Patch Set 14 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"cae97ef0f6320d8d3f7adb7c4261387c97636f9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-10 22:59:16.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/97be5fd1430f4c6fafce994053c377a9\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/214892cf5ab141918115cf1c026d7600 : SUCCESS in 4m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6da42866e4ed444582ccce578879f23b : SUCCESS in 2m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55502811f0ab4a27918d405f75e4f5a9 : FAILURE in 14m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb8a9859bd74493494207d18ac7eafc6 : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a0c61a59a8ae49ca98380ca461b6b270 : FAILURE in 6m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ff01d8a79e8d4c57bbb771df0dfbd440 : FAILURE in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a2cc2aaf8014801ae4edadeb7ee154d : TIMED_OUT in 30m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/24759599c0e446fbbf88a82685dfd00c : SUCCESS in 4m 49s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/2b8f694e2057430589e68fa1985dc667 : FAILURE in 4m 08s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/7f36afdfd614417394d836ff129baf2c : SUCCESS in 6m 12s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/21b4acdf9e134738a3b7fac961064b7a : SUCCESS in 3m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/598e5cdf944d43ac90b912d9202518c3 : SUCCESS in 9m 24s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/dd7b8ee8e0e04e7fb117624dec74f9d7 : SUCCESS in 14m 31s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/fab2aff42d6a4034bc9857512f2bbf65 : SUCCESS in 13m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/56ef76fb4682484c81c9e82f6f3cd82a : SUCCESS in 2h 33m 15s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/555c2d47476b4aeaaa424f72baf7b5f4 : SUCCESS in 2h 26m 06s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/a92ec7af03904392ad80ebaab68324f5 : SUCCESS in 1h 16m 01s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e14522692f43418d803ed481362e5e1c : SUCCESS in 2h 05m 41s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/20d1a73535d848f68fab0b8a104ec258 : SUCCESS in 1h 18m 47s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ec776426fa154e8fbfa0f427e1005baf : SUCCESS in 2h 30m 54s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/9ba280009c3f4ce0a2a125ccae898432 : SUCCESS in 1h 37m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b6a2b14892ef4f35b81a5cfc7b48ddaa : SUCCESS in 55m 06s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9ebd441fc6514cb581abfc454f4f4ddd : SUCCESS in 48m 20s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/0e731a57413d4da094fae6903f0581fb : SUCCESS in 1h 20m 03s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ce818d09aff6416aae5bc3988b4b22c2 : SUCCESS in 40m 29s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7be5eb89858d48ca8de399cb5d03e743 : TIMED_OUT in 2h 34m 28s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/56d9ecf355b54c3f811fadfecb7221ea : SUCCESS in 2h 52m 49s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/64587438856f457087c846cd48a885db : TIMED_OUT in 2h 34m 38s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/3ecaeffb4e3e4060945a3501d92a3ee2 : TIMED_OUT in 2h 35m 09s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"97c87ed46983deac743a373ea05a6d5776654d1f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-02-11 01:23:36.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"c1c88caae2ad45b9486b10c0df620b1b1cf92b79","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-02-11 01:24:22.000000000","message":"Patch Set 15:\n\n(4 comments)","accounts_in_message":[],"_revision_number":15},{"id":"d823d7e6232e00373af0f1dad06f5e9973a04a80","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-02-11 01:24:45.000000000","message":"Patch Set 16:\n\n(2 comments)","accounts_in_message":[],"_revision_number":16},{"id":"695ff2233833b2e33a7e775e5a43c05ab6035e91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 04:28:17.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/60ac3b7179be4d729b1b706f67d64710\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a3a25b24113543a1a6de9b26368a59bb : SUCCESS in 5m 01s\n- requirements-check https://zuul.opendev.org/t/openstack/build/95216934c9d94b3a9a9fd48aa26d009f : SUCCESS in 2m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df3ea5b27b304759b6fa145b22eb32a7 : FAILURE in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4301eb65886042bcab0d4831edf54319 : SUCCESS in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/45284cb45d0a43ada2d97ba55989df18 : FAILURE in 4m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3971ef9112734fb183b657c0573dbf08 : FAILURE in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c45d96eba1d6484fa1c5795e3f846a9e : SUCCESS in 11m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5448dffd26934d83a04be8907eed1a31 : SUCCESS in 5m 03s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/a9051df55536420e96aa36495976b25d : FAILURE in 5m 26s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/bfb43d4c86ad4e2188c013168d935bdc : FAILURE in 6m 56s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2f06b9098cc24f73a99966153e087cd5 : SUCCESS in 4m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/08a185cfa84e43ecbe8576522c76263f : FAILURE in 8m 08s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e563d37bcffb47e98821e55e50f2f5e0 : FAILURE in 6m 55s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/492e53fa1b8c4796987a6f444622028c : FAILURE in 7m 03s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0b51a690d006497a9d01587fb713ba45 : FAILURE in 1h 42m 50s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/704865c50544421f9fc641317b4a6b59 : SUCCESS in 1h 48m 06s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/d9046913e5a44503b6363e04cfa4752e : SUCCESS in 1h 03m 21s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/204ac9c9166142968cec3cc2fc1d8cfd : SUCCESS in 2h 14m 42s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a3ede6e54e2f4ab083c022c8aafcd2db : SUCCESS in 1h 01m 19s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a0b3319cd8354096bb915b40ef72d0f9 : SUCCESS in 2h 33m 59s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/77e21f5e8ca4401d9a86735c48f06184 : SUCCESS in 1h 19m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f6307e264dde4ce89b01967985672a51 : SUCCESS in 35m 40s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/be2d6e7d3ab747279ddcfc2785674ef5 : SUCCESS in 59m 45s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/84656246aa004cc0a69469a1fe44e217 : SUCCESS in 1h 02m 45s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f80551c4e6d345a981740c1bf3563cc3 : SUCCESS in 55m 27s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0aadfa07dcfd46649a755aaf90fd9a22 : SUCCESS in 2h 20m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a9aa5aa9e6254e83be588105e9754324 : TIMED_OUT in 3h 03m 15s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6623e7245f764de58f0eb39deeaa0dc8 : TIMED_OUT in 2h 33m 12s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/4db171ea7ada4d7ba74f24c95f8edc65 : TIMED_OUT in 2h 35m 06s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"f09d859a91de3e3929dfcd38aa4d7249b5a22305","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-02-12 02:35:59.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"dd886d816a4b35099fb90564131140ff99a14edf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-12 05:49:47.000000000","message":"Patch Set 17: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0cef316dc5104ea19d335587c59415eb\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e9560eaa880848319e8e4664ca98ef70 : SUCCESS in 5m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d70e97c8f3a14e2c833612bb3e28d79e : SUCCESS in 2m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12021fcf0ae745318624f721b3337ff7 : FAILURE in 12m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6e6e47755124bc4b17413221a5cbb4a : FAILURE in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/78228fb9c18f440485d6aa2ddefb04c1 : SUCCESS in 4m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1ec3684e0eea44bda649832f02970726 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd11406810ac4c03ac4bf88e7a4a93d8 : SUCCESS in 13m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/97c1b7daca6141039c360b33e428b27d : SUCCESS in 5m 59s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/e41b745393d947c98db44a0396ff76c9 : SUCCESS in 4m 30s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/b5be9a6dc371473da12744491df1dd82 : FAILURE in 6m 43s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/36fbd6549706465ca98855dfbe87c603 : SUCCESS in 3m 39s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a33a814eb3eb4100bea07f1e3cc939e0 : FAILURE in 8m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c037fb62267d4d6b83da7a29450fc382 : FAILURE in 7m 23s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/bf88fda3555f4c06a45b182b8b26c516 : FAILURE in 6m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2731e2a75cff46f5ac17537f917a8634 : FAILURE in 1h 17m 50s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/c16f0d17f1b344c58f154071ab8fbbd6 : SUCCESS in 1h 51m 08s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/dbb839e35e13414894a51e4e8e26cf95 : SUCCESS in 55m 59s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/09eaa82f2c444f57b92f13c2805ede0a : SUCCESS in 1h 43m 46s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/46469612eb624878a1eb0d0c2a27abdb : SUCCESS in 1h 07m 05s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/42f383f8bf1846f8ba8d2c40d370b699 : SUCCESS in 1h 55m 33s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/4ce5fb7eea2643e495e480f5afaf4fcc : SUCCESS in 1h 04m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2c890d748fd847b9830f9f0a4f2abb5a : SUCCESS in 31m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d9759d04a9be442898e833ad8bc5e843 : SUCCESS in 58m 55s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/7f549c3742cb40298eef0500228bfe16 : SUCCESS in 1h 11m 43s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7378a70bd8ea4ebdb71353a663972057 : SUCCESS in 47m 06s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b27c92db36dc4476ad41ea9d381cf02e : TIMED_OUT in 2h 34m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/184753a9c7d9497287540cd11989d585 : TIMED_OUT in 3h 04m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5b1e1174239143dea8a5c8e59cb96f2a : SUCCESS in 2h 32m 09s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/4de49ca453254b9da6555024f6b5c431 : TIMED_OUT in 2h 33m 47s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"93c2f0b225b345a8ab04f4cd8f024ea8c371e8f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-02-13 17:58:40.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"05f4b913bfc18093a6c502b31272b22a62e99664","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 20:44:52.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a763138b5da949fcb9452c3eecca0f62\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c740bc898c734fb8865a88dfa2ca685f : SUCCESS in 3m 53s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5b24aed5114a40d488adb06e68ff204b : SUCCESS in 2m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/031e520f9bbf4d11b322fd7ab7ee0586 : SUCCESS in 11m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/767b4c43f8f94e0e95c94f24889672c8 : SUCCESS in 6m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/def3c61aa56c4c979b1b8a1c20b29741 : SUCCESS in 4m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/86bd6c6b01744332a51469b3f47985e9 : SUCCESS in 7m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8638ee82ae7a405d9926a0012060c724 : SUCCESS in 10m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a759d77fb174648a2b120d867411e6a : SUCCESS in 5m 29s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/5c5642d36a14485794e4d99b5fbe3a44 : SUCCESS in 4m 43s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/5343d6d1a2b946f2ab6e5914f75dcf28 : SUCCESS in 7m 24s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9f728f68f5cd482baa5aa9b7f21a7277 : SUCCESS in 4m 24s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1aa3d63e7ddb4e9dbbc8dada80398416 : SUCCESS in 7m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/666e5458283345d7b713ba3b062d567f : SUCCESS in 8m 38s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/88f024d325ef4e1ab736c12ddca40bd5 : SUCCESS in 6m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0405257a745348489c44a4eb5d9f50fa : SUCCESS in 1h 38m 17s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/7406e30853c84dd0bfb330618aa4b55e : SUCCESS in 1h 47m 12s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/df41ef8baaca48edb3c2748f09529aee : SUCCESS in 56m 23s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5fe7c86cd05542969dbfc76c00f9925c : SUCCESS in 1h 42m 50s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/80ad3408331d445eba97c4fd04a02dde : SUCCESS in 1h 23m 08s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a0947383507d4a05910729e210128c3f : SUCCESS in 1h 47m 30s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3b9d3f82b4ef4b3d87e5d92bf8b36449 : SUCCESS in 1h 25m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fc9e7686817e4e68bd7a44b9860dc255 : SUCCESS in 41m 50s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3bf4358d83574ec387c9edd9980b14ca : SUCCESS in 41m 23s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/564679e16dcb471b9338deb93d3ec3aa : SUCCESS in 1h 03m 56s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f86a93c543c84b4a80cf8b9b10eca768 : SUCCESS in 34m 00s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/133993f4e8ff4964a292e17e97ef5a42 : SUCCESS in 2h 17m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b5c8a108b8794502bec49970cb85fe4f : SUCCESS in 2h 40m 54s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f2f78f475edf4fa7ae906a3e4acc7f6c : TIMED_OUT in 2h 32m 51s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/82f6ab7d282147cbbffd9d315b92e1ec : TIMED_OUT in 2h 33m 28s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"4a1e4483f69b488ee338794d86235690f7c00d30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-02-13 23:47:42.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":19},{"id":"c74f2911853af7ec9a9c0bc8b76778944bcbcc68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 02:22:39.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f728a5310bd241d8a726482f1ef7efbb\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9a3225c850484a32a87d1f018f4d9cf2 : SUCCESS in 4m 11s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d8840b42c2f048fd838e472738c572f6 : SUCCESS in 2m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/31579cbab21f4f4d819f2ccd1be02191 : SUCCESS in 14m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4007f8194f9147ef9ac4d0baf2eedc8c : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d4c65ac89e654ee4b6bb6f0d8fdc8468 : SUCCESS in 4m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c0029e3cb50149588fd137536e727720 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e1b1137885445da8866f92947861c30 : SUCCESS in 10m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b43fdbcb0424a3fa104a8278172d81a : SUCCESS in 5m 10s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/4f1f22768bc243e891f9972001d999a3 : SUCCESS in 4m 54s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/61c99b33a81343ac8cf7f08d0e274fdd : SUCCESS in 7m 42s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/45f1e3e4f502459284b0e8944b5b97fd : SUCCESS in 3m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d7550b5e112f4b6bb634eebe5dc1ac42 : SUCCESS in 7m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b790b3750e744d42a6e22c574e14a168 : SUCCESS in 4m 54s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8c2886d53dd143a7b6bb2f36aee9f623 : SUCCESS in 6m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1ec7a041754b4442b9efebd8b6646cf4 : SUCCESS in 59m 37s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/1421e0e3d34e4fdeaab0dfa40e245e59 : SUCCESS in 2h 00m 27s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/be1a33d4ca7c4252be99c5e68a10f13a : SUCCESS in 1h 02m 29s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f0145d500332468cb460ee53ec4125d2 : SUCCESS in 1h 37m 09s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a4fdb240c97c4da8a7b2438f5f141eac : SUCCESS in 1h 07m 28s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/863ad6f374be4911bc2ed586411cb91b : SUCCESS in 1h 39m 56s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f6bb389596bd4243a5cbce0c733c4d1b : SUCCESS in 1h 09m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ade269ded6c349bc84b2e1aa50957c4e : SUCCESS in 36m 05s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/42186db80d3d48bcbd8b0684ac73705e : SUCCESS in 38m 01s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/88e26f13ee244406aff1d59808fd8869 : SUCCESS in 57m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8672a570ec7b426ab4d8e758f38ba9d4 : SUCCESS in 30m 36s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/83c11cec1424467da0b865dbffe3668f : TIMED_OUT in 2h 33m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/069d124adbda497ab0e7fe5a3a7f560d : FAILURE in 2h 01m 20s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6920216c7bee4543a766941952224e04 : TIMED_OUT in 2h 33m 36s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/ac0811027763489481bee38199461c0d : TIMED_OUT in 2h 33m 32s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"b617ace4df3ba2d1d29db54b16557eb69f33ac0d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-02-19 13:49:22.000000000","message":"Patch Set 19: Code-Review+2","accounts_in_message":[],"_revision_number":19},{"id":"156cd0489e09df4e04bb2c82a1cdd6cb8c43a27a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-02-26 00:28:46.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":20},{"id":"ca937ca0afac5a7c6d46db485d2f65e56a40f551","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 03:32:48.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8c71f675184b4bb4bd06220e8f2cf3bb\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/34abc85f00aa49edae0f0e1765af2788 : SUCCESS in 5m 07s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f2610e1ba1124cd0864c0116e6c84a55 : SUCCESS in 2m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc788f25b9b544aaae8ed310e439ef5a : SUCCESS in 12m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4474df42daea483eb81f08330b9fcfe1 : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d04ae201c93465bb49e3816ffb201ff : SUCCESS in 4m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9d7d593f1bb5456093e3f36878d3f786 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a062ec597704ea797796fdbfcf8f4d7 : FAILURE in 5m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86dd4ce49f734d2da11b4ba8ecd0450b : SUCCESS in 4m 46s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/f42e4118198045d69ddcb8d258c8b69c : SUCCESS in 4m 21s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/f12e0fb186b44477afa3f0f53629ee5a : SUCCESS in 6m 38s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8b9e4e4f4b3343f9b949e79ffceed4e0 : SUCCESS in 3m 26s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/91c8d6d19f8445ae8e1d531ab8c7d774 : SUCCESS in 6m 09s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8e15fded70ef434d93ff7ab61c425c3f : SUCCESS in 6m 19s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5a377e5f68c74a1e91d753295df9c151 : SUCCESS in 6m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c12aecfe661a4e64abc2b785d4e3b666 : SUCCESS in 1h 24m 37s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/0065dbb08ba74b0c931affe4b9f894c3 : SUCCESS in 1h 32m 48s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/b9c7da3118af402d95c1eca01599c9be : SUCCESS in 59m 59s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/037a65b3145f4d299277690e6ee34abd : SUCCESS in 2h 04m 28s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/061446519fa24afc84cac3d7c8f5c43c : SUCCESS in 1h 19m 23s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ccdc12198b5041388bf37835a0b9ea06 : SUCCESS in 2h 14m 09s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b87cff1b04b64b26b632816291a2ffca : SUCCESS in 1h 28m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/72b6d154015b499e867e4d3e45a05f73 : SUCCESS in 43m 52s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9de5516b3199455fad27800c4eff1ab5 : SUCCESS in 40m 24s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/75163f9f2116462b84694b9d1a60536c : SUCCESS in 1h 16m 03s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/50c79749a62a42eca2c2bec42e1589d2 : SUCCESS in 38m 56s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6af118844017417daa0f41996638972a : SUCCESS in 2h 11m 19s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/094cf73f3b1a48df9a89017b2e3bda89 : SUCCESS in 2h 57m 53s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7f6bb818a93248389955a181b26942bc : TIMED_OUT in 2h 34m 36s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/1a503161c43a4d1ba5149ee3bc5a5dc3 : TIMED_OUT in 2h 32m 54s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"6f0040eeb4166c34699c850f89770ce2b806d306","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2025-02-26 13:10:37.000000000","message":"Patch Set 20:\n\n(2 comments)","accounts_in_message":[],"_revision_number":20},{"id":"f7a0c9a9f71d6cda995ca2f27513c3c78fb35174","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-02-26 16:11:14.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":21},{"id":"849adba68f06acd76f7c7990203cbb8b35f7e9d9","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2025-02-26 16:12:11.000000000","message":"Patch Set 21: Code-Review+2","accounts_in_message":[],"_revision_number":21},{"id":"e7d5b6a7042cdf318c636c99748209288ec08bf6","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-02-26 17:36:52.000000000","message":"Patch Set 21: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":21},{"id":"66e576cffac82b975598d59ff4a75995552d81f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 19:15:01.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/519b528e95f54050bdf83f954efe00d4\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9c462e58fa214aee82506ed2db19f5a2 : SUCCESS in 3m 32s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f4c95689ad6f44efa6da5ca311967f26 : SUCCESS in 1m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/992825ef67c04736a8a26652d44b1f70 : SUCCESS in 11m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/959cba1cebab4480a2b1ba6af75bd2c3 : SUCCESS in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/62796d3b756b4010929a95dfa7746caf : SUCCESS in 4m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a8260fd1e21f4d4d8dcb4ebc94b810d9 : SUCCESS in 5m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc6fe1f77c42482b999393bdc6552551 : SUCCESS in 4m 43s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/388d4bfc74f44b73906ca3eb09a81c41 : SUCCESS in 5m 49s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/13799b52fc1e4354884d86fb45d1ceb7 : SUCCESS in 7m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f7b042718a8040a99d899a92cc11c1c5 : SUCCESS in 4m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3a8ab994e76f4597ad76bcecef7cc2ea : SUCCESS in 9m 24s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/688c46b8a85c4523b9734a4601888942 : SUCCESS in 6m 29s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a410df51f4e24b8791bd6774d5532711 : SUCCESS in 7m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ca8e7022655c4220a6c08a1c72bd271e : SUCCESS in 1h 19m 51s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/da76fcbd79994792af61dadf973f7277 : SUCCESS in 1h 48m 52s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/f129c57efe8243108c8061a88313257a : SUCCESS in 1h 10m 28s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cb2a3c51ed58424cb42f3071991dea08 : SUCCESS in 1h 45m 52s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aa0539a9cfa3492e801caaf2fa142fc6 : SUCCESS in 1h 07m 17s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e5f80e0db8614d2ba789a0e2c14231ef : SUCCESS in 1h 53m 50s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/da25f8bd12514c24bdf08e133d323376 : SUCCESS in 1h 35m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f685936bcf644412a281a936f556adb1 : SUCCESS in 50m 10s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/75c63719f153427dbcdc9a5ed00816bf : SUCCESS in 43m 32s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/5df4fcc48e2841bf81fd866abb636d6e : SUCCESS in 1h 18m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0e5751d8d94140ab872441d3fe92d9cf : SUCCESS in 33m 09s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1b355ffd101d424ba82f17156e240cd2 : SUCCESS in 2h 10m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5f12b25d7fd44811b8641f45e7e382ca : SUCCESS in 2h 59m 30s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1cad7ea7f2f344c79c31efcc7dcbba02 : TIMED_OUT in 2h 34m 53s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/3973c148c9b0451c8004006a1aa8131f : TIMED_OUT in 2h 35m 14s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"ebfbf733cbdc08fc244c73ed769a751cdeb7d975","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 19:15:43.000000000","message":"Patch Set 21: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":21},{"id":"36f75fe5805bf819afbac8a6080b27e86ce56e8a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 21:01:13.000000000","message":"Patch Set 21: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27042c78731044c185fe245c06e6385d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3dd4575a9e6f4838905e434c6712395b : SUCCESS in 4m 32s\n- requirements-check https://zuul.opendev.org/t/openstack/build/aee02c6ca2a0415ba7097b8c434cf0aa : SUCCESS in 2m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b6e47cd6bcd4a918201068f46c0753a : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/71e44d8362c4433e8f6d65777f05429b : SUCCESS in 4m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3be5319d89d445ab9548096da41957ee : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee54c03a306b41328a1dfb049ae8bc57 : SUCCESS in 5m 06s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/66b88f8c82ec4769bbceec78e047710c : SUCCESS in 9m 30s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/28eb8d3f274b4997810804259e3997ab : SUCCESS in 8m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8775100d854445e28bee19d4555dab6f : SUCCESS in 1h 25m 15s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/846dbb2647ef4832a871f989fa8880e4 : SUCCESS in 1h 45m 04s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/fa5f4d8c47994b379b88151abfca6a8c : SUCCESS in 1h 08m 37s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cade747687ef4a50b525123804084110 : SUCCESS in 1h 44m 08s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/84b38661f09b4e6885e6d226c9c0da19 : SUCCESS in 1h 00m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/33ec747549474f39accf6a76b33be476 : SUCCESS in 36m 11s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5a0fd36168c9483ab17e848960c86bda : SUCCESS in 40m 29s","accounts_in_message":[],"_revision_number":21},{"id":"47b35d053bd0a4c2ccc646f6e95e2cd691463a09","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 21:01:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":21},{"id":"554fb2736cdf3daac974a5245ce0f837c9a82b7e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 21:02:35.000000000","message":"Patch Set 21:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f780a803c5b54fe6a470c61e107ffedf\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/0f981830d9bb45d0be45d24c25ea2a31 : SUCCESS in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd0e01fcd34c4d6e85ed3af7b116fed6 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":21}],"current_revision_number":21,"current_revision":"54c059d94da78bcec5b76547f68bc9f787ee3a80","revisions":{"61ee264cae2b81dd83dd6f6da8bf215f93112800":{"kind":"REWORK","_number":1,"created":"2024-05-22 00:13:03.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/56/920156/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/920156/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/920156/1"}}},"commit":{"parents":[{"commit":"b43620cfa8e509647f99835e6eb43416a7f214f8","subject":"Merge \"Fix listener update when using SRIOV VIP\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b43620cfa8e509647f99835e6eb43416a7f214f8"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-05-22 00:10:26.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-05-22 00:12:56.000000000","tz":0},"subject":"Add support for SR-IOV member ports","message":"Add support for SR-IOV member ports\n\nDepends-On: https://review.opendev.org/c/openstack/octavia-lib/+/914773\nChange-Id: I737e3228eccacccf5370a666c793a0714dbe0b24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/61ee264cae2b81dd83dd6f6da8bf215f93112800"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/61ee264cae2b81dd83dd6f6da8bf215f93112800"}]},"branch":"refs/heads/master"},"c2770e94b2fc0dac04839bc86bc0d901b40f1c8c":{"kind":"REWORK","_number":2,"created":"2024-05-30 04:16:06.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/56/920156/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/920156/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/920156/2"}}},"commit":{"parents":[{"commit":"b43620cfa8e509647f99835e6eb43416a7f214f8","subject":"Merge \"Fix listener update when using SRIOV VIP\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b43620cfa8e509647f99835e6eb43416a7f214f8"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-05-22 00:10:26.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-05-30 04:15:20.000000000","tz":0},"subject":"Add support for SR-IOV member ports","message":"Add support for SR-IOV member ports\n\nDepends-On: https://review.opendev.org/c/openstack/octavia-lib/+/914773\nChange-Id: I737e3228eccacccf5370a666c793a0714dbe0b24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c2770e94b2fc0dac04839bc86bc0d901b40f1c8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c2770e94b2fc0dac04839bc86bc0d901b40f1c8c"}]},"branch":"refs/heads/master"},"00070ad338821df2c08205ca665033116fa5b18a":{"kind":"REWORK","_number":3,"created":"2024-06-04 15:48:48.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/56/920156/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/920156/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/920156/3"}}},"commit":{"parents":[{"commit":"541682314693ab543b148a9ed7703192f1673e00","subject":"Merge \"String conversion to f-strings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/541682314693ab543b148a9ed7703192f1673e00"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-05-22 00:10:26.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-06-04 15:48:40.000000000","tz":0},"subject":"Add support for SR-IOV member ports","message":"Add support for SR-IOV member ports\n\nDepends-On: https://review.opendev.org/c/openstack/octavia-lib/+/914773\nChange-Id: I737e3228eccacccf5370a666c793a0714dbe0b24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/00070ad338821df2c08205ca665033116fa5b18a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/00070ad338821df2c08205ca665033116fa5b18a"}]},"branch":"refs/heads/master"},"1ab809e0364c7f426fe9f622536d3d2532126406":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-08-13 14:04:56.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/56/920156/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/920156/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/920156/4"}}},"commit":{"parents":[{"commit":"1b09ba8f378a61bc1d913be26fd7e0a28e950c27","subject":"Add spec for custom SGs on VIP ports","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1b09ba8f378a61bc1d913be26fd7e0a28e950c27"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-05-22 00:10:26.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-08-13 14:04:56.000000000","tz":0},"subject":"Add support for SR-IOV member ports","message":"Add support for SR-IOV member ports\n\nDepends-On: https://review.opendev.org/c/openstack/octavia-lib/+/914773\nChange-Id: I737e3228eccacccf5370a666c793a0714dbe0b24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1ab809e0364c7f426fe9f622536d3d2532126406"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1ab809e0364c7f426fe9f622536d3d2532126406"}]},"branch":"refs/heads/master","description":"Rebase"},"8bd97bbf1a7275d8bea7dacccb0e4eec3f883f1a":{"kind":"REWORK","_number":5,"created":"2024-08-14 22:44:56.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/56/920156/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/920156/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/920156/5"}}},"commit":{"parents":[{"commit":"f68cafbb7201522e6140e2dc05317d06225bb82a","subject":"Merge \"Remove sysvinit/upstart support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f68cafbb7201522e6140e2dc05317d06225bb82a"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-05-22 00:10:26.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-08-14 22:44:54.000000000","tz":0},"subject":"Add support for SR-IOV member ports","message":"Add support for SR-IOV member ports\n\nDepends-On: https://review.opendev.org/c/openstack/octavia-lib/+/914773\nChange-Id: I737e3228eccacccf5370a666c793a0714dbe0b24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8bd97bbf1a7275d8bea7dacccb0e4eec3f883f1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8bd97bbf1a7275d8bea7dacccb0e4eec3f883f1a"}]},"branch":"refs/heads/master"},"025ef591004f0c3f36a88bd3531bedf44d50702a":{"kind":"REWORK","_number":6,"created":"2024-08-22 22:12:06.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/56/920156/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/920156/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/920156/6"}}},"commit":{"parents":[{"commit":"b71e752c0fff2939e4347a22e55c7cbb3e9c7f9a","subject":"Merge \"Fix test_update_load_balancer_status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b71e752c0fff2939e4347a22e55c7cbb3e9c7f9a"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-05-22 00:10:26.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-08-22 22:11:57.000000000","tz":0},"subject":"Add support for SR-IOV member ports","message":"Add support for SR-IOV member ports\n\nChange-Id: I737e3228eccacccf5370a666c793a0714dbe0b24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/025ef591004f0c3f36a88bd3531bedf44d50702a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/025ef591004f0c3f36a88bd3531bedf44d50702a"}]},"branch":"refs/heads/master"},"f5a70736ac6ea04630b2b31f6bcb5b11d2ff3ece":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-01-15 04:06:25.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/56/920156/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/920156/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/920156/7"}}},"commit":{"parents":[{"commit":"697c89ec234fd8ff42ebb97cf282960cc65683f4","subject":"Merge \"Update gate jobs as per the 2025.1 cycle testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/697c89ec234fd8ff42ebb97cf282960cc65683f4"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-05-22 00:10:26.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-01-15 04:06:25.000000000","tz":0},"subject":"Add support for SR-IOV member ports","message":"Add support for SR-IOV member ports\n\nChange-Id: I737e3228eccacccf5370a666c793a0714dbe0b24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f5a70736ac6ea04630b2b31f6bcb5b11d2ff3ece"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f5a70736ac6ea04630b2b31f6bcb5b11d2ff3ece"}]},"branch":"refs/heads/master","description":"Rebase"},"8dc3417356487327a8d595549054cce68b82a7ba":{"kind":"REWORK","_number":8,"created":"2025-01-27 22:12:04.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/56/920156/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/920156/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/920156/8"}}},"commit":{"parents":[{"commit":"697c89ec234fd8ff42ebb97cf282960cc65683f4","subject":"Merge \"Update gate jobs as per the 2025.1 cycle testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/697c89ec234fd8ff42ebb97cf282960cc65683f4"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-05-22 00:10:26.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-01-27 22:12:02.000000000","tz":0},"subject":"Add support for SR-IOV member ports","message":"Add support for SR-IOV member ports\n\nChange-Id: I737e3228eccacccf5370a666c793a0714dbe0b24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8dc3417356487327a8d595549054cce68b82a7ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8dc3417356487327a8d595549054cce68b82a7ba"}]},"branch":"refs/heads/master"},"aad39cda3a6ceba130f07558d76b6781e66f7f4c":{"kind":"REWORK","_number":9,"created":"2025-01-28 01:33:55.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/56/920156/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/920156/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/920156/9"}}},"commit":{"parents":[{"commit":"697c89ec234fd8ff42ebb97cf282960cc65683f4","subject":"Merge \"Update gate jobs as per the 2025.1 cycle testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/697c89ec234fd8ff42ebb97cf282960cc65683f4"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-05-22 00:10:26.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-01-28 01:33:39.000000000","tz":0},"subject":"Add support for SR-IOV member ports","message":"Add support for SR-IOV member ports\n\nChange-Id: I737e3228eccacccf5370a666c793a0714dbe0b24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aad39cda3a6ceba130f07558d76b6781e66f7f4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/aad39cda3a6ceba130f07558d76b6781e66f7f4c"}]},"branch":"refs/heads/master"},"151e3d51f9ef43fe8293fa3630ecf8afd6db9d8b":{"kind":"REWORK","_number":10,"created":"2025-01-30 21:06:19.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/56/920156/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/920156/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/920156/10"}}},"commit":{"parents":[{"commit":"697c89ec234fd8ff42ebb97cf282960cc65683f4","subject":"Merge \"Update gate jobs as per the 2025.1 cycle testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/697c89ec234fd8ff42ebb97cf282960cc65683f4"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-05-22 00:10:26.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-01-30 21:06:17.000000000","tz":0},"subject":"Add support for SR-IOV member ports","message":"Add support for SR-IOV member ports\n\nChange-Id: I737e3228eccacccf5370a666c793a0714dbe0b24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/151e3d51f9ef43fe8293fa3630ecf8afd6db9d8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/151e3d51f9ef43fe8293fa3630ecf8afd6db9d8b"}]},"branch":"refs/heads/master"},"d49b399cdfaaeef219ab3d1520f29cec7c5a4d8f":{"kind":"REWORK","_number":11,"created":"2025-02-05 16:16:39.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/56/920156/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/920156/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/920156/11"}}},"commit":{"parents":[{"commit":"697c89ec234fd8ff42ebb97cf282960cc65683f4","subject":"Merge \"Update gate jobs as per the 2025.1 cycle testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/697c89ec234fd8ff42ebb97cf282960cc65683f4"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-05-22 00:10:26.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-02-05 16:16:13.000000000","tz":0},"subject":"Add support for SR-IOV member ports","message":"Add support for SR-IOV member ports\n\nChange-Id: I737e3228eccacccf5370a666c793a0714dbe0b24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d49b399cdfaaeef219ab3d1520f29cec7c5a4d8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d49b399cdfaaeef219ab3d1520f29cec7c5a4d8f"}]},"branch":"refs/heads/master"},"748198456680a8cc2dae06dd86c404aced5cbf63":{"kind":"REWORK","_number":12,"created":"2025-02-07 16:52:21.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/56/920156/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/920156/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/920156/12"}}},"commit":{"parents":[{"commit":"697c89ec234fd8ff42ebb97cf282960cc65683f4","subject":"Merge \"Update gate jobs as per the 2025.1 cycle testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/697c89ec234fd8ff42ebb97cf282960cc65683f4"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-05-22 00:10:26.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-02-07 16:52:19.000000000","tz":0},"subject":"Add support for SR-IOV member ports","message":"Add support for SR-IOV member ports\n\nChange-Id: I737e3228eccacccf5370a666c793a0714dbe0b24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/748198456680a8cc2dae06dd86c404aced5cbf63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/748198456680a8cc2dae06dd86c404aced5cbf63"}]},"branch":"refs/heads/master"},"3e9b37167a8720a1b06fce2f3a6236c5465340f2":{"kind":"REWORK","_number":13,"created":"2025-02-07 23:14:23.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/56/920156/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/920156/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/920156/13"}}},"commit":{"parents":[{"commit":"7aef02891274b5c18d21e1b4aa16e8105e3e8821","subject":"Fix network no-op driver for concurrency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7aef02891274b5c18d21e1b4aa16e8105e3e8821"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-05-22 00:10:26.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-02-07 23:14:06.000000000","tz":0},"subject":"Add support for SR-IOV member ports","message":"Add support for SR-IOV member ports\n\nChange-Id: I737e3228eccacccf5370a666c793a0714dbe0b24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3e9b37167a8720a1b06fce2f3a6236c5465340f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3e9b37167a8720a1b06fce2f3a6236c5465340f2"}]},"branch":"refs/heads/master"},"eb346748e5aaebb2458919c89c609ec2fb08ab23":{"kind":"REWORK","_number":14,"created":"2025-02-08 01:48:55.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/56/920156/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/920156/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/920156/14"}}},"commit":{"parents":[{"commit":"7aef02891274b5c18d21e1b4aa16e8105e3e8821","subject":"Fix network no-op driver for concurrency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7aef02891274b5c18d21e1b4aa16e8105e3e8821"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-05-22 00:10:26.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-02-08 01:48:43.000000000","tz":0},"subject":"Add support for SR-IOV member ports","message":"Add support for SR-IOV member ports\n\nChange-Id: I737e3228eccacccf5370a666c793a0714dbe0b24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eb346748e5aaebb2458919c89c609ec2fb08ab23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/eb346748e5aaebb2458919c89c609ec2fb08ab23"}]},"branch":"refs/heads/master"},"ec661c8f82fcb8a148f71106c2061c9b535340ed":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2025-02-10 18:05:38.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/56/920156/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/920156/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/920156/15"}}},"commit":{"parents":[{"commit":"878fd5c60f25345119f70ee6c48f5643d24a8f91","subject":"Fix network no-op driver for concurrency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/878fd5c60f25345119f70ee6c48f5643d24a8f91"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-05-22 00:10:26.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-02-10 18:05:38.000000000","tz":0},"subject":"Add support for SR-IOV member ports","message":"Add support for SR-IOV member ports\n\nChange-Id: I737e3228eccacccf5370a666c793a0714dbe0b24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ec661c8f82fcb8a148f71106c2061c9b535340ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ec661c8f82fcb8a148f71106c2061c9b535340ed"}]},"branch":"refs/heads/master","description":"Rebase"},"9c1b7896396231b067b5018ad64b10887192658d":{"kind":"REWORK","_number":16,"created":"2025-02-11 01:23:36.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/56/920156/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/920156/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/920156/16"}}},"commit":{"parents":[{"commit":"4b981154ad9fc1f480d661d4ac5b9b05adb1259c","subject":"Fix network no-op driver for concurrency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4b981154ad9fc1f480d661d4ac5b9b05adb1259c"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-05-22 00:10:26.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-02-11 01:23:20.000000000","tz":0},"subject":"Add support for SR-IOV member ports","message":"Add support for SR-IOV member ports\n\nChange-Id: I737e3228eccacccf5370a666c793a0714dbe0b24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9c1b7896396231b067b5018ad64b10887192658d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9c1b7896396231b067b5018ad64b10887192658d"}]},"branch":"refs/heads/master"},"1228a5a70d87a9c7f9855157b5dba70ad8956849":{"kind":"REWORK","_number":17,"created":"2025-02-12 02:35:59.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/56/920156/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/920156/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/920156/17"}}},"commit":{"parents":[{"commit":"4b981154ad9fc1f480d661d4ac5b9b05adb1259c","subject":"Fix network no-op driver for concurrency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4b981154ad9fc1f480d661d4ac5b9b05adb1259c"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-05-22 00:10:26.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-02-12 02:35:46.000000000","tz":0},"subject":"Add support for SR-IOV member ports","message":"Add support for SR-IOV member ports\n\nChange-Id: I737e3228eccacccf5370a666c793a0714dbe0b24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1228a5a70d87a9c7f9855157b5dba70ad8956849"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1228a5a70d87a9c7f9855157b5dba70ad8956849"}]},"branch":"refs/heads/master"},"da32885d0fcf69288c14c9c0bdd53276a05bd623":{"kind":"REWORK","_number":18,"created":"2025-02-13 17:58:40.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/56/920156/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/920156/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/920156/18"}}},"commit":{"parents":[{"commit":"4b981154ad9fc1f480d661d4ac5b9b05adb1259c","subject":"Fix network no-op driver for concurrency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4b981154ad9fc1f480d661d4ac5b9b05adb1259c"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-05-22 00:10:26.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-02-13 17:58:30.000000000","tz":0},"subject":"Add support for SR-IOV member ports","message":"Add support for SR-IOV member ports\n\nChange-Id: I737e3228eccacccf5370a666c793a0714dbe0b24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/da32885d0fcf69288c14c9c0bdd53276a05bd623"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/da32885d0fcf69288c14c9c0bdd53276a05bd623"}]},"branch":"refs/heads/master"},"397afb1d67d06a89ea627d8d7685e66bbd131bf8":{"kind":"REWORK","_number":19,"created":"2025-02-13 23:47:42.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/56/920156/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/920156/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/920156/19"}}},"commit":{"parents":[{"commit":"4b981154ad9fc1f480d661d4ac5b9b05adb1259c","subject":"Fix network no-op driver for concurrency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4b981154ad9fc1f480d661d4ac5b9b05adb1259c"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-05-22 00:10:26.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-02-13 23:45:26.000000000","tz":0},"subject":"Add support for SR-IOV member ports","message":"Add support for SR-IOV member ports\n\nThis patch adds support for using SR-IOV VFs for the member ports on\nAmphora based load balancers.\n\nChange-Id: I737e3228eccacccf5370a666c793a0714dbe0b24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/397afb1d67d06a89ea627d8d7685e66bbd131bf8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/397afb1d67d06a89ea627d8d7685e66bbd131bf8"}]},"branch":"refs/heads/master"},"d650ab7722f379aabd5a8cedd300c3c7d843d7e0":{"kind":"REWORK","_number":20,"created":"2025-02-26 00:28:46.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/56/920156/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/920156/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/920156/20"}}},"commit":{"parents":[{"commit":"1584e3e595ec817befde6be63e2ff8570430966b","subject":"Fix network no-op driver for concurrency","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1584e3e595ec817befde6be63e2ff8570430966b"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-05-22 00:10:26.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-02-26 00:28:23.000000000","tz":0},"subject":"Add support for SR-IOV member ports","message":"Add support for SR-IOV member ports\n\nThis patch adds support for using SR-IOV VFs for the member ports on\nAmphora based load balancers.\n\nChange-Id: I737e3228eccacccf5370a666c793a0714dbe0b24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d650ab7722f379aabd5a8cedd300c3c7d843d7e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d650ab7722f379aabd5a8cedd300c3c7d843d7e0"}]},"branch":"refs/heads/master"},"54c059d94da78bcec5b76547f68bc9f787ee3a80":{"kind":"REWORK","_number":21,"created":"2025-02-26 16:11:14.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/56/920156/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/920156/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/920156/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/920156/21"}}},"commit":{"parents":[{"commit":"d54fb850043109c3d75e3d37b0d59a7fbc888548","subject":"Merge \"Fix network no-op driver for concurrency\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d54fb850043109c3d75e3d37b0d59a7fbc888548"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-05-22 00:10:26.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-02-26 16:11:12.000000000","tz":0},"subject":"Add support for SR-IOV member ports","message":"Add support for SR-IOV member ports\n\nThis patch adds support for using SR-IOV VFs for the member ports on\nAmphora based load balancers.\n\nChange-Id: I737e3228eccacccf5370a666c793a0714dbe0b24\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/54c059d94da78bcec5b76547f68bc9f787ee3a80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/54c059d94da78bcec5b76547f68bc9f787ee3a80"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
