)]}'
{"id":"openstack%2Foctavia~921356","triplet_id":"openstack%2Foctavia~master~I3dd573300a8dfbe68eaeb736b47f6abb61b5e85b","project":"openstack/octavia","branch":"master","topic":"notvalidafter-deprecated","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2024-06-06 15:06:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3dd573300a8dfbe68eaeb736b47f6abb61b5e85b","subject":"Replace use of deprecated cert.not_valid_after","status":"MERGED","created":"2024-06-05 08:02:51.000000000","updated":"2024-06-06 15:09:12.000000000","submitted":"2024-06-06 15:06:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"921356-notvalidafter-deprecated","meta_rev_id":"315b4c518a7538e165692222450ba511d7db6e73","_number":921356,"virtual_id_number":921356,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-06 15:06:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"all":[{"value":1,"date":"2024-06-05 16:34:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":2,"date":"2024-06-05 16:38:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2024-06-06 12:00:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2024-06-06 12:00:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-05 10:46:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-05 16:34:01.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"},{"updated":"2024-06-05 16:38:36.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-06-06 12:00:04.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"d9e94e75d4a28992dc0e611b97eada80dd5c738a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-06-05 08:02:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f0e35909225e7bc9675db3c2b70a015aa0e7f01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 10:46:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2eb0451496954ad594919b4b3ec99f49\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d63a5815411342d2906fecb3e152caec : FAILURE in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c55590be2dff4b78921a7f1d8dc5776c : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c00fa6d5cc514d259486987e8a910eeb : FAILURE in 6m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bc70f6dbc95e4a0c9ee5ad02782e4618 : FAILURE in 6m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ef192211d40d4563b7f593bc0b0a1f66 : FAILURE in 9m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9002ad4771045c2becaf9baf4545dec : SUCCESS in 11m 25s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/890a4b2d62024e05b2d0d42b3b063a9b : FAILURE in 5m 30s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/e3ebb1edffd64af3a3ccfb754d36379e : FAILURE in 6m 16s\n- octavia-tox-functional-py311-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/446526e19a7840c8bbd68d7e9afb80f2 : FAILURE in 8m 01s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/33d7f7283216411483102cf78d16df90 : SUCCESS in 5m 19s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4cbeae1ad193400cae6657718cdc61f1 : SUCCESS in 8m 02s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5545e49a169b4518a94fd807926646bd : SUCCESS in 8m 15s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/93ada90e26d54b29b119912baee47899 : FAILURE in 4m 15s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2625f239c548429189a45f22d7a9f320 : SUCCESS in 1h 29m 19s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/b60dfed978fc453bad042f3fea8569dc : SUCCESS in 1h 47m 21s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/5bd6a322909f45ecbc83e456aa435a70 : SUCCESS in 1h 01m 14s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/87b933c271204a4ab419462774d3aeda : SUCCESS in 1h 53m 04s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/025256e4104b4e6686a58d86e92674e1 : SUCCESS in 1h 01m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/18df96ab308e48689508812722de6b38 : SUCCESS in 56m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1133ec5668d04ddd81b11b678cf07890 : SUCCESS in 43m 24s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/9e44682ed80446ab9ba5e49b118cc824 : SUCCESS in 1h 18m 32s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0de64749e93c4e12af0fadfac4f859d3 : SUCCESS in 33m 38s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/df1b3dfc140e48c8a92db988da1ce28a : SUCCESS in 2h 33m 18s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0a1ab03fa7a84348ab064ed16abe9203 : SUCCESS in 2h 26m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d323324bf7ea439780b97e9f46bfec0f : FAILURE in 22m 35s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/aceae8a27f874d35a8453ed38f86147c : TIMED_OUT in 2h 35m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bfc6f740f8305d32d1be98fafb85f471c2f815b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-06-05 13:21:53.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"3e79257517612879db00541a207cfa6ee00f3537","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 16:26:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61b5d2b05d514f3da4e05ed9aac0602b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/53d41bd11ee44acca5e30aaa62b4c492 : SUCCESS in 12m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b2d4d2fb16c4527a54311b4fbb1fef1 : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a07c42a6f8954e26977bfadff894df15 : SUCCESS in 4m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cce8af37456f4e89983059a0a50127a4 : SUCCESS in 5m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/40f875bec4a24b9db84046ec6b668ca2 : SUCCESS in 8m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab938122c71e466a87899b5b76f884a8 : SUCCESS in 10m 49s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/a0577e30148141c2a001229335c5fdb3 : SUCCESS in 5m 30s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/492f30c9e2104b9d9aa55d66bea9c897 : SUCCESS in 4m 51s\n- octavia-tox-functional-py311-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/c3abcea084364b47aba5fa01d8788b2b : SUCCESS in 6m 13s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1097dbb4a92c4dab8dc0d4777744728c : SUCCESS in 4m 49s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7a36a558556f4b76bca5a3b35ba5b1ef : SUCCESS in 8m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f28a73169aed4ed68b45695825f4d680 : SUCCESS in 7m 23s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1a76e52dbbf043739fa935ab16094cd9 : SUCCESS in 4m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cdd322d9480242a28476d4eeeb3c5a26 : SUCCESS in 1h 48m 24s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/ca8c855f1216470984cfced06a2c12c6 : SUCCESS in 1h 50m 29s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/52f7b9c8844a4c1581f7b2ebf3535f17 : SUCCESS in 1h 06m 16s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f0e16f83c42748c7b681728f92a90da3 : SUCCESS in 1h 58m 16s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/34ee936889a24adeb164d5ac8d092f26 : SUCCESS in 1h 06m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e9d875d52cca4c90aa7f4572ef87dd24 : SUCCESS in 38m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ee5c3f1e60c143b7a977e0e501e09000 : SUCCESS in 40m 43s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/d5c929254af540a99827fd3a44299b13 : SUCCESS in 1h 06m 27s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1dca1fdea8e840238f88ea148fb00f62 : SUCCESS in 36m 00s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/71e98ab7857b41c2b1567966a3d073aa : SUCCESS in 2h 30m 08s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6a79fd5bd57349cdaef5c9cfacb8326e : TIMED_OUT in 3h 03m 11s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/039f03d66532405e8bf0d57223d213c9 : FAILURE in 22m 22s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/f945f285b5a04f7fa76b33d3621c0e2e : TIMED_OUT in 2h 33m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f2bbca60788cf808590174173e046bed4e56ee1b","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2024-06-05 16:34:01.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"6b13c332c84ddebb987bc5e9fe2ad2e37d3d32ec","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-06-05 16:38:36.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c33c78cb605bdb2abb1cddd076ac313d38a99f1f","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-06-06 12:00:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"19c627d672db48c64b914fd79046b38e4c8480eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-06 12:00:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c60d0922832374b878242003778320cd0e04f239","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-06 15:06:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47085ffdb6894ecf98710fed2dd4395c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf2a4aaee5e64af59029d74cd74431fb : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad5cf58534484441815d1fef429e28a3 : SUCCESS in 7m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4d49579e27de4222be3a8dbcbc324ba4 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4551804e7eaf4a54a03de12dfff5e15f : SUCCESS in 12m 13s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6a75c6bd55124783b8e8e4a863898903 : SUCCESS in 4m 17s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/213837a0a6424d69b7edb800d115fa82 : SUCCESS in 5m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fab44eaf39d240788a4e865d4b1914b7 : SUCCESS in 2h 59m 49s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/c3004a0c2dbc41f2b2b343a703d22983 : SUCCESS in 1h 47m 41s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/51d849ebbbb84a6ab3015f6f71eff239 : SUCCESS in 1h 28m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/39fb5805a7834aa6bd4a796fc3b433e7 : SUCCESS in 35m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/870d58f17e2b4c3ea7bdecd77987ca63 : SUCCESS in 54m 35s","accounts_in_message":[],"_revision_number":2},{"id":"c75117797d222d8c9353ceaa318931e34db4261a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-06 15:06:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"315b4c518a7538e165692222450ba511d7db6e73","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-06 15:09:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8205dc624d344040a957c010b89e65fa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78e1478d26c14f49b8fc5027fb9010eb : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"310cf313bc9917d5983cf55527ba5bce97bb96e9","revisions":{"588622aa87b6d6abec559847c2190babec1f9bcc":{"kind":"REWORK","_number":1,"created":"2024-06-05 08:02:51.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/56/921356/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/921356/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/921356/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/921356/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/921356/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/921356/1"}}},"commit":{"parents":[{"commit":"541682314693ab543b148a9ed7703192f1673e00","subject":"Merge \"String conversion to f-strings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/541682314693ab543b148a9ed7703192f1673e00"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-06-05 08:01:13.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-06-05 08:01:13.000000000","tz":120},"subject":"Replace use of deprecated cert.not_valid_after","message":"Replace use of deprecated cert.not_valid_after\n\n... attribute. See [1]\n\n[1]: https://cryptography.io/en/latest/x509/reference/#cryptography.x509.Certificate.not_valid_after\n\nChange-Id: I3dd573300a8dfbe68eaeb736b47f6abb61b5e85b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/588622aa87b6d6abec559847c2190babec1f9bcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/588622aa87b6d6abec559847c2190babec1f9bcc"}]},"branch":"refs/heads/master"},"310cf313bc9917d5983cf55527ba5bce97bb96e9":{"kind":"REWORK","_number":2,"created":"2024-06-05 13:21:53.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/56/921356/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/56/921356/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/921356/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/921356/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/56/921356/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/56/921356/2"}}},"commit":{"parents":[{"commit":"541682314693ab543b148a9ed7703192f1673e00","subject":"Merge \"String conversion to f-strings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/541682314693ab543b148a9ed7703192f1673e00"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-06-05 08:01:13.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-06-05 13:21:43.000000000","tz":120},"subject":"Replace use of deprecated cert.not_valid_after","message":"Replace use of deprecated cert.not_valid_after\n\n... attribute. See [1]\n\n[1]: https://cryptography.io/en/latest/x509/reference/#cryptography.x509.Certificate.not_valid_after\n\nChange-Id: I3dd573300a8dfbe68eaeb736b47f6abb61b5e85b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/310cf313bc9917d5983cf55527ba5bce97bb96e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/310cf313bc9917d5983cf55527ba5bce97bb96e9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
