)]}'
{"id":"openstack%2Foctavia~921429","triplet_id":"openstack%2Foctavia~stable%2F2024.1~Ide2d5e637e5feb2dad43f952d5be1a195da1ae37","project":"openstack/octavia","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-06-25 16:32:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ide2d5e637e5feb2dad43f952d5be1a195da1ae37","subject":"Add check for duplicate members in batch update","status":"MERGED","created":"2024-06-06 09:52:02.000000000","updated":"2024-08-02 12:35:11.000000000","submitted":"2024-06-25 16:32:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"921429","meta_rev_id":"68c76be0b0a7a68b9463c6a206c2b4686ae73945","_number":921429,"virtual_id_number":921429,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-25 16:32:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},"all":[{"value":1,"date":"2024-06-07 11:43:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":1,"date":"2024-08-02 12:35:11.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":1,"date":"2024-06-25 13:02:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},{"value":2,"date":"2024-06-25 14:47:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-06-11 06:56:26.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},{"value":1,"date":"2024-06-25 14:47:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-06 12:59:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-07 11:43:41.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2024-06-11 06:56:26.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2024-06-25 13:02:00.000000000","updated_by":{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},"reviewer":{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},"state":"REVIEWER"},{"updated":"2024-06-25 14:47:37.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-08-02 12:35:11.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"}],"messages":[{"id":"84cc380c9b7ce9c1a5c033ce8b84b8bb12aafbf4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-06-06 09:52:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d5b5948a417b32a1e4a3a9b660e24e8470c67638","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-06 12:59:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a567c269bd6a4cb08859d2ce2aba3af1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f7daf808ac164f9fbf9e9c09fa664a47 : SUCCESS in 13m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1136c2877f794e74b67426164ddb1118 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/56c7494216a14362b09a6107b861bf0a : SUCCESS in 4m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/63ecd3ac4da845c883d0ed3c1d3ce166 : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd2b356737554323880ac6b0423c8f82 : SUCCESS in 11m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/510e33c34b934426b73ed21d7155272c : SUCCESS in 5m 24s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e0ff907711c7486db101bb365d1309d6 : SUCCESS in 4m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/00276054c2b543b3acb0de88a7a444a2 : SUCCESS in 6m 58s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2c4cc3bbe6f045659e16d8d4b2d6eabc : SUCCESS in 6m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/28d3f1afd46940a3a0cf549d73adaa0a : SUCCESS in 7m 07s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0ee51a5a166a4714b27f3738ec069b2a : SUCCESS in 1h 52m 12s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/c8290ebfd4f144d9a9d31bf67b60b8a4 : SUCCESS in 2h 08m 19s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/cb8c470b6c3b4b42a711ed78cef0fd4c : SUCCESS in 1h 17m 50s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/40a9cb7a4a2942649df05b347a7b2767 : SUCCESS in 1h 42m 54s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/5a4c3200006e472fa569cd18ce401826 : SUCCESS in 1h 31m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1b9cc9aa04d548129ed81b99bf8d4758 : SUCCESS in 36m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/67d42b10a3c348a1bfe516b43bb98cdc : SUCCESS in 1h 05m 36s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/9c3de4235fde4696a8d9207ca86706f6 : SUCCESS in 50m 31s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/4596bc7557bd4546951fa924271c077e : SUCCESS in 1h 17m 51s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d14a7b470ef046e6a6cf3cb43f3e1566 : SUCCESS in 44m 35s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a22e400a4f8c4c8089e3f7d5cb39fbae : TIMED_OUT in 2h 34m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/dc84cfc2d59a4b78aa2afa592b6aea28 : TIMED_OUT in 3h 03m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/85fb5bace37a496b9bb7ca4b16e57538 : FAILURE in 39m 05s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/6e0b52cdb3b24ec28e4bcc506a15b2ef : TIMED_OUT in 2h 34m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9f05c1422151395a02f673ffb914debb5bcbc223","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-06-07 11:43:41.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"175f349c4232f5733b04d9d982f309363db9cb53","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-06-11 06:56:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2f8058b9b7b00e8f8ee9c3c19ce02af766f40ae5","author":{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},"date":"2024-06-25 13:02:00.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"53ab6dd960425aafc4667003575210e652ccd813","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-06-25 14:47:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3f340eb227e976b2b85ba3d30f3e4aa697f9eb6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 14:48:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8ddad533f141b5a1f9ab394cfed6c8eb79f78522","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 16:32:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e79fee2716ef4280a70963f65aec797b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0e226dfa7484021b5d2b20d0163aa4e : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a741b22c86b4650b8bc39922c503dbc : SUCCESS in 3m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7673f9398122408882f12fef37649463 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cefc5c08e954ad7b836e676db89623d : SUCCESS in 9m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/942d798129364e8881207e164b666b5d : SUCCESS in 4m 23s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/cc09f41b85ae4a5baf5613277c8df7b4 : SUCCESS in 3m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/30d2eaf301d2422496fc0a384d8c7557 : SUCCESS in 6m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bdb7d81716f44669940b474bbde8cdb2 : SUCCESS in 1h 42m 25s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/03c0bcc4dfa940f19d0e0da23f3cf561 : SUCCESS in 1h 43m 17s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/453d6a5ce34046879fdb51bdf0c98c69 : SUCCESS in 59m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e8c3712a5b0243438c93c770e15ba861 : SUCCESS in 34m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4f4bf223dda84c819a1feb2da69ff3e8 : SUCCESS in 41m 19s","accounts_in_message":[],"_revision_number":1},{"id":"0a3e28bd19e7fd5bde5515fc0cf68321237802b8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 16:32:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fbcb7284b6dee90b42d5e707b446c89c2f58c342","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 16:33:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2819acfab6034e04a24ad81fcec775cc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf34b9df5e704563a62a78bdd657241e : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/11efc95f8b24444f9ba2760afed33cd2 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1},{"id":"68c76be0b0a7a68b9463c6a206c2b4686ae73945","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2024-08-02 12:35:11.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b61c340b096f0a2a90a7db8209639da11e25a4c6","revisions":{"b61c340b096f0a2a90a7db8209639da11e25a4c6":{"kind":"REWORK","_number":1,"created":"2024-06-06 09:52:02.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/29/921429/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/29/921429/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/921429/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/921429/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/29/921429/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/29/921429/1"}}},"commit":{"parents":[{"commit":"ab7712ede59b4585be36358dee7271acfe978a63","subject":"Merge \"Fix listener update when using SRIOV VIP\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ab7712ede59b4585be36358dee7271acfe978a63"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-11-10 09:13:46.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-06-06 09:51:32.000000000","tz":120},"subject":"Add check for duplicate members in batch update","message":"Add check for duplicate members in batch update\n\nIf a user requests a batch update of the members and includes the same\nmember twice in the body of the request, taskflow triggers an exception\n(because of a duplicate Atom) and the load balancer is stuck in\nPENDING_UPDATE. Now the API denies such calls and returns a\nValidationException if a member is updated more than once in the same\ncall.\n\nStory 2010399\nTask 46777\n\nChange-Id: Ide2d5e637e5feb2dad43f952d5be1a195da1ae37\n(cherry picked from commit d7b090702523cb07ae4aed0338f9e80fe30791a6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b61c340b096f0a2a90a7db8209639da11e25a4c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b61c340b096f0a2a90a7db8209639da11e25a4c6"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
