)]}'
{"id":"openstack%2Foctavia~921433","triplet_id":"openstack%2Foctavia~stable%2F2023.1~Ide2d5e637e5feb2dad43f952d5be1a195da1ae37","project":"openstack/octavia","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-06-25 17:21:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ide2d5e637e5feb2dad43f952d5be1a195da1ae37","subject":"Add check for duplicate members in batch update","status":"MERGED","created":"2024-06-06 10:15:48.000000000","updated":"2024-06-25 17:22:41.000000000","submitted":"2024-06-25 17:21:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"921433","meta_rev_id":"1b323f5b3cde38ea791794dd74eebc0e8149adbd","_number":921433,"virtual_id_number":921433,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-25 17:21:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},"all":[{"value":1,"date":"2024-06-07 11:43:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":1,"date":"2024-06-25 13:01:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},{"value":2,"date":"2024-06-25 14:48:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-06-11 06:56:10.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},{"value":1,"date":"2024-06-25 14:48:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-06 13:21:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-07 11:43:31.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2024-06-11 06:56:10.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2024-06-25 13:01:45.000000000","updated_by":{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},"reviewer":{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},"state":"REVIEWER"},{"updated":"2024-06-25 14:48:05.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"ea8dc7e9018e55167c81caf4bbae98ba0f756f82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-06-06 10:15:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fca0b0a328c0357f04c938bab8de08da30465a0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-06 13:21:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d56a99426f740cf8facaa9d84ec903d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b1ed30a428f2457da78a2a9c94083492 : SUCCESS in 12m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b17e895a1d324ddbabd527e668649cd7 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82cd69a7c113481686f3a61afee2b0ec : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be2c2a44555b44b7aa2516bc973c9115 : SUCCESS in 12m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d805e73d185a4a7a91c022a26f5a9924 : SUCCESS in 5m 33s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/31ae53379610452495f03389fa1a416a : SUCCESS in 4m 39s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/dfbf3d09c5474be7a5b675806382ceb1 : SUCCESS in 8m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e05df716b4f24f2bbafb57a93ab66be0 : SUCCESS in 4m 33s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4e008a10207a4e7194d6aa395d9b0d4b : SUCCESS in 8m 24s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9aa808afc68a44f18c64d6386024f556 : SUCCESS in 7m 45s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c0628e627b464f7d84eced1e6d422261 : SUCCESS in 6m 44s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ded5096aa6444a6082fb95a6c9e7220e : SUCCESS in 1h 59m 36s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/06f9e31626f14fc5b76f7421c3e85a58 : SUCCESS in 2h 00m 25s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/1d92d74a86af4367b4d482ed9f0c3cac : SUCCESS in 1h 10m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9aa0a5a431244a299c6ef998259a5ef3 : SUCCESS in 42m 16s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/9738b7ac3a154d899d2ef6057ca421eb : SUCCESS in 1h 02m 55s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5732bd059a8247baabcd588a5be48e23 : SUCCESS in 39m 53s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7d7f121b2051440c80d569d412b912cf : TIMED_OUT in 2h 33m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8ce3707930504ed6b2d41a83471db73d : SUCCESS in 3h 03m 18s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4ae00a39cce242be8c8157a05cc758b0 : FAILURE in 31m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9f2808c186034369bcb16ad7676cc8f9 : SUCCESS in 2h 25m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/7de6fe07623745d091af325e13f6abe1 : SUCCESS in 2h 30m 23s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/4f14223f78054710b5c7e137b6cb9b44 : SUCCESS in 2h 29m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c17f2450eb01433d05c87f8760c8874ee379d8b3","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-06-07 11:43:31.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"74177fca398a9cccbf0f02c49cc92c6d5d54d046","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-06-11 06:56:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c634ce40893c7f05e457c12f96fc4ff64a143858","author":{"_account_id":2711,"name":"Zhang Hua","display_name":"Zhang Hua","email":"joshua.zhang@canonical.com","username":"zhhuabj"},"date":"2024-06-25 13:01:45.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6e68a920ffe7986ba886087a714a346d08166160","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-06-25 14:48:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0025995197308145656aab78d21d01dcb634ee16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 14:48:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"80288391106c8b78162a3df8711bc995f3bddcde","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 17:21:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbf2bc567bab4f3898147a25233424f8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e8efe8305534a39a321cc23202ff7f0 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2bd21b5cb84a4cc087b52bb81cb93d26 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d61e4d49d91748d8bfd3197f79847ba5 : SUCCESS in 10m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be205e6a62e548cbb86b42b462c13326 : SUCCESS in 5m 32s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ce6a50dd8f044d7eaf49d4f5c0cb41a0 : SUCCESS in 4m 38s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/01248113a2fb4cc4b0778097f8650bc6 : SUCCESS in 8m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6ceaafac30c7417dbfe9dec5828340c9 : SUCCESS in 2h 26m 05s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/6bfe181a8c3148f0939636181b61878e : SUCCESS in 1h 51m 51s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/4a94672a20a94b5f8ec0b8c0ba5ef1e4 : SUCCESS in 54m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/37e1a93e770543ccaa5bc55e7f84e634 : SUCCESS in 51m 15s","accounts_in_message":[],"_revision_number":1},{"id":"81769da2dc03c74bde7bc31635190f9b6b34b593","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 17:21:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1b323f5b3cde38ea791794dd74eebc0e8149adbd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 17:22:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b96510cfb7642b18d407742afd3ec60\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fbd53e44ebb4f0083ae7f596e997db4 : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d2fe406d6a42443791f525c1b891dedd : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f7abb00fb9da14b498852f37a106d82b778a9b8b","revisions":{"f7abb00fb9da14b498852f37a106d82b778a9b8b":{"kind":"REWORK","_number":1,"created":"2024-06-06 10:15:48.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/33/921433/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/33/921433/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/921433/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/921433/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/33/921433/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/33/921433/1"}}},"commit":{"parents":[{"commit":"583ce8dc1b8cba2fbc78911312ce447c97a3b695","subject":"Merge \"Fix health monitor information retrieval in API response\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/583ce8dc1b8cba2fbc78911312ce447c97a3b695"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2022-11-10 09:13:46.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-06-06 09:56:13.000000000","tz":120},"subject":"Add check for duplicate members in batch update","message":"Add check for duplicate members in batch update\n\nIf a user requests a batch update of the members and includes the same\nmember twice in the body of the request, taskflow triggers an exception\n(because of a duplicate Atom) and the load balancer is stuck in\nPENDING_UPDATE. Now the API denies such calls and returns a\nValidationException if a member is updated more than once in the same\ncall.\n\nStory 2010399\nTask 46777\n\nChange-Id: Ide2d5e637e5feb2dad43f952d5be1a195da1ae37\n(cherry picked from commit d7b090702523cb07ae4aed0338f9e80fe30791a6)\n(cherry picked from commit b61c340b096f0a2a90a7db8209639da11e25a4c6)\n(cherry picked from commit 70ecbcc8be77a77f880f4d6a827bb5b57dc55087)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f7abb00fb9da14b498852f37a106d82b778a9b8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f7abb00fb9da14b498852f37a106d82b778a9b8b"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
