)]}'
{"id":"openstack%2Foctavia~921752","triplet_id":"openstack%2Foctavia~master~I6ecbd6175004500855afe3b3657c97e72469ccac","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"34750":{"account":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"last_update":"2024-06-12 02:21:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6ecbd6175004500855afe3b3657c97e72469ccac","subject":"Update cryptography requirement","status":"MERGED","created":"2024-06-11 10:26:02.000000000","updated":"2024-06-12 02:23:33.000000000","submitted":"2024-06-12 02:21:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"921752","meta_rev_id":"7d77c6ccda92f4befb9f0e4b02d8eb3af10e6318","_number":921752,"virtual_id_number":921752,"owner":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-12 02:21:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2024-06-11 11:27:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2024-06-12 00:24:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2024-06-12 00:24:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-11 11:27:48.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2024-06-11 13:30:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-12 00:24:20.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"82a42a26b6a23737354d0752260946fc8863959b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"date":"2024-06-11 10:26:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"461a0f169dc33fd0ada107be98f77f567f74e53e","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-06-11 11:27:48.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b0e13aa9bfc31639bdb4b359e25fe586a3a7356d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-11 13:30:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/065ab3a5110346f19331fe4a42db2f60\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d700c2806c50467bb1e394abac25cc95 : SUCCESS in 1m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a80bea4b3b54c92be8e6fcedba2c5f7 : SUCCESS in 13m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f467213abf4402fb6bb9fdb45f95d77 : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f72ca76228934e19b8f1ccf8a3e95939 : SUCCESS in 6m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/66388a2dc8c94b8c840b278afb73f418 : SUCCESS in 5m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4f915e5e91f34029a69c3f07496ccdf8 : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cce55d2eb2d441fe9b22626785e15a81 : SUCCESS in 11m 12s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/2656544576b74037b044ef440599cc13 : SUCCESS in 5m 25s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/f7a6b0d3c2ac4c4fbd84607bbdc8f654 : SUCCESS in 7m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/adc4684a7f444c67aeb3865bf7c32950 : SUCCESS in 5m 24s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4a22e3ca64c54da1860613ed4d2b4841 : SUCCESS in 9m 44s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4a35cb785db74beea8f406fe7eca80e1 : SUCCESS in 7m 03s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/68a9ecf53231481bb61d01e68f2063f2 : SUCCESS in 5m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e74e518531e84ded9835edbb184a128d : SUCCESS in 1h 40m 08s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/d50eb06dc9fb416e8559b2eddae462e1 : SUCCESS in 1h 49m 01s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/bce95ecf33ad4a6895d2e65384132e50 : SUCCESS in 57m 18s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f9260260fa3f4c99b0fd901d3e13eb7d : SUCCESS in 1h 53m 08s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/1c782e29800a41608d0c4ce4c5fa1dcc : SUCCESS in 57m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6abdf5f3abac4fea8b4217489cbe33f5 : SUCCESS in 35m 37s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/492a389e081940bfa44e038d0c753417 : SUCCESS in 1h 11m 02s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9c64bcfe1a4444dbaeb900a1359635dd : SUCCESS in 49m 25s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/809988a791a6420a9a2172b12ebe6aa2 : TIMED_OUT in 2h 34m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1e27a06ce40041ff9b94cc80c85835c5 : TIMED_OUT in 3h 02m 51s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9b461d87f3264c23b60f88f11cf9c73d : FAILURE in 34m 58s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/20c5acfc28504243b0383bf290456522 : TIMED_OUT in 2h 34m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"09ed6498831abccbcb6f544b932fa66a6eb04908","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-06-12 00:24:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7c422b981b9c54eb4bdeff4f031a086ef0062e8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 00:24:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7a70912c6e91a2b12255469b0fe70d82f6dc1bab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 02:21:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/645c71e7de8f452999cb9aa9578b35d9\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/39f9486ad63c4b81843efd924c0f96f0 : SUCCESS in 1m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be9cc35a1f3b4f9fb03de02dc870425d : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4a4ce90a9f6b42ea9d1691c55bd73351 : SUCCESS in 4m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/742fc5ee148349bba57390cb502eda2d : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcbbcbbda53a442099f32b057d67fc08 : SUCCESS in 11m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/77d2b34ab060469fbeae084c47d290b6 : SUCCESS in 4m 10s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3a1a12e887714023bd1a08215d60be5d : SUCCESS in 5m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bcfa093c48024f928414530be7b6c452 : SUCCESS in 1h 14m 22s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/7a3c6177508e49bab3e20d0859a8c15d : SUCCESS in 1h 56m 25s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/74fba3ac3e1240b59115fc3460b26bc0 : SUCCESS in 1h 09m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2994f8a5fd4046b3b879489dfebdcf4f : SUCCESS in 41m 09s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/64aa974052a5435d8fb80bd154be919b : SUCCESS in 44m 49s","accounts_in_message":[],"_revision_number":1},{"id":"a737dbb49681acae2f3456a11b33b14597d9635b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 02:21:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7d77c6ccda92f4befb9f0e4b02d8eb3af10e6318","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 02:23:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4c57d3849f4438c8047226fbe969a8f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b48fd1c95d4c4a22bb6c77ae95e79e0c : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"830134808dc00438ad6dd0db9984bacea4cfd30d","revisions":{"830134808dc00438ad6dd0db9984bacea4cfd30d":{"kind":"REWORK","_number":1,"created":"2024-06-11 10:26:02.000000000","uploader":{"_account_id":34750,"name":"Karolina Kula","email":"kkula@redhat.com","username":"karolinku"},"ref":"refs/changes/52/921752/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/52/921752/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/52/921752/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/52/921752/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/52/921752/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/52/921752/1"}}},"commit":{"parents":[{"commit":"fcc61e9ca8e79674cf641eefcb65972e60a25e7d","subject":"Merge \"Replace use of deprecated cert.not_valid_after\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fcc61e9ca8e79674cf641eefcb65972e60a25e7d"}]}],"author":{"name":"karolinku","email":"kkula@redhat.com","date":"2024-06-11 10:17:40.000000000","tz":120},"committer":{"name":"karolinku","email":"kkula@redhat.com","date":"2024-06-11 10:26:00.000000000","tz":120},"subject":"Update cryptography requirement","message":"Update cryptography requirement\n\nRecent change [1] introduced usage of function not_valid_after_utc\navailable in crypthography since 42.0.0 version [2]\n\n[1] https://review.opendev.org/c/openstack/octavia/+/921356\n[2] https://cryptography.io/en/42.0.2/x509/reference/#cryptography.x509.Certificate.not_valid_after_utc\n\nChange-Id: I6ecbd6175004500855afe3b3657c97e72469ccac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/830134808dc00438ad6dd0db9984bacea4cfd30d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/830134808dc00438ad6dd0db9984bacea4cfd30d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
