)]}'
{"id":"openstack%2Foctavia~922673","triplet_id":"openstack%2Foctavia~master~Ice6e4ee98edb446a8b0547ced477598dcf22f791","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-09-12 09:42:56.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-09-12 07:56:23.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"Ice6e4ee98edb446a8b0547ced477598dcf22f791","subject":"Fix SR-IOV when VIP interface is used for members","status":"MERGED","created":"2024-06-24 21:15:58.000000000","updated":"2024-09-12 09:44:09.000000000","submitted":"2024-09-12 09:42:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"922673","meta_rev_id":"e5fa6b6b72f2bf1acb9dc4e3fccf22781068b4ba","_number":922673,"virtual_id_number":922673,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-12 09:42:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2024-09-12 07:56:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2024-09-11 16:49:37.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2024-09-12 07:56:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"recommended":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2024-09-12 07:56:23.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-25 00:00:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-31 07:58:36.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2024-09-11 16:49:37.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"6c6e0ccc28673e900df37d2153324eb833e6c25a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-06-24 21:15:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a000f18e9017db87fdc1790142ca4993400856dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 00:00:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f607a83144874caab5080e977b1f7a9d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b57c315425d4300bbb80ed82bf991a0 : SUCCESS in 10m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd7fc563887b48baa3170f93c5c5f205 : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7da0c82ec78e44f4a78e0482b21bf122 : SUCCESS in 4m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/36c6245664964736a021da1e09e51a25 : SUCCESS in 4m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7dacc4b94c05465aa50dbeed5d6083bb : SUCCESS in 10m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/463a645aa22242b2943cbf0275e7c13c : SUCCESS in 10m 27s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/e374019c7d0f4212a5a6515778e1dccd : SUCCESS in 4m 53s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/22ebeefa7c33477899b2083f80691f61 : SUCCESS in 7m 20s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b8aeaec4021f482a854902a69487dd54 : SUCCESS in 4m 15s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1ba8e5a7ca5d4c71a78bd658fa9ce147 : SUCCESS in 8m 39s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c01cd080dda44f51b08567fef363616f : SUCCESS in 7m 53s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/833c1399f2754be18fef062254af2de0 : SUCCESS in 7m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3ef957d02aaf4b36b75f5fb722f139f4 : SUCCESS in 1h 37m 17s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/a01d56908d1042358d335a3638f1d370 : SUCCESS in 1h 52m 52s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/fb85cd83fb604824888739e4e9f7e88e : SUCCESS in 1h 00m 37s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/5a2bbf47b24e40e8b45bafa6bd139e99 : SUCCESS in 1h 54m 21s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/7849645e91ba484ba18dafab810e0c24 : SUCCESS in 1h 31m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9c49fa2d81ae490eb498e7ac6b94875a : SUCCESS in 39m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/aaa56aaad54c4a51a0db4d6e096bca8b : SUCCESS in 43m 06s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/6a32fb85920a4ff4b594273ff7d13f5e : SUCCESS in 1h 08m 10s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ab982ccb1a834d93a6ba774000b8312e : SUCCESS in 37m 46s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/32d9b6415aa9491db3a50d26b9e3a936 : SUCCESS in 2h 21m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c979655a090d4dbaa7da3cf82aa62efe : SUCCESS in 2h 32m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/956640c2d1c34aee9a11897e9ed7caca : FAILURE in 26m 21s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/a60031f0511b49ea9b8a5958ac3e2741 : TIMED_OUT in 2h 34m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"762e796ad149ae22524d7b55872d27b4146b723d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-06-25 14:39:16.000000000","message":"Patch Set 1: Backport-Candidate+1","accounts_in_message":[],"_revision_number":1},{"id":"bfefb50fa6f4c6b9c0a23a1a6f97cdd3e0eb5021","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-07-31 07:58:36.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fe9300425ddcd78c7778e642765fddf4cb368f12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-08-14 22:52:34.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"d08167ed133963aad2dd5adc63584d9066a67a92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-15 01:38:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4e99a856fa294d618d281b02318aac63\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ba7a847ff97416ea3ed1a3b8772c748 : SUCCESS in 8m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e15a4e8bb25c4ac18ccd50adaa3130d0 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8008a4f4db7041c087c79ecb0580e36e : SUCCESS in 4m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/36392b022de446afbd96d21972d1bf83 : SUCCESS in 5m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/de17ff5846df41e6a4e1521e140cad49 : SUCCESS in 8m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2016092c4c234c0bb3a9e8c0f28d0fe6 : SUCCESS in 11m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa2ce001b12442b99d616f216915606a : SUCCESS in 5m 47s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/61c99c48b6d549dbba7c46052f6b6e06 : SUCCESS in 3m 47s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/00ad9016c03e4fd7bac5d49455f1dc17 : FAILURE in 5m 51s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/76f80c168eed433a80c9a7bf5e9c4657 : SUCCESS in 4m 11s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a6e2d8a3bdc141a5bf3c542e22eace62 : SUCCESS in 7m 33s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1fa02080e1ea4c2bbcab557671bb5981 : SUCCESS in 5m 57s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8292f8586cbb4cd29a66cc526036abd0 : SUCCESS in 5m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3241f5737ad64b55a9a712c5efe2e889 : FAILURE in 16m 31s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/3524c6a7855947639269eb2844610e8e : FAILURE in 18m 28s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ed8969ab207f41ef86c3c705c3c0c85f : FAILURE in 18m 16s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3f2c23f7d9f84227bcd2cde42871ef44 : FAILURE in 18m 22s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/7911032586994290bbe4355953f4a2d8 : FAILURE in 18m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ac2a99ba260643d688e9b9777ee60ef3 : FAILURE in 18m 17s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c0c092d1a23645ea80bb2d7da8a84fc5 : FAILURE in 18m 42s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/2c99f4f434eb4d61a991f198b68643bf : SUCCESS in 55m 02s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/213b19baa19a44c0b2adfd5e0b8a9e2b : FAILURE in 18m 23s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a0b160dc2b7247f09486a90ebd70b66c : TIMED_OUT in 2h 32m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/22f6b3639229405c8a57a683857cf211 : SUCCESS in 2h 45m 04s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b92bd03599ab4de69739d3983e14c20a : SUCCESS in 2h 19m 19s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/82770605eebc4eb98bf3f25dbc04cc1e : FAILURE in 9m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d6342a27fcaddfab79288ff061e6f114f9d9686c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-09-11 15:51:25.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"a88a08b71e4c40a1c344777d3f7daa8cf20cb73f","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-09-11 16:49:37.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f15c94a706c0645a381860eb7ca61ab36e9f626d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 18:48:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6142a2c695124121bca650ecf3e52cec\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9342f63a4f6640bd833580712ac36f00 : SUCCESS in 9m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01c7d44a6eb3404face7e8c51a210b07 : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2bed015ac5d24da0b5ad56582e2cd335 : SUCCESS in 4m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d60072615db24560abc3e0c0b22692fb : SUCCESS in 4m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/521ad05c21054ac4a1fd8eb79d41beb6 : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/160e32bd7c754db79121c1b59da7f00f : SUCCESS in 8m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fdcd89c34fd14211a639f302365eb7f0 : SUCCESS in 4m 36s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/bf3b6402ed3c49bfba695148da92b0a1 : SUCCESS in 5m 25s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/c551a95c142347c4a8a6b17ec4c35c02 : SUCCESS in 10m 26s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c318636058ae46bead6c0ab52a7cd36a : SUCCESS in 4m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e01a7f9473b94fc895ed009d9447cf2d : SUCCESS in 7m 50s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/93b470a0b15c42be8c072beb9117d421 : SUCCESS in 9m 17s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/39ec44664fbf45bf8a40c865ea9cf65a : SUCCESS in 6m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f0fba53a85c447579b4d8d6ca87d6482 : SUCCESS in 1h 39m 22s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/2f80e137f1de471c98c24526ec89254b : SUCCESS in 1h 43m 52s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/4863757cfa0246b0aecf301154a7c34b : SUCCESS in 56m 50s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/22621eb966a9465ab10bea9edbdbabe8 : SUCCESS in 1h 46m 54s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/543f93e371b0424da5c827b7bd385155 : SUCCESS in 59m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/41ce115c7dce48e4bfbbddcc9763bfdd : SUCCESS in 31m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8203c1aeb141451fbe509304aab9e17d : SUCCESS in 38m 28s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/eb4d00e17a274fcf9556939fa5ff659c : SUCCESS in 1h 09m 14s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/76839d9c119a495f9e86c2f9889f941f : SUCCESS in 37m 08s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e47d0f406eab4649aea57842b444f4b2 : SUCCESS in 2h 29m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/22384946c4044b8ba8d8bf38e1373142 : SUCCESS in 2h 55m 43s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3b499794b24a4f0ea407cae39b9517eb : SUCCESS in 2h 11m 36s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/8c2a7e83bc4541daaeceb475b31b1bba : TIMED_OUT in 2h 34m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7515565696a7b34bc4780218d5275bcf826a8087","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-09-12 07:56:23.000000000","message":"Patch Set 3: Backport-Candidate+1 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6e9bff3eeaf83a38dbf3fbbf5654103b3ee6209e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 07:57:00.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"edeb5100758c28435af0afb458fa1adcd2d3a322","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 09:42:56.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f64a43b2a3b40f7817cb93a78687163\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e6ff84a7b384cca84c6a434f231f7d3 : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/577410af24bb452c8627e3e4aff19329 : SUCCESS in 5m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d8e1b2e0cf5e43e8badb1cf7b2ab3e6a : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13d877575eaa4584a0c505adbb5f507b : SUCCESS in 10m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd9e417f77004394a6bc4abf7a2d0e1b : SUCCESS in 5m 01s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d1a90c96031b402382115f4bfc15f395 : SUCCESS in 3m 12s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/91e57602af6e4a049d5ec4b1da228f64 : SUCCESS in 6m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4b0c1645fddc4fc79426c5d892ae9397 : SUCCESS in 1h 21m 23s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/52ff16fdf81640f2abdf035f86d97a1e : SUCCESS in 1h 45m 18s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/886b93d05b85445f8abc2b627629ddb5 : SUCCESS in 1h 05m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/107825aee42b4e58ba5497260d2ae1a1 : SUCCESS in 32m 36s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c3dc4e88299d4ea58a20e3c5dc8a915f : SUCCESS in 1h 01m 21s","accounts_in_message":[],"_revision_number":3},{"id":"8ab4eb8682846a06deba9b8c1815f6d3e322449b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 09:42:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"e5fa6b6b72f2bf1acb9dc4e3fccf22781068b4ba","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 09:44:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc1bc88cbb6b4f8e93f7442b041090a6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94a45764cfc04db487d7dc36453d8bf7 : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4abc02e2e8c244d088e5c575e680db6a : SUCCESS in 50s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8b8d969d0001cdf801478abf87dd8f103179b9b7","revisions":{"f40831f65978c2bb71901a4b4abf99e662aca52a":{"kind":"REWORK","_number":1,"created":"2024-06-24 21:15:58.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/73/922673/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/922673/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/922673/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/922673/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/922673/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/922673/1"}}},"commit":{"parents":[{"commit":"7376a4ad0c305113a2aabc091f1bb761a532b14d","subject":"Fix functional-py31{1,2} tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7376a4ad0c305113a2aabc091f1bb761a532b14d"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-06-24 21:12:00.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-06-24 21:15:57.000000000","tz":0},"subject":"Fix SR-IOV when VIP interface is used for members","message":"Fix SR-IOV when VIP interface is used for members\n\nThis patch fixes a bug in the VIP SR-IOV implementation that would cause\nload balancer memebers that use the SR-IOV VIP interface to not receive\ntraffic.\n\nCloses-Bug: #2070312\nChange-Id: Ice6e4ee98edb446a8b0547ced477598dcf22f791\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f40831f65978c2bb71901a4b4abf99e662aca52a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f40831f65978c2bb71901a4b4abf99e662aca52a"}]},"branch":"refs/heads/master"},"e581ee970a366391741259701c74c7ee32f57397":{"kind":"REWORK","_number":2,"created":"2024-08-14 22:52:34.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/73/922673/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/922673/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/922673/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/922673/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/922673/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/922673/2"}}},"commit":{"parents":[{"commit":"f68cafbb7201522e6140e2dc05317d06225bb82a","subject":"Merge \"Remove sysvinit/upstart support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f68cafbb7201522e6140e2dc05317d06225bb82a"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-06-24 21:12:00.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-08-14 22:52:33.000000000","tz":0},"subject":"Fix SR-IOV when VIP interface is used for members","message":"Fix SR-IOV when VIP interface is used for members\n\nThis patch fixes a bug in the VIP SR-IOV implementation that would cause\nload balancer memebers that use the SR-IOV VIP interface to not receive\ntraffic.\n\nCloses-Bug: #2070312\nChange-Id: Ice6e4ee98edb446a8b0547ced477598dcf22f791\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e581ee970a366391741259701c74c7ee32f57397"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e581ee970a366391741259701c74c7ee32f57397"}]},"branch":"refs/heads/master"},"8b8d969d0001cdf801478abf87dd8f103179b9b7":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-09-11 15:51:25.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/73/922673/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/922673/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/922673/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/922673/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/922673/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/922673/3"}}},"commit":{"parents":[{"commit":"9873200f68aedd5fba2698aebaec20593fc3531f","subject":"Merge \"Remove use of sphinxcontrib-*diag packages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9873200f68aedd5fba2698aebaec20593fc3531f"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-06-24 21:12:00.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-09-11 15:51:25.000000000","tz":0},"subject":"Fix SR-IOV when VIP interface is used for members","message":"Fix SR-IOV when VIP interface is used for members\n\nThis patch fixes a bug in the VIP SR-IOV implementation that would cause\nload balancer memebers that use the SR-IOV VIP interface to not receive\ntraffic.\n\nCloses-Bug: #2070312\nChange-Id: Ice6e4ee98edb446a8b0547ced477598dcf22f791\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8b8d969d0001cdf801478abf87dd8f103179b9b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8b8d969d0001cdf801478abf87dd8f103179b9b7"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
