)]}'
{"id":"openstack%2Foctavia~923792","triplet_id":"openstack%2Foctavia~master~Ie46a5d9620c8ef2b6da04976a742df9c71752e08","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-07-31 09:08:35.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-07-31 09:08:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie46a5d9620c8ef2b6da04976a742df9c71752e08","subject":"Stop using DHCP for o-hm0 interface in devstack","status":"MERGED","created":"2024-07-09 21:20:53.000000000","updated":"2024-07-31 09:09:45.000000000","submitted":"2024-07-31 09:08:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923792","meta_rev_id":"d6c88747a1febd13e7a46409fa8de11970d8871e","_number":923792,"virtual_id_number":923792,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-31 09:08:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-07-31 07:21:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2024-07-17 14:20:16.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-07-31 07:21:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-10 00:01:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-10 19:24:14.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2024-07-12 12:13:18.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"},{"updated":"2024-07-17 14:20:16.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2024-07-31 07:21:25.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"d917f1880763ca08092624770415b5239d499bed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-07-09 21:20:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e11a272b95544ff0bdfea2f7c5d4e718aa69ad98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 00:01:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/566bdca67e4947ef8da966dc14610feb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/acdb153272bb471d8cae9bfad848aee1 : SUCCESS in 7m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8de0bd8323dd4901b69a67e927c24c22 : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c13ae493e54d47bbb3e67994c84af522 : SUCCESS in 5m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/13850b77b7564ba4b96e133ebecc34ae : SUCCESS in 4m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/862b794061ce4a318bf014132e92d78f : SUCCESS in 9m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f402ccae01a467884710404b9a82393 : SUCCESS in 9m 42s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/711308defbfb4ad5bf5898e8d538388f : SUCCESS in 4m 45s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/566aa0bb6ae648eaa9a3435916e1086c : SUCCESS in 6m 20s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1c4b2c81bfca4de889c95988cd0e9304 : SUCCESS in 4m 25s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fed79bdfe7d44b068abe7f0498fafa9f : SUCCESS in 6m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/30bd0014331c4c7780068bc1d030c1be : SUCCESS in 6m 54s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/6b2b766b6b62498db8828ccd557d7395 : SUCCESS in 6m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a1cbc2d421e345cb92f2f4daaa28a66a : SUCCESS in 1h 00m 43s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/89bf0e1e3a3b4b0da435c37b3f189b18 : SUCCESS in 1h 42m 47s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/446fc4d8e1a74031ae839fd0185e8deb : SUCCESS in 58m 33s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/5ce5a9af3f3544f3980f6bc06a05fb0a : SUCCESS in 1h 49m 44s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/dfc24c015d4040d7847c5d3f881691ec : SUCCESS in 54m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7e2b9ad40e2e4d96a22ce6878de09600 : SUCCESS in 32m 40s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a99143eb166b470c80187be1b68b87ed : FAILURE in 31m 24s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/1a3d95fcd6e04f1185fd1746b4c73d0a : SUCCESS in 1h 05m 09s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fc78286c5ef24c22a989316882ce1477 : SUCCESS in 31m 00s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f36c2697d0ae465491976adc10771017 : TIMED_OUT in 2h 32m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/02aad55aef044cea9e5f82441f2dcc39 : SUCCESS in 2h 37m 09s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/35aa1e80885a4cf4961a2248e1ac3bfd : FAILURE in 31m 47s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/61feb622594642e6970dc59d73678df6 : TIMED_OUT in 2h 34m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"540f6541e1ab2c8d4642ba4a73ac642dc44eee49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-07-10 17:33:20.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a2c591f7cdcef5d4f60f1f1c1ed36c9898b8de48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-07-10 18:46:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8f296e98ec47f45c52eaacb437f21c94130806a8","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-07-10 19:24:14.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"9e7077fe46a39810af53b265e42dd73bf433a072","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 22:52:11.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cbd509d12ccd4f85a2b3a985692d8c45\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cdd3f39401b548bd9293145141c0fccd : SUCCESS in 9m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31d3459453cd4fe792719f088597dcf3 : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6dc152648f274c8294b863be09b1ded9 : SUCCESS in 6m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/75329080e3be47e097e58573f9e2764f : SUCCESS in 6m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3a5b6f9bdb854b189a717fe3b494834d : SUCCESS in 10m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5b6ebe9624c43908bbd6c9ea475a0be : SUCCESS in 11m 48s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/0ba05c11e57c437ca7c40cbc368bb017 : SUCCESS in 5m 12s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/e269d561c5734b1d8da1928b39f3965d : SUCCESS in 8m 11s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8232aca661134819acf1de8b86d2f44b : SUCCESS in 4m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4f890697ad364e2882cd5856ce73dceb : SUCCESS in 6m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e378a55a400f441c8c63bfc514e345e9 : SUCCESS in 7m 08s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/34f6122b3ae64ba48e989dbf8ac13c35 : SUCCESS in 6m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/035322ab4b384c408c5ccbfe06dbbca6 : SUCCESS in 1h 28m 01s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/1f6e872e15a24440ab15f78f210a195a : SUCCESS in 2h 26m 44s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/79e172f3e38845a1aa41c358d6fa1b84 : SUCCESS in 1h 19m 33s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c6878abddeb54ac0800b6323a918d71a : SUCCESS in 2h 21m 46s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f6d2f6fd913347b5babeb553bad35683 : SUCCESS in 1h 14m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c88f4c3e35ec45b69dd504767504f226 : SUCCESS in 43m 29s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ce34ef8d70c0411eb3b13b9ca07a124f : FAILURE in 45m 35s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/1dbb57db83184f1c9fd28a3427869f65 : SUCCESS in 1h 01m 45s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/39225446808b4698ae4d9a724be24984 : SUCCESS in 39m 20s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/77482c1f64b248638edbea0b14cbbbc7 : TIMED_OUT in 2h 34m 45s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8cb5bb6b48834d7a96e47a8c98001faf : FAILURE in 25m 14s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/52668ffe19ba4fb7bd4b8532daa2389c : FAILURE in 18m 20s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/10025f9e5ea045eaad88f00e02fdc2c3 : TIMED_OUT in 2h 34m 43s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6f14444167a179e93b284b5700dcefaec0639cc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-07-11 18:23:31.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"3d3092c4bcef3dbb07a8801260a72b90cfec1985","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-07-11 18:23:35.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"4861d277057e6d6ca50309e546ec2da36ac1b8af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-07-11 20:04:49.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"439d047a8096ed2f11d63a00640f756e5fc3a106","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 22:59:27.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc7e53e6f4d8421ba0dbf7039c51508f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/578bb0f918894dad9af5df34e705bbf8 : SUCCESS in 11m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50a62e7594ca400a81216ea0d7fb4571 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a90ac43415344350aa4c0865356d73b8 : SUCCESS in 4m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a375d79550924c4bb42da93b2cb960e1 : SUCCESS in 4m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e0433d86734a49b99819c03f6c9bd33e : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5106f5aec3b4df8b4f9c8322952bd41 : SUCCESS in 11m 29s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/8c06eaa118144aa89917319b753db3ec : SUCCESS in 4m 36s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/24138651fac141929b52a1e72a4d0765 : SUCCESS in 7m 56s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/599648cc30394d8bbe8af004ab5f8075 : SUCCESS in 4m 35s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/942a0eec6ee142e98c778432e89dd8ad : SUCCESS in 9m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/68bdcdce14564e1a89c026262b64323d : SUCCESS in 7m 23s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/252e5085a4d84be5bb95a574b9ac62d2 : SUCCESS in 9m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e5edb2f9d23c47cbb497e4cf00de2cec : SUCCESS in 1h 20m 31s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/db973195fbc34ad484c7cd010953b3e7 : SUCCESS in 1h 55m 08s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e9e045e94db4423fb2ce9bed180e73d6 : SUCCESS in 1h 09m 26s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c161f49e6b3e40a1a1a6bfc2a8b511e1 : SUCCESS in 1h 58m 59s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/844a475caed5410bb847d970a68ccb06 : SUCCESS in 1h 09m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b24985fc2b144fad9c417f7f63a9955f : SUCCESS in 41m 10s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/61aedc8b2c7f49939375cc4ba03a57af : SUCCESS in 50m 01s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/d03182e2d03942e0b8b4d62f1110a302 : SUCCESS in 57m 17s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2b3ca34ccef440a184c20703082a4e2e : SUCCESS in 40m 35s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0c0961673ba041ddb21cb27344426c9d : TIMED_OUT in 2h 33m 15s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/20ef85328229481bb2e5da5e57221efa : SUCCESS in 2h 51m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7f7e95c7e61545989f537704194ba8e2 : FAILURE in 17m 56s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/ac20e672511348e187fa4babde10b044 : TIMED_OUT in 2h 34m 07s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"159cacab09719987960aee46c1b28f0db16f4f34","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-07-12 12:13:18.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5b4ff252cc958609a2894e03a48998e438eff809","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-07-15 16:48:40.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"4924c1b026c45d28e354142c6216fa7ae0bf71d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 19:45:31.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d898cb21e394defa9bbd4addb363c32\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1db14e1913fe4d55b816a1452cd57407 : SUCCESS in 11m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ba7cc114ed34c1a8476b700aedbb443 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b62276361ac47579de7c8b0498e443f : SUCCESS in 4m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/59c421630d514d2b95c1ac5d64a5449d : SUCCESS in 4m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ad0ef059beed4d45823f3c8ee52c5a52 : SUCCESS in 10m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2484e03bbb164a3482c68664a519073c : SUCCESS in 11m 07s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/615da7269312459daa4c7c07f2770dfa : SUCCESS in 5m 20s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/0af62a5fd1e94e1ba338cad5cb50e423 : SUCCESS in 8m 11s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/292b9095b9744a6591dde85849b32358 : SUCCESS in 4m 02s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8d1be253a9064b48b00ab5a532d86c2a : SUCCESS in 8m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/db75d623ab8145f7a8e047ce264df763 : SUCCESS in 7m 56s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b442b012a4eb47fbb12229feec51d4ed : SUCCESS in 7m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9a9db5f1195441e78ff4bd4973bcb837 : SUCCESS in 1h 33m 03s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/fc160abc180e496783873e864821b892 : SUCCESS in 1h 40m 29s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/d36ce1b979d24a36a0f17e3f3da482ea : SUCCESS in 54m 08s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/6235493a5ddb4305a92f1a6beec793f3 : SUCCESS in 1h 48m 45s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3f9365bc95bb48c69ef59118bb5deab7 : SUCCESS in 1h 02m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b83bd60882f142f0a68d93a6c63b510f : SUCCESS in 47m 27s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/56a6c34dae674c49bddf2d86fe44f9bb : SUCCESS in 38m 38s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/2f8fc9a9eea341a485f058567b9ec736 : SUCCESS in 1h 01m 45s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2a802e0682624034b613dcccd8a76595 : SUCCESS in 29m 34s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/eb4026c0fdf149bcaac72ad6a11893e9 : SUCCESS in 2h 27m 10s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/886c9dcef3724fb7a68611f61a7339ff : SUCCESS in 2h 53m 45s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/490eafcd00c64955921eabb8df2963fa : FAILURE in 21m 27s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/9144a80506044d428394987caa1ee36e : SUCCESS in 2h 18m 29s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f0ae5db10a91acb15a4915f07bdf474cbd52fd65","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-07-17 14:20:16.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"a893b59205ffda5b4374934ebd7c92df1379980d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-07-31 07:21:25.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"1162effcb01f34bcbe7de6a78ff55f6b926e8ac2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 07:22:02.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"057d92f3304f1a72280a66343b95add9dc1cf320","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 09:08:34.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b553203a50d9432a877317f935bfb9a5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4de2037079c4990a72d6a07bc5b5e85 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/334776fa68ba4f8e876066a7daa3b7aa : SUCCESS in 4m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0cf670671f6647a58f7d32fcdee62118 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5674dbce07514b1d843829385c5fd5ea : SUCCESS in 11m 58s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/bdccbd02815a4195ba502f1b292b0140 : SUCCESS in 3m 58s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/28196441c39a48348af468755394e4fb : SUCCESS in 6m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d5d56b052a8c49ae9301cb2da998b126 : SUCCESS in 1h 04m 44s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/7cbd78daf1ee430ab82301f255e2277d : SUCCESS in 1h 45m 58s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e80b58ed154c456599603ff57fceba3e : SUCCESS in 1h 06m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/37e9d91b40424a85834d0a091892c216 : SUCCESS in 33m 27s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/49825e1e571a425caa4eb0d78beef6c7 : SUCCESS in 42m 02s","accounts_in_message":[],"_revision_number":6},{"id":"28fd1c81c394328d1a4841a386705243ddca30d6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 09:08:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"d6c88747a1febd13e7a46409fa8de11970d8871e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 09:09:45.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/810ea163cfb1447799921a7f7b6fbabc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0074f3128a5542448a8e2892dea115b5 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"f5b73d8e00ede78eea5e450194bb3d2adb5e8403","revisions":{"9dff425ce76295364f28e0ef2ca95657210a6724":{"kind":"REWORK","_number":1,"created":"2024-07-09 21:20:53.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/92/923792/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/92/923792/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/923792/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/923792/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/923792/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/92/923792/1"}}},"commit":{"parents":[{"commit":"1c027c840c57c425519542218898fd2d33fda064","subject":"Do not install qemu in the Octavia devstack plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1c027c840c57c425519542218898fd2d33fda064"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-07-09 21:18:05.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-07-09 21:18:05.000000000","tz":0},"subject":"Stop using DHCP for o-hm0 interface in devstack","message":"Stop using DHCP for o-hm0 interface in devstack\n\nWe know the IP address allocated to the o-hm0 interface by neutron, so\nthere is no need to use DHCP on this interface. Some distrobutions are\nchanging DHCP clients and the current code fails on those new releases.\n\nCloses-Bug: #2072593\nChange-Id: Ie46a5d9620c8ef2b6da04976a742df9c71752e08\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9dff425ce76295364f28e0ef2ca95657210a6724"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9dff425ce76295364f28e0ef2ca95657210a6724"}]},"branch":"refs/heads/master"},"7aac429830846acaeef02bf13ac428b7f3717afe":{"kind":"REWORK","_number":2,"created":"2024-07-10 17:33:20.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/92/923792/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/92/923792/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/923792/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/923792/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/923792/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/92/923792/2"}}},"commit":{"parents":[{"commit":"1c027c840c57c425519542218898fd2d33fda064","subject":"Do not install qemu in the Octavia devstack plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1c027c840c57c425519542218898fd2d33fda064"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-07-09 21:18:05.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-07-10 17:33:09.000000000","tz":0},"subject":"Stop using DHCP for o-hm0 interface in devstack","message":"Stop using DHCP for o-hm0 interface in devstack\n\nWe know the IP address allocated to the o-hm0 interface by neutron, so\nthere is no need to use DHCP on this interface. Some distrobutions are\nchanging DHCP clients and the current code fails on those new releases.\n\nCloses-Bug: #2072593\nChange-Id: Ie46a5d9620c8ef2b6da04976a742df9c71752e08\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7aac429830846acaeef02bf13ac428b7f3717afe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7aac429830846acaeef02bf13ac428b7f3717afe"}]},"branch":"refs/heads/master"},"952944c1f46bfe1094a1040d179aec6f552e9973":{"kind":"REWORK","_number":3,"created":"2024-07-10 18:46:20.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/92/923792/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/92/923792/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/923792/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/923792/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/923792/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/92/923792/3"}}},"commit":{"parents":[{"commit":"1c027c840c57c425519542218898fd2d33fda064","subject":"Do not install qemu in the Octavia devstack plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1c027c840c57c425519542218898fd2d33fda064"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-07-09 21:18:05.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-07-10 18:46:03.000000000","tz":0},"subject":"Stop using DHCP for o-hm0 interface in devstack","message":"Stop using DHCP for o-hm0 interface in devstack\n\nWe know the IP address allocated to the o-hm0 interface by neutron, so\nthere is no need to use DHCP on this interface. Some distrobutions are\nchanging DHCP clients and the current code fails on those new releases.\n\nCloses-Bug: #2072593\nChange-Id: Ie46a5d9620c8ef2b6da04976a742df9c71752e08\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/952944c1f46bfe1094a1040d179aec6f552e9973"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/952944c1f46bfe1094a1040d179aec6f552e9973"}]},"branch":"refs/heads/master"},"2224d6b2dc5fdafa30fb12508d42dbb6a9381966":{"kind":"REWORK","_number":4,"created":"2024-07-11 18:23:31.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/92/923792/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/92/923792/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/923792/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/923792/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/923792/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/92/923792/4"}}},"commit":{"parents":[{"commit":"1c027c840c57c425519542218898fd2d33fda064","subject":"Do not install qemu in the Octavia devstack plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1c027c840c57c425519542218898fd2d33fda064"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-07-09 21:18:05.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-07-11 18:23:08.000000000","tz":0},"subject":"Stop using DHCP for o-hm0 interface in devstack","message":"Stop using DHCP for o-hm0 interface in devstack\n\nWe know the IP address allocated to the o-hm0 interface by neutron, so\nthere is no need to use DHCP on this interface. Some distributions are\nchanging DHCP clients and the current code fails on those new releases.\n\nCloses-Bug: #2072593\nChange-Id: Ie46a5d9620c8ef2b6da04976a742df9c71752e08\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2224d6b2dc5fdafa30fb12508d42dbb6a9381966"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2224d6b2dc5fdafa30fb12508d42dbb6a9381966"}]},"branch":"refs/heads/master"},"cb118a790671bd41eae79dd3a2764371445834bc":{"kind":"REWORK","_number":5,"created":"2024-07-11 20:04:49.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/92/923792/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/92/923792/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/923792/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/923792/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/923792/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/92/923792/5"}}},"commit":{"parents":[{"commit":"1c027c840c57c425519542218898fd2d33fda064","subject":"Do not install qemu in the Octavia devstack plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1c027c840c57c425519542218898fd2d33fda064"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-07-09 21:18:05.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-07-11 20:04:34.000000000","tz":0},"subject":"Stop using DHCP for o-hm0 interface in devstack","message":"Stop using DHCP for o-hm0 interface in devstack\n\nWe know the IP address allocated to the o-hm0 interface by neutron, so\nthere is no need to use DHCP on this interface. Some distributions are\nchanging DHCP clients and the current code fails on those new releases.\n\nCloses-Bug: #2072593\nChange-Id: Ie46a5d9620c8ef2b6da04976a742df9c71752e08\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cb118a790671bd41eae79dd3a2764371445834bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cb118a790671bd41eae79dd3a2764371445834bc"}]},"branch":"refs/heads/master"},"f5b73d8e00ede78eea5e450194bb3d2adb5e8403":{"kind":"REWORK","_number":6,"created":"2024-07-15 16:48:40.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/92/923792/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/92/923792/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/923792/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/923792/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/923792/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/92/923792/6"}}},"commit":{"parents":[{"commit":"1c027c840c57c425519542218898fd2d33fda064","subject":"Do not install qemu in the Octavia devstack plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1c027c840c57c425519542218898fd2d33fda064"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-07-09 21:18:05.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-07-15 16:48:23.000000000","tz":0},"subject":"Stop using DHCP for o-hm0 interface in devstack","message":"Stop using DHCP for o-hm0 interface in devstack\n\nWe know the IP address allocated to the o-hm0 interface by neutron, so\nthere is no need to use DHCP on this interface. Some distributions are\nchanging DHCP clients and the current code fails on those new releases.\n\nCloses-Bug: #2072593\nChange-Id: Ie46a5d9620c8ef2b6da04976a742df9c71752e08\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f5b73d8e00ede78eea5e450194bb3d2adb5e8403"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f5b73d8e00ede78eea5e450194bb3d2adb5e8403"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
