)]}'
{"id":"openstack%2Foctavia~925327","triplet_id":"openstack%2Foctavia~master~Ie41c943d9bd730816c1b471a00e682aab8976ffa","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"12404":{"account":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"last_update":"2024-08-06 15:02:20.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ie41c943d9bd730816c1b471a00e682aab8976ffa","subject":"Add OVN DB sync to cmd","status":"ABANDONED","created":"2024-07-31 09:16:05.000000000","updated":"2024-08-06 15:02:20.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"578f7485c06671fb99babbccac59c6c5add47989","_number":925327,"virtual_id_number":925327,"owner":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-07-31 12:21:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":-1,"date":"2024-07-31 23:49:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"all":[{"value":-1,"date":"2024-07-31 09:16:38.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-31 12:21:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-31 23:49:42.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"079c1ced1c1bfb68b8d7ab0d3c9db0b3be4e3ba4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2024-07-31 09:16:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"34197ce75296223756a2a96223a969bd905fddf2","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2024-07-31 09:16:38.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"d70572464f9727169a700cd5b4598b6c122ce5ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 12:21:16.000000000","message":"Patch Set 1: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e6ae4c684050463b8873508d6bff7247\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e19832f62a1b45eabfdca415020709e3 : SUCCESS in 1m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e31291b2ad6d4c0e8d5dd85863aa2f4d : SUCCESS in 12m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bdeb0cf39e347c785550bcb9a01a262 : FAILURE in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ec463012f5b4c8f862c3147e74f2a7c : SUCCESS in 4m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4b5cc4e5e4d64e73ad961be7e4ae0eb1 : SUCCESS in 4m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/26053dd2bd3e4968b6ab8967870c0a9d : SUCCESS in 8m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e42d5acfd0a44cbc928a3b2890581f62 : SUCCESS in 10m 41s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/275a9030b3f5474c9290e1b07a628cf5 : SUCCESS in 5m 00s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/0d90012ccbb046279af3f031e1e706de : SUCCESS in 7m 48s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a40d4194da2d41c6b0916f58e5dd8e7b : SUCCESS in 4m 20s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8fb09d0976ae427a9bade8140062fffd : SUCCESS in 8m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/161ae38669e84140920f9cbfe2f773ba : SUCCESS in 6m 55s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a851308f373e4433ab5834b9f7e9d542 : SUCCESS in 6m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3cc068a61a5c4196b665278404565619 : SUCCESS in 1h 42m 46s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/18c2fec82eb64d78bb3f03cb6bcc1a65 : SUCCESS in 2h 14m 22s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/3e387fb374674db2a50ed7d6cdb68b6d : SUCCESS in 1h 16m 27s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a20071969a53413fae7a03219a4c5293 : SUCCESS in 1h 51m 04s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b19c1fda372d4ec09e6375b1c1f229a5 : SUCCESS in 1h 25m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0f8fa6367abd45a0b40fd5476a9b8da3 : SUCCESS in 34m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fb3a6828444d439ca0e07203f1782bf0 : SUCCESS in 1h 06m 27s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/2a5ca5f6947344a6849563e1c747d9e1 : FAILURE in 14m 45s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1256ca4118fd424f952a9f6b302dced6 : SUCCESS in 48m 29s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ed2c6f6b4293487eb32464fcd01f72a1 : TIMED_OUT in 2h 33m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c3e743bd10a84fe6b6c7c8aa2fdcad6e : SUCCESS in 3h 00m 56s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a4c7fa1d6eb44101b8968cb4eb8880f5 : FAILURE in 32m 44s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/2bff629a2dbc422da488cf0403ef726c : FAILURE in 12m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"94593baf75889cc7bed369b69c3f091f67df88de","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-07-31 23:49:42.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"578f7485c06671fb99babbccac59c6c5add47989","tag":"autogenerated:gerrit:abandon","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2024-08-06 15:02:20.000000000","message":"Abandoned\n\nturn to https://review.opendev.org/c/openstack/ovn-octavia-provider/+/925747","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ff79e16a56c82007e0f33c1597c7ba565a9ece5c","revisions":{"ff79e16a56c82007e0f33c1597c7ba565a9ece5c":{"kind":"REWORK","_number":1,"created":"2024-07-31 09:16:05.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/27/925327/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/27/925327/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/925327/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/925327/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/27/925327/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/27/925327/1"}}},"commit":{"parents":[{"commit":"1c027c840c57c425519542218898fd2d33fda064","subject":"Do not install qemu in the Octavia devstack plugin","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1c027c840c57c425519542218898fd2d33fda064"}]}],"author":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2024-07-31 09:13:41.000000000","tz":480},"committer":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2024-07-31 09:13:41.000000000","tz":480},"subject":"Add OVN DB sync to cmd","message":"Add OVN DB sync to cmd\n\nThis allows Octavia to sync OVN SB/NB DB with Octavia DBs.\n\nTODO: Add Unit test\nChange-Id: Ie41c943d9bd730816c1b471a00e682aab8976ffa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ff79e16a56c82007e0f33c1597c7ba565a9ece5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ff79e16a56c82007e0f33c1597c7ba565a9ece5c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
