)]}'
{"id":"openstack%2Foctavia~925532","triplet_id":"openstack%2Foctavia~master~Ifbec670e1afa86725d2659a185f6c185abbefb16","project":"openstack/octavia","branch":"master","topic":"secure-rbac","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-08-22 11:21:56.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-08-22 11:21:56.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-08-22 09:25:03.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2024-08-22 11:21:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifbec670e1afa86725d2659a185f6c185abbefb16","subject":"Fix tests for oslo.policy new defaults enable by default","status":"MERGED","created":"2024-08-01 19:55:10.000000000","updated":"2024-08-22 11:23:13.000000000","submitted":"2024-08-22 11:21:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"925532-secure-rbac","meta_rev_id":"9b556984e032ea6b64df22e1fcddcc941f62befa","_number":925532,"virtual_id_number":925532,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-22 11:21:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2024-08-14 16:14:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2024-08-22 09:25:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2024-08-22 09:25:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-01 23:05:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-02 20:19:37.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-08-05 06:45:23.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2024-08-21 18:08:30.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"2962b39d0fb816fdd86f2d8e51ec7f42d249b673","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-01 19:55:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"79db2ed3cd65895a929a12f524506354aeb84601","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 23:05:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b75392b5e2ea4d589f07a54323613dda\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/5ac28c1081a74770a1f1a405a65a6627 : SUCCESS in 1m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9108d57d2b7541f987ea6213245d1bba : SUCCESS in 10m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/426218c7d9df4e5aac28bab256922f3c : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2a8cb8842cc4781936e087283768e16 : SUCCESS in 4m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/87a6e2e144f2415fa74ee5e6f3e2834f : SUCCESS in 5m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6e90d2763caf42f182e65b37de174871 : SUCCESS in 10m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f8f412235aa4952ae4f71b7374e6c3b : SUCCESS in 11m 18s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/0b46ba050b22440192c706cfa750d927 : SUCCESS in 4m 57s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/3c59cb584950492c861ce55fb1756faf : SUCCESS in 7m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/77afab70702c4a32ab0edc7aece68d34 : SUCCESS in 5m 24s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5a691f6993cb4ded87c13d71b42b69ca : SUCCESS in 7m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/76e566ce58f849fa965588987a0637be : SUCCESS in 7m 00s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c3d979d25b444512b32ff7e5ac24ac89 : SUCCESS in 9m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/938fbc4b79ec4a298b85a3bd416c8028 : SUCCESS in 1h 23m 07s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/40ac0387acce4a1a8d4a733ab3d0ca65 : SUCCESS in 2h 22m 20s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/83a6418d0fe74ee98412fe240dd29a85 : SUCCESS in 1h 23m 41s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/cdf256580f1d445aabe0bfb87acb0ffe : SUCCESS in 2h 26m 13s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f5c78a463f694eb3bc8b9c1c93ff492c : SUCCESS in 1h 23m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a83e28b5973f423f82d7a2cadd325d70 : SUCCESS in 37m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6a849ef645784706a3fb13b12d7ad4f2 : FAILURE in 37m 02s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ae23b66f5d574cdfac7a6f5d97bd977f : FAILURE in 10m 44s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e28684b73ae249478be47e7eafe3d963 : SUCCESS in 35m 02s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d749b6b86f6c4e30a69d7799172fb1b0 : SUCCESS in 2h 26m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9090573a638b42bc8f6b6cff23c161bc : TIMED_OUT in 3h 02m 48s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5a2b67857e664a8686f5ebf9732d178f : SUCCESS in 2h 24m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/a811f8875f5348b286f2269e6cd3ef8f : FAILURE in 8m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6cb752ea251681a2b9dce1f1a85a41925bf916b3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-01 23:48:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"63208d29294b2905d337b08f8d61ab1785f986fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-02 02:51:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe238f1ff1ab4137bfff04d55b112921\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/15d22765468c4611b7c7ee65dfd1a3ad : SUCCESS in 1m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d5f7b7d65c204015ae48851a38f89a6c : SUCCESS in 9m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0035438871484717a547407f4bb858be : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/701c7bdd5f004f24bb01fa29e08b970b : SUCCESS in 4m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0e4ff50cdb7449fd9fbc76941a256a5f : SUCCESS in 3m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0f537e2b6dd14b3eadfdc8ea9618b7ca : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d17ad8477d504c4cb3a6242e21f47e7d : SUCCESS in 7m 57s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/ea186d201e4749a59f87b0e328790628 : SUCCESS in 4m 03s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/ba3d13259d6a406c9ad4eb1e18e5b646 : SUCCESS in 5m 11s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ec770ae28d6a454d94967fab53fdea5e : SUCCESS in 3m 15s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/beaa93c47328499dbe8b8648dac803a2 : SUCCESS in 6m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6ed9e56a67914b33a6c6177ac486d84d : SUCCESS in 7m 12s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4f91a11657c942ef9ca3f5ee97a2df8b : SUCCESS in 6m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/72c13bbb717e4ad3869f0b639d5463da : SUCCESS in 1h 33m 08s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/4e81b0d4436a46748ed31c5ef0b783d3 : SUCCESS in 1h 48m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/d798284401884531b0dac17babfdb838 : SUCCESS in 1h 04m 36s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/4190896b587b4c7b927480af8dd868dd : SUCCESS in 1h 50m 48s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/7a10ad72a67b456d9843a957af8c9589 : SUCCESS in 57m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/616e779c0a014c578e45dd7ab55af9a5 : SUCCESS in 36m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/993544e75f314d459f91c0a9d0436e8b : SUCCESS in 42m 05s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c5331cf395f74828b9941412a0f2a60c : FAILURE in 10m 23s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c6d0f430b730464dbf72c460f99e7708 : SUCCESS in 32m 47s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/d18809f9f53b4a3b86a164422fd57051 : TIMED_OUT in 2h 33m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d2e899e59b9f4c48b74eb7cbc1228048 : SUCCESS in 3h 02m 10s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e38e47f13160493b9ccd6495ad08ab2c : TIMED_OUT in 2h 33m 10s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/d8423f6857af4491a39bca8abce919e2 : FAILURE in 10m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"88dcd63c61e1ee8ae5cf2004ce3b79fadd449ce2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-02 18:12:20.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"63c552d52bbfac56a58801ea2a05d3654dee10af","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-08-02 20:19:37.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"815e326c4aa64e01a3518efe4928d701f5531234","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-02 21:19:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/77c43a56676e44a6bb8df88f67b5d8c0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/735fd215b0dd4e4ebc96f37ab29b52b7 : FAILURE in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f032bb1ded90456bb68df6137735a893 : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c5e218df3a04a2aa5242ec3fc65b8ad : FAILURE in 5m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6335998bee09479d98b16d0d1db07655 : FAILURE in 4m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/32674a33a7144147bf9c527f9897c6b4 : FAILURE in 7m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3354d81ec8047ed97979850a26eb9ce : SUCCESS in 11m 06s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/5381dfc794e94beea8b7fea23a6984c2 : FAILURE in 4m 41s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/1f67751e3b494b3b91fefe7ab72ed81c : SUCCESS in 8m 04s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b6efe7ac37854ae28e4cd05371ab2c86 : SUCCESS in 4m 36s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fef8b02f1a6548feb2ce61b88f8d2e09 : SUCCESS in 8m 07s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/25d912a61c83476b89097467feef4184 : SUCCESS in 6m 52s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/ab162821e57043479c19619a8c26aa85 : SUCCESS in 7m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9f57d670e87743e696a1a7461e31b462 : SUCCESS in 1h 13m 14s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/5a4902e0cf2f4bd195d0dbf43f540d72 : SUCCESS in 2h 22m 43s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/9f7643cb8c8c45dc9917a2fb550d8c88 : SUCCESS in 1h 02m 56s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/cb77464261ea49a881853785d50ba0bb : SUCCESS in 2h 19m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b56c246921f0489aa85b0895e81232d1 : SUCCESS in 1h 23m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2415c20940a44f08b0100fad1616deec : SUCCESS in 40m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/beb429109f814b569c3afb770fd81371 : SUCCESS in 44m 40s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/824044e0aef4433d95dabd24e43e92d7 : FAILURE in 14m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b1396742099d4cad94bc8605c00774cd : SUCCESS in 40m 08s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b09e4f2236aa43cb937a227210aa8815 : TIMED_OUT in 2h 32m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9dee81e2f4fd457898ed8e0e3bbcb6e2 : TIMED_OUT in 3h 04m 07s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c10b2cd9779a4af48dd2ebf29411432f : SUCCESS in 2h 22m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/12e9abf5a8c6484e9376345e6903a8b1 : FAILURE in 12m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"497596b073598d8eefb005e7678f212b8cb3c7ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-03 00:04:42.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"3b6ef450a33f8d29bc7965f3fcdd6406ca6776fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-03 03:09:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af1cef1b428641da85988a121bd0230e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1219b911e354f388912bd9e9ab76c42 : SUCCESS in 7m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c2abf1cc93c4ddc8dfb34463cf166f2 : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c982dc0ca5d4ad98cc976650cb5c8c1 : SUCCESS in 3m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c83d8423339242f98036cbde88d6a6eb : SUCCESS in 4m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cf310f1ddb94459e9f4e047f1247afa3 : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44dab99444ef4ed5aa626e562b7f04d7 : SUCCESS in 8m 13s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/b4c09b4060de4c20a680340c9507168c : SUCCESS in 3m 54s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/2412fb591e994fa88beb6d7391022238 : SUCCESS in 7m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/219f5a1e4668429db74c8017efdeab64 : SUCCESS in 4m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/17a7ec53a2014fc38c402912d162d73d : SUCCESS in 7m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ca6ac57d3e454ed2aaf4ed8dbadbd477 : SUCCESS in 7m 15s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3c9267f0c8d344838d33ab99b816b8f6 : SUCCESS in 9m 05s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4a45c01f76534663be9bf4ad3aba2900 : SUCCESS in 1h 28m 51s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/38f923f2f3da4587b4f3d2d8618e07ef : SUCCESS in 1h 47m 28s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/bc8359e3784c45a5a5e62f55be3d4217 : SUCCESS in 1h 05m 57s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/8e317597d6b745efbf1bb510f23e8f49 : SUCCESS in 1h 54m 13s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f8f57c78d0904069be9a9cc5287ae5ce : SUCCESS in 1h 07m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8bf7106258684ddab6c295adf788b5de : SUCCESS in 38m 22s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/db6e9598998f4d1598b78ded2c77418b : SUCCESS in 36m 32s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/1f1d16cc3b9f42a79268980490aa257e : FAILURE in 13m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/68825e0946154102a33bdf5006f69372 : SUCCESS in 30m 02s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2dd9c9b706144cd5aae948734bb2c537 : TIMED_OUT in 2h 32m 47s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ead16b0de09640ccbe9d2134dbcfe1d5 : TIMED_OUT in 3h 03m 16s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c6a785e4727443d39fc3907b2e390ee3 : SUCCESS in 2h 31m 06s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/b7794f8dd49a4f93bbb61949cc5b440d : FAILURE in 10m 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0b8a20f39ce05d4cdd7a92feae5e792e1e6ed0a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-08-07 21:29:53.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"07b9faabf63311d427f63a94c213f7806990eb40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-08 00:43:25.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5dd1fe34250e40bc8a68d5758b0644e1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8dc883e5b4964460b61a4548f5b28429 : FAILURE in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92ff99bc1ca94ccfbd04d59c31ae6723 : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/73c4c1c76d0a412797f6d5be81981b39 : FAILURE in 4m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/48c6e00bd5c24eb99e5d10d088575177 : FAILURE in 4m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7b731057d6dc46f1a42175db1d7d38d4 : FAILURE in 7m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9dec6c84585432bb504e2685c8278e6 : SUCCESS in 10m 38s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/f5a9456d182a4a64835399393fa576d2 : FAILURE in 4m 13s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/51b727d114a944ddbc0ef6fd74c4975a : SUCCESS in 7m 07s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6f00704534c646ccb786de47544b1b75 : SUCCESS in 3m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/50dafe299df3467b9bcda757e053668d : SUCCESS in 5m 56s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9499f8376ae04352984d5c68a89724cd : SUCCESS in 5m 13s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8bd00405ec8049d4a0617ef31ed529e1 : SUCCESS in 5m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5bf0e611217f43c7abe283bc936f02cf : SUCCESS in 1h 00m 39s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/ef62b2a364114c03a427a415cb7f1a9a : SUCCESS in 2h 06m 27s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/6f4d861463fe4d54a0996fb5fc87131f : SUCCESS in 1h 24m 15s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/35fa0526263545a6acf5b6b13e0c5f8a : SUCCESS in 2h 04m 06s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/176f8d4bdf9d409cabc312b047730078 : SUCCESS in 1h 10m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f0d1c7a39ed74163becb275155d886d1 : SUCCESS in 42m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0117ac3b15094dfca3e8e51fa63238fe : SUCCESS in 48m 09s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/7ef224020fca4ac7a2b694f1c5c6bdc0 : SUCCESS in 58m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f2285c2b86ab4e4d9fc6d4aa25abf0c0 : SUCCESS in 40m 27s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7c282b5850864ff39665e76a8063fd3f : TIMED_OUT in 2h 33m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1a06ac567f9d4bb6b0dd105a4b5f0d34 : TIMED_OUT in 3h 02m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f9d419da1536489db0ad6b89761272fb : SUCCESS in 2h 32m 29s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/7225274913cc4469ae56c30c33246f48 : TIMED_OUT in 2h 33m 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4866e1ffc38fc1e9374847051f27b1afdbf03ca1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-08-08 03:49:49.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"799d1b7eca8aa40e0ef160323ff4482072264bff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-08 06:47:35.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a2f0923e80a4e8bbc6718c3982e7659\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c4c6ef096e1e49b8b9be00a8f680dec9 : SUCCESS in 10m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3a7c8f48a504654ab510fb06c2856d4 : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b25a43ed57bc40d3a5f7eacd8b83669f : SUCCESS in 5m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/644953a8526144b28d6bd6d49952a4a7 : SUCCESS in 4m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2b02ffdb0dad4c5fa13cfa2a9d1dcebf : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d68d1240e04487e8c27902c2e61140f : SUCCESS in 9m 09s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/9e9f0fbbb5b047f989a76232c8859a08 : SUCCESS in 4m 54s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/7bbb46cbe9904f9b8800528ddd27b231 : SUCCESS in 6m 31s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/48123b8eef764c898862b637dc8e07d7 : SUCCESS in 4m 20s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d01100defa0148969de483178f8548fd : SUCCESS in 7m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4b4a57c9467246ebb7fdad4307798c1f : SUCCESS in 7m 20s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/230e0c0dd7f94583993cf0c116788730 : SUCCESS in 7m 15s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/967811afa179413cb6db8cf7021f127f : SUCCESS in 1h 08m 07s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/fd9b548d5cf445949e008afa371b46d5 : SUCCESS in 1h 58m 05s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/7b9082d5acac4a5580c1a77661bf4606 : SUCCESS in 1h 04m 00s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/586580900181463a8d9884c6c3da4786 : SUCCESS in 1h 57m 52s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e68b4088d58848fcbf29496410eb0e9a : SUCCESS in 1h 18m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/204cebbb415d4cbebd37094619457e8f : SUCCESS in 50m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b21a8007ba534dad957cd1c5acf3e60d : SUCCESS in 1h 01m 57s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/f72bd7e2c94a48bf9ee3ff5fbfcac7ef : SUCCESS in 1h 09m 28s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/87c7bd4e8afa4efe9bc07d76456c08dd : SUCCESS in 45m 29s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/289ecd58159444509e018e6007defacd : TIMED_OUT in 2h 34m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2a2c52de757b415ba1bdfb70286fde6f : SUCCESS in 2h 55m 09s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/96e8e4dd6f1f49f49fe1659c3cb0ceb2 : TIMED_OUT in 2h 33m 55s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/59f5692133444f4499ce19c62e36b258 : TIMED_OUT in 2h 34m 32s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"381dc7a5e09e34db9168cfae6be8389b5561d8f8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-08-14 16:14:46.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"12bc6e0e7065924be20fdccd1d187003c9ec1fcd","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-14 17:42:43.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e53501d245cd5fd7817b9bbbb7302dc07da13586","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-08-22 09:25:03.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"5bc4c8849e15d91bbfecad84cff63fda7c669d8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 09:25:47.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"a49fe4ea5fd618b0148596db306e4d4885e9e930","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 11:21:55.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f49a8bb3295457ab27f353120450655\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85c0ecc7c3414b9188eeec953e2b9dfd : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3680052b7ccf4b8d9da6a38391176871 : SUCCESS in 4m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2e006cda6f3e41748604bfaa1309dd71 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d10bd830061a4218868a028818ed38e9 : SUCCESS in 10m 24s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ab4af4c9e7a1418586d63c99bf2088a1 : SUCCESS in 4m 25s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d1a0fc6d187b46188f0ad5e38841aeab : SUCCESS in 7m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9fbf15a6938e4349b35409473522b191 : SUCCESS in 1h 24m 33s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/9b1e9f8f959c453d8c86cd184b71bf90 : SUCCESS in 1h 53m 44s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/66b07c860df04f03a4fac2197c155b24 : SUCCESS in 58m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bedf42b4125a4339b47733aa596bb44b : SUCCESS in 36m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6c03c665c96a4ce79daabe4d3928f14c : SUCCESS in 44m 34s","accounts_in_message":[],"_revision_number":5},{"id":"1df4dff8305d92cfaffa388642d44a6abfa2eb62","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 11:21:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"9b556984e032ea6b64df22e1fcddcc941f62befa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 11:23:13.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fbfca35c3554b7e8936e95dfb24c01f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be8c7dd9f2a64279857e31bb9fa9c49a : SUCCESS in 55s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"240347fc3e066a998b0ad7c472019d7ca6c9b4da","revisions":{"919fbf5eb129f7641c0e2646282f3a4f7384b206":{"kind":"REWORK","_number":1,"created":"2024-08-01 19:55:10.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/32/925532/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/925532/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/925532/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/925532/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/925532/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/925532/1"}}},"commit":{"parents":[{"commit":"c2c5a4ce820abb9251556ca0f31c7c2ac34c66bb","subject":"Merge \"Add specs to resize a load balancer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c2c5a4ce820abb9251556ca0f31c7c2ac34c66bb"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-08-01 19:54:27.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-08-01 19:55:08.000000000","tz":-420},"subject":"Keep new RBAC disable by default","message":"Keep new RBAC disable by default\n\noslo.policy has enabled the new RBAC config options\nenforce_scope and enforce_new_defaults by default[1][2].\n\nUntil octavia is ready with the new RBAC, we need to continue\ndisable them by overriding their default values.\n\nNOTE: there is no change in behaviour, octavia continue with the\nold RBAC as default.\n\nref: https://review.opendev.org/c/openstack/requirements/+/925464\n\n[1] https://review.opendev.org/c/openstack/oslo.policy/+/924283\n[2] https://review.opendev.org/c/openstack/releases/+/925032\n\nChange-Id: Ifbec670e1afa86725d2659a185f6c185abbefb16\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/919fbf5eb129f7641c0e2646282f3a4f7384b206"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/919fbf5eb129f7641c0e2646282f3a4f7384b206"}]},"branch":"refs/heads/master"},"7a0230fdd2b53885312c6b8dd332b4e8a8f78ddc":{"kind":"REWORK","_number":2,"created":"2024-08-02 18:12:20.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/32/925532/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/925532/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/925532/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/925532/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/925532/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/925532/2"}}},"commit":{"parents":[{"commit":"c2c5a4ce820abb9251556ca0f31c7c2ac34c66bb","subject":"Merge \"Add specs to resize a load balancer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c2c5a4ce820abb9251556ca0f31c7c2ac34c66bb"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-08-02 18:10:55.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-08-02 18:10:55.000000000","tz":-420},"subject":"Fix tests for oslo.policy new defaults enable by default","message":"Fix tests for oslo.policy new defaults enable by default\n\noslo.policy has enabled the new RBAC config options\nenforce_scope and enforce_new_defaults by default[1][2].\n\noctavia is ready with the new RBAC but we need to fix\nthe test and some system scope cleanup in base rules.\n\nNeeded-By: https://review.opendev.org/c/openstack/requirements/+/925464\n\n[1] https://review.opendev.org/c/openstack/oslo.policy/+/924283\n[2] https://review.opendev.org/c/openstack/releases/+/925032\n\nChange-Id: Ifbec670e1afa86725d2659a185f6c185abbefb16\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7a0230fdd2b53885312c6b8dd332b4e8a8f78ddc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7a0230fdd2b53885312c6b8dd332b4e8a8f78ddc"}]},"branch":"refs/heads/master"},"b52ae9c4988b15ca30b24371cbf2e343c60d6c7c":{"kind":"REWORK","_number":3,"created":"2024-08-03 00:04:42.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/32/925532/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/925532/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/925532/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/925532/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/925532/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/925532/3"}}},"commit":{"parents":[{"commit":"c2c5a4ce820abb9251556ca0f31c7c2ac34c66bb","subject":"Merge \"Add specs to resize a load balancer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c2c5a4ce820abb9251556ca0f31c7c2ac34c66bb"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-08-02 18:10:55.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-08-03 00:04:31.000000000","tz":-420},"subject":"Fix tests for oslo.policy new defaults enable by default","message":"Fix tests for oslo.policy new defaults enable by default\n\noslo.policy has enabled the new RBAC config options\nenforce_scope and enforce_new_defaults by default[1][2].\n\noctavia is ready with the new RBAC but we need to fix\nthe test and some system scope cleanup in base rules.\n\nNeeded-By: https://review.opendev.org/c/openstack/requirements/+/925464\n\n[1] https://review.opendev.org/c/openstack/oslo.policy/+/924283\n[2] https://review.opendev.org/c/openstack/releases/+/925032\n\nChange-Id: Ifbec670e1afa86725d2659a185f6c185abbefb16\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b52ae9c4988b15ca30b24371cbf2e343c60d6c7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b52ae9c4988b15ca30b24371cbf2e343c60d6c7c"}]},"branch":"refs/heads/master"},"643209e51bdec828f2138889b5ab8a93e7ffc891":{"kind":"REWORK","_number":4,"created":"2024-08-07 21:29:53.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/32/925532/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/925532/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/925532/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/925532/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/925532/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/925532/4"}}},"commit":{"parents":[{"commit":"c2c5a4ce820abb9251556ca0f31c7c2ac34c66bb","subject":"Merge \"Add specs to resize a load balancer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c2c5a4ce820abb9251556ca0f31c7c2ac34c66bb"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-08-02 18:10:55.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-08-07 21:29:51.000000000","tz":0},"subject":"Fix tests for oslo.policy new defaults enable by default","message":"Fix tests for oslo.policy new defaults enable by default\n\noslo.policy has enabled the new RBAC config options\nenforce_scope and enforce_new_defaults by default[1][2].\n\noctavia is ready with the new RBAC but we need to fix\nthe test and some system scope cleanup in base rules.\n\nNeeded-By: https://review.opendev.org/c/openstack/requirements/+/925464\n\n[1] https://review.opendev.org/c/openstack/oslo.policy/+/924283\n[2] https://review.opendev.org/c/openstack/releases/+/925032\n\nChange-Id: Ifbec670e1afa86725d2659a185f6c185abbefb16\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/643209e51bdec828f2138889b5ab8a93e7ffc891"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/643209e51bdec828f2138889b5ab8a93e7ffc891"}]},"branch":"refs/heads/master"},"240347fc3e066a998b0ad7c472019d7ca6c9b4da":{"kind":"REWORK","_number":5,"created":"2024-08-08 03:49:49.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/32/925532/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/32/925532/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/925532/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/925532/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/32/925532/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/32/925532/5"}}},"commit":{"parents":[{"commit":"c2c5a4ce820abb9251556ca0f31c7c2ac34c66bb","subject":"Merge \"Add specs to resize a load balancer\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c2c5a4ce820abb9251556ca0f31c7c2ac34c66bb"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-08-02 18:10:55.000000000","tz":-420},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-08-08 03:48:51.000000000","tz":0},"subject":"Fix tests for oslo.policy new defaults enable by default","message":"Fix tests for oslo.policy new defaults enable by default\n\noslo.policy has enabled the new RBAC config options\nenforce_scope and enforce_new_defaults by default[1][2].\n\noctavia is ready with the new RBAC but we need to fix\nthe test and some system scope cleanup in base rules.\n\nNeeded-By: https://review.opendev.org/c/openstack/requirements/+/925464\n\n[1] https://review.opendev.org/c/openstack/oslo.policy/+/924283\n[2] https://review.opendev.org/c/openstack/releases/+/925032\n\nChange-Id: Ifbec670e1afa86725d2659a185f6c185abbefb16\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/240347fc3e066a998b0ad7c472019d7ca6c9b4da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/240347fc3e066a998b0ad7c472019d7ca6c9b4da"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
