)]}'
{"id":"openstack%2Foctavia~925625","triplet_id":"openstack%2Foctavia~master~If5090f0e4d7672d92a72642d96171b4fd6f490c7","project":"openstack/octavia","branch":"master","topic":"secure-rbac","attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-08-30 19:46:06.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-08-21 16:16:11.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"If5090f0e4d7672d92a72642d96171b4fd6f490c7","subject":"Update oslo.policy version for new defaults","status":"NEW","created":"2024-08-02 18:18:14.000000000","updated":"2024-08-30 19:46:06.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e79c7848cd503261149398af03a48638d76a50e6","_number":925625,"virtual_id_number":925625,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-08-22 05:16:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-02 19:38:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-05 06:45:31.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"ed4841c32aff312272934ec8c1511bb01cc5d2b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-02 18:18:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5d07a63499ce0e249329cbe6d5bdcc9431c7c918","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-02 18:18:57.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"719ab90236000a72d212c15f6b12071ac78cf250","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-02 19:38:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5f1e9b5202d2463482d3a44cdc0df077\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d81480b24a704142a83f9552dd7df4b6 : SUCCESS in 3m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dba60698286447b988f41a194627c76a : FAILURE in 10m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a53d375eae994dd98031d4a118ac103c : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1258654fa8684b139aac9248edf401de : SUCCESS in 4m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3a08c53306c545e69e80c709b39dffbc : SUCCESS in 5m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/274bbcbd59184d629212a2743de37c3a : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b69c79e3fe84b56abc059b47dee28d9 : SUCCESS in 10m 46s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/b1a03e430d714d16973bf420d0688327 : SUCCESS in 4m 54s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/a974a77621fe4a9c95f9dc9292ec6298 : FAILURE in 8m 16s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c2eceb535ce94cd5a455825b57d73160 : SUCCESS in 4m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cee635bc98414d65b0eca37120784662 : FAILURE in 7m 44s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1210f4b371ab43aea4669f339747e34e : FAILURE in 8m 17s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1529b76bdd5e4030a14934744662942e : FAILURE in 7m 46s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cb05483243164684b12e45b0fd52f8df : SUCCESS in 1h 18m 12s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/6cc33ed9118945159b91e5c7072b9687 : FAILURE in 23m 50s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/b53e74a11db1460689622aa0ffefc8a3 : FAILURE in 31m 32s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/4c098d76a98e4a5c801b8aee792989cf : FAILURE in 25m 24s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b3b9abe3fe5b47c993361465680cce0c : FAILURE in 24m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/63ae36d15d894553afb518b7171012f5 : FAILURE in 27m 14s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/8e238dd53d5e4678883fa2619990b38c : FAILURE in 11m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/753176e8fda84ac5b176dd2702ce3aa1 : FAILURE in 16m 57s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f08368778ade4f5aa1589a78b34ca022 : FAILURE in 24m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cd05c62fe36548aebb138a52ca51908f : FAILURE in 29m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d9c694ea32eb494c8d44f3b10aab0c94 : FAILURE in 30m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/782e56ab86df4237a48246e1a2aec211 : FAILURE in 13m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c4d388cdb32534fdee11aa62b21b1ff0fc99868b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-05 01:39:12.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"669c56c1513ec127e349da2e7fafbad723ab8240","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-05 03:28:19.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/42ce55189329463c948751eaf073323c\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/bba8f837a90f4dd2a3bdb11992d839d3 : SUCCESS in 2m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ecad93492d14852900f550b804285ab : FAILURE in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b80570f0a9c4681bd08dc1e383579d6 : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f4ab999440aa45d9b1a70817c14b47ce : SUCCESS in 4m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a319a4fd7c14447bb6c8a70588ca2026 : SUCCESS in 4m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0dfd5db1930c45aa8a605acee9a66e65 : SUCCESS in 9m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c1e0d6ee8794c14b19c62eddaf31acc : SUCCESS in 12m 22s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/491192cd0bdb40a8860b6a93ccfca668 : SUCCESS in 5m 18s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/1fd44732aa764ffb9418c2cdc4df491a : FAILURE in 7m 12s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7a996fd2c9d94072a88115d520ee6a58 : SUCCESS in 4m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5fb3773767bf48a9a32f2db0e640f895 : FAILURE in 7m 02s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a82635c192d74c03aff04f75dda1bdda : FAILURE in 5m 56s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/557ea944e7064bf69ccced822e49f1ba : FAILURE in 8m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fa1f8d52abd444b89f329753e266137d : SUCCESS in 1h 43m 36s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/db8c68a51e4c43d98082b0f76bee88c8 : FAILURE in 23m 00s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/2c2fbe9c89d046c8adb5e094d7169fad : FAILURE in 24m 02s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/940ee6d6d4e641c8822eddae03108b61 : FAILURE in 23m 28s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/7b1c7ee76a6846fda3cf49d2029ca201 : FAILURE in 17m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c850bf47d9e743fb9f4482d4553df4f8 : FAILURE in 18m 53s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/cf4c0feb4210432f9c91d51be9fcb510 : FAILURE in 10m 27s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/614f7a827a264687a5def11a19f416b9 : FAILURE in 17m 59s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f369e325fb1548968d134abee1336d43 : FAILURE in 18m 19s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/57453f543546439f884b05ae4aa4476f : FAILURE in 21m 14s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9ab6254ffd7c4592a743119cdba43bda : FAILURE in 17m 51s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/a489022aa56e476f99e8ded59950a732 : FAILURE in 8m 11s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ea230ccb85ba2fce6d1ce20744e489da5dd53e8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"real_author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-08-08 15:03:39.000000000","message":"Patch Set 4: Patch Set 3 was rebased on behalf of \u003cGERRIT_ACCOUNT_8556\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"_revision_number":4},{"id":"ee4d9c210f3249e994df3019269bd622b53a1c76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-08 16:39:44.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/45d60e96d6e04d1b9af47196ac2e8fa8\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/474eabc4e0c54629b2b47bb48617e26c : SUCCESS in 1m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6251c0de7ed84218adeb9abffca04702 : SUCCESS in 9m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f47387ba844481fac24b1186faf2d57 : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e1da3d8d930141d18ad91fa0da5fe6cc : SUCCESS in 4m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c7a5470bf8384026b853228aed1896a9 : SUCCESS in 4m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8c0e8a0c5d024f9bb2bdb6dd746d4e05 : SUCCESS in 11m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f62984d5f6e247688c9870f124e288fd : SUCCESS in 9m 08s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/c44e8412c501411d83fc11df7d47a07f : SUCCESS in 6m 33s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/103a1f7ab1db4d92bbda4776c1f0dea5 : SUCCESS in 6m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/49bf7afa9ae84361a6ae9c5507b42471 : SUCCESS in 6m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/740a46b0eace47198b2ee31ce50e9a41 : SUCCESS in 7m 24s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fbbe3f306ba948a0afebfed8ccbca914 : SUCCESS in 8m 34s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e6b73ff7b968493ba3663e0ab10f98d3 : SUCCESS in 8m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/02c9a3e12ec744bd93cc408fee7fcc04 : SUCCESS in 1h 34m 00s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/cf6319a1005341758b66eb5c43a5f9de : FAILURE in 19m 52s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/876267a2b5d74caeadd178e9f2bbb0bf : FAILURE in 24m 40s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/895b7b86cc614968b65d9c650517f28a : FAILURE in 24m 48s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/064f9ef065ae4bc1a19cafc08b4f9371 : FAILURE in 25m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/34f192ca9aa94e9f8d2fb4b9d57fc674 : FAILURE in 26m 14s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/fa547fbaae594e47a7c24c4aef935c90 : FAILURE in 37m 44s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5dfe8ada3c264c5fbdb52f0ac04b5bfa : FAILURE in 24m 53s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/fbdf1e99f7c240eba19e100ff9fd2940 : FAILURE in 25m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/05696e351cc941b781afe3983b8aa3ba : FAILURE in 25m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2246abf538fd42a6a746827eb7401357 : FAILURE in 20m 02s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/2ec69e3f2b0543a394b6d701d7f91b04 : FAILURE in 36m 33s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"55e33387086e7e092b058cc275faac15111e1e31","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-08-21 16:16:11.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1cf47396c481bfcc72e21a97c1717cff5c99658b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 17:55:58.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fbe79fd86ea342ca902c04b8b481701f\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3963a0ed985445bdaa54e8b8025c2525 : SUCCESS in 3m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/753a8fa1c55144c1bb88ad2f04d475b2 : SUCCESS in 12m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b9ef50613c246289a699fa5e636e873 : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b428aa23c22b42e9bdc530dd9fbec074 : SUCCESS in 4m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b9a4072ea417476e99891e8e31a34e7d : SUCCESS in 5m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/384e9365b5b94c1e91bcfd33ac151c5c : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1597101c3664d35b5c2cf56e87e1c74 : SUCCESS in 12m 45s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/9cd7696e8fb54efaa2dec911c0e5d8ee : SUCCESS in 5m 07s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/c5c7eea156e140f1851b495d5fab164c : FAILURE in 7m 05s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/66459cf97a334be1b668646cdd6b07b1 : SUCCESS in 4m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/217245af966042c186b60b9b30ee0809 : SUCCESS in 8m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bd0c51009ed84adaaa6410bf7da04a83 : SUCCESS in 8m 14s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4b611b996f18482e839ec24295adafa4 : SUCCESS in 7m 15s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4b93b0ef14334026b45fa3d7589ffd33 : SUCCESS in 1h 38m 54s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/5afbb331d00741bb99dcdde79bcabf83 : FAILURE in 23m 47s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ed28203d776746d2b7522b5b136f132b : FAILURE in 26m 56s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/8e01bfe2a0c54754a0e7d69707b908df : FAILURE in 23m 57s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/90691a809dbd4de8b079664cc6697d2d : FAILURE in 27m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8ab062fcc1ff4b05ab17ee2ecb9c0e8b : FAILURE in 26m 24s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/7ddc7f52469a4c2cab53538f98e7d676 : FAILURE in 42m 06s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4930ca4a2117456e9c98ace55ac7c32e : FAILURE in 24m 04s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7f17c26bd3be4afeb56334e7968386d8 : FAILURE in 25m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4d0d66af877b40eca9ded73fc3530bf9 : FAILURE in 33m 37s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7bec93898d4e46e094fed0cb3c3d9622 : FAILURE in 16m 16s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/1855fc33d1eb4cbfac4938eede422665 : FAILURE in 31m 31s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"70c3bca91fb1efb646827bdf787bffa8ff749d5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-22 03:55:00.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"948f927d00ef1d54ec7e5ba612c06afbd668dfbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 05:16:13.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/df63a59a00ed47e3b88a95b3fd0f135e\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/259fb6ac8dcc4d428a031c8b46f35dc2 : SUCCESS in 2m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c1baacaccdf4ba58e40b916993d496f : SUCCESS in 11m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36c6b9f761c54e34b3e764a44a82476a : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2e5e3cb11a94edd854b5ac283bfc8f8 : SUCCESS in 4m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ba1595711c4b4b7eb05ba2eb5f8902e2 : SUCCESS in 4m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/55a7d1c2fd01449091ab2c2a8c308869 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf7fd9b933c044fe9ef6701bcb0c4b0d : SUCCESS in 11m 11s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/27a1a4fbf2f346b2a1246bf1c22e70f6 : SUCCESS in 4m 53s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/9944466f44f541c789dcc161ac8e6dc9 : FAILURE in 7m 59s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6f967056838043e7aea4eb301ddca6e0 : SUCCESS in 4m 39s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a95d88f926ef4f5082f2d593c107d5de : SUCCESS in 8m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/36d3f769b56842c0841ba60e887d9198 : SUCCESS in 7m 28s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b3c29bfcb7eb4d67b0011c1554154c39 : SUCCESS in 7m 33s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/755f9edc2d314e48b46be137dc747f3c : FAILURE in 1h 15m 52s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/5d48a8f4dc42470ebd14de67b7836cc1 : FAILURE in 17m 02s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/8ec6a1d236d24de38fd7395dc7471d59 : FAILURE in 19m 30s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/31a36e893a89411fa0d86f053f27f010 : FAILURE in 17m 05s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ae82a2058fb14a83bb709495a35842c7 : FAILURE in 16m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/29e00f037fe34d069e78c21537443ce8 : FAILURE in 18m 38s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/62a71566822444ab8f370ec3611fdff8 : FAILURE in 39m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cc3b4feded9942efab4ae9bbee7d4436 : FAILURE in 16m 38s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1ea04862d6b54ecf8c53ac8cdbaf1deb : FAILURE in 16m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a9f7d924b4514cb199eef6fe3605b138 : FAILURE in 39m 42s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ad4e1412a2bf4679930e6f75e3f99c2f : FAILURE in 16m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/b59b9320d13942bf9a1f7eb85c25db0f : FAILURE in 29m 30s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5388143c5266ec68451a007c01c325efc3869d6f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-22 20:08:51.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6a726faea47824fa6d34b8043e0bbbe4468b0a03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 21:39:02.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1279e78c9fb943559c64c2b488e261a3\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a38ad097e4b649689161e180d913329e : SUCCESS in 2m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c4e241851a942ca956f9286df196deb : SUCCESS in 10m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb12425759c74e1cae063378c2a4720e : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc08934e46114b04b6933c03a18e474b : SUCCESS in 6m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/facf50c70f7f429a9d3427d904cbdd87 : SUCCESS in 4m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8572a78dddaf4cc8bb5a4215860fdd3e : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/850d23114988436296bb1ef86f030240 : SUCCESS in 11m 38s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/265b6246394f4f17ae2c0200cf7248b1 : SUCCESS in 5m 12s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/52d5f6cc693f4031809ef6b66467e5e3 : SUCCESS in 7m 06s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b334b308ccd24f2bbaf462ab32a96134 : SUCCESS in 4m 36s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a1f1ce31f9e5483998b2e1eeb41f2b9d : SUCCESS in 8m 14s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4dce95e38e3746b7934004c3a4a5e02f : SUCCESS in 6m 55s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a0152737e8ee4030b3a23fb173cd8403 : SUCCESS in 6m 25s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/949399a9b3a14b2da2cc1299fd6b0231 : SUCCESS in 1h 29m 18s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/0cbd93061b7f404cb52e1b90b55f35d5 : FAILURE in 17m 02s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/dfc8a090b17e4242a497b705458c2705 : FAILURE in 17m 04s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/d7c1b578e9984d8c8e99b99528e80e03 : FAILURE in 17m 11s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f12f41ef92ae408b82db00560e52085a : FAILURE in 19m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/699108c83cb8484283d40feca757a364 : FAILURE in 18m 06s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/0a17117983a5440785a566f788c70b63 : FAILURE in 35m 52s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/07cbcfc5798249fbbf487b8627371d35 : FAILURE in 19m 47s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a86a31f7ee504d93982a44b6db30fa16 : FAILURE in 20m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6bdf4d6155e24b188800ffe7ab1f59ce : FAILURE in 27m 03s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8fa1e92e92fb42ea86ffe89beca38a8b : FAILURE in 29m 53s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/991231359fc341e8a83391764d96d370 : FAILURE in 33m 57s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6360b5e0bdca9faac660bde56bb17d9667f54b10","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-30 18:21:33.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e79c7848cd503261149398af03a48638d76a50e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 19:46:06.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dd13774f03474345bcd5fe24eef63d14\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/23a9f624260b4d108769044bbff42a7d : SUCCESS in 2m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6596a323c9341208df75cc893df2086 : SUCCESS in 10m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a1a5ec6ace74db28d497db50223b5b7 : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/88337e73d1dd4f56a4b368ea981d114a : SUCCESS in 4m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/10573e3d54144cbea086a7b53cffadbf : SUCCESS in 4m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cf7211232bb4485496ad7c2df6631acf : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a78492bebafd4f55bc4339d7372f0fc7 : SUCCESS in 10m 29s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/e19272a47b2549fdaa140c037689bf03 : SUCCESS in 5m 01s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/5ebc9b041a1045758737f75011416624 : SUCCESS in 7m 49s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/66cc27ee515a48cfbf2db94b08809a90 : SUCCESS in 4m 23s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8f1ca15cd35f4e468b9d4bcf2e8d8005 : SUCCESS in 7m 04s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/47e47ddc268548f086e6d97a8afcd24b : SUCCESS in 7m 23s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/dfd66d0036f24ba98aef31c13976409d : SUCCESS in 10m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e526c470e60a4cf1aa89d88be50f438d : SUCCESS in 1h 18m 26s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/6e4ef90ed0e14dc881b2197b3a0a1c86 : FAILURE in 23m 59s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/fa5d45d5c6af4bcbae3785c6b7f4cfff : FAILURE in 30m 09s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a8f4b9f15b5e41e6a86e0e9b96132957 : FAILURE in 16m 35s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/66d47ab2adee4267bee527018ff27994 : FAILURE in 23m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7ac1fd853aee4b60984cfa20e9b3accd : FAILURE in 16m 59s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/dd5a671e839b4c9a9bcb53e60edf8856 : NODE_FAILURE Node(set) request 300-0025070382 failed in 0s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ff9f3dedf3f6427887d23b7d2513c5f0 : FAILURE in 21m 43s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6346d8dad9e244178b16df7d9102da46 : FAILURE in 20m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6ca08d7a3360458c90c13d456bfdbbae : FAILURE in 23m 57s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/33245c2e1fa14ac4b01a7f626eadd4e9 : FAILURE in 20m 09s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/e2174b3403364242bb36ef45255f4a23 : NODE_FAILURE Node(set) request 300-0025070387 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"62d7e7db3681cb360351398a348357e9f14969b0","revisions":{"42d6265d50210135f8b744505a8f2235c5b93a0a":{"kind":"REWORK","_number":1,"created":"2024-08-02 18:18:14.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/25/925625/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/25/925625/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/925625/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/925625/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/925625/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/25/925625/1"}}},"commit":{"parents":[{"commit":"7a0230fdd2b53885312c6b8dd332b4e8a8f78ddc","subject":"Fix tests for oslo.policy new defaults enable by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7a0230fdd2b53885312c6b8dd332b4e8a8f78ddc"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-08-02 18:16:12.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-08-02 18:16:12.000000000","tz":-420},"subject":"Update oslo.policy version for new defaults","message":"Update oslo.policy version for new defaults\n\noslo.policy 4.4.0 enable the new defaults and scope\nchecks by default which means those will be enabled\nfor all the OpenStack services.\n\nThis bumps the oslo.policy version in requirement file\nalso to ship new default enabling.\n\nChange-Id: If5090f0e4d7672d92a72642d96171b4fd6f490c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/42d6265d50210135f8b744505a8f2235c5b93a0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/42d6265d50210135f8b744505a8f2235c5b93a0a"}]},"branch":"refs/heads/master"},"a219d209d0ef08cb58b4aa09a94d87e1aa438db1":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-08-02 18:18:57.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/25/925625/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/25/925625/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/925625/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/925625/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/925625/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/25/925625/2"}}},"commit":{"parents":[{"commit":"7a0230fdd2b53885312c6b8dd332b4e8a8f78ddc","subject":"Fix tests for oslo.policy new defaults enable by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7a0230fdd2b53885312c6b8dd332b4e8a8f78ddc"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-08-02 18:16:12.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-08-02 18:18:24.000000000","tz":-420},"subject":"Update oslo.policy version for new defaults","message":"Update oslo.policy version for new defaults\n\noslo.policy 4.4.0 enable the new defaults and scope\nchecks by default which means those will be enabled\nfor all the OpenStack services.\n\nThis bumps the oslo.policy version in requirement file\nalso to ship new default enabling.\n\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/925464\n\nChange-Id: If5090f0e4d7672d92a72642d96171b4fd6f490c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a219d209d0ef08cb58b4aa09a94d87e1aa438db1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a219d209d0ef08cb58b4aa09a94d87e1aa438db1"}]},"branch":"refs/heads/master"},"dfa0be428644e033ff532608b96bb78317a3d4ba":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-08-05 01:39:12.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/25/925625/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/25/925625/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/925625/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/925625/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/925625/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/25/925625/3"}}},"commit":{"parents":[{"commit":"b52ae9c4988b15ca30b24371cbf2e343c60d6c7c","subject":"Fix tests for oslo.policy new defaults enable by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b52ae9c4988b15ca30b24371cbf2e343c60d6c7c"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-08-02 18:16:12.000000000","tz":-420},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-08-05 01:39:12.000000000","tz":0},"subject":"Update oslo.policy version for new defaults","message":"Update oslo.policy version for new defaults\n\noslo.policy 4.4.0 enable the new defaults and scope\nchecks by default which means those will be enabled\nfor all the OpenStack services.\n\nThis bumps the oslo.policy version in requirement file\nalso to ship new default enabling.\n\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/925464\n\nChange-Id: If5090f0e4d7672d92a72642d96171b4fd6f490c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dfa0be428644e033ff532608b96bb78317a3d4ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dfa0be428644e033ff532608b96bb78317a3d4ba"}]},"branch":"refs/heads/master","description":"Rebase"},"ed893978dcf1aef8db497bbf8f7387a36494374a":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-08-08 15:03:39.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"real_uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/25/925625/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/25/925625/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/925625/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/925625/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/925625/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/25/925625/4"}}},"commit":{"parents":[{"commit":"240347fc3e066a998b0ad7c472019d7ca6c9b4da","subject":"Fix tests for oslo.policy new defaults enable by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/240347fc3e066a998b0ad7c472019d7ca6c9b4da"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-08-02 18:16:12.000000000","tz":-420},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-08-08 15:03:39.000000000","tz":0},"subject":"Update oslo.policy version for new defaults","message":"Update oslo.policy version for new defaults\n\noslo.policy 4.4.0 enable the new defaults and scope\nchecks by default which means those will be enabled\nfor all the OpenStack services.\n\nThis bumps the oslo.policy version in requirement file\nalso to ship new default enabling.\n\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/925464\n\nChange-Id: If5090f0e4d7672d92a72642d96171b4fd6f490c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ed893978dcf1aef8db497bbf8f7387a36494374a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ed893978dcf1aef8db497bbf8f7387a36494374a"}]},"branch":"refs/heads/master","description":"Rebase"},"62d7e7db3681cb360351398a348357e9f14969b0":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2024-08-22 03:55:00.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/25/925625/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/25/925625/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/925625/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/925625/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/925625/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/25/925625/5"}}},"commit":{"parents":[{"commit":"240347fc3e066a998b0ad7c472019d7ca6c9b4da","subject":"Fix tests for oslo.policy new defaults enable by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/240347fc3e066a998b0ad7c472019d7ca6c9b4da"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-08-02 18:16:12.000000000","tz":-420},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-08-22 03:55:00.000000000","tz":0},"subject":"Update oslo.policy version for new defaults","message":"Update oslo.policy version for new defaults\n\noslo.policy 4.4.0 enable the new defaults and scope\nchecks by default which means those will be enabled\nfor all the OpenStack services.\n\nThis bumps the oslo.policy version in requirement file\nalso to ship new default enabling.\n\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/925464\nDepends-On: https://review.opendev.org/c/openstack/octavia-tempest-plugin/+/926867\n\nChange-Id: If5090f0e4d7672d92a72642d96171b4fd6f490c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/62d7e7db3681cb360351398a348357e9f14969b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/62d7e7db3681cb360351398a348357e9f14969b0"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
