)]}'
{"id":"openstack%2Foctavia~926182","triplet_id":"openstack%2Foctavia~master~I4a15c1a19c5bb0b39494381763f68c056861b003","project":"openstack/octavia","branch":"master","topic":"custom_sg","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2025-02-27 06:17:13.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-02-27 01:00:58.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}},"hashtags":[],"change_id":"I4a15c1a19c5bb0b39494381763f68c056861b003","subject":"Added RBAC for vip_sg_ids","status":"MERGED","created":"2024-08-13 08:57:17.000000000","updated":"2025-02-27 06:17:13.000000000","submitted":"2025-02-27 06:17:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"926182-custom_sg","meta_rev_id":"1ac29b45f6e819d2e708c2b712bda9026b5e1deb","_number":926182,"virtual_id_number":926182,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-27 06:17:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-02-27 01:00:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2025-02-24 13:14:16.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2025-02-24 13:14:16.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-13 10:34:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-05 11:42:07.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2025-02-19 01:28:37.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"3033dac61ff9ce03f234e5a9b378201e6eff7875","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-08-13 08:57:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cf3efcf19ac2f92b3829e3ef717cb6397c40ec6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-13 10:34:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c0a81c4dcfa74321ab10edf397d6d19c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/19d58abd83ba4fc2b532ac6b8914cac0 : FAILURE in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c842e1c265246bab140df9a2cde8765 : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/71de219af09541b08af7ed7c46dedb14 : FAILURE in 4m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f1064cee7eaa433f8446f0ac68ab622c : FAILURE in 4m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/50c0732fed894fe7991d07b60028ce91 : FAILURE in 7m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2101c8e241f64e96a39ca1e8efa36c87 : SUCCESS in 12m 42s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/9609cbac527f41029e3db0d1623b1915 : FAILURE in 4m 23s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/c6ab6e53b49c46929b62ad854843a64b : FAILURE in 4m 55s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d44cbc3cc77847848c43df895c0eb5c4 : SUCCESS in 4m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/21ffaffb38584b0aadb4eece259d680c : FAILURE in 4m 46s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0b23269828854f1e8d49ec7b6236ce9a : FAILURE in 5m 13s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/98a67bf657744591ad955f24c56bbb96 : FAILURE in 8m 11s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f21cfb18ff1f43efaa9c8d0db50aef97 : FAILURE in 21m 00s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/6c41782d68a64ddc8015bae1906ffccb : FAILURE in 27m 09s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/051b10b7661443ed8afcb640bdc1ade3 : FAILURE in 41m 33s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/2b497fc29fbc4e48a3d27d1c0cfbddb1 : FAILURE in 25m 14s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/db1c2c1d2923437696c8359a291464b1 : FAILURE in 40m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/53182c7f1ca44dc48e15c91de271e068 : FAILURE in 41m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/af726e8cf39a41dea3d3bac6d0bdc222 : FAILURE in 54m 05s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/f01356dae43940328d36928f70962b43 : FAILURE in 46m 41s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e35d02992c0b47288f29504f77facba7 : FAILURE in 33m 27s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/69b5dca4088c48cc8770d25b2c8e1a4b : FAILURE in 51m 19s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/df6d7fe8e1704c3383fa64120842b575 : FAILURE in 53m 37s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f6ed1556dec940c8a90a3ad9362255cb : FAILURE in 51m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/38f44e5de3dd47b0836fe16c44eed551 : FAILURE in 31m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4202c2e8b83af1e23a78706e65017966d8a79891","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-10-11 14:26:55.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d0f1bc160b6a54428c3191369fedc2dfd9d88b9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-11 17:26:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb44918507c445329011f3092ce58251\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/128281edb74646a49ff4d56f3c640cb5 : FAILURE in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ba6af5e0bfb46638103d4bcbe86075a : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/157810243d81402aaa94610c51c187bc : FAILURE in 3m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1ddbaa2121fb49a695d256054c07664c : FAILURE in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc5f2ee666884276b1c61bf0a7fcc7eb : SUCCESS in 11m 00s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/09980e28b1284005a7e9891d248d18f9 : FAILURE in 4m 49s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/6c7e2b2ad3654501b355845d1a9379bd : SUCCESS in 6m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6159d0e06d854667ae28f011f17c79f4 : SUCCESS in 5m 24s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b88dff3478114451ad7315960ec592ed : FAILURE in 5m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/75be7a62382f47199a454b2771a7eb58 : FAILURE in 5m 08s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/266cf42341b4486896ee7b78fc727484 : FAILURE in 5m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/eee85b22d0dc4459890fd3d4c56dad10 : FAILURE in 1h 35m 04s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/d2ace630a2404a0fb72d838ae9ab756b : SUCCESS in 1h 43m 47s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/1db684eeb3064eb891f30da2491b56ce : SUCCESS in 1h 05m 38s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/933fdf2d5e3747f9abd6bee2ef570da4 : TIMED_OUT in 2h 33m 46s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/99284fde7e0c4677b971afea65b8573d : SUCCESS in 56m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0e4298493220493dafaead82f93dbe4a : SUCCESS in 32m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/618bf53df9d4404e84bca6a260a34585 : SUCCESS in 44m 52s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/efd9eabeb6334458b4f90d1fd424e123 : SUCCESS in 49m 27s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/32acb1ebc9c748068cd6e03c723b49cf : SUCCESS in 47m 12s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1756bdd868164f8b8b59dc7b59f0638f : TIMED_OUT in 2h 33m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c304efaab2ab4509b0deb47bdb0d04f3 : FAILURE in 2h 04m 02s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/997e23fd47ed494497d37413601ba538 : TIMED_OUT in 2h 34m 00s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/8a4078dc83fb4057a132d902dea25bd4 : TIMED_OUT in 2h 33m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f4d52c81794010273e5963b3bf9f2193782b1cbe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-10-14 08:46:02.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"4bd66e3c0722b5fd0d616a23a3fed192bbc36dd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-14 11:50:37.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/96347afda3c247528f5f5cd040b9261e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52520ce53a07496f98409e6553476fe8 : FAILURE in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fca8a5b126b9475cb0b27ec46d43457e : FAILURE in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e3a594f9380443a8a97063c393dbe0dc : FAILURE in 4m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d08ab592f9c5410ab754544bf1f0151b : FAILURE in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c1c7c15d071415aa5c770d11ff8889b : SUCCESS in 10m 38s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/611aa9a45ca7436fa64208058f74bf37 : SUCCESS in 4m 51s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/dfe03ffee30d4f8b9fbffc121419acb5 : SUCCESS in 6m 58s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8dda1d29e1184cd890a67c5b5f2fbc0c : SUCCESS in 4m 07s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c384f2f8bf8b42cdb3ac11af94b5527c : FAILURE in 6m 28s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d1e6e53c44544178895b809644502eeb : FAILURE in 6m 16s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f5b2021ab83743458fc0b81ab319ec57 : FAILURE in 7m 08s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c29bc7588ad44406aac1b7d61da16722 : FAILURE in 1h 29m 36s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/03d705ea8ecd44408ff8d6d6103be9f6 : SUCCESS in 1h 58m 10s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/4262a04a7f9347e6a574204f3a004754 : SUCCESS in 1h 30m 49s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/842b8511540549988e3d5fd3c35b0306 : SUCCESS in 2h 21m 48s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/eb013388a3324a738f7e665fef3ba7ef : SUCCESS in 1h 33m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/64a031afd74c44a5aff6d179340dd8f5 : SUCCESS in 41m 50s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0d547b9fc3864b95a3dadcfbbc86be6c : FAILURE in 46m 32s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/e668e5afe4ec43948cee0cd16e7cca7a : SUCCESS in 1h 00m 02s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/733b7fc063e74ad1b2267d2e85deca32 : SUCCESS in 49m 09s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b342bcd976474aeebcefc5fa961c9646 : TIMED_OUT in 2h 32m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f8272f53c33842c68908da19006cc836 : TIMED_OUT in 3h 02m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a62c826348284c0a9ebda285a9835954 : SUCCESS in 2h 31m 06s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/9e928298875c451381a4aaba42a4685f : TIMED_OUT in 2h 33m 53s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/api/drivers/noop_driver/driver.py","accounts_in_message":[],"_revision_number":3},{"id":"e803c281b5410111555e37f90a7b1776d0634a7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-10-15 12:50:00.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"49589f19e69fabe3042f6d29cecaebbd93378da3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 15:49:17.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fe14e906d7c14691b1a0fb30d02737ca\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6bdcf1a5ef9f49dfa869c05506c3e6e7 : FAILURE in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdc1037bcff540db8975cbb9ae332d65 : FAILURE in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a2291ae8ce5042dd8471272f4210c447 : FAILURE in 4m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/21dd36353d7043529c8627d2a62aaf8e : FAILURE in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/245783b4233b4404b7b6bcd8bbc4d77e : SUCCESS in 11m 47s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/f287c4ada91e43ecbfc65d64b8c3c94b : SUCCESS in 4m 36s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/3f549cea32a347a882ffcfbdbab8cdf0 : SUCCESS in 5m 29s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ee7dbc6fcc6547888b31ecf4a97a2b0e : SUCCESS in 5m 24s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/70bd00fa24b6498ba83a665b92aa18a4 : FAILURE in 5m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8b9f5e129de3400e9ee9b92137ca33fc : FAILURE in 6m 48s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/e8eb054708794ba98dacfeb3e038ecdd : FAILURE in 3m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/85ab948c59654778afdf964f2a6d8355 : FAILURE in 1h 24m 00s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/07b527617a8047f7b3dd1178241acc8d : SUCCESS in 2h 24m 14s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/d84e99dfa0d8400baa1aa853b4cf4909 : SUCCESS in 1h 33m 57s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/1b0a27a711284b2c8d5bdae10d05a9b8 : SUCCESS in 1h 49m 45s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a333fec3184943d78f0047e51ab978ca : SUCCESS in 1h 40m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/35a31699a9a6448db806175788e9e48a : SUCCESS in 35m 09s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f85cd293325b432a80288e9be29c6722 : SUCCESS in 41m 03s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/fc37453302d9497fbb19d4d87bff0968 : SUCCESS in 1h 07m 43s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4a757b0c2d7343ff804a6e8e55a0aee9 : SUCCESS in 32m 25s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/feacb3a213194314a9ae9ca9f07353eb : SUCCESS in 2h 17m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e9cdadb7d577457bb815563776e0649e : SUCCESS in 2h 56m 37s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6a683ce88f8a4b3d96d0083e23cf1908 : SUCCESS in 2h 27m 36s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/198365cb7ab34864acdd0fcc52062e3e : TIMED_OUT in 2h 33m 20s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/api/drivers/noop_driver/driver.py","accounts_in_message":[],"_revision_number":4},{"id":"fc0a62033150623d7863964eb82ea692ccb1d764","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-11-08 06:19:37.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"2045d30c88ea302bfb4ac8b9620dc5eabcf783f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-08 09:24:25.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7b486ae4c9654afc8157637195dd62a8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc703ea3a8df4b14ab5ac092c50569e6 : FAILURE in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80361fedbe3e49cd813c01b8b38869e5 : FAILURE in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b7aaa3c7972741ac9fff9bbe13824a48 : FAILURE in 4m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/02173a94df054e38b3477817f28eaa51 : FAILURE in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13d6591d26254332b53359a82e65ac77 : SUCCESS in 11m 22s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/58ed576ebcda4b2dbd6c3d8a40aa0223 : SUCCESS in 4m 48s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/674728710ba74b1bb7abec0603a0cf9a : SUCCESS in 7m 19s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8f14bc4c7b1d411fb76953f0703a2e2d : SUCCESS in 5m 00s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/304548f0a4824653a3a9739b8de2d156 : FAILURE in 5m 28s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6f526ad4bdc94a8ea3a3737d80d1d918 : FAILURE in 6m 22s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c9f2b92456d748d599cde4340d37bacb : FAILURE in 5m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/50c82fd0122d4ef2807074271617bb24 : SUCCESS in 1h 27m 18s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/334958b4a10e4f64be324d122d3331d6 : SUCCESS in 1h 53m 42s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/c4132aad61604b70953bd7d170ccac67 : SUCCESS in 1h 10m 29s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/d34199a429194353964cc15717d3478a : SUCCESS in 1h 58m 18s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/380eb314dd874f9896dc7355e3742779 : SUCCESS in 1h 04m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/29334f095e2042128212a6f458583044 : SUCCESS in 53m 14s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2edd91f18d894357979305d5be471b0c : SUCCESS in 56m 09s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/5a434a8a5b6945efb3c07794d4bf41d0 : SUCCESS in 59m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4887a70ba2164fd4b87c7243abe9f5c4 : SUCCESS in 40m 32s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/59035ba692af44d5b670cecf5287b800 : TIMED_OUT in 2h 33m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/22847626d6e54a9f98321774ad6671af : TIMED_OUT in 3h 02m 54s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/90581ce3a51d43ada67914f617dce2d0 : SUCCESS in 2h 27m 15s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/fbf1d6d43f7643618e47e73a0fbf4dbd : SUCCESS in 2h 00m 54s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/api/drivers/noop_driver/driver.py","accounts_in_message":[],"_revision_number":5},{"id":"61cfdaed61df4d4bf7fecef4814b7f437c29d593","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-01-07 18:28:05.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1f14a8cab790a625b2b4336a5df2c394cc8c2bad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-07 21:20:49.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d06106e1d3f6418b8a34b6dd063cf659\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e16bd1dd0674485859dc2ad29745cda : SUCCESS in 12m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc96b3b689ed42638c4cb423c1bb5d77 : SUCCESS in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf427a82ad244be1817cb76715213e38 : SUCCESS in 3m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8c1d0d1723fe4d4a9508d7cabf80d129 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a3d81e8dadf4f7fb38bf578ba4f6744 : SUCCESS in 10m 32s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/c323f0c0e1194dc4a1fa33ed49604525 : SUCCESS in 4m 31s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/1870bef09eab4e0b915936ca20342139 : SUCCESS in 7m 55s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/92690b2046b14ea8bc56d4c683ca15fc : FAILURE in 5m 54s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/12de5dd8bcb84817a5d6cb002ee21056 : SUCCESS in 10m 11s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/40680aeeff704d919f8486f9756baacf : SUCCESS in 8m 07s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/308f9c6c86b249a3b5ef800f579475f7 : SUCCESS in 8m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e2f93be42ed148fa82cfba70499e1716 : SUCCESS in 1h 19m 46s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/dc2c3d8a15344ae5a0d29a98894e6fa3 : SUCCESS in 2h 22m 50s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/751d4ac56fe44dcdb3ecbca5b95425fd : SUCCESS in 1h 24m 03s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/0899adb6e8bc4a51a30470d39ae66344 : SUCCESS in 1h 48m 40s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e023190d12c54d989ac5569275606bbf : SUCCESS in 1h 04m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bb699b59dd6a4068a18c9014223d2839 : FAILURE in 32m 48s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/42eaeb4fd1af4b808821f24c5b988277 : SUCCESS in 33m 13s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/b501866abe754d02811f701113ad456f : FAILURE in 57m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/041bdac9bcc04405aaf2552e3498d377 : SUCCESS in 32m 54s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/dce9a4d1a48c46bab4a27c70b095c094 : TIMED_OUT in 2h 32m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/819b8f1efaee4b07b10cda34bcde42ee : FAILURE in 1h 54m 46s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f2d4a0fca136491fb86643b5bbcc8f46 : SUCCESS in 2h 19m 20s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/20615ca49cd9427a90d11f11349bd949 : TIMED_OUT in 2h 33m 37s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c9f78dd35f60b469e36279154ac2271e05f509a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-01-08 21:12:13.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"5f502c92f0380b895498132b4830ec3502482536","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 00:18:55.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7e71577d9c284856b8a4835cbe4c7132\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e8655dbb5f014d0ab03d50d15be93151 : SUCCESS in 16m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa65dbc1ab03445ebbe6be17de041b7e : SUCCESS in 6m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/718070138c274afaa58e5e7c5a13ca53 : SUCCESS in 4m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bf5ec2e931b74fbaa9bcf24dee93b6ba : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f9219bc0cff4d70b6694595eea7a338 : SUCCESS in 10m 34s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/7e7ef636b773485d9489d8903f9ea7f3 : SUCCESS in 5m 12s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/fd8735647c984f88b761469d035df6bf : SUCCESS in 6m 51s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7e3790ef070d454e9c74681917ed7808 : FAILURE in 4m 42s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/948a914b337e45b2a871287438c35631 : SUCCESS in 7m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0cecc50390fe4c379ace2c44c0380e43 : SUCCESS in 7m 16s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/022602246930451c895a9040784139fd : SUCCESS in 6m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ddd18384eba94fb5b7a64b12c1d76c7a : SUCCESS in 1h 20m 25s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/8d72203ae8c24777b1ba18558f213b85 : SUCCESS in 1h 44m 08s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/6333f5f87b414aa095e03d3e41f1b55f : SUCCESS in 1h 03m 12s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e3d31bf36adb49df9259d87272d2d5f6 : SUCCESS in 1h 34m 11s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/55a665c0a7154a00a10f4e9e9e280329 : SUCCESS in 54m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4be6f03bac53403988742ad6e8802ece : SUCCESS in 34m 22s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d0666d7cf3b840ea861fedc5e223d0d4 : SUCCESS in 36m 43s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/1126285112b340619d3bf51cdbf38a96 : SUCCESS in 57m 48s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2a032030803a4fa1826cc6edebc8198a : SUCCESS in 30m 51s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3ae2fc41e6924c0f88877b1a3ed43880 : TIMED_OUT in 2h 33m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e3bc7e66a2504582a80494be880291d4 : SUCCESS in 2h 48m 45s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7052956385854094b2742735fa743af7 : SUCCESS in 2h 23m 16s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/47705594333c4458bee74f13ae69a685 : TIMED_OUT in 2h 33m 16s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"3120129dcbf669baf1094b28a57cf7f15a86df7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-01-16 10:22:30.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"96cc95bc9966a70359463b45fa68cee3efd8d13a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 13:16:27.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9767566dc38b47798e2782b98cfb85e8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee1306ced61e4132b3b52f7cfad10202 : SUCCESS in 13m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a7c3c3f60314c0c8d2c48f9656177fd : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e1d3a5d499014300ab7758cbf0566e42 : SUCCESS in 4m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1abc66f082a14691868b54c8a19ea303 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23410f4bf1b34807aa43a1de5ee16f0e : SUCCESS in 11m 08s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/b9c51b4faf3b4e9e8c8a3ff26c056282 : SUCCESS in 4m 33s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/c1e6606684fe46dfb002960caaa93851 : SUCCESS in 6m 58s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4b1dfc604c65496a8f08e6b954588b15 : SUCCESS in 4m 13s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6185c33cf9944ec395887f20a14a9f15 : SUCCESS in 7m 04s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c14e60fd76ad4891ba33792de2f23f55 : SUCCESS in 8m 54s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/85ecd9e629b14d9498fbcf894dee433b : SUCCESS in 8m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/99f7abaec2c9450299f5ebab47a66344 : SUCCESS in 1h 22m 56s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/9d653843d3e34fbe9b4667545b8e4633 : FAILURE in 1h 58m 31s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/226d2f222e7c4669a514f34ffa9b07d8 : SUCCESS in 1h 08m 12s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d96901bc6ae74086abe68b4ef1023959 : SUCCESS in 1h 48m 46s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/76d32603622740728b2fcd8b5989a00d : SUCCESS in 1h 05m 39s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3d450be2afdd43348a463de677e4d49d : FAILURE in 2h 07m 34s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/215e54b3e69f4ea8985fd0704ee07f23 : SUCCESS in 1h 34m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3d2c1c786d984d37b708201e96ef9502 : SUCCESS in 43m 00s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/35c19cb754764c40993d3da73fc2b023 : SUCCESS in 1h 06m 46s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/15c8c508c2fa4c948a2ef492a3d48272 : SUCCESS in 1h 11m 45s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/89653f2e118242338d22670824027b9c : SUCCESS in 51m 57s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3adc2ac63cc246dfb9a89c0d7512393f : TIMED_OUT in 2h 33m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0e5c2e909af44d029b9ec69ef034201c : SUCCESS in 2h 51m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/10cbf4b499444bbe86057ee8301a5453 : SUCCESS in 2h 26m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/d5241b3e38554f969af480b34d60480f : SUCCESS in 1h 55m 11s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"97f591d23096cd6e264c8fa227d7fbdc23c1dd1a","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2025-02-05 11:42:07.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"33b676fbe3918926de1ca1c312644227fbd77f1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 14:49:16.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3072a100f7541bcb65e8ece74e1b960\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d487aa44d5f44cc8a7ee7098aac271f0 : SUCCESS in 11m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53acce494b094e77b7dbb65488a9696d : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/22297f8fab2d454c9b00c331a8719a0c : SUCCESS in 5m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/43b4c9fb0b894a2582e397bc48d634fc : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/659144e82d9b4092a3a6774b69f627f2 : SUCCESS in 10m 19s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/93034379a2e9487fbac5557dcc446fad : SUCCESS in 5m 36s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/48032b3d14264a99ac44ecfce462b713 : SUCCESS in 7m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b010e625393a4a3d81bf30e700678683 : SUCCESS in 3m 15s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b471732a330d4a429d99d9a30b531860 : SUCCESS in 6m 56s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fc5428411f8b4e339ea09e37cd1b061e : SUCCESS in 6m 20s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8acb1c18764241a3aee1b12d95c0a03f : SUCCESS in 8m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cb44dcb360344604a1cb2985000c7a34 : SUCCESS in 1h 24m 36s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/491917aba49a4138a474d8ebd651d25e : SUCCESS in 1h 44m 26s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/cf99d77f39e744bcbaf8bdb85877d81e : SUCCESS in 1h 28m 04s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2083f9ef80664e38809b68da2100d453 : SUCCESS in 2h 21m 08s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/70c5808d04dc4eaeb915346e4c8c168d : SUCCESS in 50m 53s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/41b0da0509e14151bf55a53fc45c3318 : SUCCESS in 2h 22m 22s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/cfaede2139a447558eb29385a8e37689 : SUCCESS in 1h 29m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7bf73664f90b48d7a58a65ed721a44a1 : SUCCESS in 42m 55s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c127a8ad9daa481b978f30ca60abff60 : SUCCESS in 43m 48s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/88d71470f7cf40d29056b290396325e9 : SUCCESS in 42m 03s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c374501cf535415a9c35fa505dccbdd7 : SUCCESS in 53m 12s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3d6a7fcb7def4ae8958257ccb907d08b : TIMED_OUT in 2h 35m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6cf60071107b45ea93c40b06616ca002 : TIMED_OUT in 3h 04m 37s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6cce4ec491a84e958e18e2ab9ce4a248 : TIMED_OUT in 2h 34m 54s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/2823b36a1bf646a59c266b2208dd66e5 : TIMED_OUT in 2h 34m 54s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"b43d5d3bb78ede275e0fb518ecc9c0d9d739ebbb","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-02-19 01:28:37.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"220e5c402156f463ab59c62989e811b86e5a76db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-02-19 09:05:16.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":8},{"id":"d4f1596f9579e0433183249c2fef6a6ed32c043b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 12:16:55.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0f47cabd4b404e7a81fbe67061d53fbb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57607f813221454aae8fbd63bca8eeb8 : SUCCESS in 16m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd54a456c8f74879bd9ffe260848cc48 : FAILURE in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dab8848810a5430b832164280d6f955e : SUCCESS in 4m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/56b166820e3a420faab427f64d76cf64 : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11f6976abd4e45f688c6a2976fda8b93 : SUCCESS in 11m 04s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/200fdb831a914b3cb33790589a55d96d : SUCCESS in 4m 24s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/bebc7ed4d8364c5799b00e6d80a0f7f5 : SUCCESS in 7m 06s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c873f77c94634fabacbf827e2d2cdb99 : SUCCESS in 4m 04s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5951cf900cbf4019be3bfde349a027a5 : SUCCESS in 9m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a5f008efeda74064b806a49371320e74 : SUCCESS in 6m 42s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/284220a58daa4f6993dbceb5c1701845 : SUCCESS in 6m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/93c8b07f53d3420bbc4039cd9be9adda : SUCCESS in 1h 32m 21s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/01cb63482c5646d09774c5b711d4d80e : SUCCESS in 1h 44m 30s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/dc347483a99a42adb930353c973945a3 : SUCCESS in 56m 26s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d607fe99e9648e2b36ed2b2fad91c7c : SUCCESS in 1h 41m 31s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3e84c684e925419a8ec416d598e28034 : SUCCESS in 1h 16m 39s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/8068c96030344c24943e606f4d2f7d34 : SUCCESS in 2h 28m 31s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a4bd2f3b001c4f41b86831133519878a : SUCCESS in 1h 25m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/931bb7b6008a4af787edcc4d47054c7b : SUCCESS in 33m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/061ce00186b0408b9b6b3a6449a01464 : SUCCESS in 39m 50s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/09f9e4a490574241b9e15fce9f439dbc : SUCCESS in 46m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/221fbc485e1d47829c6ad2c8ac0655c3 : SUCCESS in 35m 29s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/832b1ae8c4a94355ab52f6dcb135ffc9 : SUCCESS in 2h 23m 17s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ceb3a91eb1cc41bf95e9e3110f82331a : SUCCESS in 2h 31m 24s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/101cc699bcf541cd974594f2479fcc3e : TIMED_OUT in 2h 33m 19s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/ef4a9c5823b7418fbf7286fa2ff70955 : TIMED_OUT in 2h 33m 22s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/network/drivers/neutron/allowed_address_pairs.py\n  Comments left for invalid file octavia/tests/unit/network/drivers/neutron/test_allowed_address_pairs.py","accounts_in_message":[],"_revision_number":8},{"id":"c960e340c773f899d65a6ee7774518ce2662e93c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-02-19 13:19:40.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"4b042e1a9b0e4322572beba2681e2b3e5dfcb76e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 15:59:55.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/214a8698948f487e96d78d96b4e5e997\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2905d584d9cc430fa1fc21426e5ef3d8 : SUCCESS in 13m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e981ef5f18f43fab24ddee0658353ca : SUCCESS in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/62df52c9ef3044ca88130ffe75d33e0e : SUCCESS in 5m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/541a75900ace44088bec3a65c99a540e : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/372069935cb34ebbb6affd384cfcc3ad : SUCCESS in 12m 52s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/1894611669a14bfea601bdcdf1953cbd : SUCCESS in 4m 43s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/3e3e1a4f9ee14ed2bba9642a6432b1ea : SUCCESS in 6m 58s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ce6b32831176485792f82eb0c208b983 : SUCCESS in 3m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b7d02ac07d5349cea1c2bb3cb04de5cf : SUCCESS in 8m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/77c44ce6c36d45d99a2662bbf9dd5797 : SUCCESS in 7m 05s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/58b14732a59243bca816f3bccdbba8eb : SUCCESS in 7m 00s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2c8dbafd98364b888e3b9150e1c36322 : SUCCESS in 1h 51m 33s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/85288bb538bb42e282bac895d834eb53 : SUCCESS in 1h 49m 11s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/8300e0e02c984ac992a4110d0ae2ae28 : SUCCESS in 1h 02m 14s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d80ac340e6c6433caa09799e9f42884e : FAILURE in 2h 33m 39s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c515b02816044da093b00d4455f81c04 : SUCCESS in 58m 15s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/2476b309287743348b2f5fc829e4e8a4 : SUCCESS in 1h 49m 12s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/1647853d8188409192c42cc04a2ece32 : SUCCESS in 1h 28m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b475a71496324530b81b159f9c8bb65f : SUCCESS in 40m 36s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3c84c6686817403c9eb01a5a5d1942e6 : SUCCESS in 42m 44s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ab3aef56bc92447da99e5cdd34995eb8 : SUCCESS in 1h 18m 50s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3047e10f07f44591af3eb2978e28881f : SUCCESS in 40m 05s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/73ee15c0bd434aba817349375ef40f2b : SUCCESS in 2h 08m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b528bd9794034e13bfa44bd5dd327f35 : FAILURE in 2h 14m 36s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/12eb409e7f9f4c7e9f9166dc0b7d603f : SUCCESS in 2h 18m 19s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/e869688fe77a400aaa5a4815b2ca72c5 : TIMED_OUT in 2h 34m 50s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"8d3de89f2aa7945fcbb278c12bacbb3464480908","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2025-02-24 13:14:16.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"24962bd2d5bdbf9e72f584b851b439fdb53a741e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-24 16:18:34.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bb1a72fbd1714d9783408c7b69bf4ec6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/26d885780d484415aff1b8dfb874c594 : SUCCESS in 11m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9ba600340694d1085e8a8f32abe1f7d : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/93b41fc67aef49efb59bf5c810b9efbe : SUCCESS in 5m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4e87f9f456534e2a848b746ac9509cf4 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db6686e623744f40b4312194a4fa8843 : FAILURE in 4m 29s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/775be3a642e44bd6a335d73970d47f13 : SUCCESS in 4m 56s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/ec951134a3384e4e854884154f0a93fe : SUCCESS in 6m 37s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/19efe3402c3945a28a3f96d75901d8d4 : SUCCESS in 4m 47s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/889f65c09a1947039b2af37989b76902 : SUCCESS in 9m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d413302c05324b47981528c25f7864dc : SUCCESS in 6m 59s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1717adc553fe4e22b481b73cf1ef7cc1 : SUCCESS in 6m 26s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/593533db09ed4861a0f06b8f2cb0c3af : SUCCESS in 1h 28m 37s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/33a2a0e43f1e4a6a8dab7db019aaef45 : SUCCESS in 2h 02m 37s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/935252428da4454686adbae7c9534eed : SUCCESS in 1h 07m 05s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4115fde0ba6c456b8ebd510147521999 : SUCCESS in 2h 11m 38s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b4b9299070141419305372ddba90e79 : SUCCESS in 59m 20s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/519a1c586dd04bd2af95c18f6dd82cab : SUCCESS in 2h 19m 18s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/0518c97d1ffd4930b96a1d3256954545 : SUCCESS in 1h 03m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8bcb44df610a40d5adaceacbaa7ad047 : SUCCESS in 49m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e001e492722e45cfb46db5d66ccf9494 : SUCCESS in 35m 49s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/82f79ed5a2c641d2acbadaf66871646c : SUCCESS in 1h 13m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c610bc020b4f4300ac97cdac82bba39c : SUCCESS in 29m 12s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/cb6276fd912646e0bf3d85c3ead23076 : TIMED_OUT in 2h 33m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/480f6977ef264519b2c186f06b41ad1a : FAILURE in 2h 08m 12s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/729a43824cce4c99a85cae8fd58df6f7 : TIMED_OUT in 2h 33m 22s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/7bff3e9a10414d49b009c02e45ca6dc3 : TIMED_OUT in 2h 33m 38s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"91c3edf4de20414b2f267e081c2dff53aa1de669","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2025-02-26 15:55:55.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"4dcb26cff4469468f9aa2e203d70d6373a7c0eba","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-02-26 17:39:06.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"9255ead8e0e14bc48f97e44369fec11209361e87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 18:37:41.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ddec43336e274cb3b4f1674f0d093552\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb287945b5224dffac9c14fa20ffb57f : SUCCESS in 10m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7de0aefeaae04b1ebeb399324820a7f9 : SUCCESS in 5m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/440c49e4bc56490da295d6c8ad3ad0d7 : SUCCESS in 4m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f4f80f2e881243299d2943ffd885a476 : SUCCESS in 5m 06s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/90e8c861c0a94389802b573a2dfc87d1 : SUCCESS in 4m 41s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/c69696abf558431aa872b4561fb901c0 : SUCCESS in 8m 55s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/34616c2afe79421ca7f37af3048b0296 : SUCCESS in 5m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/46db5aba24524683aaf25b5de983f079 : SUCCESS in 5m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bfd4a353e5894b39b840643e8d37b22f : SUCCESS in 7m 18s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a3f64a4b8da74bb9854af87f2115a132 : SUCCESS in 6m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d07e438335ba4540aabb199ce1568e87 : SUCCESS in 1h 09m 31s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/6310003c63f845f5a9d30fc9696772e5 : SUCCESS in 1h 45m 18s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/0983f4866d0e4d83a209c62331aa42b1 : SUCCESS in 59m 19s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/88ba52b85cf74e2c983dbb393b8444ff : SUCCESS in 1h 44m 27s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8bd2310c7e684788af1ea0ce02807eef : SUCCESS in 1h 00m 56s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c1d2e323ad504cc88a0528902fcb9dd5 : SUCCESS in 1h 58m 36s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/8cb3ed5ece9f4993a5f8f67fff2d2fb3 : SUCCESS in 1h 14m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9a9ba65eec794385bc5bdfa2048f45f0 : SUCCESS in 42m 14s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/20c846b3188b4b939d18004e01cc282e : SUCCESS in 47m 44s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/38564c265c0742cea2f17229a53e73e4 : SUCCESS in 1h 22m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/516655c4dde14b6aa05e804e93c59f1b : SUCCESS in 33m 59s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8e22eaf584c843188090916df8eeac4b : SUCCESS in 2h 17m 03s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7ba0ae697c114a65b6f5e1aa9053fe14 : SUCCESS in 2h 40m 11s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d7ce3e4674664747819237852aff1f95 : TIMED_OUT in 2h 33m 35s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/9d39a3193c474177b79f40b1aebd2e9e : TIMED_OUT in 2h 33m 31s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"69bc86505f2f1ca60e65eee113fc5c49910e4d03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 18:38:18.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"e0edd91e553773f749d2ff64c1a09c42c83990aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 18:50:02.000000000","message":"Patch Set 9: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b8630aa069be4f9f966a1c5a816482e4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42625f939bef43b78593883ea4decdf4 : SUCCESS in 7m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/66fb8b7c650247fab1ed993700994d8b : CANCELED\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4eb521ff6eb7464ca769fe400c0e0226 : CANCELED\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/99591dac53594c1898a1be3ab2e22974 : CANCELED\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2d2b2335cd3a4c61af2c13586ff9a837 : CANCELED\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4bd1d6ee904a44f9a03a393645c9ab3b : CANCELED\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/9f637ac037a94821ad355833087a65fa : CANCELED\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/a9e0fea54f664be48e7f6101efcb1b9f : CANCELED\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30b54d0ef8524de6a40fce18c31448fd : CANCELED\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b7a216fcd53142dba8fe04d6b88f25f4 : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2f192a80f483422cbd63f57a8a9b59c2 : FAILURE in 11m 14s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d56ed593c1c040c1a231e5c8e8d38244 : CANCELED","accounts_in_message":[],"_revision_number":9},{"id":"2e93f0f65383c758d26ba53598611c4af4c9c552","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-02-26 18:54:09.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"d0a29a1d117296bfe69b490735fdd7a8b416461c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 21:42:08.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/be5cc0e477854179840f30b6b4ed778b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c3382dbc3b4e497497493cc0481d8f79 : SUCCESS in 14m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39c461791da347a698e12db6bdd7176d : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/15f4f5ad3ce046b5b61ea727a7cd39d9 : SUCCESS in 5m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f7e4888abd7143b5ac7477a9cef013d0 : SUCCESS in 5m 08s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/c0fb4bb91bad41afae4dae4140cccf09 : SUCCESS in 4m 39s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/be670d668f134026bdb8184749b3a447 : SUCCESS in 7m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/85cb9fc2df3a49f09a9034aca3e48763 : SUCCESS in 4m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c3b007eddb0b49e3ab19c4af09919af8 : SUCCESS in 8m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c8943fdf97924f2eb9474ce4ee143be3 : SUCCESS in 7m 58s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b5a0e7c0f2c044f191b4025c621c199a : SUCCESS in 7m 57s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/65ab0decb10d4c868e74df7665a7ad11 : SUCCESS in 45m 01s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/e74fd7035c2a4f0eb16cd8b39ba30bd1 : SUCCESS in 1h 44m 18s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/33cce763a0664ece9a985feeb16d6299 : SUCCESS in 1h 07m 12s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2b46357c1e654a079679085410903213 : SUCCESS in 1h 41m 57s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ae881c96e1cc486db623ff8fbae220c8 : SUCCESS in 1h 06m 09s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/12f08ddfe4fa4eaeb92b1ccdd4f442ca : FAILURE in 1h 52m 36s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/566880b364734344906a05a21db2cea1 : SUCCESS in 1h 08m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/669bb53b6ba349508c4aa97b9123b454 : SUCCESS in 39m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6005ca65657345d49d2473e087777d0c : SUCCESS in 47m 10s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ed46bc67fd264f2b9cf70226644ab4e4 : SUCCESS in 1h 02m 18s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ac222783fbe04558b05d8eec3418c2de : SUCCESS in 38m 35s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/cd62c5a407a642d08b5733d3135863c4 : SUCCESS in 2h 24m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7bf61e109a234ad8b997e61523d7e43b : SUCCESS in 2h 35m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6dfc15108b2e4874b846d2efd5ed1b5a : SUCCESS in 2h 25m 15s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/f3e04ca9397149508cbe660a9e79f820 : TIMED_OUT in 2h 34m 52s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"971a06637dce847c1c3993f64fc0385d64eaca62","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-02-26 21:49:34.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"1a6c95be43b90565e8d848bba861a74c0f58545b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 00:39:43.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d28a88b0634342bfaea23ef9dc756452\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a4244b700b634065be902994058022c6 : SUCCESS in 12m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd9833e43ac24398b93d82e5943fb495 : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c6f1cbd04b534e5fa6daf4be80996431 : SUCCESS in 4m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1f5667274c8e4b40b108bd6679f2eaa3 : SUCCESS in 7m 00s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/bb26716937c84b9fb6821747542f346c : SUCCESS in 4m 19s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/841396741275498ca5aed9318702b7d2 : SUCCESS in 6m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5aeeb162e5054609a8b34142713ebd41 : SUCCESS in 3m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c9e81c6b181f467d9d263e24423aa55c : SUCCESS in 7m 39s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b61700a423d04cc487f2739242f44de4 : SUCCESS in 9m 45s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/086336319a4e4a1ba343186e5d286573 : SUCCESS in 5m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a742384bc108408c8cf1a1f056d86c95 : SUCCESS in 1h 38m 52s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/d8fb55d5099c459b928f4dc57d6e649a : SUCCESS in 1h 34m 38s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/8bee4f3fcbbf44a794b06237ea4b7694 : SUCCESS in 1h 04m 36s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4ceca1e4b37446159d57494867aefa27 : SUCCESS in 2h 14m 02s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ac8717ac0a54ad5b2cc032007de674c : FAILURE in 1h 29m 25s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/42275196fd024b5f8cc208fd2209f564 : SUCCESS in 1h 36m 20s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/fd272f81e3244914905ad6ea5aaa82d7 : SUCCESS in 1h 30m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/01c6761a3268460daf8cbd3977ec9004 : SUCCESS in 34m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0336ac327d7c466ba4c9fa3e3ff3262c : SUCCESS in 1h 09m 52s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/bd3fb3badb8747eaab6903d52848dbf6 : SUCCESS in 1h 05m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f422b1eb7dd74088a90dae93471a4759 : SUCCESS in 33m 24s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9149f8ee46074a54b97db78a5e9079ba : TIMED_OUT in 2h 34m 36s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/195aa29c93494e08a5914a81514f2761 : SUCCESS in 2h 30m 02s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5af6a81fd52a47d598e2c21452b2f1dc : TIMED_OUT in 2h 33m 09s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/c4574b0a39f742eab4bf2242abee66d1 : TIMED_OUT in 2h 33m 10s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"fc4178d08075af4f9eade57039dbe5fdf5e2d27d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-02-27 01:00:58.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"b4e41a72c4320ba8b344c91094777ad3af7b24cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 04:06:10.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0af5f5e525c4e33a5aad2acd6761ff5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f275c996b6c749c7a31afb8348aeb453 : SUCCESS in 13m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbbe26b1ce4d4319a56aae2b7c2d0aa0 : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5021f7cfa98b4a53b4050ee0a0f6f6b0 : SUCCESS in 6m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/abe5368e2010427ca0d511b43acc96e1 : SUCCESS in 4m 19s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/138bf682505a455394ae02dbc8ad212d : SUCCESS in 4m 19s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/a48fc5e4b688494b951a59281211c6e5 : SUCCESS in 7m 41s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/82fcd694c129466abbfaaafc9a53a58a : SUCCESS in 3m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8edc1ecc827d4f4c87ecd3a81b3b9461 : SUCCESS in 8m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7d6f2d41e8a34550b0067079abe76c40 : SUCCESS in 7m 23s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/336932336fab45dd889be84a58b2aecf : SUCCESS in 6m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c7913168c42241619e3168026d9f5446 : SUCCESS in 1h 17m 29s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/3206023f5a824105912daa6a5005adf2 : SUCCESS in 1h 51m 40s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/7cd843b4de2043ca9eff9e7078d55cc1 : SUCCESS in 1h 11m 06s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c6a61f95b6cd4f93b412a4f76ee99837 : SUCCESS in 1h 45m 14s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f6493fef9c6b414fa0939cb86b4a0c4f : SUCCESS in 58m 32s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/705630ff6aa041188f9da52ed0dfc161 : SUCCESS in 1h 53m 23s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a62f49dbb2db402a9c5a41f8627d9840 : SUCCESS in 1h 06m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b43f763efe6a43c4a7127c825965740a : SUCCESS in 39m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9ebb0963681045feb0fac2732570cf7a : SUCCESS in 47m 20s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/11bb1362868f4a88b7d740108b7a8033 : SUCCESS in 1h 03m 42s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/82c3f8300c8242de83e72436afe3b801 : SUCCESS in 33m 36s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ba251d88b1ce4796ae4be518d4d65db3 : TIMED_OUT in 2h 32m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ab47bed7983d4f07a54cd41dd7997be1 : TIMED_OUT in 3h 02m 56s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/303040a089f14d60ba15c4be3f995d38 : SUCCESS in 2h 31m 50s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/a6b7497641c247cc8d3bb3503ca4890d : TIMED_OUT in 2h 35m 01s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"769b385b95093fac97df0be2d4e644d130a205c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 04:06:50.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"1022060f4c464a883148476951b86b81125ce271","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 06:17:12.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2097dcfe194a4224bad3b9ec9ecb44b5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a1cc4eb9b2743cdbd025906231de267 : SUCCESS in 5m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ae5e86858517403b86a13e17d2f5072e : SUCCESS in 4m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c89277e1662447cf8d92e19dc98056f2 : SUCCESS in 6m 19s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b7a0088761054278884c3c07fc18f1bf : SUCCESS in 4m 27s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9f2f336f0305439c9ef21ffa64fa1af9 : SUCCESS in 7m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1bb5af0b268e4b39adf5910ad9e70b64 : SUCCESS in 44m 09s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/da2de41a7a4c4b9cbeca2061d22d5d5b : SUCCESS in 2h 07m 09s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/a11f63d5e48d401887d68755a4c6caab : SUCCESS in 1h 14m 50s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/168af3b668304a7eacbdbe987f373edf : SUCCESS in 2h 08m 58s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/df158fb719aa4b6e88e889d3725d08ab : SUCCESS in 58m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5555fc5444f049289df541a5590a3161 : SUCCESS in 29m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/69b3f823921442a1843da2e47059d33f : SUCCESS in 56m 15s","accounts_in_message":[],"_revision_number":9},{"id":"1ac29b45f6e819d2e708c2b712bda9026b5e1deb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 06:17:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"5396c627f4183259f9a2a6c4962fda591571795c","revisions":{"5770a0c25f69bd0a2119976546c6c08d428bf640":{"kind":"REWORK","_number":1,"created":"2024-08-13 08:57:17.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/82/926182/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/82/926182/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/926182/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/926182/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/926182/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/82/926182/1"}}},"commit":{"parents":[{"commit":"a4d33a326da18631c7a465eb2e3c9fe60db91200","subject":"Handle custom security groups in controller","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a4d33a326da18631c7a465eb2e3c9fe60db91200"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-05-21 08:30:59.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-08-13 08:28:36.000000000","tz":120},"subject":"Added RBAC for vip_sg_ids","message":"Added RBAC for vip_sg_ids\n\nChange-Id: I4a15c1a19c5bb0b39494381763f68c056861b003\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5770a0c25f69bd0a2119976546c6c08d428bf640"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5770a0c25f69bd0a2119976546c6c08d428bf640"}]},"branch":"refs/heads/master"},"69b4fc26f25e4b2edc1d381b7640b5faec6a8462":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-10-11 14:26:55.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/82/926182/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/82/926182/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/926182/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/926182/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/926182/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/82/926182/2"}}},"commit":{"parents":[{"commit":"38953ad9c470ae21964cecf6dbaa9c8407b6ca15","subject":"Handle custom security groups in controller","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/38953ad9c470ae21964cecf6dbaa9c8407b6ca15"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-05-21 08:30:59.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-10-11 14:23:26.000000000","tz":120},"subject":"Added RBAC for vip_sg_ids","message":"Added RBAC for vip_sg_ids\n\nChange-Id: I4a15c1a19c5bb0b39494381763f68c056861b003\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/69b4fc26f25e4b2edc1d381b7640b5faec6a8462"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/69b4fc26f25e4b2edc1d381b7640b5faec6a8462"}]},"branch":"refs/heads/master"},"980e1c4bfb62aa77c52020f9a086876bc9f039f8":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-10-14 08:46:02.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/82/926182/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/82/926182/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/926182/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/926182/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/926182/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/82/926182/3"}}},"commit":{"parents":[{"commit":"e3ea9a7e411fd750d774a96b7c9935a5f800fb0a","subject":"Handle custom security groups in controller","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e3ea9a7e411fd750d774a96b7c9935a5f800fb0a"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-05-21 08:30:59.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-10-14 08:43:12.000000000","tz":120},"subject":"Added RBAC for vip_sg_ids","message":"Added RBAC for vip_sg_ids\n\nChange-Id: I4a15c1a19c5bb0b39494381763f68c056861b003\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/980e1c4bfb62aa77c52020f9a086876bc9f039f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/980e1c4bfb62aa77c52020f9a086876bc9f039f8"}]},"branch":"refs/heads/master"},"c5c44e066c61d47ae03ab55c4d0aa067b3f2abca":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-10-15 12:50:00.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/82/926182/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/82/926182/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/926182/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/926182/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/926182/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/82/926182/4"}}},"commit":{"parents":[{"commit":"8f61be345ecc45a587b4664fc0372bde5489f4f7","subject":"Handle custom security groups in controller","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8f61be345ecc45a587b4664fc0372bde5489f4f7"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-05-21 08:30:59.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-10-15 12:42:11.000000000","tz":120},"subject":"Added RBAC for vip_sg_ids","message":"Added RBAC for vip_sg_ids\n\nChange-Id: I4a15c1a19c5bb0b39494381763f68c056861b003\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c5c44e066c61d47ae03ab55c4d0aa067b3f2abca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c5c44e066c61d47ae03ab55c4d0aa067b3f2abca"}]},"branch":"refs/heads/master"},"96879fc816e40fe7a01f1e3b6375a0eff68e1fcd":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-11-08 06:19:37.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/82/926182/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/82/926182/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/926182/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/926182/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/926182/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/82/926182/5"}}},"commit":{"parents":[{"commit":"93f7ac09fc06b2bce4e27bd2255208254c5a6676","subject":"Handle custom security groups in controller","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/93f7ac09fc06b2bce4e27bd2255208254c5a6676"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-05-21 08:30:59.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-11-08 06:04:14.000000000","tz":60},"subject":"Added RBAC for vip_sg_ids","message":"Added RBAC for vip_sg_ids\n\nChange-Id: I4a15c1a19c5bb0b39494381763f68c056861b003\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/96879fc816e40fe7a01f1e3b6375a0eff68e1fcd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/96879fc816e40fe7a01f1e3b6375a0eff68e1fcd"}]},"branch":"refs/heads/master"},"6f21682bd9ee6efa07613be7234c50b3eb2f5b57":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-01-08 21:12:13.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/82/926182/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/82/926182/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/926182/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/926182/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/926182/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/82/926182/6"}}},"commit":{"parents":[{"commit":"52531b1fb677a37082c4db2da9ecbd3c3543f6cc","subject":"Handle custom security groups in controller","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/52531b1fb677a37082c4db2da9ecbd3c3543f6cc"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-05-21 08:30:59.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-01-08 21:09:48.000000000","tz":60},"subject":"Added RBAC for vip_sg_ids","message":"Added RBAC for vip_sg_ids\n\nChange-Id: I4a15c1a19c5bb0b39494381763f68c056861b003\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6f21682bd9ee6efa07613be7234c50b3eb2f5b57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6f21682bd9ee6efa07613be7234c50b3eb2f5b57"}]},"branch":"refs/heads/master"},"25149d4f12da8842feca45ed90c3e296426adfb1":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-01-16 10:22:30.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/82/926182/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/82/926182/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/926182/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/926182/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/926182/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/82/926182/7"}}},"commit":{"parents":[{"commit":"3fc7f5e7260d73b750d068880651d872a9e73d51","subject":"Handle custom security groups in controller","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3fc7f5e7260d73b750d068880651d872a9e73d51"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-05-21 08:30:59.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-01-16 10:04:01.000000000","tz":60},"subject":"Added RBAC for vip_sg_ids","message":"Added RBAC for vip_sg_ids\n\nChange-Id: I4a15c1a19c5bb0b39494381763f68c056861b003\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/25149d4f12da8842feca45ed90c3e296426adfb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/25149d4f12da8842feca45ed90c3e296426adfb1"}]},"branch":"refs/heads/master"},"09664903c1c0dacfcd1737a1b03171bfe6315e98":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2025-02-19 09:05:16.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/82/926182/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/82/926182/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/926182/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/926182/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/926182/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/82/926182/8"}}},"commit":{"parents":[{"commit":"228f496426a4cc51231688061f837e8165247fd4","subject":"Handle custom security groups in controller","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/228f496426a4cc51231688061f837e8165247fd4"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-05-21 08:30:59.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-02-19 09:02:35.000000000","tz":60},"subject":"Added RBAC for vip_sg_ids","message":"Added RBAC for vip_sg_ids\n\nChange-Id: I4a15c1a19c5bb0b39494381763f68c056861b003\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/09664903c1c0dacfcd1737a1b03171bfe6315e98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/09664903c1c0dacfcd1737a1b03171bfe6315e98"}]},"branch":"refs/heads/master"},"5396c627f4183259f9a2a6c4962fda591571795c":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2025-02-19 13:19:40.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/82/926182/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/82/926182/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/926182/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/926182/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/926182/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/82/926182/9"}}},"commit":{"parents":[{"commit":"0e440c9e7b79c45959e1d857ea7077922b0a0237","subject":"Handle custom security groups in controller","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0e440c9e7b79c45959e1d857ea7077922b0a0237"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-05-21 08:30:59.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-02-19 13:15:11.000000000","tz":60},"subject":"Added RBAC for vip_sg_ids","message":"Added RBAC for vip_sg_ids\n\nChange-Id: I4a15c1a19c5bb0b39494381763f68c056861b003\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5396c627f4183259f9a2a6c4962fda591571795c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5396c627f4183259f9a2a6c4962fda591571795c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
