)]}'
{"id":"openstack%2Foctavia~926564","triplet_id":"openstack%2Foctavia~master~Iae2b075de5412bb183db6a21f7f8376801853e00","project":"openstack/octavia","branch":"master","topic":"bug/2077348","attention_set":{},"removed_from_attention_set":{"6577":{"account":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"last_update":"2024-09-05 13:56:36.000000000","reason":"Change was submitted"},"10273":{"account":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"last_update":"2024-09-05 13:56:36.000000000","reason":"Change was submitted"},"15025":{"account":{"_account_id":15025,"name":"Michael Johnson","email":"Michael.Johnson@sas.com","username":"mikjo"},"last_update":"2024-09-05 13:56:36.000000000","reason":"Change was submitted"},"37241":{"account":{"_account_id":37241,"name":"Evgeniy Bykov","display_name":"Evgeniy Bykov","email":"chypakabre@gmail.com","username":"evbykov"},"last_update":"2024-09-05 13:56:36.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-09-05 13:56:36.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-09-05 11:33:43.000000000","reason":"removed on reply"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2024-09-05 13:56:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iae2b075de5412bb183db6a21f7f8376801853e00","subject":"Fix loadbalancer stuck in cascade delete","status":"MERGED","created":"2024-08-19 17:42:41.000000000","updated":"2024-09-05 13:57:43.000000000","submitted":"2024-09-05 13:56:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"926564-bug/2077348","meta_rev_id":"48767b83d99ed77679db949d070260dddf843a6d","_number":926564,"virtual_id_number":926564,"owner":{"_account_id":37241,"name":"Evgeniy Bykov","display_name":"Evgeniy Bykov","email":"chypakabre@gmail.com","username":"evbykov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-05 13:56:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-09-05 11:33:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2024-09-04 22:29:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-09-05 11:33:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-09-05 11:33:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":15025,"name":"Michael Johnson","email":"Michael.Johnson@sas.com","username":"mikjo"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"reviewers":{"REVIEWER":[{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":15025,"name":"Michael Johnson","email":"Michael.Johnson@sas.com","username":"mikjo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-19 20:47:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-26 10:58:14.000000000","updated_by":{"_account_id":37241,"name":"Evgeniy Bykov","display_name":"Evgeniy Bykov","email":"chypakabre@gmail.com","username":"evbykov"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2024-08-26 10:58:14.000000000","updated_by":{"_account_id":37241,"name":"Evgeniy Bykov","display_name":"Evgeniy Bykov","email":"chypakabre@gmail.com","username":"evbykov"},"reviewer":{"_account_id":15025,"name":"Michael Johnson","email":"Michael.Johnson@sas.com","username":"mikjo"},"state":"REVIEWER"},{"updated":"2024-08-26 10:58:14.000000000","updated_by":{"_account_id":37241,"name":"Evgeniy Bykov","display_name":"Evgeniy Bykov","email":"chypakabre@gmail.com","username":"evbykov"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2024-08-26 10:58:14.000000000","updated_by":{"_account_id":37241,"name":"Evgeniy Bykov","display_name":"Evgeniy Bykov","email":"chypakabre@gmail.com","username":"evbykov"},"reviewer":{"_account_id":10273,"name":"Adam Harwell","email":"flux.adam@gmail.com","username":"rm_you"},"state":"REVIEWER"},{"updated":"2024-09-04 10:52:00.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"state":"REVIEWER"},{"updated":"2024-09-04 22:29:07.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"a1915fe309964dbb6e9eac9adec3c0a2698fddec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37241,"name":"Evgeniy Bykov","display_name":"Evgeniy Bykov","email":"chypakabre@gmail.com","username":"evbykov"},"date":"2024-08-19 17:42:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2275bddd6ed4692944c7071b956656af99afa53a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 20:47:47.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/335d312aa34e4d7ba51aad857406a8b1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8849df23366a4cdda6001b163aef5e93 : SUCCESS in 10m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3b26337b85b46b38710b403c5c9150e : FAILURE in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa5216f7135d4adb8c3d80b10bf82d67 : SUCCESS in 4m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e17aec134b524ef4983d08f0bb4b8b60 : SUCCESS in 4m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6da5ed4fb66c4716a60d1db45ad175f3 : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b672cd85d2a4b5c9ff7e9f3b53f2206 : SUCCESS in 12m 44s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/272d284bfcb94a788ddc83a78eb35ee0 : SUCCESS in 4m 38s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/f068c36e2b154c179e087ce91e703a18 : FAILURE in 7m 01s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/07e1c0af9e3740ca801d7c64317fc575 : SUCCESS in 3m 56s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/df5a7f201eec444f80da1ff33612e69c : SUCCESS in 7m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f5b2cb1910294526ae968b719408cb1d : SUCCESS in 8m 44s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/21c5bc3236614354a62a2a6f4d2e5dba : SUCCESS in 8m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6ac8c3b1486d4a10817cb5ce03b2b8b6 : SUCCESS in 1h 28m 43s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/9096a3162ff2411cba942432d239bbad : SUCCESS in 1h 46m 02s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/5d7f16f221aa4f2da4a1629e3eef87f4 : SUCCESS in 1h 27m 05s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/d21d6015245644c3bf3e16c2ed55fcc8 : SUCCESS in 2h 17m 57s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/bee1ef2c62214d1eacdf26a244299af8 : SUCCESS in 59m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/663a14fe285a4432967db08f30b1ab5b : SUCCESS in 35m 35s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/67c4ec2c65014f4f819b1fb4c329c2e0 : SUCCESS in 56m 52s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/b580ad86868c47c5a8d8f691aa48a813 : SUCCESS in 1h 06m 41s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/970b8e76fbfb482a932e447f1d98d59c : SUCCESS in 32m 58s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/662eafe648a2401cb4818e9560a7ca6a : TIMED_OUT in 2h 33m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/edb141f34df5446d8a1fc8be92c20bbb : TIMED_OUT in 3h 02m 46s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/aacb391029284ec79cf5e39c0005f650 : TIMED_OUT in 2h 34m 00s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/aac0359077744de7818dc0fb4eefd641 : SUCCESS in 2h 29m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"01f6495bbf35e2029dd6558e3981cb1df008b952","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37241,"name":"Evgeniy Bykov","display_name":"Evgeniy Bykov","email":"chypakabre@gmail.com","username":"evbykov"},"date":"2024-08-20 08:09:42.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"20841c00607e0b7e7c2b484a5650bfd4c61b58ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 11:09:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6bc0afb27c9544f0ac8de0b8f673660b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c43fc9c732f940fd8599e4f3bc872a41 : SUCCESS in 11m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1743240d0f174c9cb6efc6abb7f77a6f : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b067194b653147249b39af613d9e069c : SUCCESS in 6m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9ef0c58931544b71b09c8142a872e799 : SUCCESS in 5m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dd280c60bfdb458a8fc1a1b1ce126394 : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d65fc6b8df304fd3b1596c8bb89b09b3 : SUCCESS in 12m 25s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/cde7cb9cd6cb4579a3e7f4dc4ee26b73 : SUCCESS in 5m 55s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/71f5efba8f454d868763672160d65012 : FAILURE in 8m 38s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5712f565803340d3b29d9105d62735b9 : SUCCESS in 4m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1b34114f029c4ffbb6758aa20b04216d : SUCCESS in 8m 44s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/21ae7241060a499fb5a2402200063abc : SUCCESS in 8m 35s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2fa0368e9dda4854993831f8bc835f64 : SUCCESS in 8m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3c2a6e992d0b41df8bcfda80c0187cea : SUCCESS in 1h 29m 24s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/fd9a560c81b54abb877d6012b5386d13 : SUCCESS in 2h 23m 35s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/fee8d6b1a3e242a283c4089718603593 : SUCCESS in 1h 29m 00s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/efc88945eaaa4c95b2bb2c3af470879f : SUCCESS in 1h 53m 18s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/72217d50774a457e954d950e4ac97fd5 : SUCCESS in 1h 01m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b91ff2257efe4bdea8361672ee7e5c9f : SUCCESS in 39m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/17114822de784df183c8c280134633d6 : SUCCESS in 45m 09s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/1e34ef4c5c2a4e09897dfdd4edaa2154 : SUCCESS in 1h 08m 41s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5acd9a066ce74453926a68f1623665b7 : SUCCESS in 36m 15s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b0f6107468bd4b32a6a619babd2fd5b2 : SUCCESS in 2h 03m 59s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/59dd693f94a54552b2c8ca64b0c74f5f : SUCCESS in 2h 58m 21s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8a1109b43f5a46ea829748189049a733 : TIMED_OUT in 2h 35m 25s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/e715fddc77fe4411a05f9f30d5f8ee54 : SUCCESS in 2h 28m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4579395d1e69ed751bc7c0cc4ecdbf2870b978e1","author":{"_account_id":37241,"name":"Evgeniy Bykov","display_name":"Evgeniy Bykov","email":"chypakabre@gmail.com","username":"evbykov"},"date":"2024-08-20 12:01:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4929c29e1bcae191061ebd8a4c9ad024b9c42c70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 15:03:53.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/998166724d88418dbdee25b87988985e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d20801c9cefb45eea4a9fcdb3b67fdfe : SUCCESS in 10m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b8f5166ce754483811fef653ba20fbc : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/406e358d89a24529b3ea4709994ab668 : SUCCESS in 6m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/943ee6f35fa14229986255f54f9636a1 : SUCCESS in 4m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6f634c580eb440b592214ca83e55afb0 : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f7124ecd482484b9a8705a405092732 : SUCCESS in 11m 36s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/e1765eb02d5f4917a4a1571cae1b3736 : SUCCESS in 4m 58s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/214b21e457614378bb15160b95397b9f : FAILURE in 7m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/195974a1aeff40c7a64e6a939a283dc3 : SUCCESS in 5m 18s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/31716d78a67f43a2b57ae50892b4504e : SUCCESS in 7m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/feb820f1cabe46499265ec058499a2c5 : SUCCESS in 8m 20s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1faf450200e74fc2b2738777258c037f : SUCCESS in 6m 53s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/44987e7d7d054380988ac8bb2923f92c : SUCCESS in 1h 32m 16s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/5957e93e3c274ca89f5a4fbc642dd023 : SUCCESS in 1h 46m 01s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/3cb18d432b9040379092db15b63f00c3 : SUCCESS in 53m 46s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/93d2ebc8b2694ecb85505210c35b0293 : SUCCESS in 2h 31m 56s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ac1c28f96e2945b88071dadf4a392b47 : SUCCESS in 1h 24m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/231e3c8069e64c6fb07a034b563c1db6 : SUCCESS in 51m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8829f10eda594a76819ee192712b6b44 : SUCCESS in 52m 52s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/23bc1d1ababf4d3793b5e6fdeca5b1c7 : SUCCESS in 1h 12m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/273f1fa844a24fea94c4ac03797c468b : SUCCESS in 40m 46s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2a1be66d2b7a46db9cd45b36d0316225 : TIMED_OUT in 2h 34m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/78e40ce426234b0f8504065fe5f8b18e : SUCCESS in 3h 00m 39s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/245369c5442e4dc5899827597310e523 : TIMED_OUT in 2h 34m 44s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/f97425937b9e4579877005a0d7c74a90 : TIMED_OUT in 2h 34m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b12ecfaf38dcb47437d84b0c8ff577c9748ee734","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37241,"name":"Evgeniy Bykov","display_name":"Evgeniy Bykov","email":"chypakabre@gmail.com","username":"evbykov"},"date":"2024-08-25 20:34:54.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"6e6eb38c81a8d4db1cace23d598cbc377bfb9be2","author":{"_account_id":37241,"name":"Evgeniy Bykov","display_name":"Evgeniy Bykov","email":"chypakabre@gmail.com","username":"evbykov"},"date":"2024-08-25 20:37:36.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"094726de314062853f09a7d67c53b9b68c0bf763","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-25 23:41:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d0b6a11179f40d69d838d3f6978ad7d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c29f4f081bcc4f9e913b93a2af683184 : SUCCESS in 9m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4849f80aa55f44f1bd160377ebf019a0 : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7711c26541c1436db04c5edb1e4d03d8 : SUCCESS in 4m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3636c9050d58418494a1a0a9182af4d0 : SUCCESS in 4m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1a1fdd3d9eab496da651124dbfa58106 : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fa2291dd8124411a0882eb8fb3c172a : SUCCESS in 9m 09s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/031f486bdb974a0eb76ab4d87eece283 : SUCCESS in 4m 07s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/5b93b1e9223a4cb29be8c271253d355d : SUCCESS in 7m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/adaa3e0554ef4b58bc0c2289af5eca52 : SUCCESS in 3m 53s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a72ad8ff8b1f4effb0ada902aacaa5e1 : SUCCESS in 7m 19s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3e60420f736b47fbb29fc5ff21a43dd7 : SUCCESS in 5m 14s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/9814397fed3647c787ded66b6c967f53 : SUCCESS in 7m 18s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5dd2aa8e2d464a39971277d2ccf7e56c : SUCCESS in 57m 57s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/c3d67a7c0d8e4c2794e470abf2c8c6f9 : SUCCESS in 1h 45m 37s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/a7dca073cf2e4d80b6b110a3abbf90e6 : SUCCESS in 57m 49s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3eaca605204448fdb3269957558c75a5 : SUCCESS in 1h 50m 07s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c11ff00219074770b1f67a82c5419784 : SUCCESS in 1h 00m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0d4aa47c61cc4269ad5e1703b98eda35 : SUCCESS in 34m 10s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/36f15443d4984c1aafbd3450bb5dea0e : SUCCESS in 39m 39s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/75adc46e468b47598318730023b1de19 : SUCCESS in 52m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/265a560a809f4297b0ab023075106da5 : SUCCESS in 34m 46s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/56c275e87c6f4888a633ffe15af2f45d : SUCCESS in 2h 07m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/07f60874dc85459c92a72c5c4484c386 : SUCCESS in 2h 59m 11s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/773c9669f913430bbae20299a76df414 : SUCCESS in 2h 08m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/c636cba8094d49ce8dd47cb8e003aec8 : SUCCESS in 2h 14m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f28dca1bf55ae8531dba53d2fe970a6236da664e","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-09-04 10:52:00.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"898a24483eef0f30ae489203caf7e5b2015fcf73","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-09-04 13:35:25.000000000","message":"Patch Set 3: Backport-Candidate+2 Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"43122ef1365303e0770023992a498c57e22e14a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37241,"name":"Evgeniy Bykov","display_name":"Evgeniy Bykov","email":"chypakabre@gmail.com","username":"evbykov"},"date":"2024-09-04 14:02:27.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Backport-Candidate+2 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"3f8cb7207382b43701865b1cf6b8e8d15a85c9a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 14:57:04.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3d33162ad1a44b56a2a286946114e9f9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ee47ffd005341f19c5174a94ce586ba : SUCCESS in 10m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5e0c708de8b4388a710e26658c9fe25 : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da4e1bb0dc4e4205a44aa59dd685d22f : SUCCESS in 4m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/98512b5fcd124899a401f017d1b925bc : SUCCESS in 4m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4a153686a5ca41fdad5d6e26531eff33 : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b10390503294f278f1bd9949ab34f32 : SUCCESS in 11m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d4b61a9869d74b7092174fc2a23a9e01 : SUCCESS in 6m 25s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/a85124d9a1eb4f719c6271ebeeb7802d : SUCCESS in 5m 11s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/feb6f378e4894b43982874ad85dd5594 : SUCCESS in 7m 50s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2caf39f7f26049d49800ee1c53143e51 : SUCCESS in 4m 23s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2eda6d7923e241d6977edfeeb1c999f7 : SUCCESS in 7m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fbf04964dd904420b82e0dc196657e66 : SUCCESS in 8m 51s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/4f6a1955a6a54c8d9dac68be54b931eb : SUCCESS in 9m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f9140a0182604ed78cd0411272f05973 : FAILURE in 19m 13s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/a3a8044799d44ab1a10df3ff1e859438 : FAILURE in 21m 44s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/7690f02a8ef74c6fb8377d1355d0377a : FAILURE in 22m 03s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/2aacbed44a164990af32c25ce5770d7b : FAILURE in 22m 19s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e94280d4855440d584e703bdafad5ba2 : FAILURE in 32m 50s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/02cd8aecb9be4199824b69f8a2d57443 : FAILURE in 36m 51s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e949972e92754cdba19c1c0d200f6724 : FAILURE in 49m 14s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/6220e0f1159c44f0bb5aa82d47311923 : NODE_FAILURE Node(set) request 300-0025103481 failed in 0s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/785a3fa9593d49069b98ebe74eef73dd : FAILURE in 21m 53s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/92828b92891844c0b84d4925a2b6ba15 : FAILURE in 22m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/227118788215409c8862544abfb275b5 : FAILURE in 32m 22s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f0088708eef549eea70a994256c335d0 : FAILURE in 21m 55s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/f094f372e8fc4230927a24bbd63378c8 : FAILURE in 42m 44s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"bdc5a91748a14cc5eace3fd8ce1b05f4fb019e58","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-09-04 15:09:28.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"313cc57cdac680fc292d3595de559a0e717d3fd7","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-09-04 15:16:00.000000000","message":"Patch Set 4: Backport-Candidate+2 Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"19bc34279aa7a5b7f555395ad46235551915f53c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 18:17:32.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3d5fe27f3494e14b16365a42eb18223\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f29ec8f15404314aaad289c0ac82e38 : SUCCESS in 11m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2835a8486214af49cdb43b756096b02 : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f9da2bf85e64dbf9c778ea37ea51995 : SUCCESS in 4m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/89ed31a414494fdf95bdbf326682e6a4 : SUCCESS in 4m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ae1b158c7003466c86b26d14550ac29d : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5255ce72dccc46fdab741b2981673470 : SUCCESS in 9m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc5eba909c074633b293065cd6f27f57 : SUCCESS in 4m 47s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/9d52ff2e2ea34bc7abcd302b3ecec987 : SUCCESS in 4m 13s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/d25cde3875664f9daa6079d0f5747ac8 : SUCCESS in 7m 36s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/66b59ec7ebf44638a4f714704e53e9a7 : SUCCESS in 4m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4d6246e117ba482fb8f26c497aa67082 : SUCCESS in 8m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e33696a98d0a40178d22e96ca0bb169a : SUCCESS in 7m 43s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a1a31ba54bb84af78ef076f161083039 : SUCCESS in 9m 09s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f57a04036c3f48ecb98109c5834f316b : SUCCESS in 1h 20m 25s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/597ccd8fcaef4eb5a14ec4b25d04c19a : SUCCESS in 2h 26m 14s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/3af603dd1a5a43bd8b8c1742a655849e : SUCCESS in 1h 14m 41s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/bd17771d19c342ffbafaae65b315e608 : SUCCESS in 1h 47m 10s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ecbf9d837b7b4be88b3e3d0d18fd2b14 : SUCCESS in 1h 07m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3ce09f9195494448965c25dbfa4df38f : SUCCESS in 36m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/43e2116d49024c359aaa057b195ac7b7 : SUCCESS in 37m 51s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/6916cf9d69b641b4987fd0b4e72ed005 : SUCCESS in 1h 14m 40s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4c71862e77264ac4a87a456a02a8e982 : SUCCESS in 34m 23s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/058c78aa246944f7880e909dc85a917b : TIMED_OUT in 2h 33m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d68cbf0aa41f464aba6f2ef607ee9d52 : TIMED_OUT in 3h 04m 17s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b31a6caed3b94bbfb974120d8add0763 : TIMED_OUT in 2h 33m 07s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/916474dbed2c4162beb8692f36ed80c7 : TIMED_OUT in 2h 33m 21s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e126cd0d48c94b98b03d859a8705b9c19923220a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-09-04 22:29:00.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Backport-Candidate+2 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"46292eada907e0cda49be724e6933a314ca369f6","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-09-04 22:29:07.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2098cb601e19fe6bb1694dbe5c0c3621b39e58ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-05 01:31:24.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4b581b5712a45f9b7a63f9fc8facce9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c73da12f30c469db387acf056cc435d : SUCCESS in 9m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75edb857b28c4a5fbb8bc21d2562200a : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/999a55cb059c490a9fe00bfe64d92203 : SUCCESS in 4m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/04cae573c3c3428d983100de982a4beb : SUCCESS in 4m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e01abd3ea1e2469882fcc47886efa080 : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2295becf9dd4ab4880a1ad60331f3b9 : SUCCESS in 11m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/43a65ed64d7a4d6d8c89c30c6d0ff988 : SUCCESS in 4m 53s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/e3008032d38e4a3aa4d148db23be4aa5 : SUCCESS in 4m 40s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/870a37dbadc446fe9a9092ad4ae94821 : SUCCESS in 8m 24s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/35875442c6d24c9a842a0103a7340407 : SUCCESS in 4m 10s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/734a57330e6948858097bb600d264fb4 : SUCCESS in 7m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c29c754a4f0f4257af1346efea1e7872 : SUCCESS in 6m 38s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/96be0c44e18c47a0939eae3553810c00 : SUCCESS in 6m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5f68bea61f624b8e8ef59b668f600b69 : SUCCESS in 1h 34m 00s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/ef8c6f6293124ea4b102294f47d23b8f : SUCCESS in 1h 48m 11s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/3df23384f6ba4528b178d6c19a7178b7 : SUCCESS in 1h 00m 58s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/17d0fa2252954e088e20245e98b581be : SUCCESS in 1h 51m 48s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/6b09a3936c1449e79fc697223f251c30 : SUCCESS in 1h 09m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bcb22121a0b64e5cb2fff84b3930fb07 : SUCCESS in 36m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0d47235077d646bb9844cc2cd2818c7c : SUCCESS in 40m 03s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/f2ad8619e6274b3e9a20e3aad5ae9813 : SUCCESS in 58m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f3e94e97b5c044d4a35475be8f46a62c : SUCCESS in 32m 59s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8162c89382074146998da384441f911e : SUCCESS in 2h 10m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/732c55c56c6f4eb88206761f5832935b : SUCCESS in 3h 00m 55s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0263148c840b45c5865d5bd5c2eaaae4 : SUCCESS in 2h 10m 09s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/6b07abbb6bcd4dcab87d99f76f51feee : TIMED_OUT in 2h 34m 36s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"19cd0e5cfd4d0b0b5e96c4d144858b8888b6f49a","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-09-05 11:33:43.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"1957c3774c72857d1ef40afbce7511612ff0de89","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-09-05 11:33:49.000000000","message":"Patch Set 5: Backport-Candidate+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"7464c7863af7b3ab1774336a055cc96896ee7560","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-05 11:34:28.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"33b601f4de9df8289f53d7fcf3f36f82d75477a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-05 13:56:36.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1efefdba4aaf47c7b9023fa76e511257\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a03d809f1ea47c2bd92b636a90f02f6 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/95e43909dd794e0b8d89d164af5402db : SUCCESS in 4m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/da81f3a018734e98938535971d69c385 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/330e13bb6c0d4e31b71c1c6d3ef957f2 : SUCCESS in 13m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db7462363d9c41a2ae795caa87b16f8a : SUCCESS in 5m 03s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/85fcbf023b824b68b3eac7c3a37e732a : SUCCESS in 3m 40s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/19709b0911994b24a31155cb04cb9202 : SUCCESS in 6m 21s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/183bebecebec4f44a5f1db8602b4310d : SUCCESS in 1h 21m 27s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/d19ae77c6ee9458f874cfc4091b3b8bb : SUCCESS in 2h 20m 24s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/318417fabd244824a21ca8f7bb93ffa8 : SUCCESS in 1h 22m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/80f7cb6b84504cf891d937babc4cd122 : SUCCESS in 47m 29s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/70087dc1c350463bba8b0b995cf15f0c : SUCCESS in 57m 19s","accounts_in_message":[],"_revision_number":5},{"id":"665bfcf8f637ee6aa21b84a9f9eeeb6835f74bad","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-05 13:56:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"48767b83d99ed77679db949d070260dddf843a6d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-05 13:57:43.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73e7131a1cd34c3ca1fe442d266a1dfc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbc647491f034e8aa66d2bd44891e41a : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3bf00296f5644c178e2979e3bac2e70c : SUCCESS in 42s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"ec9a50599012b99deaf8bb26683fedcd381db1a6","revisions":{"6f11c3b382895468a0547cb1a5a6a7ebe785202b":{"kind":"REWORK","_number":1,"created":"2024-08-19 17:42:41.000000000","uploader":{"_account_id":37241,"name":"Evgeniy Bykov","display_name":"Evgeniy Bykov","email":"chypakabre@gmail.com","username":"evbykov"},"ref":"refs/changes/64/926564/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/64/926564/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/926564/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/926564/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/926564/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/64/926564/1"}}},"commit":{"parents":[{"commit":"f68cafbb7201522e6140e2dc05317d06225bb82a","subject":"Merge \"Remove sysvinit/upstart support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f68cafbb7201522e6140e2dc05317d06225bb82a"}]}],"author":{"name":"Evgeniy Bykov","email":"chypakabre@gmail.com","date":"2024-08-19 17:20:38.000000000","tz":180},"committer":{"name":"Evgeniy Bykov","email":"chypakabre@gmail.com","date":"2024-08-19 17:42:16.000000000","tz":180},"subject":"Fix loadbalancer stuck in cascade delete","message":"Fix loadbalancer stuck in cascade delete\n\nFix loadbalancer stuck in PENDING_DELETE in cascade delete\nwith TERMINATED_HTTPS listener if TLS storage not available\n\nCloses-Bug: #2077348\nChange-Id: Iae2b075de5412bb183db6a21f7f8376801853e00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6f11c3b382895468a0547cb1a5a6a7ebe785202b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6f11c3b382895468a0547cb1a5a6a7ebe785202b"}]},"branch":"refs/heads/master"},"cb9f6754bb7fde319319d346dca9dba00843e6b6":{"kind":"REWORK","_number":2,"created":"2024-08-20 08:09:42.000000000","uploader":{"_account_id":37241,"name":"Evgeniy Bykov","display_name":"Evgeniy Bykov","email":"chypakabre@gmail.com","username":"evbykov"},"ref":"refs/changes/64/926564/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/64/926564/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/926564/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/926564/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/926564/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/64/926564/2"}}},"commit":{"parents":[{"commit":"f68cafbb7201522e6140e2dc05317d06225bb82a","subject":"Merge \"Remove sysvinit/upstart support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f68cafbb7201522e6140e2dc05317d06225bb82a"}]}],"author":{"name":"Evgeniy Bykov","email":"chypakabre@gmail.com","date":"2024-08-19 17:20:38.000000000","tz":180},"committer":{"name":"Evgeniy Bykov","email":"chypakabre@gmail.com","date":"2024-08-20 08:08:18.000000000","tz":180},"subject":"Fix loadbalancer stuck in cascade delete","message":"Fix loadbalancer stuck in cascade delete\n\nFix loadbalancer stuck in PENDING_DELETE in cascade delete\nwith TERMINATED_HTTPS listener if TLS storage not available\n\nCloses-Bug: #2077348\nChange-Id: Iae2b075de5412bb183db6a21f7f8376801853e00\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cb9f6754bb7fde319319d346dca9dba00843e6b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cb9f6754bb7fde319319d346dca9dba00843e6b6"}]},"branch":"refs/heads/master"},"dcef779c5e2ca3161987cacabb7f7137f727ed27":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2024-08-25 20:34:54.000000000","uploader":{"_account_id":37241,"name":"Evgeniy Bykov","display_name":"Evgeniy Bykov","email":"chypakabre@gmail.com","username":"evbykov"},"ref":"refs/changes/64/926564/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/64/926564/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/926564/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/926564/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/926564/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/64/926564/3"}}},"commit":{"parents":[{"commit":"a4317a80d61ce2d8e699ea7ca17f161ecc4db8ee","subject":"Fix failover when using SRIOV VIP","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a4317a80d61ce2d8e699ea7ca17f161ecc4db8ee"}]}],"author":{"name":"Evgeniy Bykov","email":"chypakabre@gmail.com","date":"2024-08-19 17:20:38.000000000","tz":180},"committer":{"name":"Evgeniy Bykov","email":"chypakabre@gmail.com","date":"2024-08-25 20:34:44.000000000","tz":180},"subject":"Fix loadbalancer stuck in cascade delete","message":"Fix loadbalancer stuck in cascade delete\n\n- Fix loadbalancer stuck in PENDING_DELETE in cascade delete\nwith TERMINATED_HTTPS listener if TLS storage not available\n\nCloses-Bug: #2077348\nChange-Id: Iae2b075de5412bb183db6a21f7f8376801853e00\nSigned-off-by: Evgeniy Bykov \u003cchypakabre@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dcef779c5e2ca3161987cacabb7f7137f727ed27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dcef779c5e2ca3161987cacabb7f7137f727ed27"}]},"branch":"refs/heads/master"},"720562ca360c6dce4d64011105333610873a80d1":{"kind":"REWORK","_number":4,"created":"2024-09-04 14:02:27.000000000","uploader":{"_account_id":37241,"name":"Evgeniy Bykov","display_name":"Evgeniy Bykov","email":"chypakabre@gmail.com","username":"evbykov"},"ref":"refs/changes/64/926564/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/64/926564/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/926564/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/926564/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/926564/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/64/926564/4"}}},"commit":{"parents":[{"commit":"a4317a80d61ce2d8e699ea7ca17f161ecc4db8ee","subject":"Fix failover when using SRIOV VIP","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a4317a80d61ce2d8e699ea7ca17f161ecc4db8ee"}]}],"author":{"name":"Evgeniy Bykov","email":"chypakabre@gmail.com","date":"2024-08-19 17:20:38.000000000","tz":180},"committer":{"name":"Evgeniy Bykov","email":"chypakabre@gmail.com","date":"2024-09-04 14:01:43.000000000","tz":180},"subject":"Fix loadbalancer stuck in cascade delete","message":"Fix loadbalancer stuck in cascade delete\n\n- Fix loadbalancer stuck in PENDING_DELETE in cascade delete\nwith TERMINATED_HTTPS listener if TLS storage not available\n\nCloses-Bug: #2077348\nChange-Id: Iae2b075de5412bb183db6a21f7f8376801853e00\nSigned-off-by: Evgeniy Bykov \u003cchypakabre@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/720562ca360c6dce4d64011105333610873a80d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/720562ca360c6dce4d64011105333610873a80d1"}]},"branch":"refs/heads/master"},"ec9a50599012b99deaf8bb26683fedcd381db1a6":{"kind":"REWORK","_number":5,"created":"2024-09-04 22:29:00.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/64/926564/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/64/926564/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/926564/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/926564/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/64/926564/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/64/926564/5"}}},"commit":{"parents":[{"commit":"a4317a80d61ce2d8e699ea7ca17f161ecc4db8ee","subject":"Fix failover when using SRIOV VIP","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a4317a80d61ce2d8e699ea7ca17f161ecc4db8ee"}]}],"author":{"name":"Evgeniy Bykov","email":"chypakabre@gmail.com","date":"2024-08-19 17:20:38.000000000","tz":180},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-09-04 22:28:59.000000000","tz":0},"subject":"Fix loadbalancer stuck in cascade delete","message":"Fix loadbalancer stuck in cascade delete\n\n- Fix loadbalancer stuck in PENDING_DELETE in cascade delete\nwith TERMINATED_HTTPS listener if TLS storage not available\n\nCloses-Bug: #2077348\nChange-Id: Iae2b075de5412bb183db6a21f7f8376801853e00\nSigned-off-by: Evgeniy Bykov \u003cchypakabre@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ec9a50599012b99deaf8bb26683fedcd381db1a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ec9a50599012b99deaf8bb26683fedcd381db1a6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
