)]}'
{"id":"openstack%2Foctavia~926565","triplet_id":"openstack%2Foctavia~master~I5a82b9639e7ccf8f3810879c92007386ef02a3e6","project":"openstack/octavia","branch":"master","topic":"octavia-sriov","attention_set":{},"removed_from_attention_set":{"6577":{"account":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"last_update":"2024-08-22 22:12:05.000000000","reason":"Change was abandoned"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-08-22 21:58:07.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}},"hashtags":[],"change_id":"I5a82b9639e7ccf8f3810879c92007386ef02a3e6","subject":"Fix sample data model for member vnic_type","status":"ABANDONED","created":"2024-08-19 17:48:27.000000000","updated":"2024-08-22 22:12:05.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"186b1803992389ca6a48b479f70c3892dc8f338c","_number":926565,"virtual_id_number":926565,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"all":[{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":-1,"date":"2024-08-22 21:58:07.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-19 18:05:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-20 09:29:05.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"state":"REVIEWER"}],"messages":[{"id":"9bbf2e03834e0f131a3053d49f58d3ee9aa39833","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-08-19 17:48:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1ae0c4d3df44c754b7aba10c63f0347b9f49b708","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 18:05:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a111ceee6e6d4bf58503b18288313fcc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3e2f07a0cdd476fba29a338781f3de6 : FAILURE in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4ef2c69b1b941dab34df760de241a4c : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1292854aaa3947348d3ad255df3cca46 : FAILURE in 5m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ea62ab83cfcb49b5baf64471248b0e33 : FAILURE in 6m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/edd1aed1377d48389d4037cc4ac2f357 : FAILURE in 7m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eccb92cbc204483b81a158be73a2e8c7 : SUCCESS in 11m 16s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/1231b144194f48cc80eb1bd8c3060b88 : FAILURE in 4m 13s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/82bec02d939c4fb58b6dc8c3ff5cc6f8 : SUCCESS in 6m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/af325c5cc1734d59952b767d59f45c90 : FAILURE in 6m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/461530d0d275459aaf9ce2b58e8f9980 : FAILURE in 6m 43s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a2892462960d461b9ba2efd5596a95ef : FAILURE in 5m 45s","accounts_in_message":[],"_revision_number":1},{"id":"f778bb99eb5b32142f35c87b5c7051f8b86389a7","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-08-20 09:29:05.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d979c544eefd4419fe8f29a437f01063bcc5a77c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-08-20 14:06:20.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"9b94f287e008592c050590f0d6395fa3055e73ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 14:46:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d690c1ad0c904929bb4085c557cd5744\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a882705a7f874ce7bab2cc0c0cd2e724 : FAILURE in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f74df70e76174e8abcdff52eb6a49cf0 : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/82429171769e4573b35302ff748beba7 : FAILURE in 4m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/422dbb716caf42aca6e5e1cef553f5b5 : FAILURE in 4m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0df5653969754cf4af8dd439aca849f4 : FAILURE in 5m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/100711d6dee44c42a4d79f2de5ae8210 : SUCCESS in 11m 34s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/afe2554d11b14f7bb1e33d3de28680bd : FAILURE in 5m 24s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/f29a0946332c4ad1acfacaf2c33f7103 : SUCCESS in 9m 01s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c9c2f9af21474c2493feefd8ee9d86f5 : SUCCESS in 7m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b983fbd7dcae4baaa81b04d05221f169 : SUCCESS in 9m 37s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2b1075dcdd8a436593c857d658235d56 : SUCCESS in 8m 57s","accounts_in_message":[],"_revision_number":2},{"id":"ffdf50b50f3c28a3f2e7e371d2deb53724cfa281","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-08-22 21:52:17.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"3ab0e2471d3dad03bd436b076c111f60c8bb93ad","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-08-22 21:58:07.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"186b1803992389ca6a48b479f70c3892dc8f338c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-08-22 22:12:05.000000000","message":"Abandoned\n\nMerged into 920156","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f71e8e427eb67b6fc3ab0c5668bf9a26956ccdb8","revisions":{"02613cc2514fba6d401ec664be681cd914f053ac":{"kind":"REWORK","_number":1,"created":"2024-08-19 17:48:27.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/65/926565/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/65/926565/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/926565/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/926565/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/926565/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/65/926565/1"}}},"commit":{"parents":[{"commit":"f68cafbb7201522e6140e2dc05317d06225bb82a","subject":"Merge \"Remove sysvinit/upstart support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f68cafbb7201522e6140e2dc05317d06225bb82a"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-08-19 17:46:15.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-08-19 17:48:25.000000000","tz":0},"subject":"Fix sample data model for member vnic_type","message":"Fix sample data model for member vnic_type\n\nThe recent octavia-lib release updated the member provider data model to\ninclude the vnic_type field. This patch updated the testing sample data\nto reflect this new field.\n\nChange-Id: I5a82b9639e7ccf8f3810879c92007386ef02a3e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/02613cc2514fba6d401ec664be681cd914f053ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/02613cc2514fba6d401ec664be681cd914f053ac"}]},"branch":"refs/heads/master"},"3d80cd16fcc8f018d1751fb5b76b6523282126cf":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-08-20 14:06:20.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/65/926565/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/65/926565/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/926565/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/926565/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/926565/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/65/926565/2"}}},"commit":{"parents":[{"commit":"f68cafbb7201522e6140e2dc05317d06225bb82a","subject":"Merge \"Remove sysvinit/upstart support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f68cafbb7201522e6140e2dc05317d06225bb82a"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-08-19 17:46:15.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-08-20 14:06:20.000000000","tz":0},"subject":"Fix sample data model for member vnic_type","message":"Fix sample data model for member vnic_type\n\nThe recent octavia-lib release updated the member provider data model to\ninclude the vnic_type field. This patch updated the testing sample data\nto reflect this new field.\n\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/926399\n\nChange-Id: I5a82b9639e7ccf8f3810879c92007386ef02a3e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3d80cd16fcc8f018d1751fb5b76b6523282126cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3d80cd16fcc8f018d1751fb5b76b6523282126cf"}]},"branch":"refs/heads/master","description":"Edit commit message"},"f71e8e427eb67b6fc3ab0c5668bf9a26956ccdb8":{"kind":"REWORK","_number":3,"created":"2024-08-22 21:52:17.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/65/926565/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/65/926565/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/926565/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/926565/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/65/926565/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/65/926565/3"}}},"commit":{"parents":[{"commit":"b71e752c0fff2939e4347a22e55c7cbb3e9c7f9a","subject":"Merge \"Fix test_update_load_balancer_status\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b71e752c0fff2939e4347a22e55c7cbb3e9c7f9a"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-08-19 17:46:15.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-08-22 21:52:16.000000000","tz":0},"subject":"Fix sample data model for member vnic_type","message":"Fix sample data model for member vnic_type\n\nThe recent octavia-lib release updated the member provider data model to\ninclude the vnic_type field. This patch updated the testing sample data\nto reflect this new field.\n\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/926399\n\nChange-Id: I5a82b9639e7ccf8f3810879c92007386ef02a3e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f71e8e427eb67b6fc3ab0c5668bf9a26956ccdb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f71e8e427eb67b6fc3ab0c5668bf9a26956ccdb8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
